)]}'
{"id":"openstack%2Fdesignate~969490","triplet_id":"openstack%2Fdesignate~master~I27904f83bdefe1dff03b913cfc9b0fcfef81d451","project":"openstack/designate","branch":"master","topic":"bug/1690184","attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-04-30 11:19:40.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2026-03-01 12:06:27.000000000","reason":"Change was marked ready for review"}},"removed_from_attention_set":{"33685":{"account":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"last_update":"2026-05-01 08:33:37.000000000","reason":"\u003cGERRIT_ACCOUNT_33685\u003e replied on the change","reason_account":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"}},"20104":{"account":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"last_update":"2026-04-30 09:12:52.000000000","reason":"\u003cGERRIT_ACCOUNT_20104\u003e replied on the change","reason_account":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"}}},"hashtags":[],"change_id":"I27904f83bdefe1dff03b913cfc9b0fcfef81d451","subject":"Allow share pools for domains.","status":"NEW","created":"2025-12-03 16:05:11.000000000","updated":"2026-05-01 10:06:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"053cb13cec055bed4f323cec640ac7ffc0ec5631","_number":969490,"virtual_id_number":969490,"owner":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-01 10:06:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-03 16:58:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-22 13:07:14.000000000","updated_by":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"REVIEWER"},{"updated":"2026-02-22 13:07:14.000000000","updated_by":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2026-02-22 13:07:14.000000000","updated_by":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2026-04-28 09:15:11.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-04-28 09:15:11.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"}],"messages":[{"id":"cb0144b6c798201094e43e4ce38554c701c4ee61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2025-12-03 16:05:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a798274133d3c7e3a615eaf0c7f025874e43cd8","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2025-12-03 16:09:05.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"e12567aa11877f17ed804730cee7743959eebbd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 16:58:54.000000000","message":"Patch Set 1: Verified-1\n\n(23 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/152f4bcf58e145b6af0202109c117fea\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d998f89f880c4ad69516775bef7b9b99 : FAILURE in 4m 18s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f3099c5997af41c99bd9c7a65a613d53 : SUCCESS in 2m 17s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/fcc80513ef6946f7816c76eb73333e4a : FAILURE in 12m 16s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f0671be634ce41208096da49dd801a0e : FAILURE in 12m 14s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/f3fa3c2a233d4de1b2b8a429dd926164 : FAILURE in 28m 35s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/17dc2b7f96234737b840e5b6b1407e25 : FAILURE in 11m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8d9ee55d7a0a4c4cb7e6da20de636907 : FAILURE in 44m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/782a21c120b74a6bb0761a2bce2fd008 : FAILURE in 44m 18s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/5b75aed5e18047c8a9ddfe7fb4ebecf2 : FAILURE in 43m 17s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/b8911e6ad612439ab0d940e9c5b4709a : FAILURE in 20m 54s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/efebcff758de42848da9461b3af7d076 : FAILURE in 24m 55s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/94bf9941ddcb4771b36c1c391b840c47 : FAILURE in 11m 58s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/75e344cf0e2c4e4896c415482ff8c08a : FAILURE in 25m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54bb8bf04c5847039ae3e80cc51c4695 : FAILURE in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b624b860155b45dcbaf0fa759a4e916e : FAILURE in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62e1d05427564a238d757bce12e90a31 : FAILURE in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2de0e477d2774a0da7a2235b7d02a8c2 : FAILURE in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2b2a67c8dd74c01a832e3d8b7a586e4 : SUCCESS in 8m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b100a6c54502411b9174be4e1b0c6c46 : FAILURE in 27m 21s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c5ce0d0b617544eaa029386d678ff4e1 : FAILURE in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"71f85cb52452545c1ebda2a1c7d93f2c9ab39fbe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2025-12-07 01:20:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2ebe3c3323672237ca111ac3f18843dd57cbd252","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2025-12-07 01:27:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6e17bc7275fcdbd0c5b27d217291019ea80cb11c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2025-12-07 01:30:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"92218ae63d53df477d6043015ed40ee32328b2a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 02:34:54.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c538085632de4175a42bedc19c63d193\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6ddbe73f43e84b86a18a43ec1053c947 : FAILURE in 3m 38s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2f94b17000ad40ed8dbd2df986ccde56 : SUCCESS in 1m 27s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/fd21d09871bc468b81b017d53d37a093 : FAILURE in 38m 11s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6a1e42cc567a46319ccf7ca5a9c702ba : FAILURE in 47m 19s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/bb3b9ce03f334e07bf6394abd762c8e4 : FAILURE in 35m 08s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/c6e90d80f7a54690a2bbdbe7074550bf : FAILURE in 29m 35s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4480b34b96be4287ab4b11722797d15d : FAILURE in 44m 32s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0c8b2b2186f44fcd9325b30f8a5d26dc : FAILURE in 40m 30s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/0a3f99279b2645b58391f7057b12393b : FAILURE in 58m 29s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/82cc54c5eae14b61bd15353fe8088797 : FAILURE in 45m 21s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/d78203112b7d4063a372d1928db35f39 : FAILURE in 35m 47s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/ac97ecabc6d54c9a987c67f2a6589b1e : FAILURE in 35m 21s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/06ca7b65ed8a43cea7a32dbe7691f629 : FAILURE in 21m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94f2924839d94cd3a57de59f19c88dfd : FAILURE in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69a74cadbdd54582a291eeb46b55ef2d : FAILURE in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/267c30872b1d4892bbe6a14d6392101d : FAILURE in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/99e846f5e1b64fc6aa29d282a5aca7a2 : FAILURE in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/debadb87b0484f5ea44e7d30245160c8 : SUCCESS in 8m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5559933472cc49ac96a3f327df834038 : FAILURE in 31m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/65ae8214996a438dbd7313aa52550450 : FAILURE in 5m 01s","accounts_in_message":[],"_revision_number":4},{"id":"2f6f71ffc32499aa9d81de96b51e3aba2c36a5e4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2025-12-07 06:41:42.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"1d4b5f6b081073be1498626a6d3c8600443da755","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 07:39:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8f3ea7850874d28873d1c579d7ba871\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c1ce478f326f451baae38bba4b2ec03a : FAILURE in 3m 40s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fe5d52a0b5d043c8ae4d2b0dc88968af : SUCCESS in 2m 42s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/45abb5015be243d8a9ee7ab99f8ca95e : FAILURE in 36m 43s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0de745701c484768babdaa921331eba2 : FAILURE in 36m 24s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/c0d9d062a6b642b7b185b6d111be5f22 : FAILURE in 37m 01s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/6867b05e8e97457d91aeedf937502e9d : FAILURE in 31m 17s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/08fba0f21c064161a99e76c190852cf0 : FAILURE in 37m 03s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f06742990def41418c66acb8e2884066 : FAILURE in 32m 03s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/f2b76e306e4a4068825a864a523337b9 : FAILURE in 37m 58s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/95411d8c961e4a7d8dccd1e3c462df83 : FAILURE in 53m 06s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/52eefec39ea141cdb175ea1cb6b32e4c : FAILURE in 20m 07s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/7d6052d944594655914ea089f037d7b2 : FAILURE in 34m 59s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/160ed295e43144e8a06397ea9d9478a6 : FAILURE in 21m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c10d8e4ea5554757944ef240cbc9ffe0 : FAILURE in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/838a90d46bbd4cb9a39e3f13515816df : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/179d47085c4d4750b556ffa01e53f5d9 : FAILURE in 7m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/44e1928656bf48a5903a0d7215bbac05 : FAILURE in 8m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db805689310b48d9ac6f3a3feab9588e : SUCCESS in 9m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/90cf24111d3443c29c32bc5e10375739 : FAILURE in 30m 11s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/06ccbe37ec78400c826bcd3932811f7a : FAILURE in 9m 02s","accounts_in_message":[],"_revision_number":5},{"id":"f51d1d7970fb6bfed088a4aa6b379a2dcafda952","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2025-12-07 08:20:28.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"f98ce6e981be88ed83f57c36aa0ba63799fe304b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2025-12-07 08:43:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1afe2519b73ac554610bd2251cfa48967e1dd8f6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2025-12-07 09:01:42.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4eea08b0d077f25d27e4b21db3138c76c4ff2f58","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2025-12-07 09:21:37.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d202111697c0e98e14662f9b9e4315df9b10b47e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 10:15:59.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b575b2cd20bd4cc3a8fd500251538557\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/216f12ff3b3a44c692c1750462bc7cd6 : FAILURE in 3m 52s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8ff698e99ed34948a3876bdfac77a252 : SUCCESS in 1m 23s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/c63fed0ca724418fba3b0d500c18267f : FAILURE in 23m 18s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/dfb4a5e7e6c14a4b929a1dc9ccd2e1f6 : FAILURE in 15m 37s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/784d14fddc8d4bce874a5e6b2a403401 : FAILURE in 16m 10s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/34220f92940644e090978858c008620f : FAILURE in 25m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d00b488392cc45a18d7db24ec605219c : FAILURE in 31m 18s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/92924664dbee428ca0fc273e47dfbb1e : FAILURE in 47m 03s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/febbe24b219c42ffa63163b5ea23c78a : FAILURE in 36m 41s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/58fa7ed11b72419793b3587194a86abd : FAILURE in 31m 22s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/d39572ebc4064a26acdf5a7584470845 : FAILURE in 10m 18s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/637cd36d0be948ed81a28da4a75ea7c4 : FAILURE in 12m 14s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/09f8ec907d664956bed7293b48fceb7a : FAILURE in 23m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b0cc972acca41f185f9bbbbc4358cf6 : FAILURE in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e8c06d683c2461b8b5ae63458bdcb67 : FAILURE in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eec9877646844a658f26b3dcaba7c7d6 : FAILURE in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd561f2f1fde4c4fbe58d4e3b03d9b24 : FAILURE in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4dee4fe32b342248aaf5522a50dec81 : SUCCESS in 8m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5587173b8cad4d9babfaa2f7e54c70f7 : FAILURE in 36m 48s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cec4e77e6fc048b0941ff44a8ee65740 : FAILURE in 4m 48s","accounts_in_message":[],"_revision_number":9},{"id":"c4168ec7fc19942ae5c27d953bbca00885477a85","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2025-12-07 12:11:31.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"033a21b544805ab593c9479fe5a3f1456ab3d9f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 13:18:29.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6fadc2b495204d6c959c4e3f115b3293\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/44b8eca15523490597c272d165708b5b : FAILURE in 3m 54s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5b73fbe3262942b8b724149d95968975 : SUCCESS in 1m 29s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/5ca7048456f641f2808ab5ee5aa8ebac : FAILURE in 34m 51s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5ee61a8ba67343f4ae2da04202d8897a : FAILURE in 46m 38s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/bf2b999fa36f4e3ba7b0d4c46f357f1c : FAILURE in 33m 01s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/054e779ae2f04130ae4c95a6cead4402 : FAILURE in 32m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6e6215f29f8440b78779e59a9cf20c61 : FAILURE in 31m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6c3b11b20d094fb9a7bc4a47362ed3f6 : FAILURE in 55m 08s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/fb6652973433447ab16124474d69d56f : FAILURE in 31m 24s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/466673f6a109468dab15e39b670990cf : FAILURE in 54m 08s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/af0915aa923f4732999f935f03ce8ddc : FAILURE in 31m 38s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/d741fdfc6f7540aeb073a6b7339a5f01 : FAILURE in 40m 39s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/c4426b45d6ea42a7b99fc055827b44de : FAILURE in 16m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db153a67805a4be3b2a77f2236c21ede : FAILURE in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e03b293f21d547b3b23a172b300f2252 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7cb43bc5702c47ac9576de790b218a5f : FAILURE in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2d3afc14db3e4482af2172236f0a69fc : FAILURE in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fd4479480e6473d9a872861501a852e : SUCCESS in 8m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0985644e89544ae08db6a180ee946bcc : FAILURE in 29m 43s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1075d68bf15e40b6aea8fdc3876a7ec1 : FAILURE in 4m 35s","accounts_in_message":[],"_revision_number":10},{"id":"deb66b9ff7c54af8b3169963708003f7d6f4b51d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2025-12-07 14:06:08.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"f8b4e07beb9926f8147905adeec4d38feeb9da41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 15:05:43.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e148175298424f8eb9235f6657ff0d9a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/478db1c3d9a04584bcb93aad94c940c0 : FAILURE in 3m 15s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c4a8e8f76c5d436faa6f9254cbe56c35 : SUCCESS in 2m 16s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/04de31d1eb7b4f368472ed79b2186e21 : FAILURE in 25m 39s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7d08b77eacfd42d29694cad678ef5f9f : FAILURE in 51m 07s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/de04317032c64e708848ff6ce9e1c0cc : FAILURE in 26m 48s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/b867284955a043b2b56465cfc05b6342 : FAILURE in 32m 26s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9f416921ded84d26991e3c3f18aa9a44 : FAILURE in 54m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c9bbd79857a94e0798df08a30052aec8 : FAILURE in 53m 18s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/3692ad4fcf384ee9b022996ee9615ebe : FAILURE in 37m 17s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/a36a3ce3012a4381b5be85caf513d949 : FAILURE in 27m 07s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/e095309afe0d4bc388bcfef72f301502 : FAILURE in 35m 12s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/6ecf0087bc3c45a88eb6e7c6877bac81 : FAILURE in 17m 53s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/5385e3dfc42347e389057d38d716a974 : FAILURE in 24m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49063abc50f54b1f9e15d46d692be2f4 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b41f1e0629db4766b5bb533436dd7d3f : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ff3ac8a98904c42b088a5b1418bf3fa : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/326bdb68144d410f9ee2682f80f125a2 : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05587883cdf74d02ad9fdaad8046925b : SUCCESS in 8m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0779c124080546e1ad515e74c6b964b8 : FAILURE in 30m 05s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6ec2af0e54094f49b1d59d09ed805257 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":11},{"id":"092933703517df30e2ea24414aea5ddba7232125","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-01-07 06:20:47.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"bf491c58746bc8772e298fe149c6a7b832e7cb86","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-01-07 06:27:40.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"93137e8c28f61bf137e3e1db400541355600e709","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 07:44:44.000000000","message":"Patch Set 13: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c0ee95381c64539aa0a8b1f65f58a0b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/870dcc5b4db34c22a181f5cf0a08f414 : FAILURE in 3m 14s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ace77ca210e34636bf01b2580c2eee94 : SUCCESS in 2m 22s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/4839966274c34e5c94fea70c7630d301 : FAILURE in 33m 01s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7c2a3f72983447f58625befa6aeabcab : FAILURE in 37m 32s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/65636d885cbc4015b048e10c39907a7c : FAILURE in 33m 54s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/060cbdc62ea14afabffd9719335cd626 : FAILURE in 23m 29s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/bbb3bc35fba04c3ab2a8a3aee18a1cf3 : FAILURE in 1h 11m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/eae0f0ddd7794513b2fcd37460ff850e : FAILURE in 53m 31s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/c6e964f154af4580b648276ef0fff063 : FAILURE in 56m 57s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/47818d6de94c4fef824205ffd82a2a09 : FAILURE in 51m 58s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/f2451f17326742d0a8988499113930bb : FAILURE in 33m 19s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/e51a4ec865e247e59ec9da585ad2702b : FAILURE in 33m 27s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/82fe8dcee5e345739141ad2efe65fcaa : FAILURE in 35m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16c114c7d4c149f8861a2a6a303b4a8f : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/551c78f9e8f24bc69d37f1a69fd34bb8 : FAILURE in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15636171013e4206a46adf53ed3395ea : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37c06f1d82584ce2ac7cecfaa00b4dae : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3614a0543b4bd383b6531559748890 : SUCCESS in 9m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/091693b9be504a2cb2eb0ba93dc0b6d6 : FAILURE in 29m 47s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3514e7d9edcf42a7bf4acf38b6fecf87 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":13},{"id":"dd8efc23b3a4fed0150d5a45c1ddb9a01ea89dab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-01-07 14:27:20.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"132c854ddc84b37dbeeff9a444cdf83160494f00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 15:45:22.000000000","message":"Patch Set 14: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/953f953bd46247faad368a8338f289aa\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9f8cda2d7b0940a7ab66fe87a5f7dcd1 : FAILURE in 2m 34s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7864bf1acc494b97bb9b157a86d3ce63 : SUCCESS in 2m 45s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/039f0a111f1a47cfa34a1fa6e5fcfc0c : FAILURE in 22m 25s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d60c15c7dd9b41adaffecd3f86fd23b1 : FAILURE in 39m 26s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/bd54bf0d934f4532b8958b7c62855d94 : FAILURE in 34m 31s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/7a60feed9ad7435297e7c8b2418e20a3 : FAILURE in 14m 01s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0dc453da253844a8a5d588a4ecbc45ae : FAILURE in 1h 10m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/922f6aa6f638480d9ed3fb058400e41f : FAILURE in 35m 59s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/aaad67a547e84f36bcfd743caf4e77ec : FAILURE in 28m 48s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/a4a731422d16455ca65344220be1154e : FAILURE in 30m 24s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/27475a8ad6bd479a81ca4e8a0a2821f9 : FAILURE in 33m 12s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/33c0554e3029458a963e06e26fb00159 : FAILURE in 28m 33s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/65d92785cf164b5bb3824497b5202926 : FAILURE in 31m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da3d3fe30f2a42858d7669e8bb33d3d2 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2776fce96682445382c384993dc94647 : FAILURE in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27b73f9bbfa04c3eb1bb3d0380742657 : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c88b41a7681440a2bd375a7543dec34b : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16a74c9f89a54c7cb87254549376eb51 : SUCCESS in 7m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d3161d3b336144eda0823e0e4d39d879 : FAILURE in 28m 57s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b39f72d72314411e88c01d08589eb46b : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":14},{"id":"9ef3bacc24abb73f7148c8cac8d261ec477373de","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-01-08 00:27:37.000000000","message":"Uploaded patch set 15: New patch set was added with same tree, parent tree, and commit message as Patch Set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"0d288c3639fc2761eb52ba9bccf30d28bde71452","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 01:42:12.000000000","message":"Patch Set 15: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/609ecb8ccc9c4c12b8457191444e693e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b0e01367b7c448ae8811d0c2df9b7c63 : FAILURE in 3m 22s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3263c6c8e1b04a7b905dd1cd88b6ea19 : SUCCESS in 1m 35s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/08444841d13f47db9467be72ac66c0f6 : FAILURE in 32m 48s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4923f7b8b3e24aefa458933a1c197caa : FAILURE in 44m 06s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/faf7b71b108a4d009a041d0a46976f66 : FAILURE in 28m 31s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/d4ce380eff774c279c15757d17ccb959 : FAILURE in 13m 44s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/081eea5f3ceb429d9fdc2e24d93c456b : FAILURE in 1h 07m 58s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/59fbf6c3dcc34d4696484cd4266cc205 : FAILURE in 27m 14s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/00c9481773574497b0ba1b851f95eab5 : FAILURE in 50m 34s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/8a832cddd4cc4650944a71dc01e1ca81 : FAILURE in 27m 45s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/f26f684e244d476ba174e7a031a3d22c : FAILURE in 33m 10s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/4f0f349fca4a4d05a1779b262a7029cc : FAILURE in 18m 50s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/4834df236fe84aac83a04c2c45f38ce9 : FAILURE in 33m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2dfe8b303b54b22949ca0076b9e22fb : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd660709a16d4e18b9853bf3ad3dba2f : FAILURE in 2m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ad1cd96c3464ce69a783236ffe517c7 : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f3fca9c24f7440eb9fecac507c601a87 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d31b890a56c94d979fe57aa7c2f1d203 : SUCCESS in 8m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d059826d81b248b899d0c8f2577fefcb : FAILURE in 28m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/21b15c52599e42c69bc8daac72b08105 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":15},{"id":"58a99c275421d8daccf0cffd8e9c6c2d2a6f253e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-01-14 01:46:47.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"e695f802d98c391f79b1249ade8a6a4cb0c703d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 03:05:25.000000000","message":"Patch Set 16: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd898999e2e743879562b4e04b7e5836\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3eee7bca9dd1481d934a80c001308944 : FAILURE in 2m 35s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/43cc91ed591a4506b805350b9400196f : SUCCESS in 1m 48s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/1a90b456c5e2483c9ff9db6e4f8edcd3 : FAILURE in 34m 27s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/bcf20266903b45a4b9ceba524675234c : FAILURE in 48m 52s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/dc79dc2cdda84a9291dab0d42d37224f : FAILURE in 35m 23s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/3585400933b44216b98d206178b88601 : FAILURE in 32m 54s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a7911e56fcf04d909ca096cd97d87ed5 : FAILURE in 1h 07m 49s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3e9fb15c5c7449139e0cfb6bfba7ff19 : FAILURE in 1h 05m 34s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/329c1013a61f4b6cbc0be1ce658efcd7 : FAILURE in 1h 13m 16s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/0f5a3cc8cac94586b5a9ee2c51f5e5d4 : FAILURE in 54m 54s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/ae434d0c4a4543bdaab41d520f775fe8 : FAILURE in 34m 27s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/b6b099da79194898952af4e04a8b5283 : FAILURE in 26m 05s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/1ad8132e79dc4acd9be758da9b0c6ba6 : FAILURE in 26m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86cdf36d75ae4cbaa9083b57df3e2d1d : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d1738bde1be4a30bfb2b6ba7a5a49ab : FAILURE in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f7eede0a6c54bcf88843517ceeafd8c : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97a3ed00e76b47fcad50503d468fc609 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fbce55bd8544eeaab14301810c3f7b3 : SUCCESS in 4m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/76174e490470456eaec386dc6e2c5d55 : FAILURE in 31m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3c9e4b73792a4b8c88a87b5b3de8317b : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":16},{"id":"b8fca50a1a2a855e24329e46c9b4c2aeee5ff090","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-01-14 07:13:43.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"e497aceaa3e3c6ed308546d7e9c4e2e646e5e5fa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-01-14 07:23:15.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"9ddbd266315168baa33161b67ccd72a2df8898cc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-01-14 07:27:32.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"d4f5e5abe91c8a19d95b8a1c86e6b999fd71965c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-01-14 08:05:25.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"3c8c994eede61399583ae67cd74c2a28e3b69284","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-01-14 08:06:30.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"165f3cc7fd813786cfe4df30df10f5e70590201e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 08:55:15.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a0f6584dbef46dda1e48d1f135eda4c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0224ac5626d644b39d53aa81dc221469 : FAILURE in 3m 59s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6759ec18cb9f40119a1fb671b54be9c5 : SUCCESS in 1m 59s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/6915c6793a694fd2bd177f605490d373 : FAILURE in 23m 24s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9149735295e249e999a573ec1dbd918d : FAILURE in 25m 33s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/eb68807936c64be59401f8184d0b35e7 : FAILURE in 24m 16s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/633f100870f140e5b16b084df434700f : FAILURE in 23m 25s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d12deb1ef79d445fa7bd0069a4bb8bec : FAILURE in 41m 50s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3e5eb9e25dfc4f12bbcea8ce74487084 : FAILURE in 42m 59s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/07812618a9884e5692f28fe85bcb5506 : FAILURE in 33m 11s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/b19d907af0734243a7da5065b76681de : FAILURE in 44m 06s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/38d8c28fe9584187ad429fe658dc8143 : FAILURE in 17m 25s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/7553e014b58a4dcd9b0eab19fcae9e9f : FAILURE in 24m 51s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/718dd4e469ae40808c10c81a7693c66c : FAILURE in 13m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48ac152a01234f8497c0d8b46494bd28 : FAILURE in 14m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b548efdd49b4c7982c53625e9431f0e : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dff2748c19cd46c89c55e5f3c1ed0a93 : FAILURE in 10m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2399055b540e4344a3a033abf2d434d4 : FAILURE in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1255ffe551d7407e95632c3b365d322c : SUCCESS in 8m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7b76399c83724d7e9a17b0aa17ee761e : FAILURE in 27m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e7a737786f39437db2a1c6493e8abe9a : FAILURE in 6m 35s","accounts_in_message":[],"_revision_number":21},{"id":"63c5d56cee00f6c984200435699397fbb67fb41a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-01-14 09:28:22.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"d1963ea047978ddbc5378926203847df37d62822","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-01-14 09:28:59.000000000","message":"Uploaded patch set 23: New patch set was added with same tree, parent tree, and commit message as Patch Set 22.","accounts_in_message":[],"_revision_number":23},{"id":"92e797aefb634f20f9dccb02e09b6065fc9f1bf1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-01-14 09:29:19.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"5c2a1de01418fa00c7ced8f41812b694955f4a3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 10:42:40.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f195d5c7c4054f2eac18d29ae88af95d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2b9741dfef37426fad9a876adaf92a36 : FAILURE in 3m 32s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a0fc3e257eb84876b26c7981a91350ff : SUCCESS in 2m 10s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/40263dc502ea4e95b8422b4f520b421d : FAILURE in 44m 27s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5b1d0b6b3e66469ea350040d04215eaf : FAILURE in 47m 33s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/7acc684064da403b8b9e3978298833ed : FAILURE in 35m 00s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/1991010db8ad4d37b0e7fb5f51bc0b9a : FAILURE in 26m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/bcb3fabe8a5742d68ab4214e0ade6c94 : FAILURE in 1h 09m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9a8f4f46808a4370b5dda253864eccf2 : FAILURE in 49m 04s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/131dee48c24246dcaeb53cb672ce06b4 : FAILURE in 50m 42s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/9fa7606ba4fc4814b58b9b1636d3f5e5 : FAILURE in 1h 05m 15s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/4677ad08fc2f4e7fac2f7cb1192a47bf : FAILURE in 23m 09s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/d1366848cdc54e8884b8ddb13597c14f : FAILURE in 35m 19s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/c500a331856f419bab17039a31c5f1cd : FAILURE in 32m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df3cc7fda5404256b0f5c0263d925c83 : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7f4dbb4b1304a038158314d68785dd2 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/239964fcd45c460fb82c2ff91b0dff0b : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8251720c12a4d70aebd6c827afc2c2a : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d16ab3cf3304bb7b9568952bf2dfbc0 : SUCCESS in 5m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/359bb64a35e943e3a4bcf13d9ee1955a : FAILURE in 31m 46s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/09dc225ad0f044dd8ed62f52994f0751 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":24},{"id":"ff7f5738e0cffbc88787fdbb3ead72fb1fae2f79","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-01-14 12:22:35.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"c9c6a12107a724743e6a87cd1b14cb0f06aafb4d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-01-14 12:24:40.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"3cd43ab4e3b8b4c81fc1d0439699b3b980f58399","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-01-14 12:36:04.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"3bd7d097b1e45f0951fd0298fd82a9a12f54f95a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 13:56:55.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/47562d65a3d24bc1ad6d20fa27b3128f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ca5919297c5c46eab4b4969cf18df3c9 : FAILURE in 3m 59s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/61b6fd6206f24fe1a246e82e852e2815 : SUCCESS in 2m 24s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/2dda7b527fc9455882e249322c6ea89e : FAILURE in 50m 26s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8a30f94e50144bd2b37432cd0a20bfcd : FAILURE in 51m 51s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/bb3a32c210534294886419b49b8aca8c : FAILURE in 46m 01s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/2e6b8189449d47f1a58989ee9ea63444 : SUCCESS in 21m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0712d53a607c4c188200a1e3956f6f30 : FAILURE in 1h 18m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/885ede3702eb444e8d9df2168bb37ad9 : FAILURE in 1h 06m 49s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/0b3b5cf99f364c1c8a9c269e8bd1c6b6 : FAILURE in 51m 58s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/bce7e93cb7a34aeda3855915bc9b8874 : FAILURE in 1h 07m 18s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/ddf09cf8dc2d46609f335228c983d515 : FAILURE in 29m 21s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/af2600d2545046a1a195ce72739673e0 : FAILURE in 42m 19s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/3715225ea2a84c4f845a88b267c33ed7 : FAILURE in 56m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4979823a6c3e4309b8e4010b05a78c17 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d66a2d0b44e3462f83236ff7ff237aa2 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5386dc4ea6244703baf00d3076799717 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b37c8acb89f04ad3bd3d90aeef836320 : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/668c82390aae4b1d8cbc57884d4ac93c : SUCCESS in 4m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ac21f73d6494433ca757420ae298497e : FAILURE in 32m 29s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/75a6d0dbdedb45d88da658b27451a4e0 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":27},{"id":"3757daa942e16610bf92c73948af38733a655fa7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-02-22 00:07:34.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"5e49c33ba02b5d0b1acb9986ecdc5b046182c7f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 01:00:26.000000000","message":"Patch Set 28: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c04d1317ed35420db0c22154fcbfdfef\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a289dbbcd97347f59e5600af6c1b1565 : FAILURE in 2m 32s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/56c0764475f64c8aa1506fd09bb9be0b : SUCCESS in 1m 37s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/437a13a59c3f4d8da9c9c2e8c4e38a83 : FAILURE in 21m 20s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/03921d89808349cdb186b367d35e0028 : FAILURE in 16m 50s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/c8dcd0f79d1848d89980ffc3a964a7c0 : FAILURE in 15m 26s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/97347199a61641bc9aed4004e7745c8c : FAILURE in 8m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/aa5883f230504258a927e495bf2513d1 : FAILURE in 42m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/abd98755df604667ac6f57a8ccc583ae : FAILURE in 21m 57s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/496dbe0376804c2d9a718c30a0447da4 : FAILURE in 46m 00s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/ceb9590310244daebb820c3b871b3121 : FAILURE in 19m 45s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/c09992a13d594b1692db3f27fd519d9b : FAILURE in 18m 14s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/ba25bcb2b688490e8bea7df4bb2aae5a : FAILURE in 8m 39s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/1908731420bb4c33a26962b29855c0ac : FAILURE in 10m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e69918c4cb794455b88405a833bd4a31 : FAILURE in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d8bc6dafe424f43a2f1c039d632a99a : FAILURE in 1m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eaf21f73d55046378117222ad8309eb6 : FAILURE in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/72569778497144cf8f5f8e3d0f338876 : FAILURE in 2m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3be5070189144899acc4fca9636c2cb5 : FAILURE in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f19575fa2e84738a85b4217e1652efa : FAILURE in 5m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d60e605936954464b9a0bfbc73fc0351 : FAILURE in 3m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/872a0ad5ca844318ac1402156346929c : FAILURE in 10m 52s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/53bcd8225cf449a0aef5dd8305c8ba08 : FAILURE in 3m 34s","accounts_in_message":[],"_revision_number":28},{"id":"8e691e82f6af81cce781fccbba335c6f29343d0d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-02-22 01:08:13.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"5e562f058a4f1055000dfb886c577670bbc7ddee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-02-22 01:29:27.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"24c3bc1329f6161dac2219b556f44b12f38ff4db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 02:49:01.000000000","message":"Patch Set 30: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99adcd7437684852bcc0634da12bf3fe\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9381b837120643628b2a9bf272001a72 : FAILURE in 2m 32s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/82e4f8158a9b4224ae849a7c72717a05 : SUCCESS in 2m 40s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/1e014e7dcdf547b7b2dfbf0a3e94d37a : SUCCESS in 48m 28s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ac75e223c6d94e7c90f0d823567636b3 : SUCCESS in 43m 23s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/1af503a2e04c407faa9c753a6a175c44 : SUCCESS in 35m 14s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/c2d4c2ae05e24b5d8c2eb0bc136bdb95 : SUCCESS in 24m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/176bcb66fdbb4f088f980ddbfe92ffbe : SUCCESS in 50m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6bc403cb90554629a360148e9d7fe13f : SUCCESS in 50m 25s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/136e1c6a302547f1bf06584ac44ced89 : SUCCESS in 1h 14m 02s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/817f06eacc94416f96d9bd9b4f1ebe0a : SUCCESS in 1h 09m 07s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/be21c79286ef4b3f86f32c20d09945e6 : FAILURE in 23m 55s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/66c14e90bb1d403997913b61af1495ea : SUCCESS in 46m 00s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/94daf1dce31746979ea9616f9d27f295 : SUCCESS in 50m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f39430851da45a5baac41ecd3209a73 : FAILURE in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eecc6b61042f4a07a2b34ad02341abd4 : FAILURE in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d3fc9159986454ca4eb280c81569325 : SUCCESS in 5m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/256219875da442cf89dde250710abd61 : SUCCESS in 4m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aac0dcabe298473b9a70e5b74f7c1053 : SUCCESS in 10m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09fdcae7e2c04095a2adf0bad9eb54c5 : SUCCESS in 5m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/234b83f62973404e9d3df4a2645cc775 : FAILURE in 2m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f391d0f72a6241358439522efccaafc4 : FAILURE in 43m 59s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1de36727961b41f5a5a06679974b7df0 : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":30},{"id":"8542a6dcb873a3a9a84135874ffcf8b0ff102d25","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-02-22 05:43:30.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"c8da88d24cc92cdab17711595129435f9b958cc4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-02-22 05:46:16.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"18fc5ed01eba65b00e97662bfd91298f04325746","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-02-22 05:51:34.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"567c1d80caa75c023d6a3e47c5947261b8265400","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-02-22 05:53:03.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":33},{"id":"5a1514ddf3d36b7d10666316106d1deaf1c79941","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 07:09:35.000000000","message":"Patch Set 33: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9bb808a21893479194b034e73e84bb4f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/25fa507d9163440ebbfd6e4492ab66aa : SUCCESS in 2m 11s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/30f274ac1f8a479fb69dab83383df08f : SUCCESS in 2m 17s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/8002258f608d4084a96cb1a5bb11977c : SUCCESS in 50m 04s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9ff1ab8d476a45cdbf3d0335f5dae6c8 : SUCCESS in 51m 59s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/5e899546802c4e229f831dd00c27ef5b : SUCCESS in 45m 01s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/815606dce0154e37934eb5f2259e0efe : SUCCESS in 35m 01s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/40e3f8d97d1645ce95844cc5f7e60610 : SUCCESS in 1h 06m 25s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3689f761427d4a84a4a2bfc257a33a43 : SUCCESS in 1h 09m 38s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/0e13a48d764c49fab34aed1574a8044d : SUCCESS in 1h 10m 57s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/f86865cf9cdf443994122c204300c044 : SUCCESS in 55m 02s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/629a6f660790476fa58f50265de83a0c : FAILURE in 34m 37s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/cb17b98ae07d4ef79ed867e875d5af37 : SUCCESS in 42m 42s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/49e2283a22cf4e9d89a906b88b4b0407 : SUCCESS in 51m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38ae014ff98a4f71ab3443f967d36ab9 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86ce877f1496427092270f92d4d699e4 : FAILURE in 2m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51013d7ea38e4336a67a00eb4aff2085 : SUCCESS in 5m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e1c2b16b7e34301accef27977ebc9ba : SUCCESS in 5m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/156ade08d4ba4cb7ab882325c9dd2818 : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db0733a959ca486c81a984c81808ab5a : SUCCESS in 8m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/609d671c46364b099e2026d3d8501949 : SUCCESS in 2m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1d09de62fc374db293afa4ec5b1b48ef : FAILURE in 30m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/44fe48698e9f4c73b643329adda11d59 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":33},{"id":"66f2aa7c545a38c3041d76a6283db27afe9b93e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-02-22 11:08:14.000000000","message":"Uploaded patch set 34.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":34},{"id":"d80ffa28927795ac026b9dcc7930c60ab3af3d32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 12:30:56.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b71ef5e3cf5427f94b83de8663cb946\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/560dab6cf5c740f4808500681712b57b : SUCCESS in 3m 27s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9a10e81be5a04c4f8a5f48cdcc058d30 : SUCCESS in 1m 34s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/b448362d57a0422e89f38f4194b7768f : SUCCESS in 38m 48s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c10526cfeab84bfb894f8c24515a15b9 : SUCCESS in 52m 57s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/6803022cfa07457cb1642fb9a0dce82b : SUCCESS in 45m 54s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/b096171eb6ca40f8a7f372ec73ac1461 : SUCCESS in 16m 37s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5e8df0799bc840aca66f8962a4953aca : SUCCESS in 1h 08m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f50b851db78d46ada9ca839c40f3be22 : SUCCESS in 1h 07m 38s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/43e7620deb4a4f4aaf701d4e278c3abe : SUCCESS in 1h 13m 38s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/b1b0b3cecb0e4686aac75b76b4b2d75d : SUCCESS in 1h 13m 36s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/15978a51a57946c8b0e9c0ef5596928e : FAILURE in 36m 23s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/1365eb3b85404093a04f700bf939deaa : SUCCESS in 41m 16s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/82e56c69ff2446b488575536e511291d : SUCCESS in 40m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99f516e1b44e405baeac8c2d46f8e36f : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/359fd378904c4dd99e7abd3d4fe238de : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf2333d0e13b4fc1ae1fbe5448f7d92c : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2cb0fe1c658e4fceaa23491a6c300302 : SUCCESS in 3m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7d93c39d5c574baf9f9f4557372a73d3 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7fd45be07574a1cb30187c7e1d311f3 : SUCCESS in 8m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/53c4d526251a46f5a82e6721579e3418 : SUCCESS in 2m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f9cbe07e385b4f71ae3193bb43390519 : SUCCESS in 34m 26s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2c700895d9104c25af5a2e8f042beee5 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":34},{"id":"f51c0efc7a9f0422353f7ce75ffbfeb334457b65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-02-25 13:22:35.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":35},{"id":"3b1458a92a4f8584a254318629a7ace502257a9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 14:52:36.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/afdf644e08e840c8878c06fb361e789b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1d4c60f5c765401291f564990d3c4b66 : SUCCESS in 3m 59s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1403768fe9f741719476e8b21ea9ded7 : SUCCESS in 3m 00s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/20e6f780ddcd4ec0891da3214f474bd2 : SUCCESS in 41m 19s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8329b289864e43f7b547da6a8250cfa9 : SUCCESS in 42m 32s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/78b2be1ad0394a41a6892e66e51d792f : SUCCESS in 39m 21s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/42ab3bfbbef94e189cfb9bab320129a6 : SUCCESS in 22m 16s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/465ef43b419342b89ec50842332b67ee : SUCCESS in 54m 03s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0f2e713e66164e87984f27c2f484bc3c : SUCCESS in 1h 17m 07s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/a9974bb42b92404a8cbfc2aa954b5ca7 : SUCCESS in 1h 19m 47s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/ffd2aacda7774289b323b986da3b17b0 : SUCCESS in 49m 13s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/fee6ba3dc5b042ec96bdeac343986c53 : FAILURE in 33m 43s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/efb14bcb29f944b8af771f6c21b9659a : SUCCESS in 52m 32s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/f0dc3982c2164724a74e152b83d97cd7 : SUCCESS in 39m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6318e950f36b4784a5e5762040a24a6e : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b4dbf2f222745518964bb7ce2317cad : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c04982a579b640e7908850cf9073b4a3 : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ded98d1a144c4557a44f9d00db9e01bc : SUCCESS in 5m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/03d93aae40ad4dfca20572e840a7861e : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9609b6a653b44397b00e62e88c7151dc : SUCCESS in 4m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dafd0760b73540cfa62118dd0ba0e0b4 : SUCCESS in 2m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d51ac5f6da7448789e58ff8ac95a518e : FAILURE in 29m 10s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7c3cb3a2d5474c1884d0dc3de55ac16c : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":35},{"id":"944908cf63e0e1259db65be5c94c1a9971423681","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-02-25 15:47:16.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"d9c2f8a2cce510b3cdcd2e6d3694a8e0bbfd2318","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 17:08:43.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0188181e4194410890abd86796665bd\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/390f06dad4824f5f8e3680ada23f0b64 : SUCCESS in 3m 19s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b9c2548090604f3dbe005a47bad46932 : SUCCESS in 2m 39s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/9af6688d66e64d518f58324eaa3b5a74 : SUCCESS in 42m 26s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/70d8254870ff4f528d972e501a422514 : SUCCESS in 49m 42s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/10085da3fce74641a4ff178f86c3fd24 : SUCCESS in 41m 30s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/3e9d1199b40f4f1d86f4e6b57d6ae546 : SUCCESS in 33m 09s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4491e678e65d48ec8fc968ace325d1e5 : SUCCESS in 1h 11m 34s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5930016f1da247789e6e436c65b2061d : SUCCESS in 1h 06m 27s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/3dcbbf0063254bafb63312220bbf5fa9 : SUCCESS in 1h 12m 44s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/f10e2041a11e42e587474c85502338b3 : SUCCESS in 1h 08m 40s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/4fbacd8307ab4e5a8a0ab42be89ceea4 : FAILURE in 30m 57s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/f57038d4b7cb47e1bcac2ef32f2b0b94 : SUCCESS in 57m 30s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/4f0fe5dcb7da445abb81c6fcbaa128eb : SUCCESS in 44m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/632bbf9911ec470cbc802797f91766b4 : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dda8c0a54ca4d479609cde2d917baaa : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/432179adede2467989274f36b76d50ac : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ed3959f924e44ff797d6ccd30ba5d804 : SUCCESS in 5m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8f0a1646de434f9d8a14c1162b05ecc5 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75134a16960b45e2a88d5afa290fdeee : SUCCESS in 10m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f9510a96d38e45fd92fcbc15fc73a1c0 : SUCCESS in 2m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/58df7a6cc3ca49e69eeb6e78e62a53c9 : FAILURE in 36m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b1b58edb36b04e9599f334d65cea5cf8 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":35},{"id":"3f5e38ab5fd665f4bcfd9040b151d776caeca818","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-02-26 10:12:44.000000000","message":"Patch Set 35: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"79519a9e58c06745cca8c842b85f3364be13ebf3","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-02-26 10:16:39.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"1e3141a7aa38d23ebaf2506a7f9c66b1487f0caa","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-03-01 02:30:50.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":35},{"id":"f8e4820574d5f9c7c7d606ea247e464aa0743240","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-03-01 02:31:31.000000000","message":"Uploaded patch set 36.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":36},{"id":"431585604636e34dc9ab57f0fcdd36ff87cb94a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 04:06:58.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6edefe224c49487f9e062a812b4b7345\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ced23a01e24d4ab5a4bf7d5a48f711e9 : SUCCESS in 4m 54s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/817a04444c5542699b301279a805d3ce : SUCCESS in 2m 53s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/80df8f67a2904ca297dee94ff7fbe09c : SUCCESS in 50m 43s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2ec92a038d724e7284548fabdffa9823 : SUCCESS in 54m 02s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/6d84e682883f471fa72d87eb152eb66a : SUCCESS in 47m 38s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/59e6e85485294e03a95cde48aad84687 : SUCCESS in 33m 06s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f83ce35b3b0f4bb088562a7bd712c2c5 : FAILURE in 1h 11m 10s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/699cf5df260548869c9e9973fc673fd5 : SUCCESS in 1h 11m 31s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/8b12216379d84d0ba2e6fa8f6610e7d4 : SUCCESS in 1h 10m 10s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/aa66966d1903424894842b190c95fd26 : SUCCESS in 1h 00m 11s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/ba815ad290e94478a612841ce1bc25b9 : SUCCESS in 45m 37s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/c6cca55b6a924e4491718d348314275a : SUCCESS in 40m 24s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/a00fe5468abb46c0b6776eab4e433b04 : SUCCESS in 41m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8adcb4578ea6480ea637d3aa9f4d9552 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f865c1ba986e4590a10377bfc11b16ef : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9974161e28a1478ea50f63d21ff077ba : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/74036f51f8264161916aef2180ccff52 : SUCCESS in 5m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/716278828ddf447baab4be5b134e7b6c : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eb7f5b321444860b4f219e0f970048e : SUCCESS in 9m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8551c6eaf13f4933a34733fb5229762d : SUCCESS in 2m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a132f29aa16d4f14992044583f1b0f8a : SUCCESS in 32m 52s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d190a833d50f4a3abadd95c9c6fbd8e4 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":36},{"id":"558429faf0a170b12d8e7bae4c8a7d185d6ddad1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-03-01 05:40:20.000000000","message":"Uploaded patch set 37.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":37},{"id":"6284d430da619a3b122a38047c269fa4525b09da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 06:35:57.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/365cc47ca1674a27bb4f9cf9d501f5dd\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f8686cf83ed84e5fa7155d549afdd650 : SUCCESS in 3m 01s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6267ace73d5749449ea2b0eb4455f924 : SUCCESS in 2m 19s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/78aa30293dec4f3fb058640edd889227 : FAILURE in 25m 26s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/43817f9a82c041019797a7080477d138 : FAILURE in 48m 38s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/51de47718da44368b85b3106cc6e599c : FAILURE in 32m 42s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/dd3caf0946ab4908898a957f251152c6 : FAILURE in 25m 41s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/387360e18ec44b29846c142d4c5b2de1 : FAILURE in 24m 32s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c47f32b519d94451bfce1b6f26f7203e : FAILURE in 32m 34s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/cb58a6a8c46742a384b297b92dbd4a92 : FAILURE in 49m 10s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e40827581d5f4a3bb7eb1c0736ebe2e0 : FAILURE in 34m 17s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/2febeb6e7f1f4149b285298b2315cd43 : FAILURE in 14m 08s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/cff3ce40c2c94c5e9a945a10d5cff14c : FAILURE in 14m 20s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/cbc4b57493184c939b6226faf35a294c : FAILURE in 19m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4248f6c578e4d8b8edcb5924387ec11 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1269998df2e4735a4eed32170513102 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16aed8a61e904635b7e5e8105819a04c : SUCCESS in 5m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4689353bdd443acac877c291cdf9413 : SUCCESS in 3m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c6932f3e32204f979f6ea971b3ec2a57 : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d1ac35a83264123bf99132a69503d92 : SUCCESS in 8m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/13c5ecb262d64a4685ba06c6e7c289a9 : SUCCESS in 1m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a4773dbbce614421acfdac2d37179f91 : FAILURE in 31m 23s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/658ab570ccfe43d6b1cc9e7199a7f77a : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":37},{"id":"06377198e08470e5efc83806d3e67aeb8c50eae2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-03-01 06:49:53.000000000","message":"Uploaded patch set 38.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":38},{"id":"3dcefc237ab3cc12e22be27422d95782cade05fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 08:09:39.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fa79076aafd496ca9d662546856518f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5777336d2a5c407487fc6b5b76531865 : SUCCESS in 5m 02s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4105a93c75754411a4cf5de63747deb2 : SUCCESS in 2m 45s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/7037f79c911243e492a41c76f2473b23 : SUCCESS in 43m 14s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/882f0c3f021e4b859837584ed2d33b71 : SUCCESS in 51m 30s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/a75a50bc68e3419f9dfb004989174965 : SUCCESS in 47m 27s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/80ae3e9bc2bf49a98177cfb30bb6ae9a : SUCCESS in 17m 10s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/88f628cce1cb4bddac5c7ba11925f1ab : SUCCESS in 56m 30s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/afdcffcf44574fdd8f246538f4e7cb1b : SUCCESS in 1h 09m 28s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/17118e6647e9494c9b412dc66dd87afb : SUCCESS in 1h 12m 57s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/8136fd3b2b69447e9da4dc4b83dc2b11 : SUCCESS in 1h 12m 06s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/cd652259ba5f4ddf92b90d68d8f8817f : SUCCESS in 50m 21s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/096e08ff8f974cae970eb9291c596164 : SUCCESS in 43m 34s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/af1f7039ff7c4f55b19b2985597ed40e : SUCCESS in 45m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48306ee1677d405c96e48f8775d400f6 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d5cc32910e64e488d59c59053febe61 : SUCCESS in 2m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9e33b27b53441608f570402c85ad2a0 : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4753d7e877ac472d89a9e6e1916828c7 : SUCCESS in 2m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/16dda8490c584bc4ab4a3076a34c8b70 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/265abbbc011e42d78fcbd6cef90a2e41 : SUCCESS in 4m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a439771822fa49d8814eef6107365ede : SUCCESS in 2m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4b179c04b7aa491ba4e5977e03b58de8 : SUCCESS in 30m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/919f587e30e545eb89d95d172227fb03 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":38},{"id":"f8315f679e15758ec0f4481947b18a49625c88a0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-03-01 08:34:08.000000000","message":"Uploaded patch set 39.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":39},{"id":"a52d458bd0cb660558871c1879a833addf7a7f01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 09:37:27.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1385bd2c4cb049ad932fae013059a056\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c5e427d1eab24bb09d39a63185f3aafb : SUCCESS in 4m 36s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fd1c41f3ed834723ab67519638267ae2 : SUCCESS in 3m 27s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/a8e025b0a6954f3a8ab131ebb1608ce4 : SUCCESS in 46m 00s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/185b5e59c34844e2a0d1590f364312d1 : SUCCESS in 45m 29s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/1225d4b4e62b4ec3adc8daed82586679 : SUCCESS in 49m 01s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/879636d7edb443bc9f861879ce1c0396 : SUCCESS in 22m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/824562d8caeb4cd79f8d4cd07a28cd80 : SUCCESS in 55m 33s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c51d467597244ac09ef1a4ac9c517520 : SUCCESS in 49m 49s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/920ab80823ae4c95a012f18df7c0ea6a : SUCCESS in 55m 18s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/81ffc477a44643cd82bf4f21ef3dc20e : SUCCESS in 58m 18s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/e11cbdc26d344244aeb969f26e7bb6d0 : SUCCESS in 39m 44s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/448de3ed945843e2ab0206186081a39f : SUCCESS in 43m 30s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/c18731b1415b4f81b5190a8b8acbe38d : SUCCESS in 42m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59e7a21f5e8b4394ae0fb522f6cc30e0 : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef335bdcf43a4fd39408170b7cd38b69 : SUCCESS in 2m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59609c80ab1a4e4b82fa00d864d71b5f : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a8cb7c166a364d5b9f77da49a7c27ebe : SUCCESS in 4m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/37baad78bb4541718a23bb428345a6b0 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a488c44ecce4a418998d4c12d8ea4ee : SUCCESS in 7m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8ec61d16423b4e3d825828f5c61fc8f8 : SUCCESS in 2m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/18d733c8efd34c97a865d47f2db8e6b5 : SUCCESS in 20m 05s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/053c1e2d150849438ad62fb711fc6fb0 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":39},{"id":"44ae4b20d306060224ce3506033f09539ef76202","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-03-01 12:06:27.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":39},{"id":"a8f74c5d9b12b31f49506d1423e997ca85a6c18a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-03-27 05:53:36.000000000","message":"Uploaded patch set 40.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":40},{"id":"9b5849cb699415c80879709175e8c3a1d6c09687","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 07:11:27.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55bfe3ff02184d29b0b54e49e2aa3a80\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3bc15f0da43646e6811b0da23767e461 : SUCCESS in 2m 21s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6480f68f4e4b467fba16848ed7955ffc : SUCCESS in 1m 49s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/21b103201ff44105a9c0eaaa24915219 : SUCCESS in 52m 19s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ac4d89c27b7441eda61fc2b79cd8b8e5 : SUCCESS in 46m 24s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/3f01840a7f124358a08d253518f91d6e : SUCCESS in 49m 13s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/df58ed88e32247ae8ef9fe4390216627 : SUCCESS in 31m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/716628c75e864dbeb471bc9b5820bbc6 : SUCCESS in 56m 16s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/73b147a4d8044c0eac81e13b59cdca30 : SUCCESS in 56m 15s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/af5e275a8d094d5ab85b090d35721c1e : SUCCESS in 1h 12m 48s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/62c69ab6d0dc4ffabd2f78a4d68d79e8 : SUCCESS in 1h 06m 37s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/642040f6d8c94081baed60e0c813b3dc : FAILURE in 48m 40s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/b43dcd530450497cbaea94ad4ac3a713 : SUCCESS in 44m 07s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/57d87558934f494db6468e4f5afd98d6 : SUCCESS in 47m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3001d0ca2f0948f7a11fd5788b03d4d4 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d860d38a7c004ca9a6f5a0de88a464e1 : SUCCESS in 2m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/edf31fa9b4ac4691ae3bdf91d7538703 : SUCCESS in 6m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e11a3ff347e546d9b444e135946680b0 : SUCCESS in 4m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b1ea2ab02d9140b393121bfe3ea9bd71 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ba2dc91fb884379bc7461162574b787 : SUCCESS in 9m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/821a5529c3a940bb93c4a0cfba31e3b3 : SUCCESS in 2m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/be82dcb3b10b43d58e5035fe9b039b18 : SUCCESS in 32m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e714d84968db4445b10bc766efa9fea7 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":40},{"id":"f461e558177e6e136dce52ca3091419ea1ca4f64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-04-27 12:40:01.000000000","message":"Uploaded patch set 41.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":41},{"id":"0c6ec726cc820695f798f8723f2d8ed49ad2c783","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 14:23:48.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a59871c8fe44da4bfef2ced71bc486d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7e2d833ea21d4844b92dfcf18786d16b : SUCCESS in 4m 41s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b5df52458eca4b30b80288c07c67f966 : SUCCESS in 4m 19s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/4ded857fcb344243a4e0564a601f3d0a : SUCCESS in 41m 13s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1e8abd381bb84fb9b2f8dc144f31b25c : SUCCESS in 58m 39s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/d4b2aca246164b9c8b6ebb12219cb583 : SUCCESS in 54m 34s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/cf5392b6413a4b59a42bbbd9fcbeec22 : SUCCESS in 25m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8b099f45cb954f2891e4d897005f2cba : SUCCESS in 1h 34m 39s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0dca346bbd7f4f9eb7f86d7a0d1a4c4b : SUCCESS in 1h 02m 21s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/f5d5c3cc6997493aa2dcae3a928e68da : SUCCESS in 1h 15m 31s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/af2481366e5741b89a3cbdf1a33f06ee : SUCCESS in 1h 11m 09s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/5af0f9b801464d86a21fc987b6044dd6 : FAILURE in 48m 37s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/17e7a8e64d9e413ab70914d9faaf484b : SUCCESS in 48m 48s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/66e811740a0f4b0fba5c984b315a5efb : SUCCESS in 56m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77174c4861a44c6cb37fd514249a5e80 : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/001dd1033e64414c9f33322dacc1217f : SUCCESS in 3m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ddbfc1357444f44b61684d19cf38ca1 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c5c63d9987a4820a53f84066cb72908 : SUCCESS in 5m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9d8f53797ef3433497fd869c6640a1b4 : POST_FAILURE in 17m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4928e5e7f6ea4ea98fea4c2039237684 : SUCCESS in 11m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/246d9bdd89624061b499a3f38cb22099 : SUCCESS in 2m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2a44791659ac4134946d939b2e6d38c4 : SUCCESS in 36m 00s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/643c80954f68417ab7f04f0d4400efeb : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":41},{"id":"3c13d130c926e04a0b9f349d54b5d0d61a2e5efb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-04-28 05:50:41.000000000","message":"Uploaded patch set 42.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":42},{"id":"63c8af69e81eb2768ad7c7133608ef84c4dee47d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 07:13:30.000000000","message":"Patch Set 42: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0658882ccd34176af2492d47a33dfb4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b30caaddbe6e41a49ce2c2e733731510 : SUCCESS in 3m 44s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2d1d1bb7fba343a090dc4265b31ffef0 : SUCCESS in 2m 26s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/bdacc4eaecdb46a0b0c1b4e74bd04938 : FAILURE in 47m 29s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/94ddcca231fc4e3b9412236f33152542 : FAILURE in 53m 10s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/f1e833893f5149d58d9c6fd5d25cb72b : FAILURE in 41m 42s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/2dc449434180426f8f1574e002ab5a01 : SUCCESS in 17m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fba40016ee2942f3acd84ea801eacb47 : FAILURE in 1h 10m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2f035d3ec6a1430d844674fb9f2d4ed9 : FAILURE in 1h 17m 20s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/13216b742ea84f228f0076ba57c82afa : FAILURE in 1h 16m 15s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/d882468fca7e44dabf0ffb25cc58445d : FAILURE in 1h 14m 28s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/3a310a693e7f46d194102b7439c7e325 : FAILURE in 46m 43s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/b00e24b68cb5472999570060856ffbee : FAILURE in 55m 14s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/8a89238dd330404faf9776520a572227 : FAILURE in 45m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8d886297e7743908e3ab7378c1d046b : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/233caa72760d4043aef0d3904c8a07c1 : FAILURE in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b08509960ff54b65b0e025b02a83f047 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5293ad9e6d7c499fab4690532ca21514 : SUCCESS in 2m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2a75c9c166d6430cbaa0a31b339ff940 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19e442b54da24439a643bcd8dcd129c3 : SUCCESS in 7m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/017513eae357493b8fc8c8051477e957 : SUCCESS in 2m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/34746d77abbf46bd9133b0e759c42074 : SUCCESS in 31m 16s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0cf8f9dedd654765bac44b29bacd90fc : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":42},{"id":"3d6ac7d069b2e139a9423c274a898968ab11f943","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-04-28 09:11:44.000000000","message":"Patch Set 42: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"e3b99890e1d1ce0852e5fc73c8158da954266a7f","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-04-28 09:15:11.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"e0b528a94d593502c06b4c60073d1b0c01f97128","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2026-04-30 09:12:52.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"3ade9f216f9a06ef3f8a436e70b09fb599ed8b1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-04-30 11:19:40.000000000","message":"Uploaded patch set 43.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":43},{"id":"d6af0958279a3e427d5cacc0bbf26cdec8f84c89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 12:43:47.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd181baa71474eb8b50e88a8af90be77\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/675603fba48e4c08ad8446740d0e32cd : SUCCESS in 2m 45s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f650d0b26d50474192815fd4e73c6e58 : SUCCESS in 2m 37s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/1ebea762b2a6432ebcf98e1e540a632e : FAILURE in 45m 11s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/00208a9ec3e543d1a5384862b61326eb : FAILURE in 43m 48s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/e3b277cf7ec44f06a3d84325d7427acd : FAILURE in 41m 13s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/94310b018068499bb82b28999579dfd6 : SUCCESS in 18m 54s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/72a10d8ba26842799e5ef5778ac21b7d : FAILURE in 1h 18m 32s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d5e2b396987d483da0dda02db882c11d : FAILURE in 59m 40s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/801b4bd2298d40c6880b1406c8d78227 : FAILURE in 1h 01m 12s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/1cbdf8c42cc44003a9057c02813eefed : FAILURE in 1h 13m 38s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/291da319e7df406089777817236224b8 : FAILURE in 55m 27s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/ec593045bcaf4aa1824cc909c41cb6ae : POST_FAILURE in 29m 42s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/7f91fc4f408b4381b039fa026efe1524 : FAILURE in 41m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6b96f8a296f4b0d9db2e7aa8409b750 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f53957ea9b2412fa7d624bb3cf215ca : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/63cfe3e5dc2641e2b26fc3e6bdf79de4 : SUCCESS in 5m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/908f90c91f024b32a98d723220809f84 : SUCCESS in 4m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/89bdef294bd44865a623fe6de413d1e5 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f45b5a669c44bccb498df81453fa321 : SUCCESS in 5m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f29c0f904c974b2ea4555da60ac9ae95 : SUCCESS in 2m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/99f202160667434590868c5fd636fb26 : SUCCESS in 30m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/21330fecbb90407a8ed58782b4467c6e : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":43},{"id":"6e0910c8ad94997cd6faee3a345418ece71de01c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-04-30 13:05:28.000000000","message":"Uploaded patch set 44.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":44},{"id":"3bb4291e3072ca689ecfa98ce4bb5abde2ae8394","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 14:16:59.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec2d68deb52244c6a459bc97c95134d5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8949ec544b204e8796f62777a88d0a45 : SUCCESS in 4m 27s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dd4e51b8eb18480d9cd8ad1cb2617190 : SUCCESS in 1m 33s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/07f5a3917e54435d93087b89f84a59f3 : SUCCESS in 41m 17s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/62e856cab1274bfc9df38a11aff76adb : SUCCESS in 46m 47s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/afaaab1667c140e9bc25f871c7d38ef9 : SUCCESS in 41m 27s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/ae021da6cba24ced9e8a159838d15a2c : SUCCESS in 17m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/04530f055cdf4e1197c6f7c83ee7f53c : SUCCESS in 1h 03m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c16093ea818e44f68b9e30549916cfe8 : SUCCESS in 1h 02m 06s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/d81b99ed9660498e9c7ddd6bffdda4f2 : SUCCESS in 1h 09m 53s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/c3c38fbb1b064c46b81e295b3f031fd9 : SUCCESS in 54m 53s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/3dfed1356c174025a57c981c32a6b913 : SUCCESS in 37m 43s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/9ef76a6c5b53418981c426da6da7ef10 : SUCCESS in 36m 26s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/05c1004145634c61aecacb5cf60d8ffe : SUCCESS in 47m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02097bf963514bd1b3ecb012a589286d : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f1a2df91a3544989a719a71262c8d80 : SUCCESS in 2m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2b1fc326bb79435ba143f501ce34fb68 : SUCCESS in 6m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20a918c17f54406095789bcb6b60295f : SUCCESS in 5m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1f4a934af0b745dbbd72d826c96076b0 : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9d7dfa7fa75405490f3f7f691e58a21 : SUCCESS in 9m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0aedee0f47d64a7e8ec36c0fbbc73e98 : SUCCESS in 2m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5288c4cf4bd1493d9e918458e11070d8 : SUCCESS in 34m 00s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4a66f8ce60084db5acb5264976521e77 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":44},{"id":"bf8e7d3f3ff6dbaf4fdf3c5ecff6c3ba8aca5dee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-05-01 07:09:54.000000000","message":"Uploaded patch set 45.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":45},{"id":"445f3050324d47b137d476fb696bab8120ed4851","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 08:29:45.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3630c2a841084198925ea47efc6ada42\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/529da4cc54f347eba454805a2fe7a00b : SUCCESS in 3m 22s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/36a477a1c0b249178231a9ae46d4bfa6 : SUCCESS in 2m 32s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/596906a07dac41958321467ffff73d56 : SUCCESS in 51m 51s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c8ff3b1a3f644a8eaf84d8ba782629b7 : SUCCESS in 48m 20s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/b34cbef7836f487fbb0562e450009c9b : SUCCESS in 49m 04s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/036952a4813447bdae9091992c29f3c9 : SUCCESS in 31m 30s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e67628f17854475fb3f70fca4b97dce5 : SUCCESS in 1h 15m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7df84e53688e449ab951028f94ccbd3f : SUCCESS in 1h 09m 26s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/e2064da5999d40cc908024203f22892a : FAILURE in 1h 13m 41s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/9dd7098d57d245788b8a3af9ea8f2c5a : SUCCESS in 1h 08m 32s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/6bd8990d22414a05bd723ab008b2b445 : SUCCESS in 47m 52s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/b23f22d191b74a72b13e37ec7de87035 : SUCCESS in 31m 56s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/db20219f5f3843f3a3813d2f90c71531 : SUCCESS in 47m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92fbbaa6df154b1085fef241707a46a4 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f01dbbe1c1c47fb800ed54c60d634ed : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/26e395c03202456998176e69baed46a8 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5bfa1500ec1f45908c365edf44bcda93 : SUCCESS in 4m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/047b6f512bf347cfa3a8d7e660ce1224 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbb75a0827db427480e223c8c5725001 : SUCCESS in 9m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9d56bb5917a34aadaa4157fed744e1e0 : SUCCESS in 2m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dfecc78743394117b52ffacef9be24fd : SUCCESS in 34m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3d4e5a03f5c847a3b361c9caaec84bd8 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":45},{"id":"552f59a7019ddaa6670122efe7e90e0130fe70cc","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-05-01 08:33:37.000000000","message":"Patch Set 45:\n\n(1 comment)","accounts_in_message":[],"_revision_number":45},{"id":"16eb8a30781920ca241b1ff54c72a6a74b58a14e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2026-05-01 08:49:18.000000000","message":"Uploaded patch set 46.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":46},{"id":"053cb13cec055bed4f323cec640ac7ffc0ec5631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 10:06:40.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed212f3c9f884953bc33774f2abe85b9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/befbb52568204fa08e87ab836c3b9c89 : SUCCESS in 2m 10s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/550f3fa409cd4a4faa250a070c182752 : SUCCESS in 2m 24s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/f21426cf10b245a4b77c50bbf2b2e874 : SUCCESS in 46m 13s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/860ad8b57fa54e9abcf908d4f232ee32 : SUCCESS in 39m 20s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/bd7d6252fd7f490c9f5a4968a6611213 : SUCCESS in 50m 19s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/fcbeaaad254c4718a1681743d83eaca1 : SUCCESS in 14m 41s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c1e724f952724d80a52675272f05173f : SUCCESS in 53m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/aeda4292c78e4cbd8bcf99befc86e2c6 : SUCCESS in 1h 15m 14s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/1b198f62d27c40448e4635484febd064 : SUCCESS in 57m 53s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/1c8e5229e2c145aca7b33449a2af465a : SUCCESS in 54m 29s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/be6cc53ec9114e76a6ff36974a6f4283 : SUCCESS in 32m 18s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/4edbb9578706435a8b902abed190e174 : SUCCESS in 35m 33s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/d12d00823ae445e68f0541e70a1ca38e : SUCCESS in 42m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c02b5f7f0ca4a9ab35cc4c9662cbead : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8f34e4e0e944260a0a109a1d737cd1a : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b4907f01a136436a8a3a60bb6f3961a1 : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9608ff46818e4c19961a50898a58a1bc : SUCCESS in 6m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/648aa6173728435eb09573914ddec292 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebe520daa2284227ac62de80321143d0 : SUCCESS in 4m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f20147f7f75e4374b2281bb0b609c732 : SUCCESS in 3m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1d16a994410345b69a79a884cc1c3a0c : SUCCESS in 34m 30s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3dd45c0e945849c8bede7374e6f14222 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":46}],"current_revision_number":46,"current_revision":"0d8c30f9d62f8b7687a00b397c0a9c80b90f988f","revisions":{"4a9c704363418f7dc15cbc4ec69eccb32322c156":{"kind":"REWORK","_number":1,"created":"2025-12-03 16:05:11.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/1"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4a9c704363418f7dc15cbc4ec69eccb32322c156"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4a9c704363418f7dc15cbc4ec69eccb32322c156"}]},"branch":"refs/heads/master"},"88c607ce56ccf9b365af3a5ffd556f34b799add3":{"kind":"REWORK","_number":2,"created":"2025-12-07 01:20:59.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/2"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-07 01:17:57.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/88c607ce56ccf9b365af3a5ffd556f34b799add3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/88c607ce56ccf9b365af3a5ffd556f34b799add3"}]},"branch":"refs/heads/master"},"b6d2868f8f0ec18b393902f57fadeeb4a6a6bd5b":{"kind":"REWORK","_number":3,"created":"2025-12-07 01:27:40.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/3"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-07 01:27:26.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b6d2868f8f0ec18b393902f57fadeeb4a6a6bd5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b6d2868f8f0ec18b393902f57fadeeb4a6a6bd5b"}]},"branch":"refs/heads/master"},"40241c28feb8bc53649fb858ce8e06671efbf693":{"kind":"REWORK","_number":4,"created":"2025-12-07 01:30:57.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/4"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-07 01:30:43.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/40241c28feb8bc53649fb858ce8e06671efbf693"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/40241c28feb8bc53649fb858ce8e06671efbf693"}]},"branch":"refs/heads/master"},"330e5be86d0d766ae928a9783b41efea8f21f2e9":{"kind":"REWORK","_number":5,"created":"2025-12-07 06:41:42.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/5"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","date":"2025-12-07 06:41:14.000000000","tz":180},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/330e5be86d0d766ae928a9783b41efea8f21f2e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/330e5be86d0d766ae928a9783b41efea8f21f2e9"}]},"branch":"refs/heads/master"},"a97b60b72d1134ee76e01e1f08904696303ccd16":{"kind":"REWORK","_number":6,"created":"2025-12-07 08:20:28.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/6"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-07 08:20:11.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a97b60b72d1134ee76e01e1f08904696303ccd16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a97b60b72d1134ee76e01e1f08904696303ccd16"}]},"branch":"refs/heads/master"},"ca46db707e1467fd4c26c496e39855a55459c731":{"kind":"REWORK","_number":7,"created":"2025-12-07 08:43:49.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/7"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-07 08:43:33.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ca46db707e1467fd4c26c496e39855a55459c731"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ca46db707e1467fd4c26c496e39855a55459c731"}]},"branch":"refs/heads/master"},"bed0dccd80fc14b0ccce1108e60387c7170db23b":{"kind":"REWORK","_number":8,"created":"2025-12-07 09:01:42.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/8"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-07 09:01:28.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bed0dccd80fc14b0ccce1108e60387c7170db23b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bed0dccd80fc14b0ccce1108e60387c7170db23b"}]},"branch":"refs/heads/master"},"47054251620e7fdaf3b7ff495494575c0d7cb88a":{"kind":"REWORK","_number":9,"created":"2025-12-07 09:21:37.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/9"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-07 09:21:24.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/47054251620e7fdaf3b7ff495494575c0d7cb88a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/47054251620e7fdaf3b7ff495494575c0d7cb88a"}]},"branch":"refs/heads/master"},"a51ab132fd85e592e5bffe6278846b6f0a6f02a1":{"kind":"REWORK","_number":10,"created":"2025-12-07 12:11:31.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/10"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-07 12:11:16.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a51ab132fd85e592e5bffe6278846b6f0a6f02a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a51ab132fd85e592e5bffe6278846b6f0a6f02a1"}]},"branch":"refs/heads/master"},"27f27a0bccbb5267ac94d76ae027dec2069102f5":{"kind":"REWORK","_number":11,"created":"2025-12-07 14:06:08.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/11"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-07 14:05:47.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/27f27a0bccbb5267ac94d76ae027dec2069102f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/27f27a0bccbb5267ac94d76ae027dec2069102f5"}]},"branch":"refs/heads/master"},"50c2db443c87f8d405ab4effbd32511445e7913c":{"kind":"REWORK","_number":12,"created":"2026-01-07 06:20:47.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/12"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-01-07 06:20:21.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/50c2db443c87f8d405ab4effbd32511445e7913c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/50c2db443c87f8d405ab4effbd32511445e7913c"}]},"branch":"refs/heads/master"},"865c973c203224bf4aeb92f292b6d8e93c5cf1bb":{"kind":"REWORK","_number":13,"created":"2026-01-07 06:27:40.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/13"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-01-07 06:27:25.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/865c973c203224bf4aeb92f292b6d8e93c5cf1bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/865c973c203224bf4aeb92f292b6d8e93c5cf1bb"}]},"branch":"refs/heads/master"},"a1bad6190f75f44c85d4752b1f5bcb47993f2fcb":{"kind":"REWORK","_number":14,"created":"2026-01-07 14:27:20.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/14"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-01-07 14:27:00.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a1bad6190f75f44c85d4752b1f5bcb47993f2fcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a1bad6190f75f44c85d4752b1f5bcb47993f2fcb"}]},"branch":"refs/heads/master"},"d7141bdb9a02a3365beb812a4927bb2b5e1d0b8f":{"kind":"NO_CHANGE","_number":15,"created":"2026-01-08 00:27:37.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/15"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-01-08 00:27:19.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d7141bdb9a02a3365beb812a4927bb2b5e1d0b8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d7141bdb9a02a3365beb812a4927bb2b5e1d0b8f"}]},"branch":"refs/heads/master"},"f5827606996c57b83ef3bece05ecdabe1f6b61c9":{"kind":"REWORK","_number":16,"created":"2026-01-14 01:46:47.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/16"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-01-14 01:46:30.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5827606996c57b83ef3bece05ecdabe1f6b61c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5827606996c57b83ef3bece05ecdabe1f6b61c9"}]},"branch":"refs/heads/master"},"7ba778a80560a6fd9eda51476d594353141dbd30":{"kind":"REWORK","_number":17,"created":"2026-01-14 07:13:43.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/17"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-01-14 07:13:29.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7ba778a80560a6fd9eda51476d594353141dbd30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7ba778a80560a6fd9eda51476d594353141dbd30"}]},"branch":"refs/heads/master"},"e165b80445cd0a173b95aed302faed2ed0c402e0":{"kind":"REWORK","_number":18,"created":"2026-01-14 07:23:15.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/18"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-01-14 07:23:02.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e165b80445cd0a173b95aed302faed2ed0c402e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e165b80445cd0a173b95aed302faed2ed0c402e0"}]},"branch":"refs/heads/master"},"d427c5e7228acc3d049bdab05cd51e1e68b72dc6":{"kind":"REWORK","_number":19,"created":"2026-01-14 07:27:32.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/19"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-01-14 07:27:17.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d427c5e7228acc3d049bdab05cd51e1e68b72dc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d427c5e7228acc3d049bdab05cd51e1e68b72dc6"}]},"branch":"refs/heads/master"},"dd0b064dfa103ef710718838c34cb5069f68a85e":{"kind":"REWORK","_number":20,"created":"2026-01-14 08:05:25.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/20"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-01-14 08:05:10.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dd0b064dfa103ef710718838c34cb5069f68a85e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dd0b064dfa103ef710718838c34cb5069f68a85e"}]},"branch":"refs/heads/master"},"0779f5a1212cef184afbf37cf2529e13c3adf47e":{"kind":"REWORK","_number":21,"created":"2026-01-14 08:06:30.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/21"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-01-14 08:06:17.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0779f5a1212cef184afbf37cf2529e13c3adf47e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0779f5a1212cef184afbf37cf2529e13c3adf47e"}]},"branch":"refs/heads/master"},"cb5cb72ea60fd26e177fb30d5b008400a04c7275":{"kind":"REWORK","_number":22,"created":"2026-01-14 09:28:22.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/22"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-01-14 09:28:07.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cb5cb72ea60fd26e177fb30d5b008400a04c7275"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cb5cb72ea60fd26e177fb30d5b008400a04c7275"}]},"branch":"refs/heads/master"},"b8221139ea5bb08be0376c48117959bec53b7199":{"kind":"NO_CHANGE","_number":23,"created":"2026-01-14 09:28:59.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/23"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-01-14 09:28:47.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b8221139ea5bb08be0376c48117959bec53b7199"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b8221139ea5bb08be0376c48117959bec53b7199"}]},"branch":"refs/heads/master"},"ac78923cd7809dc7274ed51d5c5e6ec498ee2345":{"kind":"REWORK","_number":24,"created":"2026-01-14 09:29:19.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/24"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-01-14 09:29:06.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ac78923cd7809dc7274ed51d5c5e6ec498ee2345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ac78923cd7809dc7274ed51d5c5e6ec498ee2345"}]},"branch":"refs/heads/master"},"eac713f856df184e408d7de7c6cf96a9e711b108":{"kind":"REWORK","_number":25,"created":"2026-01-14 12:22:35.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/25"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-01-14 12:22:21.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eac713f856df184e408d7de7c6cf96a9e711b108"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eac713f856df184e408d7de7c6cf96a9e711b108"}]},"branch":"refs/heads/master"},"b4c74d4eec8615c25d9513d2319cdafff4cfe710":{"kind":"REWORK","_number":26,"created":"2026-01-14 12:24:40.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/26"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-01-14 12:24:26.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b4c74d4eec8615c25d9513d2319cdafff4cfe710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b4c74d4eec8615c25d9513d2319cdafff4cfe710"}]},"branch":"refs/heads/master"},"8af8b7743faaa0ff97c0cd019c0d439a4f4820bc":{"kind":"REWORK","_number":27,"created":"2026-01-14 12:36:04.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/27"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-01-14 12:35:50.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8af8b7743faaa0ff97c0cd019c0d439a4f4820bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8af8b7743faaa0ff97c0cd019c0d439a4f4820bc"}]},"branch":"refs/heads/master"},"8bd7bf70248a93b6b75dbdf3d7d94f99c7b0637e":{"kind":"REWORK","_number":28,"created":"2026-02-22 00:07:34.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/28"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-02-22 00:07:11.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8bd7bf70248a93b6b75dbdf3d7d94f99c7b0637e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8bd7bf70248a93b6b75dbdf3d7d94f99c7b0637e"}]},"branch":"refs/heads/master"},"a417d7b7d6a099395a907ff4ae572051aa7689e7":{"kind":"REWORK","_number":29,"created":"2026-02-22 01:08:13.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/29"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-02-22 01:07:58.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a417d7b7d6a099395a907ff4ae572051aa7689e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a417d7b7d6a099395a907ff4ae572051aa7689e7"}]},"branch":"refs/heads/master"},"f1e62995ad6b95a73d1dab01d40b65348da2db37":{"kind":"REWORK","_number":30,"created":"2026-02-22 01:29:27.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/30"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-02-22 01:29:10.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f1e62995ad6b95a73d1dab01d40b65348da2db37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f1e62995ad6b95a73d1dab01d40b65348da2db37"}]},"branch":"refs/heads/master"},"c00b1db64800c56e9028473a5eedde0cc483e973":{"kind":"REWORK","_number":31,"created":"2026-02-22 05:43:30.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/31"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-02-22 05:43:00.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c00b1db64800c56e9028473a5eedde0cc483e973"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c00b1db64800c56e9028473a5eedde0cc483e973"}]},"branch":"refs/heads/master"},"9e9cbd59481a7c24b2bc0fae49ec50074c9cd394":{"kind":"REWORK","_number":32,"created":"2026-02-22 05:46:16.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/32"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-02-22 05:45:59.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9e9cbd59481a7c24b2bc0fae49ec50074c9cd394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9e9cbd59481a7c24b2bc0fae49ec50074c9cd394"}]},"branch":"refs/heads/master"},"dc490102e79566f35d70b6a8ba4f592603c0809f":{"kind":"REWORK","_number":33,"created":"2026-02-22 05:51:34.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/33"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-02-22 05:51:20.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc490102e79566f35d70b6a8ba4f592603c0809f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc490102e79566f35d70b6a8ba4f592603c0809f"}]},"branch":"refs/heads/master"},"a7bf70040d33147753a3ba4efa72a10ef095e70d":{"kind":"REWORK","_number":34,"created":"2026-02-22 11:08:14.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/34"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-02-22 11:07:58.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a7bf70040d33147753a3ba4efa72a10ef095e70d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a7bf70040d33147753a3ba4efa72a10ef095e70d"}]},"branch":"refs/heads/master"},"ccaee6ad38512cd3125c9ffb3d83bc76eaf62d31":{"kind":"REWORK","_number":35,"created":"2026-02-25 13:22:35.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/35"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-02-25 13:22:13.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ccaee6ad38512cd3125c9ffb3d83bc76eaf62d31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ccaee6ad38512cd3125c9ffb3d83bc76eaf62d31"}]},"branch":"refs/heads/master"},"d6d5c621c0817861c9318c9576bcb6275e2c0d8d":{"kind":"REWORK","_number":36,"created":"2026-03-01 02:31:31.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/36"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-03-01 02:31:14.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6d5c621c0817861c9318c9576bcb6275e2c0d8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6d5c621c0817861c9318c9576bcb6275e2c0d8d"}]},"branch":"refs/heads/master"},"8c1cb7ab79bcbdfbf683b5583eb9092349ae531f":{"kind":"REWORK","_number":37,"created":"2026-03-01 05:40:20.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/37"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-03-01 05:40:05.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8c1cb7ab79bcbdfbf683b5583eb9092349ae531f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8c1cb7ab79bcbdfbf683b5583eb9092349ae531f"}]},"branch":"refs/heads/master"},"f55759f671aa3e858e52e1d379d5f0d07dcdd3f9":{"kind":"REWORK","_number":38,"created":"2026-03-01 06:49:53.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/38"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-03-01 06:49:26.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f55759f671aa3e858e52e1d379d5f0d07dcdd3f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f55759f671aa3e858e52e1d379d5f0d07dcdd3f9"}]},"branch":"refs/heads/master"},"1e1c7556557f3956ca358f3d574d96985e2bfc1c":{"kind":"REWORK","_number":39,"created":"2026-03-01 08:34:08.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/39"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-03-01 08:33:52.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nCloses-Bug: 1690184\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e1c7556557f3956ca358f3d574d96985e2bfc1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e1c7556557f3956ca358f3d574d96985e2bfc1c"}]},"branch":"refs/heads/master"},"11daa4d44ff41a42fc8dc3d3f990e29b2b609731":{"kind":"REWORK","_number":40,"created":"2026-03-27 05:53:36.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/40"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-03-27 05:53:03.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nImplements: blueprint share-pools\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/11daa4d44ff41a42fc8dc3d3f990e29b2b609731"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/11daa4d44ff41a42fc8dc3d3f990e29b2b609731"}]},"branch":"refs/heads/master"},"450ba825262fee8e6a936d4023de967f55ae37d1":{"kind":"REWORK","_number":41,"created":"2026-04-27 12:40:01.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/41"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-04-27 12:39:44.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nImplements: blueprint share-pools\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/450ba825262fee8e6a936d4023de967f55ae37d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/450ba825262fee8e6a936d4023de967f55ae37d1"}]},"branch":"refs/heads/master"},"05dfe839336228c1e7a49875aebe349b3e16c5db":{"kind":"REWORK","_number":42,"created":"2026-04-28 05:50:41.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/42"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-04-28 05:50:24.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nImplements: blueprint share-pools\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05dfe839336228c1e7a49875aebe349b3e16c5db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05dfe839336228c1e7a49875aebe349b3e16c5db"}]},"branch":"refs/heads/master"},"e8be13eb24780723f8204d6bd30d25bc8a83a482":{"kind":"REWORK","_number":43,"created":"2026-04-30 11:19:40.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/43"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-04-30 11:19:22.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nImplements: blueprint share-pools\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e8be13eb24780723f8204d6bd30d25bc8a83a482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e8be13eb24780723f8204d6bd30d25bc8a83a482"}]},"branch":"refs/heads/master"},"eb0f3a39a29c13bbe2046f00c06cf2a0103fe29e":{"kind":"REWORK","_number":44,"created":"2026-04-30 13:05:28.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/44"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-04-30 13:05:12.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nImplements: blueprint share-pools\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb0f3a39a29c13bbe2046f00c06cf2a0103fe29e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb0f3a39a29c13bbe2046f00c06cf2a0103fe29e"}]},"branch":"refs/heads/master"},"40f8152bdb309c6c6f9f8adceb4a06a34a85aa59":{"kind":"REWORK","_number":45,"created":"2026-05-01 07:09:54.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/45"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-05-01 07:09:39.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nImplements: blueprint share-pools\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/40f8152bdb309c6c6f9f8adceb4a06a34a85aa59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/40f8152bdb309c6c6f9f8adceb4a06a34a85aa59"}]},"branch":"refs/heads/master"},"0d8c30f9d62f8b7687a00b397c0a9c80b90f988f":{"kind":"REWORK","_number":46,"created":"2026-05-01 08:49:18.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/969490/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/969490/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/969490/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/969490/46"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-12-03 16:03:28.000000000","tz":420},"committer":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2026-05-01 08:49:03.000000000","tz":420},"subject":"Allow share pools for domains.","message":"Allow share pools for domains.\n\nAdd domain_id field to pool object and endpoints to share pool between\ndomains.\n\nImplements: blueprint share-pools\nChange-Id: I27904f83bdefe1dff03b913cfc9b0fcfef81d451\nSigned-off-by: mikhails \u003cmikhailsamoiloff@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0d8c30f9d62f8b7687a00b397c0a9c80b90f988f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0d8c30f9d62f8b7687a00b397c0a9c80b90f988f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
