)]}'
{"id":"openstack%2Fdesignate~971719","triplet_id":"openstack%2Fdesignate~master~Ifa7dda8f9b1c5b2890688bffc976ceeb70532991","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-01-02 16:25:52.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-01-02 20:34:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifa7dda8f9b1c5b2890688bffc976ceeb70532991","subject":"Use real abstract class","status":"MERGED","created":"2025-12-23 17:24:13.000000000","updated":"2026-01-02 20:35:16.000000000","submitted":"2026-01-02 20:34:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971719","meta_rev_id":"ab33d3d44806058261d521ce5d55f13aa6973d2e","_number":971719,"virtual_id_number":971719,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-02 20:34:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-01-02 16:25:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-12-29 09:50:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-12-29 09:30:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-01-02 16:25:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-23 18:45:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-29 09:30:55.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2025-12-29 09:50:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-12-29 09:50:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"8b997a8c0b2cd5ce507ea837afc8ea5412284ad7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-23 17:24:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fa183f05c4966b2ac101d144a1ff3bdc4db6485","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-23 18:45:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c31f5ba789f43d4b8fcd0e301da37c3\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/84e1c748ba9f4fb1b0432795e85de2b9 : SUCCESS in 50m 01s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/046ce8fd58104351bad4df1f3f8e7e2a : SUCCESS in 36m 57s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/f832c14482e2403ba2b573d292f5d797 : SUCCESS in 46m 26s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/542a9a098bc44251bf753b09560e5191 : SUCCESS in 15m 48s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/27245969f0ba4f5eb7ad13122529b798 : SUCCESS in 1h 11m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/da539e37507542298da303cbd7c139f8 : SUCCESS in 1h 05m 32s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/fe8c29f4c25a4d73b04f9904af222661 : SUCCESS in 1h 15m 05s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/22cbd73e2acc4104b03058ad85e92fbb : SUCCESS in 49m 40s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/33a8639dac334f7da406a286ee10633c : SUCCESS in 36m 10s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/38192c3a071749089cd2461db035d606 : SUCCESS in 52m 48s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/18ba9740a3eb415b941a27d8ed7fc525 : SUCCESS in 46m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/181d2d07055c411ca6a8119f97b80994 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46d44300f75a4d579b7ad3beb3e15672 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee2d946f9bcc49a49ccda3f4b5412edb : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ec8d6f0c22234e2f9f51af4849dd1bcb : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1beca4f999474220b72d40fd4b5b59e4 : SUCCESS in 9m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dc72896ffccd434f9f771863195a073b : SUCCESS in 30m 10s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5b2ae149f936495fbf35ec29f6ebced5 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9a381db1e841840861c4b2386a53231dc92b0220","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-12-29 09:30:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d5a6fd071d19e2ca6131ec7112c9f92c1ffb372b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-29 09:50:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5a38bca08f7ac9a3f5cd2c2c6608d032591458be","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-02 16:25:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f23d1875d43927409a4392e9a57b0b242ddd1188","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 16:26:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fbda5ca8d1a1cab6abb703ccc95966ea616837ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 17:10:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/89285bd0f067491fb19545b269838b09\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/f4826339a9014814b13ff5404459d04b : SUCCESS in 35m 17s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1be8ce9badc04e5a8ed31561e647f80e : FAILURE in 43m 54s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/fcbbae0d325c4a1e8e2e9a1743457590 : CANCELED\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/149842815f6846178878cc32299ddc75 : SUCCESS in 25m 08s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4384634bb6ec4a81a7959c98c564b906 : CANCELED\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e1b4437ea3674196bfd20f681e5cb456 : CANCELED\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/79c73ab428f7464dba58c7d3a1b1e5e3 : CANCELED\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/da699ae531d74fa5b517bacd3ee950fb : CANCELED\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/58b4314a27e84703ac465c30d5f89d13 : CANCELED\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/e738cda89fbe47dabb3ef07b51f5792a : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0476e605a22c4a6c953e4d3153b4aa12 : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47f1ddf305354859b408c2b073a68f36 : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/acc2dcb18e89484f86c91956c5d4c531 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e5e8b0977104ec19d7c52ab81a6bf79 : SUCCESS in 8m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9ced55bda0cd4e68b154c763029d2eac : SUCCESS in 31m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e8bfc00af918ac104ec18ccb6cb0b8116da72fac","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-02 18:03:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e8ad303986e8ade824f94ddae281e9f98e65e992","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 19:19:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/022b8b91d01c404bae1e2bd8e5958145\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/384d027aba944da888b0090408b159fc : SUCCESS in 34m 29s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/cbb069a6fac44879b5887363f5d0436a : SUCCESS in 35m 05s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/47123d6a13e348d5b7d058c6b1408f25 : SUCCESS in 40m 51s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/2338057c79464da186f782e49d30e6dc : SUCCESS in 14m 28s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8ba5e6c81bbf4423964431b76b755006 : SUCCESS in 53m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fdf744ed50b54b1e8677b8c5ed17f3e2 : SUCCESS in 1h 07m 36s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/8d7f2255d79f4d96872cd5517132d7f0 : SUCCESS in 1h 10m 57s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/9b5c41ca5c284e249b728914d3a94d55 : SUCCESS in 49m 26s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/18f9f275a0324a21a81ae9a2b084d9e3 : SUCCESS in 45m 50s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/4fb5be2c504a41f2b029ea4166f867da : SUCCESS in 51m 24s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/7e8fca6c11444e55badfa694f60fdc0f : SUCCESS in 49m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d85f4b656f14188b261313fd1543df5 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76df36468ffe463296cdf2ea5b4882ec : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03fb367e5dff42c58b7f8cab0ba1501b : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bf1502c1a20d4277a7ab957d087c7cb6 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de96db07f0cd44bdbd75b6187501dc74 : SUCCESS in 6m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9015c9bb136743dbaa6608b55c699415 : SUCCESS in 30m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/23461c509e1a40318e55313193ec64fa : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"573b3eb645bc59dc5eb754fcfea7d16704a8d9a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 19:19:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d452eb9c6acf522143840b2a45cd70f11b6549ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 20:34:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/721c709118ff4095a57dcddfb3812658\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/e90985153c7347fb991911385c2c0b32 : SUCCESS in 47m 18s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/bd92169baaf1487ea157fd979de51cab : SUCCESS in 52m 35s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/e566c70347074866af84c02371d0e9ec : SUCCESS in 51m 07s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/2f3abfe7e7c34dd490fda8bc9be4170a : SUCCESS in 22m 38s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/97b30ffa93994f71ac134c5e74483203 : SUCCESS in 51m 05s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/292e917a19ce4b3d81664c313bee99a1 : SUCCESS in 54m 03s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/a576bab221214b4881c814165c1dbc04 : SUCCESS in 1h 04m 32s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/af750b65b9b54288887857b419edc625 : SUCCESS in 1h 09m 59s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/5452f2deedca41eeb975a29ac356d21b : SUCCESS in 35m 26s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/bc76e1eaa46b4181b7acb86f30ac2d62 : SUCCESS in 35m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd8b6d573588472c859cf139cbc06960 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62361e88b10348f896d7c6630898e9c9 : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c998d7412c0e445eb30d254c55010200 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d33d4e60b0ea43faa5c06b63d2b84ea2 : SUCCESS in 7m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1f87a269e27a4479b1933463894eefe3 : SUCCESS in 32m 45s","accounts_in_message":[],"_revision_number":1},{"id":"36cdf015d84716f2c667b0a62bd13c959bf64813","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 20:34:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ab33d3d44806058261d521ce5d55f13aa6973d2e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 20:35:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe2cf0ce4e2046fa85daeed77c17f079\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/329b733f42ff409c86a6dd2242e543fd : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"54068ca7665eb1bdd7ff02aef16a536a1ddfeb0a","revisions":{"54068ca7665eb1bdd7ff02aef16a536a1ddfeb0a":{"kind":"REWORK","_number":1,"created":"2025-12-23 17:24:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/971719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/19/971719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/19/971719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/19/971719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/19/971719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/19/971719/1"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-23 17:19:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-23 17:23:21.000000000","tz":540},"subject":"Use real abstract class","message":"Use real abstract class\n\n... instead of just documenting it. It allows us to ensure that\nthe required interfaces are actually implemented.\n\nAlso remove redundant override of __init__ without no additional\nchange.\n\nChange-Id: Ifa7dda8f9b1c5b2890688bffc976ceeb70532991\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/54068ca7665eb1bdd7ff02aef16a536a1ddfeb0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/54068ca7665eb1bdd7ff02aef16a536a1ddfeb0a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
