)]}'
{"id":"openstack%2Fdesignate~971720","triplet_id":"openstack%2Fdesignate~master~I4a8cf5a54e3de768cd1883b67ce57484c32ce643","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-01-02 16:31:43.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-01-02 23:16:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4a8cf5a54e3de768cd1883b67ce57484c32ce643","subject":"Resolve deprecation warning from alembic","status":"MERGED","created":"2025-12-23 17:28:25.000000000","updated":"2026-01-02 23:16:48.000000000","submitted":"2026-01-02 23:16:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971720","meta_rev_id":"42deadf383646757db5e9859a3504f8f333cb89a","_number":971720,"virtual_id_number":971720,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-02 23:16:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-01-02 16:31:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-12-29 09:50:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-12-29 09:30:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-01-02 16:31:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-23 18:45:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-29 09:30:40.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2025-12-29 09:50:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-12-29 09:50:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"846bad7743d0de2fa57c6901180fa622e43b7579","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-23 17:28:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0c20defa23315ad2340d61a9126714bab309b0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-23 18:45:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad218cd739a6479bb48cb985f591c2ee\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/750a9319df674a97aeacc0931abfcaf2 : SUCCESS in 48m 50s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e1ada30c794f41dfbbb5b1aed5e01ea1 : SUCCESS in 50m 40s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/7da3fc5a820f4a54ad529ebe403f98da : SUCCESS in 48m 44s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/b6d69d31a1ef4f368c510d13cc5ece95 : SUCCESS in 16m 51s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/72ce282b0f964315a7d061bbef573770 : SUCCESS in 49m 36s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9eb837d6b824419e8be6474011efa9f9 : SUCCESS in 1h 08m 55s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/eaeebc3d646044f9a816b901355c89cb : SUCCESS in 54m 05s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/10911ce66e684de0a24d6bd7ce90eeda : SUCCESS in 1h 11m 28s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/788c21cdd5ce4c37b29ecdeb7106597b : SUCCESS in 42m 24s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/582aabbb46a74f1486ba5aea01577ef2 : SUCCESS in 42m 19s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/6227de44b3d04862a98227ead9b6a101 : SUCCESS in 52m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3963144b24644a24823fa061ebeb9e27 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55d15d9728b845b3977538e3cc8e772a : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f660d3f474f649c2b8a37a444fd6e350 : SUCCESS in 6m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/42f95e27787449e9b1b7e7e361b1dac0 : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7af7463def1a406c998ca5eb8d5ac3f9 : SUCCESS in 8m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ddaabec34b534aa4bbe53db27e097e2a : SUCCESS in 30m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/52cb369912574682a06efc5feec432b0 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"77c3581f58bc177f218cfc15125c12f6c1111b31","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-12-29 09:30:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9e2486c8ee435de43ed77cba844e2a271c1d156b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-29 09:50:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7d2ce9f8b62d561cf828f541401e5636fe9ff0b6","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-02 16:31:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"99bd95eddd0eb8a2f921b36a48ea52368ac80e34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 16:32:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73fbecc375b64fb073e695e431a44e250dc69b20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 17:32:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf9c89d103f84f2fadc7d973392b6739\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/9f3b69571d3546bdaf9bbb0cb59b382a : SUCCESS in 39m 57s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a001928cea984dd3bf28fd9cdf87cb04 : CANCELED\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/306d3166c47346389ae5c62e03921f0a : SUCCESS in 48m 32s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/2bc73b14a5374ea7a23cf9f6839bf19d : SUCCESS in 33m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7c537264c8f54bfa8f6bc9c69fcdb0cc : CANCELED\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e49c5ace76c44d01b53cd74dda6dcbcb : CANCELED\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e12dd7a9cc0847738809df3855e55578 : SUCCESS in 51m 41s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/078ae29d38a94654b949219cd025cb6f : CANCELED\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/1ff35a5ce1314aeeb09a744d40a7ffb1 : SUCCESS in 37m 59s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/a76768a748e443fcae06fc5a1b795966 : FAILURE in 50m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/334e436b9f204af785b7bb5ca7afe773 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae8f1412505c4a7f8b882f48f3a02216 : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7839f9d85a7b416799becb1884dc39b9 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d826fa2199b7427fa4df91e01fe56ec9 : SUCCESS in 8m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/82cbe789e03741b1b6896b7c4e8d5268 : SUCCESS in 31m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c7f44ea73203ee15511099539b3466e713af337b","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-02 18:04:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e23ba1cfb1921fcfbafbb3ca30e88f79bded9e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 19:22:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b422c2b136c94c988a2c8645d08f0a13\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/e2d36a7b9d11483496113c770df38388 : SUCCESS in 51m 34s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5dfa14a5e15c439084700b8813853209 : SUCCESS in 36m 48s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/468b88b6e384444c93cb0ce8f1816966 : SUCCESS in 39m 30s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/f760c11f0c874af29259f59fe2a274e0 : SUCCESS in 31m 06s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/1936968541c54bd7b62f9516a9cd0b2c : SUCCESS in 1h 13m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2442404d7590426bb04519045b543f67 : SUCCESS in 1h 07m 55s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/1bdba219f3c34cf5ba9c7606b19cd225 : SUCCESS in 1h 07m 27s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/2457a5cd3ad54ddd955a398a49b84ed5 : SUCCESS in 1h 03m 08s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/0232b5b57bcc4ec4b3e458062288c6e2 : SUCCESS in 52m 22s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/db27ed02593a4d58a33a7347842e3674 : SUCCESS in 43m 54s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/cb4384a416f6439690d6fef8f82db48d : SUCCESS in 44m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47ef2b11e6344560bc55a0056fdd4aeb : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4832dbc44614beaaf75da944074b0b7 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b75e88f6e6f444c1a20c88f1369175cc : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/614b14c283734affb8a4076b9e192475 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7990a1d4b7e94e59a6b93f9073e654ce : SUCCESS in 8m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/751b8be635294153b30187af73546a47 : SUCCESS in 31m 15s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/05b4f8ab7f844a15a1f8815a0d0c7394 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"5ec13e356f39681f915824219566a317a2157eaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 19:22:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d397bf35864daf0bde939b9e4ba47e408e4b91d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 20:42:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5df531798b034bc58d7aab793e3ddb66\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/e9c681be7bf849d3a833d05c8de6c0f4 : SUCCESS in 49m 06s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/32830fb3f07a4f76b1bc0555cb06945e : SUCCESS in 56m 43s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/85c134ad9fb545a0812d6c9a87634587 : SUCCESS in 47m 20s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/1ef829be216c41c58a71c80dbab18df0 : SUCCESS in 19m 50s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/077ae1415d76460e867b5642bb39fa9c : FAILURE in 1h 13m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7b72dce68fb04e9e8d119b23f6ed8552 : SUCCESS in 1h 13m 22s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/f04bea083e0742f380db87067ca3beb2 : SUCCESS in 1h 08m 19s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/0a4f89bf59ac433aacd448c40f94ab1f : SUCCESS in 52m 41s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/ade7435263b84bd996c3eaa762821014 : SUCCESS in 37m 34s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/63c970122e6e41ff9f9a55e38247969b : SUCCESS in 49m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff4a59c112f0439eb4142b1dee2f09cd : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4be036346180457ba2060299061a2d15 : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c5aa8bf482144edac5578b950631a82 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c956883ad49d4b4b93d885771748aaa0 : SUCCESS in 8m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d66313f620ba47f4a779fe5069b8bf74 : SUCCESS in 30m 40s","accounts_in_message":[],"_revision_number":1},{"id":"3d51613f4c1565b0855f42b2583bfac8bd7a9170","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-02 20:48:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8bd53e1b964d9118f0aa7ab5a17e9fda31e70085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 22:08:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83d59508e11e40c5a98210e7eda9def7\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/402636798cbb445d9564f348b3155a44 : SUCCESS in 48m 45s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ae3acf61f15a45bdaae2a90a75211c69 : SUCCESS in 47m 06s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/eec330e6c02c46658ec2e642a239a842 : SUCCESS in 48m 01s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/f0d203c1eb9747b997c053139d595787 : SUCCESS in 30m 51s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c31e4016face4ed8a63c1a0bbeae4aed : SUCCESS in 50m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2975fea2fb944e5f91a3dde15618a0ed : SUCCESS in 1h 01m 56s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/61d761d2c06443d5a69e5bc40e3a62c8 : SUCCESS in 1h 14m 41s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/754fd481d2b349928ffd02cea872d197 : SUCCESS in 1h 05m 39s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/05f68ff4fa2e4d4a99e311f5ecd3ec45 : FAILURE in 48m 30s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/9846351cf63c499a81cbf439dd6034d9 : SUCCESS in 49m 43s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/4989b22745454431b8885c8cbf12aa0e : SUCCESS in 40m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b014bbcfa9d412dbcaa3c08616d22eb : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ce1e74b57fc4364a7687dc1f5f490c9 : SUCCESS in 1m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13564e54e4994402861901ac0f5e359f : SUCCESS in 7m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f72700d0f54c4e44965ad0fd5d6fca1b : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddd5ee3a5a3e4a95b97e693e63c850d6 : SUCCESS in 11m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d4eda56fe100442f8013103827297e59 : SUCCESS in 31m 07s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9eb6a0620d68457a8d25b79b37721fa0 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"f2fef6fa37ff71fdf3917670cf2b7f9711417bea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 22:09:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23d8044efa7041285ec6dd83b25cfd33eb5b3495","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 23:16:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf8571fba51d4cb3bca9f1970c1fa20a\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/29e918b717ba4a148b8dd299f302d061 : SUCCESS in 51m 15s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e06fddd87d4e498dac39c763aa0665ff : SUCCESS in 50m 29s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/9c2ed7be04f64211a029a7b2c8b80434 : SUCCESS in 40m 29s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/0681f4d90fb044888f5a21a3fb27dd7e : SUCCESS in 35m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9e7c9676ea44414dae5ed8b82fb057b8 : SUCCESS in 54m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5f94da813c9546d0a5cc7c2f2a094193 : SUCCESS in 56m 56s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/466860e4a3984859aec68017fd32705f : SUCCESS in 52m 55s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/a4094adf57064660894ff6085ec6931e : SUCCESS in 57m 46s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/b91e619e8c394881bce1a3aaa76498cc : SUCCESS in 46m 47s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/bfe29b6160a34d67a36f705091caa39e : SUCCESS in 48m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11135b6bb11b4ea490c8d5d3c5f4f68d : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/214e3380c9e944969706d6e35489a834 : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e158f4d6b5fc446aa15fdd685720db5a : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fd998f4c19544f595b674258c68b867 : SUCCESS in 7m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/62e270198230412496fb04d86bbff7e2 : SUCCESS in 35m 42s","accounts_in_message":[],"_revision_number":1},{"id":"79fe04fd6ce114ba0df9d04477fe29545195e045","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 23:16:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"42deadf383646757db5e9859a3504f8f333cb89a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 23:16:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d6ee50488764220b0dad49685f24d2f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe4d02d989a748d9b40f3519c6060b1e : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"70ffb6c1e79958444419c847c97b41ceac022005","revisions":{"70ffb6c1e79958444419c847c97b41ceac022005":{"kind":"REWORK","_number":1,"created":"2025-12-23 17:28:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/971720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/20/971720/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/20/971720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/20/971720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/20/971720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/20/971720/1"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-23 17:26:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-23 17:28:14.000000000","tz":540},"subject":"Resolve deprecation warning from alembic","message":"Resolve deprecation warning from alembic\n\nResolve the following warning from alembic by adding the path_separator\noption:\n\n  DeprecationWarning: No path_separator found in configuration; falling\n  back to legacy splitting on spaces, commas, and colons for\n  prepend_sys_path.  Consider adding path_separator\u003dos to Alembic\n  config.\n\nThe same was done in nova for example.\n https://review.opendev.org/c/openstack/nova/+/952266\n\nChange-Id: I4a8cf5a54e3de768cd1883b67ce57484c32ce643\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/70ffb6c1e79958444419c847c97b41ceac022005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/70ffb6c1e79958444419c847c97b41ceac022005"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
