)]}'
{"id":"openstack%2Fdesignate~972795","triplet_id":"openstack%2Fdesignate~master~I94b42165a8fabf961ea6bc745c59dfe284743369","project":"openstack/designate","branch":"master","topic":"bug/1740510","hashtags":[],"change_id":"I94b42165a8fabf961ea6bc745c59dfe284743369","subject":"Use example IPs instead of localhost in install guide","status":"MERGED","created":"2026-01-08 21:26:10.000000000","updated":"2026-09-28 11:33:45.000000000","submitted":"2026-09-28 11:32:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972795-bug/1740510","meta_rev_id":"923c1e20b65285a36d1d3317d0670ce0646d14ef","_number":972795,"virtual_id_number":972795,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-28 11:32:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-09-28 10:50:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-09-28 10:50:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-08 21:40:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db0d4f1e5f4a0a84ff453d79b3f73ac52247bcff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-08 21:26:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f736b39725951257c9f9dd2afdc6bee95ace74f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 21:40:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5adc568996e4e7f99376a8f6d0d06ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4858ce9eab5c4acdb2412a97047a1b40 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1506e97b63d8467eaa07b21842b767d1 : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":1},{"id":"a4ba1e8136a08122996849c760c8502013ed15e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-24 15:11:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"fa29c54a593e2bee24095be9f75051a439f9cb69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 15:34:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8330a66693aa4594954419282799bea5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbe0532f89654142abf5d397a3c3fe8d : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e74febe69b3e4911a6649dbf7026e606 : SUCCESS in 11m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cad761c1a7084831a79be35e873fef23 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":2},{"id":"d8a5409376fb10a2bad866951769924f0e321b49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-28 10:50:12.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"d85fe385993b992e523ae1fa0e3bbebd4f5b1bd8","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-28 10:50:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3091e8b1471b2c3bbbbd820b6c2f0ba331462694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 11:24:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/158eb2665b504e959114fab4f3041ce8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20271c8d556740be9263849ddfb6d5d9 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a9e0762ba9844dcadcf74a0fa32a314 : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82370d3cd15f47d7871e9ab690a1e02e : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":3},{"id":"013be0bbb0ab7e3a746b18f77636afb5dd669b94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 11:24:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"12f1140d72d992d2a5e907705e928ac0fb636c08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 11:32:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d93c65f4feb945ed81013de6488ae5e3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c744c69449448c481771ebe31950790 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d63dc5f8807c4089b011fb02fdc307ce : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a70c201aa8c24fc1ae16cb527d0b2021 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":3},{"id":"84bd7f65af6a554005bcbb8e2805f97ca39aec66","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 11:32:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"923c1e20b65285a36d1d3317d0670ce0646d14ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 11:33:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3739e12e3a884a54af72e8cb859464a9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4ffb8d8dfbc4e32a3ceb4a1077ff81b : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7f317a5e5fced2d399cc239361edd95fc395d657","revisions":{"1de5a30b593831dc8a19ebd3ff8cdd4582556e84":{"kind":"REWORK","_number":1,"created":"2026-01-08 21:26:10.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/95/972795/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/95/972795/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/95/972795/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/95/972795/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/95/972795/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/95/972795/1"}}},"commit":{"parents":[{"commit":"1db323a3c9c2061174802c75a31df1724d2b063c","subject":"Merge \"Remove old checks for Python 2-3 compatibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1db323a3c9c2061174802c75a31df1724d2b063c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-01-08 21:24:25.000000000","tz":-180},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-01-08 21:24:25.000000000","tz":-180},"subject":"Use example IPs instead of localhost in install guide","message":"Use example IPs instead of localhost in install guide\n\nThe Ubuntu installation guide previously used 127.0.0.1 for all\nIP addresses in BIND9 and pool configuration examples. This made\nit difficult for users to understand Designate\u0027s distributed\narchitecture and how the components communicate with each other.\n\nThis change updates the documentation to use RFC 5737 TEST-NET\naddresses (192.0.2.1 for the controller node running Designate\nservices, 192.0.2.2 for the BIND9 DNS server).\n\nAdded explanatory notes describing the role of each IP address\nand how the components interact, while noting that the configuration\ncan be adapted for all-in-one or multi-server deployments.\n\nGenerated-By: Claude Code 4.5 Sonnet\nCloses-Bug: #1740510\nChange-Id: I94b42165a8fabf961ea6bc745c59dfe284743369\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1de5a30b593831dc8a19ebd3ff8cdd4582556e84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1de5a30b593831dc8a19ebd3ff8cdd4582556e84"}]},"branch":"refs/heads/master"},"5a3fd3a541afd3482746c5e7e743c32d5f49c205":{"kind":"REWORK","_number":2,"created":"2026-09-24 15:11:21.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/95/972795/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/95/972795/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/95/972795/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/95/972795/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/95/972795/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/95/972795/2"}}},"commit":{"parents":[{"commit":"689a5490baed9ca9c6b14f275e8c4dcf42a966a6","subject":"Merge \"Safe stop notinitialized coordination driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/689a5490baed9ca9c6b14f275e8c4dcf42a966a6"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-01-08 21:24:25.000000000","tz":-180},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-09-24 15:11:15.000000000","tz":120},"subject":"Use example IPs instead of localhost in install guide","message":"Use example IPs instead of localhost in install guide\n\nThe Ubuntu installation guide previously used 127.0.0.1 for all\nIP addresses in BIND9 and pool configuration examples. This made\nit difficult for users to understand Designate\u0027s distributed\narchitecture and how the components communicate with each other.\n\nThis change updates the documentation to use RFC 5737 TEST-NET\naddresses (192.0.2.1 for the controller node running Designate\nservices, 192.0.2.2 for the BIND9 DNS server).\n\nAdded explanatory notes describing the role of each IP address\nand how the components interact, while noting that the configuration\ncan be adapted for all-in-one or multi-server deployments.\n\nGenerated-By: Claude Code 4.5 Sonnet\nCloses-Bug: #1740510\nChange-Id: I94b42165a8fabf961ea6bc745c59dfe284743369\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5a3fd3a541afd3482746c5e7e743c32d5f49c205"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5a3fd3a541afd3482746c5e7e743c32d5f49c205"}]},"branch":"refs/heads/master"},"7f317a5e5fced2d399cc239361edd95fc395d657":{"kind":"REWORK","_number":3,"created":"2026-09-28 10:50:12.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/95/972795/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/95/972795/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/95/972795/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/95/972795/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/95/972795/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/95/972795/3"}}},"commit":{"parents":[{"commit":"63e6f7d6939921956b81c502ff038378ac0d573d","subject":"Merge \"Add description filter support to TLD list API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/63e6f7d6939921956b81c502ff038378ac0d573d"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-01-08 21:24:25.000000000","tz":-180},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-09-28 10:50:08.000000000","tz":120},"subject":"Use example IPs instead of localhost in install guide","message":"Use example IPs instead of localhost in install guide\n\nThe Ubuntu installation guide previously used 127.0.0.1 for all\nIP addresses in BIND9 and pool configuration examples. This made\nit difficult for users to understand Designate\u0027s distributed\narchitecture and how the components communicate with each other.\n\nThis change updates the documentation to use RFC 5737 TEST-NET\naddresses (192.0.2.1 for the controller node running Designate\nservices, 192.0.2.2 for the BIND9 DNS server).\n\nAdded explanatory notes describing the role of each IP address\nand how the components interact, while noting that the configuration\ncan be adapted for all-in-one or multi-server deployments.\n\nGenerated-By: Claude Code 4.5 Sonnet\nCloses-Bug: #1740510\nChange-Id: I94b42165a8fabf961ea6bc745c59dfe284743369\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7f317a5e5fced2d399cc239361edd95fc395d657"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7f317a5e5fced2d399cc239361edd95fc395d657"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
