)]}'
{"id":"openstack%2Fdesignate~973042","triplet_id":"openstack%2Fdesignate~master~Ifef185553ad7ddcb94317b39d2b684505deff61d","project":"openstack/designate","branch":"master","topic":"replace-scv2","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-01-12 10:37:51.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ifef185553ad7ddcb94317b39d2b684505deff61d","subject":"Use ServiceCatalog class for identity v3 API","status":"ABANDONED","created":"2026-01-12 05:57:17.000000000","updated":"2026-01-12 12:53:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5579fb9156f8930ba7d685ef36c7048fef404665","_number":973042,"virtual_id_number":973042,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-01-12 07:09:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2026-01-12 10:37:51.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-12 07:09:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"340e2e5f2bb4853930b6b62e9cbcf06fc8338244","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-12 05:57:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2420c3224e11365112ea11b354f4cdec7ef0351","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 07:09:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bb75ac1394d4c7cb94c3bd2134c93b3\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/f6fb7b96c61047a6848c1a12a62d10b8 : FAILURE in 50m 51s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3451c25d31384e8cae81497ebea0f6fa : FAILURE in 46m 29s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/257e7a2a40cf45e286d00fd11123edbf : FAILURE in 41m 57s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/ecec871ce6124a26b871bbc870e1dbd0 : SUCCESS in 32m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7e4e4c5085944ce38d4310218bbba9e3 : FAILURE in 1h 03m 39s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a67af0dda42741e1bf25654573c6e147 : FAILURE in 55m 34s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/b2ba30b76e2748c0baac52d28b97c8b4 : FAILURE in 57m 33s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/f4981d65aafa4b7084acdf0de45b2cf6 : FAILURE in 56m 00s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/6aa110e076aa4bf48a8e683b5077dce2 : FAILURE in 48m 04s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/3498003f738a43178fcbe75236c11a07 : FAILURE in 48m 11s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/69d34ccc26f44184a7a2a659f49c3eae : FAILURE in 43m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/208727672de04e0eaa5d5ea0c891f386 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4093dbe16e7c415fa9cf3660d0352343 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a89160fd27f4882ab55af4784fa9a78 : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4aad64b9584e474d8ecbca214b681ba9 : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/037411a8ea3d4e8c9de2d583c00c4060 : SUCCESS in 7m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/64d976b660134df3ae915af4378f696c : SUCCESS in 30m 45s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1ebeec0676c44889af065b1c8f391b73 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"cf116920b562dd608dc53086bfb72b9683266f7e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-12 10:37:51.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"5579fb9156f8930ba7d685ef36c7048fef404665","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-12 12:53:43.000000000","message":"Abandoned\n\nWe first need to adjust keystonemiddleware.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5514d406eec4f800efee7fd34bc6b0a5af641e49","revisions":{"5514d406eec4f800efee7fd34bc6b0a5af641e49":{"kind":"REWORK","_number":1,"created":"2026-01-12 05:57:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/973042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/42/973042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/42/973042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/42/973042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/42/973042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/42/973042/1"}}},"commit":{"parents":[{"commit":"3bc81788322f10a8ada3e685a4417f4cc88f3942","subject":"Fix designate backend coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-12 05:56:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-12 05:57:06.000000000","tz":540},"subject":"Use ServiceCatalog class for identity v3 API","message":"Use ServiceCatalog class for identity v3 API\n\n... because identity v2 API was removed.\n\nChange-Id: Ifef185553ad7ddcb94317b39d2b684505deff61d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5514d406eec4f800efee7fd34bc6b0a5af641e49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5514d406eec4f800efee7fd34bc6b0a5af641e49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
