)]}'
{"id":"openstack%2Fdesignate~974653","triplet_id":"openstack%2Fdesignate~master~I1e6ed39df5c74dd81a07c816fae60401696ac2a3","project":"openstack/designate","branch":"master","topic":"devstack","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-01-25 13:35:14.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I1e6ed39df5c74dd81a07c816fae60401696ac2a3","subject":"devstack: Drop redundant install_apache_uwsgi call","status":"NEW","created":"2026-01-25 12:16:31.000000000","updated":"2026-03-21 18:30:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2171483d5d6e968284cff0597cf75ef95b670e0a","_number":974653,"virtual_id_number":974653,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-03-17 15:12:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-21 18:30:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-25 13:35:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-21 18:30:53.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"95cf1c39506c388975e0db7492520bee3f85f567","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-25 12:16:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b5ff9ff6447ab270e4f97b876f42baba54a9078","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-25 12:17:09.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"aae1eff6ba4980f5ecb7d48a2f817d156a8b4482","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 13:35:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0ea3bcfdaaa45dcad40255302e77bb7\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/a91bca6a500840e0b03d694b14421311 : SUCCESS in 35m 48s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4bf3bf0006d34edd84ab34cfed8c311f : SUCCESS in 41m 10s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/a8c999187e144a7493aa8bcc14bbb44b : SUCCESS in 50m 20s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/12bed6981f004deeb4d1c8b998d3d41c : SUCCESS in 20m 17s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e811145d88f84ab88f2a1431d2b6a29d : SUCCESS in 1h 14m 44s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d78f9bbeefe24609a7d9d26cc777fd02 : SUCCESS in 1h 07m 34s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/d94bb11e9f0b415cafcc42cf480d5541 : SUCCESS in 1h 05m 17s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/3ff437cfcb764d6c89a9a2423362a679 : SUCCESS in 1h 08m 45s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/dd9c9acb5c004671bd8bf8c46981ca53 : FAILURE in 47m 03s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/2d2eb7997d014ba5be25a357c0ec27dc : FAILURE in 33m 12s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/abdbdd9c303b45fabd51803dfc00fe6a : FAILURE in 31m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2a35e5dc80b478f9ecbc7ec5bb42d20 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28b236b3a8964374817d57959cfa21d3 : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6593f4d2f7da4c409e3ce8c809a497be : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b7413d7a09084ee0acfe84da102bdb1a : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/685c4840d35847968b6007ab2dbb14ed : SUCCESS in 9m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e739040ed5174b29bca52482499ba059 : SUCCESS in 31m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0f5b3b72aeb3419e9ef8cc95f3368c46 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":2},{"id":"f3bb4ce3fd0c76df1dcacb9852bfe458a6e3fa4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-17 13:55:54.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c35eff6dd0e57795079cb8a30ef8bd8191bed2e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 15:12:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18206a72a2604dcfada07c20d3828571\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/7c29eac0210b4a8a9c1824305824736f : SUCCESS in 48m 25s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d8b380419b1c42b6bea3547f4c873a4b : SUCCESS in 50m 00s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/8dc359819c234178bfc8e44d02d6a6e5 : SUCCESS in 39m 35s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/051542b90a8d4f3984f40b9c2dcc7a5f : SUCCESS in 39m 14s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5f6f328ad70e476695e99fc66db8ab4f : SUCCESS in 1h 06m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a4331a7236e44eeab9ecabbfe44ee3c4 : SUCCESS in 1h 09m 38s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/bec0ee9bc1894d91b5e12ed4344d460d : SUCCESS in 1h 03m 09s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/653449eda8b74fa0874175b911813fff : SUCCESS in 1h 00m 15s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/e6db0079f90b4224af0c00b30e1e3738 : SUCCESS in 1h 02m 54s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/1429f8a14efe4417ad244b19694a842f : SUCCESS in 46m 37s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/cc2ac0298e1b4f109272d4b886c9cf40 : SUCCESS in 40m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/098620afa1bd4556bdc9d8b1bcd7ba19 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82389c5bf2294c7a998af5bff04821a4 : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ac438f983894dcd8fe7d635918c7837 : SUCCESS in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0718468fc99142d4b5fbc21cec6d35fb : SUCCESS in 6m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f0c5ba6d689b46d9aae346858d07c106 : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b12829972e4887ba9469e230f2234b : SUCCESS in 4m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3c815a5da66c44c3b33de35a8b93d480 : SUCCESS in 33m 46s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/957dd5d4f9524f29be69b26ca19a3403 : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":3},{"id":"2171483d5d6e968284cff0597cf75ef95b670e0a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2026-03-21 18:30:53.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cea9b367b4418f694edcf52937b94ae737a7540b","revisions":{"a36e63d8ecdef7f3caa92c572bc932d9f01fc62f":{"kind":"REWORK","_number":1,"created":"2026-01-25 12:16:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/974653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/53/974653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/974653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/974653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/974653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/53/974653/1"}}},"commit":{"parents":[{"commit":"7c2f720fc493a7f15cbb25a4f4b3c0a41a49cb98","subject":"devstack: Remove support for deployment under mod_wsgi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7c2f720fc493a7f15cbb25a4f4b3c0a41a49cb98"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-25 12:16:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-25 12:16:13.000000000","tz":540},"subject":"devstack: Drop redundant install_apache_uwsgi call","message":"devstack: Drop redundant install_apache_uwsgi call\n\nIt is called by stack.sh in core devstack, so don\u0027t have to be called\nagain by individual plugin.\n\nChange-Id: I1e6ed39df5c74dd81a07c816fae60401696ac2a3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a36e63d8ecdef7f3caa92c572bc932d9f01fc62f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a36e63d8ecdef7f3caa92c572bc932d9f01fc62f"}]},"branch":"refs/heads/master"},"82ca228050f56ff7e6258fb563b86340f3eae357":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-01-25 12:17:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/974653/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/53/974653/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/974653/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/974653/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/974653/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/53/974653/2"}}},"commit":{"parents":[{"commit":"7c2f720fc493a7f15cbb25a4f4b3c0a41a49cb98","subject":"devstack: Remove support for deployment under mod_wsgi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7c2f720fc493a7f15cbb25a4f4b3c0a41a49cb98"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-25 12:16:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-25 12:16:47.000000000","tz":540},"subject":"devstack: Drop redundant install_apache_uwsgi call","message":"devstack: Drop redundant install_apache_uwsgi call\n\nIt is called by stack.sh in core devstack, so doesn\u0027t have to be called\nagain by individual plugin.\n\nChange-Id: I1e6ed39df5c74dd81a07c816fae60401696ac2a3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/82ca228050f56ff7e6258fb563b86340f3eae357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/82ca228050f56ff7e6258fb563b86340f3eae357"}]},"branch":"refs/heads/master"},"cea9b367b4418f694edcf52937b94ae737a7540b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-17 13:55:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/974653/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/53/974653/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/974653/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/974653/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/53/974653/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/53/974653/3"}}},"commit":{"parents":[{"commit":"ec321809b74de000a7d618e02a1f5eca58be8254","subject":"devstack: Remove support for deployment under mod_wsgi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ec321809b74de000a7d618e02a1f5eca58be8254"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-25 12:16:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 13:55:54.000000000","tz":0},"subject":"devstack: Drop redundant install_apache_uwsgi call","message":"devstack: Drop redundant install_apache_uwsgi call\n\nIt is called by stack.sh in core devstack, so doesn\u0027t have to be called\nagain by individual plugin.\n\nChange-Id: I1e6ed39df5c74dd81a07c816fae60401696ac2a3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cea9b367b4418f694edcf52937b94ae737a7540b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cea9b367b4418f694edcf52937b94ae737a7540b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
