)]}'
{"id":"openstack%2Fdesignate~976603","triplet_id":"openstack%2Fdesignate~stable%2F2025.1~Idf1504ff00a51c0e4ff5b52d12323af2a66edbf5","project":"openstack/designate","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2026-02-13 12:45:44.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2026-02-13 12:45:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idf1504ff00a51c0e4ff5b52d12323af2a66edbf5","subject":"Pin setuptools","status":"MERGED","created":"2026-02-12 11:52:38.000000000","updated":"2026-02-13 12:46:38.000000000","submitted":"2026-02-13 12:45:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976603","meta_rev_id":"e980c7111f6ec8079f084b98540e47a07da42f6f","_number":976603,"virtual_id_number":976603,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-13 12:45:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-02-13 11:46:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-02-13 11:46:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-02-12 13:23:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-12 12:47:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-12 13:35:24.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2026-02-12 13:35:24.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"c91925b18bfd8ab2e902e393506d688ea2d0024a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-02-12 11:52:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e9045017bea396503b3bad2d1cb5be7fdf80191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 12:47:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bd362663852467db0a9b09a281e376a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/85fe0cacab7d446db45bca7648079131 : SUCCESS in 44m 41s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f372a76a1cb349129296c6197bf2b6ba : FAILURE in 14m 54s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6fd52dbe6b6645819161bcd6f175a7da : FAILURE in 12m 47s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/74085f068fee41ee98f80ec3a8571f52 : SUCCESS in 33m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8ed8e4b300d44f17a562b296cb802b60 : SUCCESS in 41m 51s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d23280142992465a9a52b2be4bd9d3ca : FAILURE in 25m 04s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5f95de8266254304b7654e2d19fe18b6 : FAILURE in 34m 39s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/c24778ef62ac4baba936a8e4875ee737 : FAILURE in 37m 04s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/6ce269f779da42cc9ae3d6268faf9030 : FAILURE in 17m 56s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a119bc6db87f47bf9047c568b125ff84 : SUCCESS in 40m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/db20574c91ed46e18785408d2023396b : SUCCESS in 48m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5a78deab2b94172b5dc8af238208be9 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30a8f63277594f10bf521d0096a2a315 : SUCCESS in 2m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6897231808441e7b3e17d49880e5595 : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/535659218ab041afae673fcfe4e81530 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09dd3b558a064b368528cd774b7330f6 : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9f986b4238a42ec86f1857d699adfda : SUCCESS in 3m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f50245ffe8a447a38ed0d0aa7328d1e4 : SUCCESS in 3m 28s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/8a122b71a974404382f271da1b848962 : FAILURE in 52m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93656469b36777748013e175c06d3a578422f116","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-02-12 13:23:34.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"918084ffa480768ec85fe5841a2a1833078a3d24","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-02-13 11:46:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5fc70d2540ee070712fa93186b3eb948170e3aba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 11:47:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c1dbe887d295b676e696bda018712fcfb119adb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 12:45:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50b57b4b89384c77b8f869d710697f00\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/215c53dd9dd6453887f40c4a74873f10 : SUCCESS in 42m 39s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/6c6f5dab7c2f4372a55d807cdc7ce02d : SUCCESS in 48m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/aada75ea0da6435bbdb170a252cffb2f : SUCCESS in 50m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b7ac7117a44e40e9bea3579ae1ecec3c : FAILURE in 53m 22s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/12591e8cd40f40e88fe81e8560166077 : FAILURE in 24m 14s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0ec172c3343a48378f470fc1072b127f : SUCCESS in 49m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6d110c7206c34f889be9ae71d644ddda : SUCCESS in 45m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c4bc27c5ccc4286aa9a28545ff422da : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a91a95a873f4b7c99b359356284d380 : SUCCESS in 7m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b938292be5f7417b86f503148f9d9eef : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/987d18ebc1654b9dbbc66f463152b0d7 : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8281e2ebef849a7aaa0618487156c1f : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":1},{"id":"d346d7cb4ad10276973db1ed2b85d0c74ba5af16","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 12:45:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e980c7111f6ec8079f084b98540e47a07da42f6f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 12:46:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e09ac80ef5f4bedbfe6e59893167b42\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fdf338d4cd14a44abe21683bfe30f51 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27773be846314907a325549b8a9d9ae3 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb09c9785e1c772c8af7e681cfe39b35652415f4","revisions":{"eb09c9785e1c772c8af7e681cfe39b35652415f4":{"kind":"REWORK","_number":1,"created":"2026-02-12 11:52:38.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/03/976603/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/976603/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/976603/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/976603/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/976603/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/976603/1"}}},"commit":{"parents":[{"commit":"bbb241f89620faa6d99ec98f78d77ca4dd32417a","subject":"Remove next pagination link for floating IP list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bbb241f89620faa6d99ec98f78d77ca4dd32417a"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-02-12 11:32:16.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-02-12 11:32:16.000000000","tz":60},"subject":"Pin setuptools","message":"Pin setuptools\n\nsetuptools 82.0.0 dropped support for the legacy \u0027pkg_resources\u0027\nlibrary. While most users of this (including the all important pbr\nlibrary) have been updated to remove references to \u0027pkg_resources\u0027 on\ncurrent master, this is not true for historical releases. Pin the\nsetuptools version used on these historic branches and disable build\nisolation to work around this.\n\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Idf1504ff00a51c0e4ff5b52d12323af2a66edbf5\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb09c9785e1c772c8af7e681cfe39b35652415f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb09c9785e1c772c8af7e681cfe39b35652415f4"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
