)]}'
{"id":"openstack%2Fdesignate~979655","triplet_id":"openstack%2Fdesignate~master~I15e7be8d0d7f91cc95defa5d500135409cfa5a4c","project":"openstack/designate","branch":"master","topic":"designate/+bug/2143729","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-04-17 12:54:04.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-04-17 11:24:53.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"167":{"account":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"last_update":"2026-04-17 18:20:04.000000000","reason":"Change was submitted"},"26319":{"account":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"last_update":"2026-04-17 18:20:04.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2026-04-17 18:20:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I15e7be8d0d7f91cc95defa5d500135409cfa5a4c","subject":"api: Fix pagination for service_statuses API endpoint","status":"MERGED","created":"2026-03-09 13:58:56.000000000","updated":"2026-04-17 18:21:08.000000000","submitted":"2026-04-17 18:20:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979655-designate/+bug/2143729","meta_rev_id":"2388fe190b45524be0fb8684b2f274964f2ff51a","_number":979655,"virtual_id_number":979655,"owner":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-17 18:20:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-04-17 12:54:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2026-04-17 11:24:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-04-17 12:54:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-17 12:54:04.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2026-04-17 11:24:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}],"CC":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 14:02:26.000000000","updated_by":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2026-03-09 15:19:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-09 16:47:21.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-03-16 08:52:20.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2026-03-31 08:57:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"be1378d51afa8442e9969c4f3a81c9cdbca900a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"date":"2026-03-09 13:58:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d4abe65ad5e916431ead507642a8de1e55e34e9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"date":"2026-03-09 14:00:49.000000000","message":"Topic set to designate/+bug/2143729","accounts_in_message":[],"_revision_number":1},{"id":"d305659a2ad7d4b47fab88802f0168cca7b249a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 15:19:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d2616307ab64ff4b5f3e954ce69fe09\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/f103bdea1c3b43c9ad4a60ac57495fec : SUCCESS in 43m 53s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2c934a57cc6b4d27a273de3afbbaac06 : SUCCESS in 36m 09s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/660efa4f7c344515899e8e497011c3c8 : SUCCESS in 43m 27s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/98c937f7f1854da2a77245767e08da86 : SUCCESS in 35m 54s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d724551b2a6149a7a33c462b7f4d30ae : SUCCESS in 1h 10m 09s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d3e1e81630b1474aa9ee17a64b636828 : SUCCESS in 1h 07m 07s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/5291ba4a24c248edb77704908a7c06f2 : SUCCESS in 1h 13m 52s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/dfe3f19b34db4105b3cf8b8ad3561510 : SUCCESS in 53m 33s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/eaf1374cff134bcba0ef8058acd2d7e0 : SUCCESS in 47m 47s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/6e0aec547eb14bbc8d5faef45fee6fb8 : SUCCESS in 42m 42s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/85e002c3fa6846e18645b45674037027 : SUCCESS in 52m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/685b332df0df44bd98a696e6887ec513 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce3cc6233b984640a08fd9216cf69be9 : SUCCESS in 2m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34d4277675a247f2969d69817fe270be : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5b154459d59e46f499b5208aa19dd3b6 : SUCCESS in 6m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/265d1222bd0c4be199b6c53ccc126f93 : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd6ca1e7dd4a41499cf9a6ee8e3a4741 : SUCCESS in 10m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/50d49b6f4e8d42cb8eb25588bc350b0a : SUCCESS in 33m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ff0c9cdf9bbe4a1599ea4a87b3a388ec : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":1},{"id":"52d8109c8b2b68280cc3f1e8eee7c2d233d9a635","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2026-03-09 16:47:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"49079c8ef68ffaf7207b783056e5004f0ac4e481","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2026-03-16 08:52:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a42d29cd9260fcf523c47597f47fc59ce5e141a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"date":"2026-03-17 21:54:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"15b32fb9a6ee67e594f9412fe4f9c90643c8a2d0","author":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"date":"2026-03-17 21:58:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5f2bb5993f9fd9c0f3b628c762fd4040281ed650","author":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"date":"2026-03-17 22:00:39.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"867390f3fcd13b6a1d2140c4f349a5d82efb2f14","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"date":"2026-03-17 22:00:59.000000000","message":"Removed Code-Review+1 by \u003cGERRIT_ACCOUNT_26319\u003e\n","accounts_in_message":[{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}],"_revision_number":2},{"id":"76cb63738391fe9181fcf6eba79b3511008f2080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 23:13:26.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/05acb5b773044824b44eb1c3c1487d6c\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/7da7219000b84a1db5abf9b7a3fb230f : SUCCESS in 50m 05s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5c3e2496f6114c1b9512de6ddb4bbe2f : SUCCESS in 40m 47s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/a9618b91a519448f8f77d588a5ccd204 : SUCCESS in 49m 01s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/8f5483a1660d44829545115730ce89be : SUCCESS in 31m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/358802e156b047bab3dfffc2c0f67aa1 : SUCCESS in 1h 03m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3ee44fd146684040959201ee9b3959db : SUCCESS in 1h 03m 26s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/9a77809d72e440e284b934741105aae4 : SUCCESS in 1h 12m 53s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/c2f9b27ac2bf4eeb969c07e7f069a5e0 : SUCCESS in 58m 55s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/175721e05b724377a6cd19076152804c : FAILURE in 51m 20s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/7fcefeeb458a4271acbc7d4cfbe27633 : SUCCESS in 34m 40s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/1e3fce716b2249f6827fff32f937fdf3 : SUCCESS in 46m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77e480697adb4d6c9bf0094f809495e9 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ea0d0ba020f4ab5876702c99eeee19a : FAILURE in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f112235270e45269850c3d74010f99e : SUCCESS in 5m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5feaedd894bf448b82d840d648136270 : SUCCESS in 4m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/64944ccf5cfd474f9862158f9345f62d : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be588d0194d2422c9be848bc28d8edd3 : SUCCESS in 8m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d8116115bcc54a72930c186f004f40f3 : SUCCESS in 37m 16s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b3f01134496c4fb58ffc5fa4ee048ee0 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":2},{"id":"fb4db65bd61ba1a0467e8c9a108121beea172972","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"date":"2026-03-18 08:54:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c2d93c8f2232ebaf2420e26bdb145f09e24fd076","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 10:23:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82a8e31a9c434c09895326f5604fcac5\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/7febf52f4b1e421f949c2a99c12a9838 : SUCCESS in 47m 37s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4caac4801b714e2da0e1f2860cfe1c68 : SUCCESS in 48m 30s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/b189da6fb63d43deb9aed58e7cb58ea6 : SUCCESS in 49m 08s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/f399231fb43f4e2595891d24c31fe388 : SUCCESS in 34m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5dd0b36922a7470faa17c4b4f6888e5e : SUCCESS in 51m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b38dc1201db9406fb85931e27a619a76 : SUCCESS in 51m 56s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/ec396da4f2e44cafa094dec002f938cf : SUCCESS in 1h 04m 12s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/2c2379e43acf49f3b3264ee8c5810602 : SUCCESS in 1h 23m 11s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/a31266e4e24b4149bed04e4e3181a041 : SUCCESS in 45m 59s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/d9c9193c17f246aab3cfd70a42828e35 : SUCCESS in 39m 05s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/b3e3ead794624fd3bf25b08cafd06cfc : SUCCESS in 53m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e9a665f4b86484f9cdb076af79f15d3 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f88a8949266245489d0f9b04c6d6665c : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51ea69971f9745cfa47b1d59f3ee5893 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/69518f037d284924b1d9a3c73f6233b6 : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/301b3f55d60e4114908bd9ea8e6da0e2 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6d7354b41f04980ac8aef7976fb3690 : SUCCESS in 9m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c9a3eb9cda0c4faca71cb412334073e4 : SUCCESS in 31m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d1f145100e7749ccb57a88ccd1a26813 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":3},{"id":"1f5d3e37c39442806f29b6dc76359d2d92fbbd56","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-31 08:57:43.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"640f00fc22e58c4400ab040487f1d073c90fb96b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"date":"2026-04-13 12:22:13.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"131bbce1b2a4a79a06302ded30daf2acab74d1f7","author":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"date":"2026-04-13 12:28:13.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"abf9084d0d9292fb7ed3f16973f28ce2c8916ad1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 13:51:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0818a911df649359121685d3c36f450\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/9d6d9332167447b9b2c510343486ab83 : SUCCESS in 49m 46s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1d5444b8220b4ca792bb4dfcbb6501fd : SUCCESS in 45m 50s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/0c8942669acd4edda99aa62caa4995d3 : SUCCESS in 53m 03s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/199a5ba0ecfe440d97a524be4c293ae5 : SUCCESS in 34m 51s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/88781c20375445ab81251011db59aae3 : SUCCESS in 1h 16m 34s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/52ac75db31a6416ca950837744ad9116 : SUCCESS in 1h 14m 22s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/bc56b9a111624b5f9b1816ff89caf857 : SUCCESS in 1h 12m 35s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/7aca740b1c6e4a608ca6ce25fb912d07 : SUCCESS in 1h 14m 36s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/d59786747c3d4aa5a13b91ff1ee65ec0 : SUCCESS in 53m 53s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/8492ea60b66546378aa116dfe6ef8059 : SUCCESS in 49m 36s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/a7aba54169a04c90867d87418b7547e9 : SUCCESS in 53m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5233adbb469648a9994d6c11d17aaab4 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f6c9d77ff034133bbd6067c72b5e859 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/161cd35e07bb4326bc656d13da6c86d3 : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/219eb58099fd4f5f9db633f0f21b9e3b : SUCCESS in 5m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ed7eef19951341d9a1dce305a8936396 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d1be966d5984b85ba477eaf0cd21f65 : SUCCESS in 11m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1e6505e93837457795960f9613972ffa : SUCCESS in 35m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/36fd0fb6565041e2b917d7e963af8041 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":4},{"id":"8897d673715808c298af7d2141892b2adccac983","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-17 11:24:53.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"6c35f120f35a4d5e2e8e28ba837a9098b2c5e21f","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-04-17 12:54:04.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"76ad43bc65c43b40f972316aab1b286266868506","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 12:54:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"131fb0a0feaaea6af74a90a5cac1b8df7a7ef0dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 13:52:43.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/da50038a8e384f82810732e51f74bcc6\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/1c909271e8aa4be4bb019e5b18e83c12 : SUCCESS in 36m 37s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c05695e79b3447698f8af25823129aac : SUCCESS in 37m 46s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/6b900b0292c34e6eb3609176d46a5b79 : SUCCESS in 36m 33s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/f61ac26326524ce48ba2278ed5a63b9d : SUCCESS in 31m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fb05cd3d5ef24372a40f307669360d64 : SUCCESS in 56m 26s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5d584c200e274ddd826ddf7dc41c5a83 : FAILURE in 57m 32s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/247ea8c0b4b4435b9e6d7a1ea45ef33f : CANCELED\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/62dfe6a6e89e42cc80a42bafc8200246 : SUCCESS in 54m 29s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/1e806236e5d14cf3b26e2386d5a5b70d : SUCCESS in 38m 40s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/d1e5e45429544773b6567c2d0dbf8037 : SUCCESS in 48m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/334787419b894e3bade106164023af72 : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1e13784236645b19418162810d55406 : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba2b3dcd99e149f29be9ce1e6ada06d8 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a11e0e1f51c0488bac8535fec262e030 : SUCCESS in 9m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/112fadc8bbb34387b7dcb2dfb34ac463 : SUCCESS in 34m 15s","accounts_in_message":[],"_revision_number":4},{"id":"fec7d47225765063fa8e916a5b5157ca2efe0f79","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-04-17 15:23:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"16b9904e96bee0f30f56fb8051f2d086cdc4bc78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 16:44:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d29e317c3924a2e9c4eefafffa81a43\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/85a23ab4cb19457693a805cb317d2d0e : SUCCESS in 41m 48s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/70d5a5629f9a4c738fa890dbb0014c65 : SUCCESS in 42m 52s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/08a89d3d01064355ba753ad3bdb62248 : SUCCESS in 41m 56s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/589564e741d74410b9ce6dd8bfd918ea : SUCCESS in 22m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/15cab233e5df4ca3b38ab2874213dc49 : SUCCESS in 1h 14m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7fa2e8374da545969d3dfac9db54643c : SUCCESS in 1h 15m 56s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/384474c91fc64fe5a9ff4ce340980da1 : SUCCESS in 56m 46s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/cd2a741e79a946429795f4712033d29d : SUCCESS in 48m 54s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/d079e4d8f8b7419eb477e885f1830522 : SUCCESS in 52m 47s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/fd76b52eeb604905b1cd3ae8c64c4a38 : SUCCESS in 48m 20s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/c8b636cceefe44ba852bd50a5b51316d : SUCCESS in 42m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d60061345d0494193948cd8b5e0b036 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aadce5928ba46038a157af109753cea : SUCCESS in 2m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f18b3c10687847bb8ccc914949e853d7 : SUCCESS in 6m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8fa01136cd244ddaced640664dafe70 : SUCCESS in 6m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cb34c3612536441bbf32a5e83d969511 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71e7f8c14f7047b08287044f90b14936 : SUCCESS in 7m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1ccdbf37b3bc4fe0850e13fce1e94190 : SUCCESS in 32m 03s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6ba9c19bf55247568bff87f944ad04a5 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":4},{"id":"8ea83a6a7a794fbc1895423e25c5237761e933e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 16:44:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a8197e19c7965db2cda5a0e46ad52560720f0e82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 18:20:04.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89fd17413e104f81bf47f41ab09ec818\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/f047d010efe0474995cc0b9b139180fa : SUCCESS in 57m 21s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/526e65af5da34107ae59875705a9ef70 : SUCCESS in 48m 36s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/7ebf61efc82547a9ac76db32cccd7772 : SUCCESS in 52m 23s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/80e0d924ee7b48aa9d6924794d072add : SUCCESS in 19m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1d3b47efc14b4498b67929edd94acebd : SUCCESS in 1h 11m 20s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/cb9b98011f334966816fa99a0a31c384 : SUCCESS in 1h 15m 23s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/dc8f9811d8ba4d7380c0da35d22a5fa0 : SUCCESS in 1h 26m 17s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/92a1cee6d8c44abbbba8fae3e6edaeea : SUCCESS in 58m 37s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/ea0769920bd94939ba6fbf980840feba : SUCCESS in 43m 37s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/fb12c765595649c3a894c604b8e1b849 : SUCCESS in 51m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d60a12759c64c03b3b25941da76e98f : SUCCESS in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1f6b2a824b04d00bd11c864169c30eb : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ef838d160f064d2abad6e93fa2b302c5 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e730c74e1ad48a1bdad72c804fac2d7 : SUCCESS in 10m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1fc1077bef8a41b78d970df4928fdb61 : SUCCESS in 33m 03s","accounts_in_message":[],"_revision_number":4},{"id":"8703ceaba3a9a7304e2c145956be8d611a1769f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 18:20:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"2388fe190b45524be0fb8684b2f274964f2ff51a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 18:21:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95b67c8b403c4035b6b0ab8395eff6dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5204c0d0b39248db8d90c609d4eb5bef : SUCCESS in 51s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b6546cb1b47131d8b1d7e12150f8861f948b58c8","revisions":{"834728e898567e339187893bd0f122cf754ad45e":{"kind":"REWORK","_number":1,"created":"2026-03-09 13:58:56.000000000","uploader":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"ref":"refs/changes/55/979655/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/979655/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/979655/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/979655/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/979655/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/979655/1"}}},"commit":{"parents":[{"commit":"ee2add08760a436d0173535699a6016ef18e03ae","subject":"Merge \"Fix infoblox backend tests for newer infoblox-client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee2add08760a436d0173535699a6016ef18e03ae"}]}],"author":{"name":"Michel Raabe","email":"rmichel@devnu11.net","date":"2026-03-09 13:57:20.000000000","tz":60},"committer":{"name":"Michel Raabe","email":"rmichel@devnu11.net","date":"2026-03-09 13:57:20.000000000","tz":60},"subject":"api: Fix pagination for service_statuses API endpoint","message":"api: Fix pagination for service_statuses API endpoint\n\nThe ServiceStatusController.get_all() method was not passing the\nmarker, limit, sort_key and sort_dir parameters to\ncentral_api.find_service_statuses(). This caused the endpoint to\nalways return all results while ignoring pagination parameters.\n\nWhen the number of service statuses exactly matched the default_limit_v2\n(default: 20), the response adapter would generate a \u0027next\u0027 link with a\nmarker pointing to the last entry.\n\nCloses-Bug: #2143729\n\nChange-Id: I15e7be8d0d7f91cc95defa5d500135409cfa5a4c\nSigned-off-by: Michel Raabe \u003craabe@b1-systems.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/834728e898567e339187893bd0f122cf754ad45e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/834728e898567e339187893bd0f122cf754ad45e"}]},"branch":"refs/heads/master"},"3648a356f41859219b253fc4216655d1d53e01d3":{"kind":"REWORK","_number":2,"created":"2026-03-17 21:54:29.000000000","uploader":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"ref":"refs/changes/55/979655/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/979655/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/979655/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/979655/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/979655/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/979655/2"}}},"commit":{"parents":[{"commit":"ee2add08760a436d0173535699a6016ef18e03ae","subject":"Merge \"Fix infoblox backend tests for newer infoblox-client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee2add08760a436d0173535699a6016ef18e03ae"}]}],"author":{"name":"Michel Raabe","email":"rmichel@devnu11.net","date":"2026-03-09 13:57:20.000000000","tz":60},"committer":{"name":"Michel Raabe","email":"rmichel@devnu11.net","date":"2026-03-17 21:54:00.000000000","tz":60},"subject":"api: Fix pagination for service_statuses API endpoint","message":"api: Fix pagination for service_statuses API endpoint\n\nThe ServiceStatusController.get_all() method was not passing the\nmarker, limit, sort_key and sort_dir parameters to\ncentral_api.find_service_statuses(). This caused the endpoint to\nalways return all results while ignoring pagination parameters.\n\nWhen the number of service statuses exactly matched the default_limit_v2\n(default: 20), the response adapter would generate a \u0027next\u0027 link with a\nmarker pointing to the last entry.\n\nCloses-Bug: #2143729\n\nChange-Id: I15e7be8d0d7f91cc95defa5d500135409cfa5a4c\nSigned-off-by: Michel Raabe \u003craabe@b1-systems.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3648a356f41859219b253fc4216655d1d53e01d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3648a356f41859219b253fc4216655d1d53e01d3"}]},"branch":"refs/heads/master"},"0ab1d2126db3825a99935402f877d3f286da6657":{"kind":"REWORK","_number":3,"created":"2026-03-18 08:54:38.000000000","uploader":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"ref":"refs/changes/55/979655/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/979655/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/979655/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/979655/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/979655/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/979655/3"}}},"commit":{"parents":[{"commit":"ee2add08760a436d0173535699a6016ef18e03ae","subject":"Merge \"Fix infoblox backend tests for newer infoblox-client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee2add08760a436d0173535699a6016ef18e03ae"}]}],"author":{"name":"Michel Raabe","email":"rmichel@devnu11.net","date":"2026-03-09 13:57:20.000000000","tz":60},"committer":{"name":"Michel Raabe","email":"rmichel@devnu11.net","date":"2026-03-18 08:54:18.000000000","tz":60},"subject":"api: Fix pagination for service_statuses API endpoint","message":"api: Fix pagination for service_statuses API endpoint\n\nThe ServiceStatusController.get_all() method was not passing the\nmarker, limit, sort_key and sort_dir parameters to\ncentral_api.find_service_statuses(). This caused the endpoint to\nalways return all results while ignoring pagination parameters.\n\nWhen the number of service statuses exactly matched the default_limit_v2\n(default: 20), the response adapter would generate a \u0027next\u0027 link with a\nmarker pointing to the last entry.\n\nCloses-Bug: #2143729\n\nChange-Id: I15e7be8d0d7f91cc95defa5d500135409cfa5a4c\nSigned-off-by: Michel Raabe \u003craabe@b1-systems.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0ab1d2126db3825a99935402f877d3f286da6657"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0ab1d2126db3825a99935402f877d3f286da6657"}]},"branch":"refs/heads/master"},"b6546cb1b47131d8b1d7e12150f8861f948b58c8":{"kind":"REWORK","_number":4,"created":"2026-04-13 12:22:13.000000000","uploader":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"ref":"refs/changes/55/979655/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/979655/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/979655/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/979655/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/979655/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/979655/4"}}},"commit":{"parents":[{"commit":"ee2add08760a436d0173535699a6016ef18e03ae","subject":"Merge \"Fix infoblox backend tests for newer infoblox-client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee2add08760a436d0173535699a6016ef18e03ae"}]}],"author":{"name":"Michel Raabe","email":"rmichel@devnu11.net","date":"2026-03-09 13:57:20.000000000","tz":60},"committer":{"name":"Michel Raabe","email":"rmichel@devnu11.net","date":"2026-04-13 12:21:20.000000000","tz":120},"subject":"api: Fix pagination for service_statuses API endpoint","message":"api: Fix pagination for service_statuses API endpoint\n\nThe ServiceStatusController.get_all() method was not passing the\nmarker, limit, sort_key and sort_dir parameters to\ncentral_api.find_service_statuses(). This caused the endpoint to\nalways return all results while ignoring pagination parameters.\n\nWhen the number of service statuses exactly matched the default_limit_v2\n(default: 20), the response adapter would generate a \u0027next\u0027 link with a\nmarker pointing to the last entry.\n\nCloses-Bug: #2143729\n\nChange-Id: I15e7be8d0d7f91cc95defa5d500135409cfa5a4c\nSigned-off-by: Michel Raabe \u003craabe@b1-systems.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b6546cb1b47131d8b1d7e12150f8861f948b58c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b6546cb1b47131d8b1d7e12150f8861f948b58c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
