)]}'
{"id":"openstack%2Fdesignate~985637","triplet_id":"openstack%2Fdesignate~master~I02fff7d8b6f1574607240e8cd11f7a239946994c","project":"openstack/designate","branch":"master","topic":"ovo-bump","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-04-21 14:46:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I02fff7d8b6f1574607240e8cd11f7a239946994c","subject":"objects: Prepare for oslo.versionedobjects 3.10.0","status":"MERGED","created":"2026-04-21 10:00:32.000000000","updated":"2026-04-21 14:47:15.000000000","submitted":"2026-04-21 14:46:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985637-ovo-bump","meta_rev_id":"2134ea9bc2c853f897295f54fa15cfad2256f854","_number":985637,"virtual_id_number":985637,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-21 14:46:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-04-21 12:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-21 13:21:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-21 13:21:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-21 11:56:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-21 12:22:57.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2026-04-21 13:21:36.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"bc21f69465c335febd4c30c6833a9e18e755a0ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-21 10:00:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2dca02d2cb2987ccb81999243ba94a9c8a73405","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 11:56:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f59f0e6801574ab2a4f074171d8467a3\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/088f79452d3e4c30862ab28ca1eaef9a : SUCCESS in 42m 33s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e712cbd389194d3e9592c098af79c382 : SUCCESS in 53m 42s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/7c69a6966e174be9901fba69a718f646 : SUCCESS in 50m 20s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/b319e4970a7645ffa61eed943f54833e : SUCCESS in 34m 00s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d84e30637b1b48ff80dc4c40b04db952 : SUCCESS in 1h 07m 34s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7530d57f3d394e6dbac1ab5bf03f0827 : SUCCESS in 1h 17m 07s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/54c0c8c66e65420086ee37c82512caf5 : SUCCESS in 1h 15m 26s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/5b8953dd8e1a4ec680c92cd272ffe959 : SUCCESS in 53m 17s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/a253594ced7b4b3ca73d7a950e0506c0 : SUCCESS in 51m 29s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/8da47e9b8efd434485c190624bdaebd7 : SUCCESS in 49m 39s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/93f71e8501ee4ef6ac655835513aaaa0 : SUCCESS in 52m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11a30c184893402aad9ceac75252e366 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17b64525774f43b1bb2130996cc0a329 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1acfb0901c44764910fd4003799367a : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d7f5d1de3e8748d6945b4ed36673e1a5 : SUCCESS in 5m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bb5c86af9bd64e0abe1ccd8c98838f8c : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b219b43b31d949139f4bac4eef26b080 : SUCCESS in 6m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d44ea73d664b4a90918052ed33dd70ef : SUCCESS in 34m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8852a4ad1fd34afea52aed8ee9a56b2e : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":1},{"id":"37c13cd919be9dbb1f7683c59006d5f9f9a27a19","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-04-21 12:22:57.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cba032bf87017363125e959105e367ce977b5097","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2026-04-21 13:21:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aa57ef62c298eb76290c446c1ce9601e843616cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 13:22:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a3f167131fff29c0073877462218c6978a200e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 14:46:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3940beae53324371ad5cd10bb31cb7a4\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/eff20e6e88244328aec13bc88c89f994 : SUCCESS in 46m 35s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/549fabdf7c1747fa887eb34524647141 : SUCCESS in 55m 15s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/419d775cda2646ab8fdd232513fb8de5 : SUCCESS in 51m 21s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/1584a2ffcbf04b08a9d5922f0dba629d : SUCCESS in 33m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/463854d15a57455fb218c338b8b632e8 : SUCCESS in 1h 16m 10s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/19a9a23f3dbd4e7696dc44276f1d5af7 : SUCCESS in 1h 15m 07s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/f5bdf1116d724e66a0be457bd092ec83 : SUCCESS in 1h 09m 48s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/e22b3b97ee854b058d35d9d58bca6bb7 : SUCCESS in 1h 17m 29s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/9d5c8927dd204c4fb2fccbb644430068 : SUCCESS in 43m 57s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/886296dafb5948a89a819af773f7d80a : SUCCESS in 49m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7ab141345ba4d9f8993c5fe6913554c : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8d2c273040f40d8a78669a5d2f6f376 : SUCCESS in 5m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4e8c9594b2b42dd86fd8996615e4fc4 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c71a884cfcf4e0bb8a51c3a0baf95e1 : SUCCESS in 8m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f97b48ec36ec435cad2ad88c23033812 : SUCCESS in 35m 35s","accounts_in_message":[],"_revision_number":1},{"id":"06e66f2aae11596008bfb0eba21f66b199712339","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 14:46:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2134ea9bc2c853f897295f54fa15cfad2256f854","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 14:47:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b83c978be1244db9a4b61894b50aa6f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47a746630a0d4dfda693ba187d72ef73 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d61d3d206d94c6dc78ecc0f7b0b8f09115a3867","revisions":{"4d61d3d206d94c6dc78ecc0f7b0b8f09115a3867":{"kind":"REWORK","_number":1,"created":"2026-04-21 10:00:32.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/37/985637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/37/985637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/37/985637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/37/985637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/37/985637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/37/985637/1"}}},"commit":{"parents":[{"commit":"0daead105053f61136d1202b051e10a8597ccb1d","subject":"Merge \"api: Fix pagination for service_statuses API endpoint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0daead105053f61136d1202b051e10a8597ccb1d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-21 10:00:11.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-21 10:00:11.000000000","tz":60},"subject":"objects: Prepare for oslo.versionedobjects 3.10.0","message":"objects: Prepare for oslo.versionedobjects 3.10.0\n\no.vo 3.10.0 has normalized all Field.coerce types to normal instance\nmethods rather than the mix of static, class and instance methods\npreviously found. Do likewise in our own custom fields.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I02fff7d8b6f1574607240e8cd11f7a239946994c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4d61d3d206d94c6dc78ecc0f7b0b8f09115a3867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4d61d3d206d94c6dc78ecc0f7b0b8f09115a3867"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
