)]}'
{"id":"openstack%2Fdesignate~986098","triplet_id":"openstack%2Fdesignate~stable%2F2026.1~I467435fc07be00757269911b467478c2dd188350","project":"openstack/designate","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-05-11 13:10:38.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}},"hashtags":[],"change_id":"I467435fc07be00757269911b467478c2dd188350","subject":"Fix pool move zone target pool id","status":"MERGED","created":"2026-04-24 14:32:02.000000000","updated":"2026-05-11 15:01:34.000000000","submitted":"2026-05-11 14:59:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986098","meta_rev_id":"aa18beea076e0fd65b7f1e08f5fc320bcddd19fd","_number":986098,"virtual_id_number":986098,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-11 14:59:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-05-11 13:10:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-05-11 13:10:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-24 16:09:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9bae389e1619cb07520bb2d5b6938e09d48c109e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-04-24 14:32:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4170fea732e5c15c064b5628fb87618b781a6c6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 16:09:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/81b1bb97459947ceac120ebe6cda4620\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/890ce6e5326b4f6ea34c07eb02f238bf : SUCCESS in 49m 07s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6504cfae4acb465489f85de290267cb0 : SUCCESS in 45m 23s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/c34c0247588c4b1aae58e7149ae43f02 : SUCCESS in 54m 20s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/f3641322a7aa41968d05a83ecf79d8b4 : SUCCESS in 18m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2257274d389a4aeebbd72d8458f6174f : FAILURE in 1h 01m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/432e73d308814ffcba82d2990fd8b58b : FAILURE in 1h 14m 15s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/db60d463926147c396f8f6464f1c40e8 : FAILURE in 1h 16m 42s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/ec98789f88294f46a16d713f103b07f6 : FAILURE in 1h 30m 37s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/79070b2f0a8347679e2511c39a829cb3 : SUCCESS in 46m 58s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/d3b52e5c5bca4af98cfa28cf8459543e : SUCCESS in 47m 28s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/c21125869fe54149a4e0ec9ebe7a4384 : SUCCESS in 54m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c40764ed81f8439d8162a38cdcd6f0e7 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24dbecd443d442e791cc44b0bd2028c6 : SUCCESS in 2m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a0f841fa3d34d7f8b971777079d168c : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12c5b8a47ba849f5bde3678399587c9d : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bdb5001e50d4df9b4ea76d3cb9c1776 : SUCCESS in 9m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00f3a3b869714205afbdfc030278fd51 : SUCCESS in 6m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bdc5af35e7db4eaeb3012eb69e03b617 : SUCCESS in 35m 13s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0599ff863ec14accbe30ad2fec43f12d : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ea3ed2115c98df7967cc7f07c14c8db723e84b0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-04-27 09:26:28.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7f258f14bf9aefec053a1d5098f0ea106b944833","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 10:59:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a39c6415da7b4cedb688d0415da05967\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/214badb3a456493e94b21b5a5732c9a9 : SUCCESS in 54m 51s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/09aea759e1ea48879ec80584de66e43c : SUCCESS in 51m 24s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/74a984ceb1db4a88b1ca522e2eb8e31e : SUCCESS in 46m 42s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/fcf0e7e994c84d7caa90166904174a52 : SUCCESS in 24m 47s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ab79e79a2efb48c0ac6decfa7e0bee08 : SUCCESS in 1h 07m 32s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/89c9a5394b6d4654a244bdd668635f66 : SUCCESS in 58m 51s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/6312892d4b0d40349da0523b252e26e6 : SUCCESS in 1h 15m 32s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/558c2a4a4ce644fa8e8e31f445e50a03 : SUCCESS in 1h 22m 54s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/518dba25585c4a199356befc802ae01d : SUCCESS in 56m 57s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/37d09ec80330450690b4ab8c7065ea25 : SUCCESS in 44m 01s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/23970994c84d4c90bb029c3bc1cfde19 : SUCCESS in 50m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1952d9e60d714528a46f37a9bf2ded21 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4198f3c2a9ca4c2f8f5c129c2453dfdd : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5641d3557025498391456c77f92c31f2 : SUCCESS in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/766f0bcc95024e8fa82e90e51be18d11 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71f76c4512df470989556443a19d3505 : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4eb78b6f995342cea7d0ec2ca5f68111 : SUCCESS in 3m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9bf75965ca154e1b9c83eb58ab3d086f : SUCCESS in 33m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/406910d04b4f4786a3ee661e22858b99 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":2},{"id":"e64fb1206fd29df5e159e8b643843bdc4e482409","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-05-11 13:10:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"454e4cf5b2578fea15df0a258ef6721df0673380","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 13:12:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"77a0d8d2dbd536711108ed9bf5df15a37dc4f48d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 14:59:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/024fe7e56c0d4e7dbe5f0607f7dac11c\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/459975749b72433eb1bdd05ac89955bc : SUCCESS in 48m 25s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4b858291c573446fab04c6996d3eb13c : SUCCESS in 44m 48s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/b316fd8302874557aadacf23809c579f : SUCCESS in 46m 17s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/c73e599e66f84657a09cca84ad140640 : SUCCESS in 28m 50s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6af8bc82b2744a9d92b7c2d13ba84f50 : SUCCESS in 1h 04m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d3f67e2734a84d84a222cd3a39075694 : SUCCESS in 1h 40m 06s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/f58f823263ee4872974b7ad6973899c5 : SUCCESS in 1h 07m 52s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/0f22a2484fe144fc8ce73810eec4aa33 : SUCCESS in 1h 06m 28s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/6eb35711e73147038d6d33f574684385 : SUCCESS in 51m 00s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/240b2ff8fd0043d3b894f4824840fa0c : SUCCESS in 51m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/677c68c892fd404b9de047e05c63a99f : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6acb77cdcc3848808b9df87c039f0d1a : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f3235658d60741ffa7bba1833da03dc9 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf410aa2a8af4f0988a6dbfaa5b0affa : SUCCESS in 8m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9549ee9627984ebc916cca87983a8902 : SUCCESS in 4m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0c535e802fc4452fa6ee91becf09a990 : SUCCESS in 32m 22s","accounts_in_message":[],"_revision_number":2},{"id":"0b495ce553ffd8d17935b37c0f29d2a58b5d2e17","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 14:59:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"aa18beea076e0fd65b7f1e08f5fc320bcddd19fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 15:01:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ddbba278cb84984a4606eb232ba1e00\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d49bdb101bc04ad79c898e97129d7892 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1581bc3f87949fb85b0ed70553dce1d : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5ba0660699d5a8c8dcd86bd52a67d9ebc8c8f63b","revisions":{"c432d68a8ff54289bfb9cfd04cf4a144931d82ad":{"kind":"REWORK","_number":1,"created":"2026-04-24 14:32:02.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/98/986098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/98/986098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/986098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/986098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/986098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/98/986098/1"}}},"commit":{"parents":[{"commit":"8f4f747f5c32b5316ca029798376e9471bfc35cb","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8f4f747f5c32b5316ca029798376e9471bfc35cb"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-24 11:48:53.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-04-24 14:30:50.000000000","tz":120},"subject":"Fix pool move zone target pool id","message":"Fix pool move zone target pool id\n\nSo far, pool move zone was assigning target_pool_id after updating zone\nin storage, i.e. zone was not moved.\n\nThis patch fixes it.\n\nCloses-Bug: #2096866\nChange-Id: I467435fc07be00757269911b467478c2dd188350\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n(cherry picked from commit 7fe4fba9a15805faf60abf8386e0ab9d93c8f0d7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c432d68a8ff54289bfb9cfd04cf4a144931d82ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c432d68a8ff54289bfb9cfd04cf4a144931d82ad"}]},"branch":"refs/heads/stable/2026.1"},"5ba0660699d5a8c8dcd86bd52a67d9ebc8c8f63b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-27 09:26:28.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/98/986098/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/98/986098/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/986098/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/986098/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/986098/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/98/986098/2"}}},"commit":{"parents":[{"commit":"20014c7b833e07a08c67e5f81aa639e2318d9d2c","subject":"Use pool-aware scheduler filters in grenade CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20014c7b833e07a08c67e5f81aa639e2318d9d2c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-24 11:48:53.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-04-27 09:22:48.000000000","tz":120},"subject":"Fix pool move zone target pool id","message":"Fix pool move zone target pool id\n\nSo far, pool move zone was assigning target_pool_id after updating zone\nin storage, i.e. zone was not moved.\n\nThis patch fixes it.\n\nCloses-Bug: #2096866\nChange-Id: I467435fc07be00757269911b467478c2dd188350\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n(cherry picked from commit 7fe4fba9a15805faf60abf8386e0ab9d93c8f0d7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5ba0660699d5a8c8dcd86bd52a67d9ebc8c8f63b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5ba0660699d5a8c8dcd86bd52a67d9ebc8c8f63b"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
