)]}'
{"id":"openstack%2Fdesignate~986110","triplet_id":"openstack%2Fdesignate~stable%2F2025.2~I467435fc07be00757269911b467478c2dd188350","project":"openstack/designate","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-04-28 09:52:42.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}},"hashtags":[],"change_id":"I467435fc07be00757269911b467478c2dd188350","subject":"Fix pool move zone target pool id","status":"MERGED","created":"2026-04-24 15:16:12.000000000","updated":"2026-05-11 15:09:19.000000000","submitted":"2026-05-11 15:09:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986110","meta_rev_id":"3c1b63148de5324b7bd2e86cbb884286b4d0091d","_number":986110,"virtual_id_number":986110,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-11 15:09:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-05-11 13:11:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-05-11 13:11:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-24 16:28:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2fbd7acf89e017358b2cfa32ecd68a724a7144bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-04-24 15:16:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c669cdabe2303b8743b920d85dbe28b5f84773a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 16:28:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6aafbd653a644041bc32138c40a2883c\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/967648518c184464bcb7ce739d904ef7 : FAILURE in 1h 10m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6410f9cfc7e142dba38e50d81923bd9b : FAILURE in 1h 07m 45s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/b54f501ea7864a16b7c47dd665c745a7 : FAILURE in 52m 05s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/ab5eb36ba1f84291b35d40d8987ac234 : FAILURE in 26m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba818547518543c482bd872a75c5cf25 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/670053e9dc0f4764bc91eb5d1d5e3133 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee7543f6d0e94b3cb313311734786571 : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a92b75a00179489c89cbb6dc1984bd18 : SUCCESS in 5m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4288c5ffb0934a01be8c7a8d854d6f0c : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb4bb0ac57ee4bd59a61de39010411b1 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f03792433d5349678052f3c53548ec93 : SUCCESS in 4m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/486a14662dbe4ed48b99d0db22f8c70b : SUCCESS in 35m 35s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8a34d88d257c412682e6a172490f4fef : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"95b52e402f50785a5f5858b05f59c7d2c5b340f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-04-27 09:29:25.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"46ce413f356be76d5bcf3035ee91c91b032315ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 10:34:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/52d64256dc2740aa9a97d0e821bb4d8e\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9d3a4a7a0a394959b43c08b3522f1d79 : SUCCESS in 57m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4848d6f7f6b54902bca54e04ea1e40b8 : SUCCESS in 59m 55s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/b48c78003ac14864aec837d3b84f0197 : FAILURE in 32m 50s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/8608aab8b1844125932d0d18865e50c4 : FAILURE in 27m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fce13a6cbc7341a3be5f6ab60c0428ee : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6308ee54c4d48b1a938ea9243f5ccd6 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46ce9825e618473e8a5e06670806c543 : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7472ddcc650c4066bc93b00f3f71a618 : SUCCESS in 5m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c881ba60be1454586cc01a8136f75ec : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d705ec1d47147eeaedb45261290758d : SUCCESS in 9m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0522d35403d2457ab9cedd98f4d5db20 : SUCCESS in 6m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/593c784c309346419615508fb63ffa60 : SUCCESS in 35m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d7c8a08e034a48e2bc225a2bf485313a : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":2},{"id":"4e56e060e4a96e597e80b9196dd0cf1f0f023f47","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-04-27 11:54:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5eb208a12f3c453926614436a2bd06ce6a385778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 13:16:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b75a4c1a841b4ed09bfe40e623ca544c\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/60eb6488eb8d4739aadb0e6d34b06f27 : SUCCESS in 1h 14m 55s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/aa08a58ef78b4b518330f5845ab6caa4 : SUCCESS in 1h 10m 06s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/2de31ef62ee642deb5b76c76cfc5d42b : FAILURE in 33m 35s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/76a99ed1dbb04854ac569e00196d8c89 : SUCCESS in 1h 11m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/103f81e5b0804d7585813536688d4949 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b062ba798574422682a1ca8aef68cb05 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/380cc9e926964f8ab8904405d40b1abd : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b065706f4d114d83840df3cdcbff44d1 : SUCCESS in 5m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb91216aa4b54b4dafe89be87fd0eda6 : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee4115ee9749468a80cf853f361fe90b : SUCCESS in 9m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ad820cf065f4b1891cf40b36bb0477a : SUCCESS in 4m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/96d3220a3dea4d26b16b32785e2b3284 : SUCCESS in 32m 30s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8344383a0a80473c88638d64f9e372ef : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":2},{"id":"50aed80983dd16ee4edb3d7af737e82b8a19762e","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-04-27 14:47:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7823c1ec0df439a025ca9051779cf5c223482c4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 16:24:18.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7d0976889084e61921165551461a504\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/48abbd38308e4184b2b9456559eb3150 : SUCCESS in 1h 23m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/905f41db973b43ee990cd9c4b99d6c6d : SUCCESS in 1h 09m 37s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/14480801dc6a47289e86439624359db0 : FAILURE in 22m 35s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/41998657d2ed430b96bc10f417dc163f : FAILURE in 30m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/015ae49fd9034f6d961c8e681e2012df : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b79e5797541437190228b105fff3e68 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3af24bb4a10b436f92405f7ce074d56a : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46ef8f48417e45acb8cc508ae5830dc7 : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29dc4655e6ae4a7e9e623e58458422d8 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efd96e92842b49a296c1f73806bda27d : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13b3a4e662c4482190b1ad04923eac7d : SUCCESS in 5m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/856c35e396464a26aa08183e65c56423 : SUCCESS in 32m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/181b5048d9f7489d94105e274ed8da71 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":2},{"id":"c8cd769a653f59a71c2c7e1d2405a875b8c8bb7d","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-04-27 16:49:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c0ff4db21b21db5bbda5d94b761eb59bec17016f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 18:12:30.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/12560e8a6d97492986dfb8a37c927e91\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/823e7b48c2b147cca557497f2384085d : SUCCESS in 1h 16m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2f08c8431075494b869e20dda5e02599 : SUCCESS in 52m 26s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/df4a27c133b24e8f8027b0e90f57b100 : FAILURE in 35m 44s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/df7b36122b404cbca7b2e3db3d63b810 : FAILURE in 57m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af9694d23f754e58b2c5dc1742efa19b : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c842ed2e7094162a94f78b08d23fc80 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8f130c3610d49e2ace214915fb5b5f8 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/67710b417544467b9eb337dfc2f7a429 : SUCCESS in 5m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f215cf8b08744fdbabbdb05ecaafbbd2 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4151f4fcf97341b7ae4a025fee248e5d : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30ced61885fc402a912055ffda78c38a : SUCCESS in 5m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/37db5b370fd94e349f730e34bf7b6a65 : SUCCESS in 32m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/44b258fca0fe477d8ca6cae3c2d18cdf : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":2},{"id":"3fc71deb68b2b5adc911fded71c60a4dfa6abe68","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-04-28 09:52:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3913554eed79043368380c7ba6b8a8c2286d2b9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-04-28 11:12:05.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"470387144ed660ae9436fe7d5fc8df1d26845344","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 12:36:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af46eacb3f894d7c821b08ad171b7342\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a80c92d4d15f4795933bf2b771bc29aa : SUCCESS in 1h 02m 19s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/04f7d81424d74f439cb2288a04b6f035 : SUCCESS in 1h 10m 57s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/22b8dc6995cb413c863f391d92d5ea85 : FAILURE in 50m 53s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/7e8a57af07ca4348b68ad90050458a89 : FAILURE in 36m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31f7e10a02524c27883f790a66f7e995 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e68afb5a2d114d778a2fc35a520744f4 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c553008ecfdb42a0b69abf93e8250920 : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/20483627d1d9442b830d86c0aab7e13a : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0625fef09fc4c11ae8770918b9297eb : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d946681fd81b47999faad1f65c444bfa : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4067ae9f91764d4fbf7138c59e48a0ae : SUCCESS in 4m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f4c74da279c04c62965a8d7e939531ba : SUCCESS in 34m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5c92bfb97f4e4d5090877672ffd12835 : SUCCESS in 7m 56s","accounts_in_message":[],"_revision_number":3},{"id":"cf8258549739d550a27b4e4457eb392ae9e3f42f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-04-28 13:00:50.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"d824950c074d7efe10bc143bf4f1bff20365072c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 14:30:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f16b82e75bc94047bc437cfd40768235\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c789c021f96f4cb5b420af636f4df0b8 : SUCCESS in 1h 17m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ddecf90bc0e1409f944b711d2873c3d7 : SUCCESS in 57m 37s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/abce785bf4ad4bbb8deb1a3268aa6b1f : FAILURE in 33m 47s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/6d3ac575fc864a3880952b9bbf67b680 : FAILURE in 33m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb18ca9c60c4471d9a90457d660c49f7 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fcb0f22afd54ce091c1f4adbc3899c1 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e20ba28b94a4028b5aa6857b805997b : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe35bb569e87440a9f893e99eff1ef50 : SUCCESS in 5m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/751b07c8c2794239ae159deaebff0a00 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f3e1e1db55142bbb58c2b69c9e50716 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f16335bcbcaf4632bdfea6d866301cfd : SUCCESS in 4m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a6488db74fdd42f1b3a06bcf11109ee3 : SUCCESS in 31m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8848937895f04b36b0d51d86e8aed237 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":4},{"id":"9fc56fdea44429704e25d337b15e8c23a3532b14","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-05-11 13:11:07.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d84eeb695b62ad51e7f4ff4f4a4f10de80ae322d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 13:13:40.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1b2c4231db05197fae7f07cf58a7b672d860e03a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 15:09:19.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aacff38464544f2dad027358ef786c25\n\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4dca07352c72487db006a153691d230f : SUCCESS in 59m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/718b1156a7a2493c8363c7183d076cc5 : SUCCESS in 56m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6db8a283cf204df2af75ba47d9d937f6 : SUCCESS in 2m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6b868b25d54424eb8f14047d353f4ce : SUCCESS in 5m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cb257dc3c85b49669a4da3eac2aa976b : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b06e299394e47ac93c97b5e6492964f : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44318505457d4bebb512f987fb1171fd : SUCCESS in 4m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/732a51bb2f5043c1849f122527c7b306 : SUCCESS in 33m 15s","accounts_in_message":[],"_revision_number":4},{"id":"3c1b63148de5324b7bd2e86cbb884286b4d0091d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 15:09:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"478dcc1e73c585be1c3038a2ac765246db08eeaa","revisions":{"4e6d97640b79edb462a30010a20811afe161c366":{"kind":"REWORK","_number":1,"created":"2026-04-24 15:16:12.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/10/986110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/986110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/986110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/986110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/986110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/986110/1"}}},"commit":{"parents":[{"commit":"4e29c8308526cdb013aee09ff8f30e44b02f0911","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e29c8308526cdb013aee09ff8f30e44b02f0911"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-24 11:48:53.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-04-24 15:15:47.000000000","tz":120},"subject":"Fix pool move zone target pool id","message":"Fix pool move zone target pool id\n\nSo far, pool move zone was assigning target_pool_id after updating zone\nin storage, i.e. zone was not moved.\n\nThis patch fixes it.\n\nCloses-Bug: #2096866\nChange-Id: I467435fc07be00757269911b467478c2dd188350\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n(cherry picked from commit 7fe4fba9a15805faf60abf8386e0ab9d93c8f0d7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e6d97640b79edb462a30010a20811afe161c366"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e6d97640b79edb462a30010a20811afe161c366"}]},"branch":"refs/heads/stable/2025.2"},"313aa6af9e0edfb1808b234918ef1619d2c1b3e4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-27 09:29:25.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/10/986110/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/986110/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/986110/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/986110/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/986110/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/986110/2"}}},"commit":{"parents":[{"commit":"c2e8b16ebaae5c17267105b094dcc92936013f78","subject":"Use pool-aware scheduler filters in grenade CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c2e8b16ebaae5c17267105b094dcc92936013f78"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-24 11:48:53.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-04-27 09:27:23.000000000","tz":120},"subject":"Fix pool move zone target pool id","message":"Fix pool move zone target pool id\n\nSo far, pool move zone was assigning target_pool_id after updating zone\nin storage, i.e. zone was not moved.\n\nThis patch fixes it.\n\nCloses-Bug: #2096866\nChange-Id: I467435fc07be00757269911b467478c2dd188350\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n(cherry picked from commit 7fe4fba9a15805faf60abf8386e0ab9d93c8f0d7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/313aa6af9e0edfb1808b234918ef1619d2c1b3e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/313aa6af9e0edfb1808b234918ef1619d2c1b3e4"}]},"branch":"refs/heads/stable/2025.2"},"c36120c698d6d532cdac7a6edf568cdd90bd9d7e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-04-28 11:12:05.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/10/986110/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/986110/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/986110/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/986110/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/986110/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/986110/3"}}},"commit":{"parents":[{"commit":"ef9ecf5de234afdb3e9d0ecfe904d6d518f85012","subject":"Use pool-aware scheduler filters in grenade CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ef9ecf5de234afdb3e9d0ecfe904d6d518f85012"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-24 11:48:53.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-04-28 11:11:18.000000000","tz":120},"subject":"Fix pool move zone target pool id","message":"Fix pool move zone target pool id\n\nSo far, pool move zone was assigning target_pool_id after updating zone\nin storage, i.e. zone was not moved.\n\nThis patch fixes it.\n\nCloses-Bug: #2096866\nChange-Id: I467435fc07be00757269911b467478c2dd188350\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n(cherry picked from commit 7fe4fba9a15805faf60abf8386e0ab9d93c8f0d7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c36120c698d6d532cdac7a6edf568cdd90bd9d7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c36120c698d6d532cdac7a6edf568cdd90bd9d7e"}]},"branch":"refs/heads/stable/2025.2"},"478dcc1e73c585be1c3038a2ac765246db08eeaa":{"kind":"NO_CHANGE","_number":4,"created":"2026-04-28 13:00:50.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/10/986110/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/986110/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/986110/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/986110/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/986110/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/986110/4"}}},"commit":{"parents":[{"commit":"9847059a65eb16aae3c837b837a39dc3b45edee6","subject":"Use pool-aware scheduler filters in grenade CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9847059a65eb16aae3c837b837a39dc3b45edee6"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-24 11:48:53.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-04-28 12:59:49.000000000","tz":120},"subject":"Fix pool move zone target pool id","message":"Fix pool move zone target pool id\n\nSo far, pool move zone was assigning target_pool_id after updating zone\nin storage, i.e. zone was not moved.\n\nThis patch fixes it.\n\nCloses-Bug: #2096866\nChange-Id: I467435fc07be00757269911b467478c2dd188350\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n(cherry picked from commit 7fe4fba9a15805faf60abf8386e0ab9d93c8f0d7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/478dcc1e73c585be1c3038a2ac765246db08eeaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/478dcc1e73c585be1c3038a2ac765246db08eeaa"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
