)]}'
{"id":"openstack%2Fdesignate~986888","triplet_id":"openstack%2Fdesignate~master~I085eabf65a26e10453e334064c1781763602e0f7","project":"openstack/designate","branch":"master","topic":"drop-python310","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-05-04 21:10:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I085eabf65a26e10453e334064c1781763602e0f7","subject":"Drop Python 3.10","status":"MERGED","created":"2026-05-01 10:35:01.000000000","updated":"2026-05-04 21:11:25.000000000","submitted":"2026-05-04 21:10:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986888-drop-python310","meta_rev_id":"1e0407821c6241d2f7f47c05a01d227aeff07297","_number":986888,"virtual_id_number":986888,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-04 21:10:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-05-04 11:41:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-04 18:15:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-05-04 11:41:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-01 11:54:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-04 11:41:22.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2026-05-04 18:15:06.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"579c5d55e7c52fd06cee014b834e3a1bf5bd817c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-01 10:35:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4586e030e76b256d99504dfd315bdafecb8aed77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 11:54:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f4e9bb877c74ba299b959a3b5dbc9fa\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b4ddcac4a0164b2ba219b41c5783a8ad : SUCCESS in 3m 58s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/873bad3b63bb45febaa6e871323ef04a : SUCCESS in 48m 52s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9ccf361ba9994906ae0d470f13016014 : SUCCESS in 47m 23s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/50bc6848985a4b4e90c1cba85da74e2e : SUCCESS in 46m 58s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/213aaf092b4248c0aa167ad4cc7c5e21 : SUCCESS in 24m 10s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6bd7273ad35940e99932eb874fe1c26c : SUCCESS in 55m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/49665479aa534a198320fbd848f85abf : SUCCESS in 57m 27s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/cf67b28ec29748c8a7cefdb266f10938 : SUCCESS in 1h 16m 30s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/7327688490e84edfa987d7def7b379dc : SUCCESS in 1h 09m 51s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/4d499de940e9433a8338fef8a0a0ea30 : SUCCESS in 48m 53s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/ba4233b6560941dd9e4cd6989799bba6 : SUCCESS in 43m 31s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/83e384e77c8e4bc6907aaa82a42c9d54 : SUCCESS in 46m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bdc604ce48948db8627f9a4433913c6 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23a86987db874066a6b139c313ccce98 : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3dc2f9a0e7de4647b493ac85888f849e : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e46a9e37ca7415d8124d71286537851 : SUCCESS in 5m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/173638335a834dd9954deffb594e62ae : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20fcfd0c9d8b46e6b9f2f0d2c2e2b97b : SUCCESS in 8m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6c0b348a5b3c42f78ecc27527cca0a5e : SUCCESS in 2m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4f86d409f024839a3b6c4e17f1c7be3 : SUCCESS in 2m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0d4d7265b66947849305e523113e35c5 : SUCCESS in 34m 30s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/574b493deb2141029dedf023933342f3 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"0cfb6b5474256960d406e1ce40d0e566a706c9b5","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-05-04 11:41:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb144ea902c5d9d6e5eef447f0161ff507fbb37d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 11:42:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5bd3387747948b9ca98881ac653f137f5bc89bd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 12:50:48.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/designate/+/980450 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"61fd89c27e019bdd922188f2ad6be3131d5d2cde","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 18:15:06.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ce07cfef600d07fd98caa7fd4e6550459e0cb6d","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-05-04 19:34:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"53b077732f819504002fa2ecfe56e81ba29752ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 19:43:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2098d4b42f25456884b36963b1fd8891\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3223c34c096345c78f2794ce0fb0d7a1 : SUCCESS in 2m 47s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/e883fcf8b9ef4788b3279e6938d168e6 : SUCCESS in 54m 44s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f91008ca856a4e80b7361dbb5d1a9953 : SUCCESS in 49m 09s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/963e969dfe9648a383c6e5417500c34f : SUCCESS in 53m 30s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/ded983c50a194b4c83090f9699914e24 : SUCCESS in 31m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/bc44801ffb6a42ef917ac8889fc131bf : SUCCESS in 59m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5754550e01084b6b9dfd47e44ea5cff7 : SUCCESS in 1h 23m 27s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/912b1cb45d12458db5728e545d82467e : SUCCESS in 1h 19m 26s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/5ed34849ae414b89be410b9a4a670c95 : SUCCESS in 1h 00m 41s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/27551a77baf34a95a6d373cc2201d097 : SUCCESS in 47m 27s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/e416da2d7ad443ab8cd7dc85f9261836 : SUCCESS in 48m 39s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/2547f38b650342a6bb96a8065281bb58 : SUCCESS in 47m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38b93732701e4f7db6f762446c4325a8 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/927d1f9d495745fda3a782ac365caed9 : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e4d0e22ffe2492c8aa572cc79dc854f : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0122be0adf3542618f1a029398b976c4 : SUCCESS in 5m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/762da199f5fd47aeb9266ad1c8725ce8 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a01cd00a7544432e8390d9ea4556db90 : SUCCESS in 8m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/420b0bca596a464eb5d243a03dc4b7fb : SUCCESS in 2m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a357df130f54318849ce48353642009 : SUCCESS in 4m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a36eb362e17c4a819009d1bbcaf2f9fd : SUCCESS in 32m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/fa474a43cb604f7e80b6c457b50939ca : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a29f8b7b8e56d8d4129e06ec010309c071870ce3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 19:44:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c36284deeadf0bd43c02daf5e3b91ee93ff41c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 21:10:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a46ac7d0f00a4d4fae0d032c00a059a6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7ee2c53eefc04cb6a66cb64c355a212c : SUCCESS in 2m 52s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/042f3740ba594d87b22fda7fb5d88c87 : SUCCESS in 46m 28s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/dd8fab85f57448059a97d646718a5121 : SUCCESS in 52m 48s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/c500e7671e494c00a883a898bd7972ab : SUCCESS in 39m 01s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/2b0d536bedee491ea604d00cd0648fb7 : SUCCESS in 34m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cc2633560fda4a9bbfe2b1e1fcd55f54 : SUCCESS in 1h 02m 59s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fbd9b8885c8646ba8050989c41fc29dd : SUCCESS in 1h 13m 59s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/4e23fb701d4e4d50a455521bf0dfea84 : SUCCESS in 1h 20m 14s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/9c36c8e344bf408e917313d31a01d8ed : SUCCESS in 56m 51s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/4b580e8b267c45aea2f25cbccbde751d : SUCCESS in 48m 29s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/3553b38949774567b53110c829e82f68 : SUCCESS in 34m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4d89b545aeb43fc84f6b9a3f29061c9 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ce03646cf014be1a48a8edfff1746dd : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/757c6e15c53443e5a5c461f2fbabe767 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/037e4bd2ff454908bcdf8f0dbe0ac3da : SUCCESS in 8m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1a75b72bc3f64f23b7d07c5adebe2416 : SUCCESS in 2m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d005176944a4643a503c8ebd109344f : SUCCESS in 5m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/32298f5376d34e3d983e149d5757f3e7 : SUCCESS in 32m 55s","accounts_in_message":[],"_revision_number":1},{"id":"0510b8a3973f905b827a591f44add98462689df9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 21:10:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1e0407821c6241d2f7f47c05a01d227aeff07297","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 21:11:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5525a42c655a4c89a4b57ff6a7cf0197\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e4519cd53f34674b130b232ec237334 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05c25f6986cd4a81a19b3386e59473a8 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c240e9a6b1144065ca880988bbdfcc4ed8c2d1d7","revisions":{"c240e9a6b1144065ca880988bbdfcc4ed8c2d1d7":{"kind":"REWORK","_number":1,"created":"2026-05-01 10:35:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/986888/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/88/986888/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/986888/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/986888/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/986888/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/88/986888/1"}}},"commit":{"parents":[{"commit":"a6e8fe6671900144410b4b15edd27761fd252fea","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a6e8fe6671900144410b4b15edd27761fd252fea"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-01 10:34:00.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-01 10:34:36.000000000","tz":60},"subject":"Drop Python 3.10","message":"Drop Python 3.10\n\nPython 3.11 is the minimum runtime for 2026.2 (Hibiscus). We also add a\nmarker for Python 3.14, since this is tested in the gate.\n\nhttps: //governance.openstack.org/tc/reference/runtimes/2026.2.html\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I085eabf65a26e10453e334064c1781763602e0f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c240e9a6b1144065ca880988bbdfcc4ed8c2d1d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c240e9a6b1144065ca880988bbdfcc4ed8c2d1d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
