)]}'
{"id":"openstack%2Fdesignate~989338","triplet_id":"openstack%2Fdesignate~stable%2F2026.1~I41b5d2380f98228a4ec58d6d7945538419e2002c","project":"openstack/designate","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-06-02 09:38:13.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}},"hashtags":[],"change_id":"I41b5d2380f98228a4ec58d6d7945538419e2002c","subject":"Revert \"Delete zone from source pool backend after pool move\"","status":"MERGED","created":"2026-05-20 08:22:58.000000000","updated":"2026-06-02 11:05:48.000000000","submitted":"2026-06-02 11:03:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989338","meta_rev_id":"c8d855a2284aa142be7b4d294baae4cbfa4bbfc2","_number":989338,"virtual_id_number":989338,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 11:03:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-06-02 09:38:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-01 18:32:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-06-02 09:38:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-06-02 09:38:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-20 09:40:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-01 18:32:51.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"d6603d0fe54026e35a091acbe5b366ea05a71503","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-05-20 08:22:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d086c1b074be5cd55e2bfb1599f2391188f47fd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 09:40:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6681132d04ab483d9b83f1fd30d46814\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/c9066fa52c5248caa58601b14de91f02 : SUCCESS in 48m 28s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/832ee6f113464714b68c10a355b060fd : SUCCESS in 44m 28s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/cd4f8c76d0dc4623a920d0939fb2c489 : SUCCESS in 46m 45s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/d010eb4afc894e9e9061efd78fb08ff5 : SUCCESS in 24m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/24e96e81def640de885da31dfbd70318 : SUCCESS in 50m 55s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b8daae87b68a4efa9bb801f03d218658 : SUCCESS in 1h 15m 23s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/dd81198a3b374534a84dc2a9c0df7d73 : SUCCESS in 55m 04s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/fd75840793a440da961d632d218353be : SUCCESS in 1h 05m 44s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/d12b802a14df409b9f2bbbc1a9adc164 : SUCCESS in 54m 47s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/5dd8f4b1f80042628cc56286edf2fa0f : SUCCESS in 43m 40s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/24afc19782644a2ab4139e0ebe9bae5a : SUCCESS in 52m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b93137564dd34b17968cb29e10dee19a : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e624cb0f7b842ccbcd1282cb98d8ca4 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fed4ce3eba92422d9ae26a4c69792a7e : SUCCESS in 5m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e0c62d4ed6e48399e82d031c84b8365 : SUCCESS in 5m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/142b728063fc4563a44ed94a019f0153 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40e6b007bd9846d6a79119d93b6eeb7d : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77a10865443c4dd5ae7801ad3114955f : SUCCESS in 6m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f56cde6276444f489edf63d8515025f6 : SUCCESS in 31m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7787abbd0079488da4c573463c7a3b48 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"5325c027987473fafc3d4f294fe5fe2ab9fffb2b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2026-06-01 18:32:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1ab07ebe4d424840bb2005ad3adc6360e8d2b5d3","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-06-02 09:38:13.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c80f0d450d9e5ae1978883f7b318d312f629ae43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 09:38:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e08bf342bf9a2537c4ee27f13cd9726e6060c6d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 11:03:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79f3e043fd2242299c2cfccee5aa66d3\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/1c8e0643186b440494496dd81dc758ff : SUCCESS in 37m 08s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c06590b1ced94051acbaf0b591d4f74b : SUCCESS in 50m 35s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/001f68fb8a214be28b99dc2f8339f53c : SUCCESS in 44m 54s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/a9a6f9f0f2804120b162b6df6210f44d : SUCCESS in 40m 48s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/988e095666be4092b0349d942a81141b : SUCCESS in 1h 14m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c0bff64fb3be486a9b3e01667edb6ba8 : SUCCESS in 1h 10m 56s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/96bebbcc25354567bc434e0208a92681 : SUCCESS in 1h 10m 38s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/058874c4589e4db3ab479b1e741ea673 : SUCCESS in 1h 18m 48s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/c87468badf82418290e1089cca122ecc : SUCCESS in 50m 06s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/11dcdcd03a2d47e4aa8bf0ac813cbd8e : SUCCESS in 51m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/072501311a9b480d9512f88cb6175c5b : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/667efb72be734bd09e9c4db0456a8134 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3e649490e5aa4b4297711350c425212f : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa4a96abcbbd4ff09ccab8a847c16c7b : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b751722cb5f6408ca8dd8aa7c4ebd9d1 : SUCCESS in 9m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eef8550224f4414c85735b0817fad3a0 : SUCCESS in 35m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8030f6e278f261b96372c65325d055b6f2130e8c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 11:03:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c8d855a2284aa142be7b4d294baae4cbfa4bbfc2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 11:05:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fe776335f1f4fd195c9e123bb5be5f5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efba5a5ff636400ea1ea8532115027d1 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/926a350087294e789304243224456726 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf4ff2d26077a79cc67d64b5dbe6ea0340c64a59","revisions":{"bf4ff2d26077a79cc67d64b5dbe6ea0340c64a59":{"kind":"REWORK","_number":1,"created":"2026-05-20 08:22:58.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/38/989338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/989338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/989338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/989338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/989338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/989338/1"}}},"commit":{"parents":[{"commit":"0234d2dfd4adb739d6f166d91cdfd4520daf564c","subject":"Delete zone from source pool backend after pool move","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0234d2dfd4adb739d6f166d91cdfd4520daf564c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-05-19 12:04:02.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-05-20 08:21:58.000000000","tz":120},"subject":"Revert \"Delete zone from source pool backend after pool move\"","message":"Revert \"Delete zone from source pool backend after pool move\"\n\nAuto-deleting the zone from the source backend during a pool move\ncauses DNS resolution failures. Resolvers with cached NS records\nstill query the old nameservers until the admin updates the NS\ndelegation and caches expire. The zone must remain on the source\nbackend to avoid downtime during this transition, as described in\nthe zone-move spec.\n\nOnce NS delegation has propagated, administrators can remove the\nstale zone directly from the source pool\u0027s backend servers (e.g.\nrndc delzone for BIND9, or the PowerDNS API). Unfortunately I am not\naware of a more UX-friendly way.\n\nThis reverts commit 0320063a0ce3016ca64f254f212de5399985b5f9.\n\nCloses-Bug: #2148552\nChange-Id: I41b5d2380f98228a4ec58d6d7945538419e2002c\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n(cherry picked from commit 3ed5e558782287bd69201051b5c7bcd0fdd2023b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bf4ff2d26077a79cc67d64b5dbe6ea0340c64a59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bf4ff2d26077a79cc67d64b5dbe6ea0340c64a59"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
