)]}'
{"id":"openstack%2Fdesignate~989339","triplet_id":"openstack%2Fdesignate~stable%2F2025.2~I41b5d2380f98228a4ec58d6d7945538419e2002c","project":"openstack/designate","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-06-02 09:38:20.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}},"hashtags":[],"change_id":"I41b5d2380f98228a4ec58d6d7945538419e2002c","subject":"Revert \"Delete zone from source pool backend after pool move\"","status":"MERGED","created":"2026-05-20 08:24:40.000000000","updated":"2026-06-02 11:05:49.000000000","submitted":"2026-06-02 11:03:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989339","meta_rev_id":"0f76b51bde68781fef601b928ab991eb7cdf8de6","_number":989339,"virtual_id_number":989339,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 11:03:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-06-02 09:38:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-01 18:32:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-06-02 09:38:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-06-02 09:38:20.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-20 09:39:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-01 18:32:49.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"330b953cd4922773396262b837ceb693031811bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-05-20 08:24:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18559303d8e01f83d33fcd22efbf80fde665113a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 09:39:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdabd4849d954bfb84424a2c69910997\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2a9f846d23d340ef806e32e997b8951b : SUCCESS in 1h 05m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/948d1910db08445f8363361415fb59c9 : SUCCESS in 1h 11m 05s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/4f8e4224e9f44dd9b5e53e8f13e55f54 : FAILURE in 49m 20s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/51700b7b280c48238e50d7ebe9f922d0 : FAILURE in 59m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bd945731b32458a9f6f20cd31f97ce5 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2be1e60944f477c84ce6b5a499b8d50 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83bd1a50083e4d908aebcbc74de59915 : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d1d646cb8c1544b490374b0e801dc04e : SUCCESS in 6m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab684b46ead04130895feb2296db1dd1 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a650c8f13dd94cfb8efb7efcd601e502 : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/992dabd350934047b72f8e2a25283689 : SUCCESS in 5m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cab8c8e159264067a3f0e2ee11f894df : SUCCESS in 33m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7a1eac8cb5b34e7eb9ffb346b3b3aca7 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":1},{"id":"2e48f5f683fd2492e25591319c833e3d9c13eb51","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2026-06-01 18:32:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4a1786318db7bebac0acf047ed1547461554846b","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-06-02 09:38:20.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"16252cf27aa816c8fd3f18711c5ff702a25e1c31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 09:40:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a47a7d6ea30bc852d07a3a8719056b424e5ad050","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 11:03:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4f8e8f14b8645e4931c3bcfd65987a4\n\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1b60f53f9a234946ba13dcc05c16fb51 : SUCCESS in 1h 13m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d08c100ef6a34b60b58307fc58f4031e : SUCCESS in 1h 13m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69c5d243493742c3bc70dec7a30713b4 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/891d4c926854471abad16a93e9ed49d9 : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dfb66c50026a4564b6a7b2cffb8d2f15 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c6d798cb634e26818dc66eb9ac3909 : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca176ff42b944d4eaaf8ffe0757d6551 : SUCCESS in 9m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b62018c7afbc41488c0306d0aee7473e : SUCCESS in 34m 11s","accounts_in_message":[],"_revision_number":1},{"id":"aae31129338720ae23b601f8e234299a64c4ce3c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 11:03:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0f76b51bde68781fef601b928ab991eb7cdf8de6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 11:05:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4477028a4ac4bbe994eafeef6ab5598\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b06624d0c86148e59894573e88ebc77e : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94acd20940f449fb961ceb59a4d9afcd : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f2948a8e749f0f33b52154622f6d99e21a3e0bb6","revisions":{"f2948a8e749f0f33b52154622f6d99e21a3e0bb6":{"kind":"REWORK","_number":1,"created":"2026-05-20 08:24:40.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/39/989339/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/39/989339/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/39/989339/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/39/989339/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/39/989339/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/39/989339/1"}}},"commit":{"parents":[{"commit":"97260b8576a235196b7f0d96b164a51392a8fc2c","subject":"Delete zone from source pool backend after pool move","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/97260b8576a235196b7f0d96b164a51392a8fc2c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-05-19 12:04:02.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-05-20 08:24:06.000000000","tz":120},"subject":"Revert \"Delete zone from source pool backend after pool move\"","message":"Revert \"Delete zone from source pool backend after pool move\"\n\nAuto-deleting the zone from the source backend during a pool move\ncauses DNS resolution failures. Resolvers with cached NS records\nstill query the old nameservers until the admin updates the NS\ndelegation and caches expire. The zone must remain on the source\nbackend to avoid downtime during this transition, as described in\nthe zone-move spec.\n\nOnce NS delegation has propagated, administrators can remove the\nstale zone directly from the source pool\u0027s backend servers (e.g.\nrndc delzone for BIND9, or the PowerDNS API). Unfortunately I am not\naware of a more UX-friendly way.\n\nThis reverts commit 0320063a0ce3016ca64f254f212de5399985b5f9.\n\nCloses-Bug: #2148552\nChange-Id: I41b5d2380f98228a4ec58d6d7945538419e2002c\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n(cherry picked from commit 3ed5e558782287bd69201051b5c7bcd0fdd2023b)\n(cherry picked from commit bf4ff2d26077a79cc67d64b5dbe6ea0340c64a59)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f2948a8e749f0f33b52154622f6d99e21a3e0bb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f2948a8e749f0f33b52154622f6d99e21a3e0bb6"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
