)]}'
{"id":"openstack%2Fdesignate~99488","triplet_id":"openstack%2Fdesignate~master~Id76bcb71369302a559d2c2c5bd6046dff05a488e","project":"openstack/designate","branch":"master","topic":"bp/standardize-logging","hashtags":[],"change_id":"Id76bcb71369302a559d2c2c5bd6046dff05a488e","subject":"Change log statements to meet I18n guidelines","status":"MERGED","created":"2014-06-11 21:42:55.000000000","updated":"2014-06-13 10:49:08.000000000","submitted":"2014-06-13 10:49:08.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b9bdf84903c36160eb650ce53adf30cbfc6f5a7d","_number":99488,"virtual_id_number":99488,"owner":{"_account_id":11626,"name":"Jordan Cazamias","email":"jordan.cazamias@rackspace.com","username":"jord7642"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2014-06-13 10:49:08.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2014-06-13 10:01:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":2,"date":"2014-06-13 09:58:16.000000000","_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2014-06-13 10:01:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-06-11 21:43:14.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2014-06-13 09:58:16.000000000","updated_by":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"reviewer":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"state":"REVIEWER"},{"updated":"2014-06-13 10:01:00.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2014-06-13 10:49:08.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"8ca6a6da1f1435bd2044cd1d468f568ae038c287","author":{"_account_id":11626,"name":"Jordan Cazamias","email":"jordan.cazamias@rackspace.com","username":"jord7642"},"date":"2014-06-11 21:42:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad6741bb1098dcadd283c1c8d3c5aedfd11f7a5b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-11 21:42:59.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"2221779cced79203fff408ac4a92f7b31c6e5eaa","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2014-06-11 21:43:14.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\n    Your patch will now be tested automatically by OpenStack testing frameworks\n    and once the automatic tests pass, it will be reviewed by other friendly\n    developers. They will give you feedback and may require you to refine it.\n\n    People seldom get their patch approved on the first try, so don\u0027t be\n    concerned if requested to make corrections. Feel free to modify your patch\n    and resubmit a new change-set.\n\n    Patches usually take 3 to 7 days to be reviewed so be patient and be\n    available on IRC to ask and answer questions about your work. The more you\n    participate in the community the more rewarding it is for you. You may also\n    notice that the more you get to know people and get to be known, the faster\n    your patches will be reviewed and eventually approved. Get to know others\n    and become known by doing code reviews: anybody can do it, and it\u0027s a\n    great way to learn the code base.\n\n    Thanks again for supporting OpenStack, we look forward to working with you.\n\n    IRC: https://wiki.openstack.org/wiki/IRC\n    Workflow: https://wiki.openstack.org/wiki/Gerrit_Workflow","accounts_in_message":[],"_revision_number":1},{"id":"2275334c7a1fc7cb1fe868ee97c9788951d32624","author":{"_account_id":11626,"name":"Jordan Cazamias","email":"jordan.cazamias@rackspace.com","username":"jord7642"},"date":"2014-06-11 22:09:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"94300a2be6c478930ec8c8b4492207263a81a06b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-11 22:09:08.000000000","message":"Patch Set 2:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"31a1e77ee548cbbc72eb6d2065269381f1ce8032","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-11 23:11:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- gate-designate-pep8 http://logs.openstack.org/88/99488/2/check/gate-designate-pep8/9e3c815 : SUCCESS in 2m 15s\n- gate-designate-docs http://docs-draft.openstack.org/88/99488/2/check/gate-designate-docs/c0b4c80/doc/build/html/ : SUCCESS in 2m 14s\n- gate-designate-python26 http://logs.openstack.org/88/99488/2/check/gate-designate-python26/e710814 : SUCCESS in 3m 57s\n- gate-designate-python27 http://logs.openstack.org/88/99488/2/check/gate-designate-python27/d186df0 : SUCCESS in 1m 33s\n- gate-designate-devstack-dsvm http://logs.openstack.org/88/99488/2/check/gate-designate-devstack-dsvm/cbcface : SUCCESS in 11m 58s","accounts_in_message":[],"_revision_number":2},{"id":"251d790a841ba4d85f1a35af6e04d009e7a3bf15","author":{"_account_id":741,"name":"Kiall Mac Innes","email":"kiall@macinnes.ie","username":"kiall"},"date":"2014-06-13 09:58:16.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks good - Thanks :)","accounts_in_message":[],"_revision_number":2},{"id":"617de5098f834ae7e637c02690b4ffc92b5c0041","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-13 10:01:00.000000000","message":"Patch Set 2: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"09f2b20314da6d1f687a1a180f2dddbb3f9aa03b","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2014-06-13 10:01:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"88e433ab071fe228749e32f905acf35ea8434e01","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-13 10:31:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"67a135934dcd33eece857838acbeada42a2856fd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-13 10:31:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- gate-designate-pep8 http://logs.openstack.org/88/99488/2/check/gate-designate-pep8/2ae8fa7 : SUCCESS in 2m 13s\n- gate-designate-docs http://docs-draft.openstack.org/88/99488/2/check/gate-designate-docs/39f81c2/doc/build/html/ : SUCCESS in 2m 30s\n- gate-designate-python26 http://logs.openstack.org/88/99488/2/check/gate-designate-python26/666d28d : SUCCESS in 3m 57s\n- gate-designate-python27 http://logs.openstack.org/88/99488/2/check/gate-designate-python27/b74cc3f : SUCCESS in 2m 10s\n- gate-designate-devstack-dsvm http://logs.openstack.org/88/99488/2/check/gate-designate-devstack-dsvm/a1826db : SUCCESS in 11m 55s","accounts_in_message":[],"_revision_number":2},{"id":"7d02e74bc239ffcf7c8315f745e9695597959396","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-13 10:49:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded.\n\n- gate-designate-docs http://docs-draft.openstack.org/88/99488/2/gate/gate-designate-docs/3f32719/doc/build/html/ : SUCCESS in 2m 18s\n- gate-designate-pep8 http://logs.openstack.org/88/99488/2/gate/gate-designate-pep8/a363d73 : SUCCESS in 2m 16s\n- gate-designate-python26 http://logs.openstack.org/88/99488/2/gate/gate-designate-python26/1be0914 : SUCCESS in 3m 25s\n- gate-designate-python27 http://logs.openstack.org/88/99488/2/gate/gate-designate-python27/4ff84c6 : SUCCESS in 2m 00s\n- gate-designate-devstack-dsvm http://logs.openstack.org/88/99488/2/gate/gate-designate-devstack-dsvm/38f7006 : SUCCESS in 11m 47s","accounts_in_message":[],"_revision_number":2},{"id":"c68c470f0cd23e05e669dba917fc6e2475f7598b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-13 10:49:08.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a0d9b4cac6c1a65efedc32304736818fb816c3c9","revisions":{"472accadf9bbbf09e0df2745a3aa125337342188":{"kind":"REWORK","_number":1,"created":"2014-06-11 21:42:55.000000000","uploader":{"_account_id":11626,"name":"Jordan Cazamias","email":"jordan.cazamias@rackspace.com","username":"jord7642"},"ref":"refs/changes/88/99488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/88/99488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/99488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/99488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/99488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/88/99488/1"}}},"commit":{"parents":[{"commit":"d9b15425ed27cd83a8a865992d96194ffa4c65fb","subject":"DevStack/Vagrant: Specify the minimum supported Vagrant version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d9b15425ed27cd83a8a865992d96194ffa4c65fb"}]}],"author":{"name":"Jordan Cazamias","email":"jordan.cazamias@rackspace.com","date":"2014-06-10 14:36:45.000000000","tz":-300},"committer":{"name":"Jordan Cazamias","email":"jordan.cazamias@rackspace.com","date":"2014-06-11 21:38:13.000000000","tz":-300},"subject":"Change log statements to meet I18n guidelines","message":"Change log statements to meet I18n guidelines\n\nChange-Id: Id76bcb71369302a559d2c2c5bd6046dff05a488e\nImplements: blueprint standardize-logging\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/472accadf9bbbf09e0df2745a3aa125337342188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/472accadf9bbbf09e0df2745a3aa125337342188"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d9b15425ed27cd83a8a865992d96194ffa4c65fb","is_merged_in_target_branch":true,"change_id":"I56cc2cc506c55deb2e11425a44b40f170492495e","change_number":98835,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a0d9b4cac6c1a65efedc32304736818fb816c3c9":{"kind":"REWORK","_number":2,"created":"2014-06-11 22:09:06.000000000","uploader":{"_account_id":11626,"name":"Jordan Cazamias","email":"jordan.cazamias@rackspace.com","username":"jord7642"},"ref":"refs/changes/88/99488/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/88/99488/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/99488/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/99488/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/99488/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/88/99488/2"}}},"commit":{"parents":[{"commit":"d9b15425ed27cd83a8a865992d96194ffa4c65fb","subject":"DevStack/Vagrant: Specify the minimum supported Vagrant version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d9b15425ed27cd83a8a865992d96194ffa4c65fb"}]}],"author":{"name":"Jordan Cazamias","email":"jordan.cazamias@rackspace.com","date":"2014-06-10 14:36:45.000000000","tz":-300},"committer":{"name":"Jordan Cazamias","email":"jordan.cazamias@rackspace.com","date":"2014-06-11 22:08:45.000000000","tz":-300},"subject":"Change log statements to meet I18n guidelines","message":"Change log statements to meet I18n guidelines\n\nChange-Id: Id76bcb71369302a559d2c2c5bd6046dff05a488e\nImplements: blueprint standardize-logging\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a0d9b4cac6c1a65efedc32304736818fb816c3c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a0d9b4cac6c1a65efedc32304736818fb816c3c9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d9b15425ed27cd83a8a865992d96194ffa4c65fb","is_merged_in_target_branch":true,"change_id":"I56cc2cc506c55deb2e11425a44b40f170492495e","change_number":98835,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
