)]}'
{"id":"openstack%2Fdesignate~996891","triplet_id":"openstack%2Fdesignate~master~I29d9a2bb9eee258286996d83a5f9730328f7c4ed","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-04 14:29:56.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I29d9a2bb9eee258286996d83a5f9730328f7c4ed","subject":"tox: Drop redundant override of install_command","status":"MERGED","created":"2026-07-11 17:10:19.000000000","updated":"2026-09-04 17:26:26.000000000","submitted":"2026-09-04 17:25:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996891","meta_rev_id":"97fe6f08482bb30a62c8988407f95136e27ca10d","_number":996891,"virtual_id_number":996891,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-04 17:25:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-09-04 12:28:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-22 10:41:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-09-04 12:28:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-11 18:32:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-22 10:41:10.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2026-09-04 12:28:24.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"c966c09252a6bc1bbd08d49ddb5bd20230343c26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-11 17:10:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b82462d43b9c7568519d72f5741b25fc67079490","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-11 17:18:30.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2d2a3124130004af28c6358f334bb9a3f969a3ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 18:32:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f28aae23e74244bd9f0a0454b822d264\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/aac4e34472294a4084c28136322cbd41 : SUCCESS in 3m 59s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/e0440db0b4de4859bba51e967845ddf0 : SUCCESS in 44m 21s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e7a912e04e704d449edc047a08a82955 : SUCCESS in 45m 46s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/99e7e6e7318d45fd96676aa5022718fc : SUCCESS in 42m 44s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/354b1329b24a44e890c8275446481549 : SUCCESS in 23m 27s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/de91c0c41bed438e810cd0732a8a4c10 : SUCCESS in 55m 18s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/495c97ab914f48e2ae3b6eeee013e5b0 : SUCCESS in 1h 07m 02s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/44c976c025074c559dbb27d0ffe35e25 : SUCCESS in 1h 08m 38s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/454d1ad5bf0e4dbaae8c4518c6e910ee : SUCCESS in 46m 15s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/a7652dde74264b68ae14a5626a31c5b0 : SUCCESS in 42m 47s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/391c7c0ff6c84edf812790b7c4431a6d : SUCCESS in 40m 22s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/8d96e492c3fc4c9c89f46d78560d9a5a : SUCCESS in 33m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e21c75379884663a7dd02f4d501d506 : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a70e8205c5fa415eab2a7eff3f9d0bcb : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b46aac798ae74098b885c58e9d790a94 : SUCCESS in 5m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/376c111f6d7f4f42a31ea7027da86ee6 : SUCCESS in 4m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b25d23b564854804a2e7a6424a85c8c3 : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a534dc1d8b445b0bcb21b75eaaec45e : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e74f7e5ee367438194b992ef6deee48d : SUCCESS in 3m 07s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8bda7f810214417ab406fbf3f2b9bb89 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":2},{"id":"350900c5cce5bad2d478fb0565297823104ecda1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2026-07-22 10:41:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"75f4fa03c61429fda0546ce293ec11be8c7693fb","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-04 12:28:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fd0fa8c7ccdbac3e7ba12794766b3d0bec634e28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 12:28:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e9c1be113cd2f1c2333ddf1010aa53cbc8b1820e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 13:32:31.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b68d6e8d5094cbaab0241cf872b7f99\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/26157993efc84a3bb56e4897dcd65382 : SUCCESS in 3m 57s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/d7753e0612ff404da0df755409055401 : SUCCESS in 39m 14s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/579b763664b242f782f04e83ea066088 : SUCCESS in 46m 30s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/3adfa0501f6848f5acbc00d8ea756d56 : SUCCESS in 45m 13s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/354a5105284f40afac3e43d797183f94 : SUCCESS in 23m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3190a91be12d4cd6b51f1e0b73fc95e9 : CANCELED\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3c1a0dcacddd4ea29546525605a491f7 : CANCELED\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/4ec2f3669e9e4798bc51d8f9e8bee56e : FAILURE in 56m 01s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/6696507c1db7449a9b0f9092515633fb : CANCELED\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/474089455a614cfab33d4199ccff0e70 : SUCCESS in 39m 04s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/73a38e36d9f7410b81066673f863d856 : SUCCESS in 44m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f16924d105a2485cb85b4a5dcc7d698d : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/20b36097b5774be28154f36c1e4e2b69 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9db7ce373414fcdba149d1cc80be075 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75275f045e324bccb57df4fba3202cbb : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1e226bd257b4b9584397958e421fe2c : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":2},{"id":"f074e0a6c166e15f3f13995cf0c0ab8940d74212","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-04 14:29:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a68a4cb91aeaa269e3dd1556051dd2631a47a795","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 15:45:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b90174141d944802910855a5b9907425\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/92c635d6bbd644638444143c73b54e60 : SUCCESS in 2m 30s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/90fa3c5ea1e8421a95c03f3ff5d29c5a : SUCCESS in 53m 37s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a7d8a44771b844bf97b2d295ff2cbd95 : SUCCESS in 47m 45s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/16d58d33bdfe403183526e2efaa6d2ac : SUCCESS in 46m 41s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/2ca79352e75342d6ba5a0fbda91e5544 : SUCCESS in 15m 11s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/945acfb106b1452199f7c84ee7f3aedc : SUCCESS in 1h 09m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/13c0caa39bef479787bb499063e7cad4 : SUCCESS in 43m 23s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/4463ddc15eaf4dc594fe7fcb6f486fa8 : SUCCESS in 47m 54s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/f4eaeed751b7448ebdf55d345c2ec06a : SUCCESS in 48m 40s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/c00c26a0ea6c41fdac4a27e74e82468f : SUCCESS in 35m 09s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/9d105d258f704392b9e04cefd2390b64 : SUCCESS in 33m 23s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/adba000df2834f6b945de3778036a96e : SUCCESS in 50m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc9f7616d8734540aea4f5168e518696 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/409a02a2e8244430a1adf693a79ae976 : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/725d2d2f8fab48f58800d1fe104bf18a : SUCCESS in 4m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5cb9cf6f29944fdaad4cb604bfc8f856 : SUCCESS in 4m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/52c21172c0d949129e60bf65233f657b : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f5a3e8053d04d2eab9feae9356147d1 : SUCCESS in 10m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fba41ddabf9c42e3916f882600b85c9a : SUCCESS in 4m 21s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8f8c0de0bea1474f9d32fb9bc061663d : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":2},{"id":"5b0654356b1a48869468ceeaf2e17521e2971c3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 15:45:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"819366d120c0da7bc44e722d08fbdb1ab65721f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 17:25:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49d5249566624aa3a6d8a56b024fcfa7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/753584f45c404210a47ab31a86d60c12 : SUCCESS in 5m 17s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/e2826fd4cae440118e4f7e63f0e656e0 : SUCCESS in 49m 44s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/db2662e656404623a11d64c088bbbfc9 : SUCCESS in 42m 55s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/12169514862041f1a51e95da6cffb5e3 : SUCCESS in 44m 02s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/1bde6761f200436a92fff6df6a332104 : SUCCESS in 32m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5dd5cbfcdfb44986abad99a3f915957c : SUCCESS in 1h 10m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/76adc708564d466da8448d7f2b94d7cf : SUCCESS in 1h 12m 27s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e82fd50c53fa4651816314b1c6ef75ec : SUCCESS in 57m 30s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/2e335341104a442bb5bd33dec54dfd25 : SUCCESS in 1h 11m 31s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/2877773007c64df48e710ef421088d2c : SUCCESS in 39m 51s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/b08b8983d01741af98705e816da2e889 : SUCCESS in 44m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef05a4f2b43a43dda8bf8e86156a4381 : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d992550b6ec84501859e8224fcdfd04f : SUCCESS in 6m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0a83835cfffe4ccf93293229f4c175ec : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f307ec9e2a024259aa2fc59c51e54350 : SUCCESS in 8m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/682fcc770ddc40b0b18bfd1e366d9afd : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":2},{"id":"0a075ab35ff9d1e2582b5cf5d1b90a499d66ee3f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 17:25:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"97fe6f08482bb30a62c8988407f95136e27ca10d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 17:26:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/055c9ca42c3c48d8a035dbb34eec0041\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0007230662c243e78e7f2bacaba690c8 : SUCCESS in 46s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84654d9bf57845e787709187822dbcd9 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79210fdeccf04448b03fac8e307ff2eb : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2ce89d01a158afb0768eb33d31bcd52bad33bb3e","revisions":{"2fa27add01e5fb283c3867ad2bf54498e27952ad":{"kind":"REWORK","_number":1,"created":"2026-07-11 17:10:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/996891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/996891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/996891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/996891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/996891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/996891/1"}}},"commit":{"parents":[{"commit":"121839fc97fd3773bba4cd3b59f5e918e131bcdb","subject":"Merge \"Fix multipool-bind9 devstack plugin for Ubuntu 26.04+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/121839fc97fd3773bba4cd3b59f5e918e131bcdb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 17:09:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 17:09:57.000000000","tz":540},"subject":"Drop redundant override of install_command","message":"Drop redundant override of install_command\n\nIt is effectively equivalent to the default value:\n\n  python -I -m pip install {opts} {packages}\n\nAlso remove the redundant override of the VIRTUAL_ENV environment.\nThe same value is already set by tox automatically.\n\nChange-Id: I29d9a2bb9eee258286996d83a5f9730328f7c4ed\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2fa27add01e5fb283c3867ad2bf54498e27952ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2fa27add01e5fb283c3867ad2bf54498e27952ad"}]},"branch":"refs/heads/master"},"2ce89d01a158afb0768eb33d31bcd52bad33bb3e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-07-11 17:18:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/996891/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/996891/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/996891/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/996891/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/996891/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/996891/2"}}},"commit":{"parents":[{"commit":"121839fc97fd3773bba4cd3b59f5e918e131bcdb","subject":"Merge \"Fix multipool-bind9 devstack plugin for Ubuntu 26.04+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/121839fc97fd3773bba4cd3b59f5e918e131bcdb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 17:09:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 17:18:30.000000000","tz":0},"subject":"tox: Drop redundant override of install_command","message":"tox: Drop redundant override of install_command\n\nIt is effectively equivalent to the default value:\n\n  python -I -m pip install {opts} {packages}\n\nAlso remove the redundant override of the VIRTUAL_ENV environment.\nThe same value is already set by tox automatically.\n\nChange-Id: I29d9a2bb9eee258286996d83a5f9730328f7c4ed\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2ce89d01a158afb0768eb33d31bcd52bad33bb3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2ce89d01a158afb0768eb33d31bcd52bad33bb3e"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
