)]}'
{"id":"openstack%2Fdevstack-plugin-ceph~1004569","triplet_id":"openstack%2Fdevstack-plugin-ceph~master~I33ff0b3f384c96738f05a2bcc7dac5da904eee3f","project":"openstack/devstack-plugin-ceph","branch":"master","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-09-08 07:59:12.000000000","reason":"Change was marked work in progress"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-09-08 07:59:12.000000000","reason":"Change was marked work in progress"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-09-08 07:59:12.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I33ff0b3f384c96738f05a2bcc7dac5da904eee3f","subject":"Add EC pool support for Cinder volumes and backup","status":"NEW","created":"2026-09-08 07:55:13.000000000","updated":"2026-09-08 09:58:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"c5411b09d7f41c69b0273ad80ea44c9db8e34470","_number":1004569,"virtual_id_number":1004569,"owner":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-08 09:58:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-09-08 07:59:08.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2026-09-08 07:59:08.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-09-08 07:59:08.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-09-08 09:58:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b70085a7fcc08d656345b78096af218f636f13a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2026-09-08 07:55:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d241462aaef1120f164608b64276dbb7889e74f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2026-09-08 07:58:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"745eb5dc5bb2488a3f0e5e457cd0b9a3054f5e7d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2026-09-08 07:59:12.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"c5411b09d7f41c69b0273ad80ea44c9db8e34470","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 09:58:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60899a8354e94150887fc7faa4a6ea76\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/4d0d280a5f1b496187dd4e00f361aec9 : SUCCESS in 1h 50m 43s\n- devstack-plugin-ceph-multinode-tempest-py3 https://zuul.opendev.org/t/openstack/build/362be031e5c5492a8a7be569080e461f : SUCCESS in 1h 58m 33s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2c0baab9d3a74263a08ed91b372f3fb9 : SUCCESS in 49m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0aaa66c444a54fdd8837166b6f47bf94 : SUCCESS in 1h 33m 24s\n- devstack-plugin-ceph-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/51beec7edddf4673a5794d246fcb12ee : FAILURE in 1h 56m 56s (non-voting)\n- devstack-plugin-ceph-cephfs-native https://zuul.opendev.org/t/openstack/build/ea32b279c87948e99491b801a85266ea : FAILURE in 41m 57s (non-voting)\n- devstack-plugin-ceph-cephfs-nfs-standalone https://zuul.opendev.org/t/openstack/build/dc1d82724c494c1aaf2292c5f7990ef4 : FAILURE in 1h 36m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/592bdbbd853a451d9a950de7b44ceaf0 : SUCCESS in 5m 51s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/eb4ed2ad60b44787950e5ed410180211 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6115ef62c20b1c54ec3632c209cf6eb862432731","revisions":{"cdb3d3c080b20ccad9452e10ed953a9b293085ce":{"kind":"REWORK","_number":1,"created":"2026-09-08 07:55:13.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/69/1004569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack-plugin-ceph","ref":"refs/changes/69/1004569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/69/1004569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/69/1004569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/69/1004569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/69/1004569/1"}}},"commit":{"parents":[{"commit":"debdbb1c953db0f75f1e2a9aef16d5e440e873fa","subject":"Merge \"Add set-require-min-compat-client to secondary ceph cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/debdbb1c953db0f75f1e2a9aef16d5e440e873fa"}]}],"author":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2026-09-07 13:21:47.000000000","tz":120},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2026-09-07 15:32:46.000000000","tz":120},"subject":"Add EC pool support for Cinder volumes and backup","message":"Add EC pool support for Cinder volumes and backup\n\nChange-Id: I33ff0b3f384c96738f05a2bcc7dac5da904eee3f\nSigned-off-by: Francesco Pantano \u003cfpantano@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/cdb3d3c080b20ccad9452e10ed953a9b293085ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/cdb3d3c080b20ccad9452e10ed953a9b293085ce"}]},"branch":"refs/heads/master"},"6115ef62c20b1c54ec3632c209cf6eb862432731":{"kind":"REWORK","_number":2,"created":"2026-09-08 07:58:20.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/69/1004569/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack-plugin-ceph","ref":"refs/changes/69/1004569/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/69/1004569/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/69/1004569/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/69/1004569/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/69/1004569/2"}}},"commit":{"parents":[{"commit":"debdbb1c953db0f75f1e2a9aef16d5e440e873fa","subject":"Merge \"Add set-require-min-compat-client to secondary ceph cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/debdbb1c953db0f75f1e2a9aef16d5e440e873fa"}]}],"author":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2026-09-07 13:21:47.000000000","tz":120},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2026-09-08 07:57:24.000000000","tz":120},"subject":"Add EC pool support for Cinder volumes and backup","message":"Add EC pool support for Cinder volumes and backup\n\nChange-Id: I33ff0b3f384c96738f05a2bcc7dac5da904eee3f\nSigned-off-by: Francesco Pantano \u003cfpantano@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/6115ef62c20b1c54ec3632c209cf6eb862432731"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/6115ef62c20b1c54ec3632c209cf6eb862432731"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
