)]}'
{"id":"openstack%2Fdevstack-plugin-ceph~966014","triplet_id":"openstack%2Fdevstack-plugin-ceph~master~I589adb435799d558b97da454e2cf754494c0feb4","project":"openstack/devstack-plugin-ceph","branch":"master","topic":"ec_pools","attention_set":{},"removed_from_attention_set":{"18002":{"account":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"last_update":"2026-08-20 14:40:56.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-08-19 18:32:29.000000000","reason":"removed on reply"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-08-20 14:40:56.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-08-20 14:40:56.000000000","reason":"Change was submitted"},"25402":{"account":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"last_update":"2026-08-20 10:41:57.000000000","reason":"\u003cGERRIT_ACCOUNT_25402\u003e replied on the change","reason_account":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}}},"hashtags":[],"change_id":"I589adb435799d558b97da454e2cf754494c0feb4","subject":"Separate replicated and erasure coded pool handling in cephadm","status":"MERGED","created":"2025-11-04 09:12:30.000000000","updated":"2026-08-20 14:42:51.000000000","submitted":"2026-08-20 14:40:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"966014-ec_pools","meta_rev_id":"40d7230ffa2a74b36a572c769e9195090361f59a","_number":966014,"virtual_id_number":966014,"owner":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-20 14:40:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2026-08-19 18:32:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2026-08-19 14:47:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2026-08-19 18:52:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-04 09:13:08.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-11-04 09:13:08.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-11-04 11:45:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-29 15:19:22.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2026-04-17 11:49:17.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"3a534f67fb37fd0ee57f615bd24e5b5cf8b89265","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2025-11-04 09:12:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f42eb6250f4258fb8fed6d6a5f65181a848bb1ad","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2025-11-04 09:13:14.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"72face82fec2bd9705e0d368e93caa5c28cbb629","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 11:45:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/820bd62ad50148ce9cd4a2ff1d7670ca\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/254d6de370a84e05b32d7f3d9ac0a540 : SUCCESS in 2h 19m 16s\n- devstack-plugin-ceph-multinode-tempest-py3 https://zuul.opendev.org/t/openstack/build/412e50e04a6540e490c60397d8ba3ff8 : FAILURE in 2h 26m 34s\n- devstack-plugin-ceph-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a6a12533ab704300a0d149f3b2202776 : SUCCESS in 1h 23m 54s (non-voting)\n- devstack-plugin-ceph-cephfs-native https://zuul.opendev.org/t/openstack/build/c0a237b948904add864d725b9d439f0b : SUCCESS in 46m 01s (non-voting)\n- devstack-plugin-ceph-cephfs-nfs-standalone https://zuul.opendev.org/t/openstack/build/cedeab9298ef4e40ac1adf9fbed1d1a5 : SUCCESS in 1h 07m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38d3236f832c47618261031be6d89458 : SUCCESS in 5m 40s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/78cbc5136c5442a5b65a47a060069790 : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a3877836053f0fbf1ee114e7803bfc04528e9937","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2025-11-19 20:19:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"69f5da15487f92ac0513097882d4d83b996b7758","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 23:29:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eade56167a76401db188a8063c8721b7\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1c003f92587942e5a6e2b60026aff24b : FAILURE in 1h 35m 42s\n- devstack-plugin-ceph-multinode-tempest-py3 https://zuul.opendev.org/t/openstack/build/21e4087978b146089a6dd751864ac4b9 : TIMED_OUT in 3h 05m 25s\n- devstack-plugin-ceph-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ac8a7e2cc85b4d4c8d40c7a8fca865c7 : SUCCESS in 1h 31m 46s (non-voting)\n- devstack-plugin-ceph-cephfs-native https://zuul.opendev.org/t/openstack/build/efd87cf858674c6c88f8d650e5a0582c : SUCCESS in 1h 17m 00s (non-voting)\n- devstack-plugin-ceph-cephfs-nfs-standalone https://zuul.opendev.org/t/openstack/build/38e605143a63404da8c72fe940b8a5c6 : SUCCESS in 1h 22m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c36f13357ec49378e189090d5234787 : SUCCESS in 6m 10s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/853a1ec030d64912b41cfd2795c2b451 : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":2},{"id":"e1104ef07303874b4290b5f1f07547dde412ec55","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2026-04-17 11:48:52.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"fd668a52029d19d1d7e0d5e1418ab2bd80da1a77","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2026-04-17 11:49:17.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"9f3312714dcc1c9d8447b57f6ebc965a3f12751e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2026-04-17 11:51:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0d498d600d5447516d99e4b254e931ab4d342db1","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2026-04-17 11:52:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"843cffddbf8f303b549c7c187fdfa165ff43144e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 14:25:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b21a46ef8306479392714a26628d9713\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1f62c4e13bcb4790945c4cff8c47d47e : SUCCESS in 1h 45m 18s\n- devstack-plugin-ceph-multinode-tempest-py3 https://zuul.opendev.org/t/openstack/build/86109c1379b0441b857451cd12fc0604 : SUCCESS in 2h 31m 24s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7bf4a7299c694bc884e278a92b8c9073 : SUCCESS in 51m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c2621820943a4258b36818ddd7de4554 : SUCCESS in 1h 59m 24s\n- devstack-plugin-ceph-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6e95bc94001a4d779c5e39bca753b058 : SUCCESS in 1h 27m 38s (non-voting)\n- devstack-plugin-ceph-cephfs-native https://zuul.opendev.org/t/openstack/build/04bcb13d39da420ead12c3fa24116601 : FAILURE in 42m 21s (non-voting)\n- devstack-plugin-ceph-cephfs-nfs-standalone https://zuul.opendev.org/t/openstack/build/26c8e4ca4e204bce88102402e4b20fe1 : SUCCESS in 1h 04m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c7a529b02d94b0bba03ba05341e8ab6 : SUCCESS in 4m 47s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/2379ddc18c8d4f329eb5d0541f52a7f8 : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":4},{"id":"1783f3555388ee4a30971496c8fc6cd77bdc8fd1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-20 04:31:33.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"67c5ee567df5765f3f9e38b95ff45498323efc05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2026-04-20 07:39:15.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"e0b4de132667ea0a2439610cd432e8a79950263a","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2026-04-20 07:39:44.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c470bca6bdf6da58b7893cf64c4767bdcab93072","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 10:16:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94df5f0e06cf4a5383c6b432badbf181\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/74d8317f4ed548be924643a2b0414807 : SUCCESS in 1h 30m 41s\n- devstack-plugin-ceph-multinode-tempest-py3 https://zuul.opendev.org/t/openstack/build/0c793e4bc81e4371b7233168eeda8aea : SUCCESS in 2h 29m 21s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/6f3dfb1dd3894dc69fa50345704da814 : SUCCESS in 26m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a3db3911bb844238be374efcae73521c : SUCCESS in 1h 15m 08s\n- devstack-plugin-ceph-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/220579d92c5648fb8581998859696f11 : SUCCESS in 1h 29m 18s (non-voting)\n- devstack-plugin-ceph-cephfs-native https://zuul.opendev.org/t/openstack/build/30d83174e2944c5f956f7590fcb0eb73 : FAILURE in 20m 49s (non-voting)\n- devstack-plugin-ceph-cephfs-nfs-standalone https://zuul.opendev.org/t/openstack/build/7f6ff0c7b7874b3a966c6cbbe27ea37e : SUCCESS in 1h 23m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75e7900eadd5484288f85c70d3eeaaae : SUCCESS in 3m 36s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ed409d5689c24d6eaa03ed7d7b43ccbe : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":5},{"id":"af61e105ddfb87c58ef928b2bdcee465da61d10b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2026-04-22 14:04:24.000000000","message":"Topic ec_glance removed","accounts_in_message":[],"_revision_number":5},{"id":"c465033252f9b4249b1e5006ccd120626bc731e3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2026-04-22 14:04:32.000000000","message":"Topic set to ec_pools","accounts_in_message":[],"_revision_number":5},{"id":"687c3cb64a9691bf5d97f94fe853cf4c77ab4f72","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-26 07:12:45.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"00d6879502ea20fbf2da56cd0285d15779b70637","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2026-05-13 12:50:07.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"7e024ff853e364834d8bad42be158f2dde57e54f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 15:32:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72474c11626745ec84d1587399b1e84d\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/775e7e9d8c8349b5955b864539f31417 : SUCCESS in 2h 36m 48s\n- devstack-plugin-ceph-multinode-tempest-py3 https://zuul.opendev.org/t/openstack/build/2391f7dfc0004e84872f4d31f1df477c : SUCCESS in 2h 26m 03s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/3d503edb97824fdeb37809ccf8099691 : SUCCESS in 43m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f4e1622f55b740e5ab6795a49103133d : SUCCESS in 57m 39s\n- devstack-plugin-ceph-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/dfe598bdf8474bf69d26819d13fed42e : SUCCESS in 49m 17s (non-voting)\n- devstack-plugin-ceph-cephfs-native https://zuul.opendev.org/t/openstack/build/b02abd44807a4518979495398331abcc : FAILURE in 22m 12s (non-voting)\n- devstack-plugin-ceph-cephfs-nfs-standalone https://zuul.opendev.org/t/openstack/build/762dbb4183b94c2d914fc42a530e146d : SUCCESS in 1h 18m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38dae0294ded4488a27652d07f7ae96c : SUCCESS in 5m 28s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/4c4e22f27755459894f8d6a1aa681293 : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":6},{"id":"9b97bf786209a4a3daf57161011f01829e81c9ed","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-05-19 15:25:12.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d9d5264d7486c2ce3bc9cdf32567b5033ef4fd71","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-05-19 17:57:23.000000000","message":"Patch Set 6: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9c949e5e09ef40fc6eaf2e68c7cf0d35511dd7e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-19 06:41:11.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"b0d6aa6c189d3032fb2daa2cc2132c39364accc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-19 08:05:05.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"208a4e91d88dc8b99d2c097db5937b69e7db8b4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 10:32:07.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc3c772fc1a84bba88eb052bf4b6fa9c\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/634de31752f5464ca42680adf8c896ce : SUCCESS in 1h 10m 21s\n- devstack-plugin-ceph-multinode-tempest-py3 https://zuul.opendev.org/t/openstack/build/c75c0a7b007847c6b9882b3864515856 : SUCCESS in 2h 03m 27s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/803dc7baa194489088cabb5457e73099 : SUCCESS in 54m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/11abe3e468eb47378e9e9c830265597a : SUCCESS in 1h 12m 05s\n- devstack-plugin-ceph-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0346eb70e2f74a5a8880a20ffd3cecb0 : FAILURE in 2h 22m 37s (non-voting)\n- devstack-plugin-ceph-cephfs-native https://zuul.opendev.org/t/openstack/build/6aed3d39e6be440fa86ec9de2c041397 : FAILURE in 30m 58s (non-voting)\n- devstack-plugin-ceph-cephfs-nfs-standalone https://zuul.opendev.org/t/openstack/build/05d184cc22384b18bac8cacc8a726b2f : FAILURE in 1h 25m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4ac480b2383418cbce909259ff30769 : SUCCESS in 6m 20s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/d039357edeb74a60afd5d48c24fdfd40 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":8},{"id":"59f02f56c159b2793a95b2c717b9635ac1861e8d","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2026-08-19 10:46:44.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1a0ca3538a36289f392bfd78b5f4638753cc2786","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-19 10:50:53.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4677b5454ecfa6eb2804368a8937886a14dcc9cc","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2026-08-19 13:05:33.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c5ac91da6118ab9fb9f8828afd8b6c7bbb6917e4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-19 14:47:20.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"620fbba4b7c52a4ade6fd3288222a97641103073","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-19 18:32:29.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)\n\nThanks Francesco and Abhishek; the Glance EC path works via the DNM patch elsewhere: https://review.opendev.org/988421\n\nOne thing to fix as a follow-up (not blocking): the Manila EC data pool goes through add_pools \"erasure\", so it gets the rbd application even though it\u0027s a cephfs data pool handed to `ceph fs add_data_pool`. I\u0027ll take care of that separately.","accounts_in_message":[],"_revision_number":8},{"id":"ae9e38c062a7b29f64c8a658073b701c93742934","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-19 18:52:58.000000000","message":"Patch Set 8: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a365df99569b95c78a61badd270982c6a386e529","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 18:53:23.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"c6e431440cb6e3558ee8a1946e32afff3bb764d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 19:11:41.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bf7a65e46724a0fbe89d9344b331d72\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/05cdf1200b4f4dc7aa773e276640f19e : FAILURE in 17m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f181dad5be254c0684990bde29f1c475 : SUCCESS in 6m 14s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/508f54686f1049f4bd5fe6ceefa4d863 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":8},{"id":"49e41a91b2a953f92ff58b9401036af643a14846","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-19 19:19:52.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0bc9c3ff0ba77d4fcb0077e0130a3975b62bd2fd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-19 19:20:06.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"aa5bdb769e0ce5cc652874a6ed403055ecb9df21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 20:12:10.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e0585f2a80c4ece960b0aad7abea308\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6fbb6a67858440debe3389fa0ae2be7a : FAILURE in 32m 33s\n- devstack-plugin-ceph-multinode-tempest-py3 https://zuul.opendev.org/t/openstack/build/f746cca6892345d483af9d2302fe0429 : FAILURE in 30m 53s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/423e26044399473faa9a32ad798f9d06 : POST_FAILURE in 32m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/caa18c0ec3c6466e94ceaad33fa9386e : FAILURE in 47m 29s\n- devstack-plugin-ceph-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/77ea416a6bb74cc3b9b8930afd19762c : FAILURE in 16m 17s (non-voting)\n- devstack-plugin-ceph-cephfs-native https://zuul.opendev.org/t/openstack/build/6ecf728cb86145d2b6539a7104f532bf : FAILURE in 33m 11s (non-voting)\n- devstack-plugin-ceph-cephfs-nfs-standalone https://zuul.opendev.org/t/openstack/build/e7126991b19347999b1dfe0d1bebf321 : FAILURE in 31m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/836e1f51c05545f0b3c8171b6bd07fec : SUCCESS in 6m 15s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/4a7ef85997444b6f80fd476b0a0bf53c : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":8},{"id":"1fca8e2ad0b6561468f04b58331afb9a6dc872bd","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2026-08-19 22:16:53.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"72c51a32e3c484d6cc06754ba305ea8a36b5b42e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 22:58:59.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37ff02ac1fa445798b274f8269a57ce5\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/46b085ef233f4692a00d9587bc5c7d96 : FAILURE in 31m 37s\n- devstack-plugin-ceph-multinode-tempest-py3 https://zuul.opendev.org/t/openstack/build/516fa7a5beb544b0a26c45a0fe1aa904 : FAILURE in 30m 26s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5974a475a4a541cab576f5e4d67996be : POST_FAILURE in 32m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0fe6861d293f4fb690b3e3f92f8e3d3e : FAILURE in 25m 49s\n- devstack-plugin-ceph-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/bba848dad370420c801658bddf4a54ed : FAILURE in 19m 32s (non-voting)\n- devstack-plugin-ceph-cephfs-native https://zuul.opendev.org/t/openstack/build/354937896fc746088077100a3a687b4c : FAILURE in 22m 31s (non-voting)\n- devstack-plugin-ceph-cephfs-nfs-standalone https://zuul.opendev.org/t/openstack/build/15ff5731d31249a1b66892f0d146327c : FAILURE in 35m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3490a8c8e3d24ccfb527bbbaf00f4530 : SUCCESS in 5m 52s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/d6c9bb884e5e4a30884fde9ae40d7080 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":8},{"id":"6a7fa766f97e2a34edd829b5dcd6b990457008b6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-20 05:37:20.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"99b75e2b24256ad0d5e6fe36e6986871e1fc6c3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 06:32:56.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/158f5b94ecf74288900c60208564b5d4\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a1184ba11223427ab56ba0ae78258e8e : FAILURE in 28m 18s\n- devstack-plugin-ceph-multinode-tempest-py3 https://zuul.opendev.org/t/openstack/build/bb8a6292c7534d26a54098205bf4ee93 : FAILURE in 31m 46s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e529002168da45c1b12a7740ba0afe0a : POST_FAILURE in 28m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2a4b570608c24fbfa6415b319f42df88 : FAILURE in 48m 21s\n- devstack-plugin-ceph-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/2c134cde97fb43abb91c0ab90239f2ef : FAILURE in 28m 24s (non-voting)\n- devstack-plugin-ceph-cephfs-native https://zuul.opendev.org/t/openstack/build/31be30d3c7734ca0a50f35b71ce43649 : FAILURE in 27m 56s (non-voting)\n- devstack-plugin-ceph-cephfs-nfs-standalone https://zuul.opendev.org/t/openstack/build/7424b9f47fe542ba8a5778be85dfbc07 : FAILURE in 27m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7651c08ac8d4e6994f539d8e90057ca : SUCCESS in 6m 36s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/348a05d1f80d4f9c9839812b3c4a753e : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":8},{"id":"c8c1e301ba7319c76ffd2bc5b1c15e5f07f77f64","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2026-08-20 10:41:57.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5f114299887e8addfbb670e109c9108112f383d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 13:00:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/754a3bcee06d46958550a86b8134fc9d\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/922d753cbefe4f7abf9a44735272cefc : SUCCESS in 49m 04s\n- devstack-plugin-ceph-multinode-tempest-py3 https://zuul.opendev.org/t/openstack/build/dfb6354dd3604efcb34ae70fbe04a79a : SUCCESS in 1h 04m 06s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e4aad970815e4870907686c796ea5997 : SUCCESS in 52m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1c4e999c9e0b4b829b593a90dc3febcb : SUCCESS in 45m 45s\n- devstack-plugin-ceph-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b912c09d9f8448a9adc4204141428950 : FAILURE in 1h 55m 33s (non-voting)\n- devstack-plugin-ceph-cephfs-native https://zuul.opendev.org/t/openstack/build/0ff9ab80d939420cb8577db33c3eacb4 : FAILURE in 20m 33s (non-voting)\n- devstack-plugin-ceph-cephfs-nfs-standalone https://zuul.opendev.org/t/openstack/build/af7bd496b3764c869e525bf6a27fa250 : FAILURE in 1h 31m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e10ad42a098e4be9b0b2fe9b19220f66 : SUCCESS in 3m 29s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e9cb8a5415064ae89aba03e8e175be73 : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":8},{"id":"769749dcf395958f29742e81b6d3c01d7283d4b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 13:02:44.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"83c112b53a94fe0723438beb11c22e2727649bca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 14:40:55.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caf70b9c5e5f49dd98fd027e6c8b9ca9\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d0e2ccd72b0b401a80efa4ad0d4f42c4 : SUCCESS in 1h 38m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec7f2fd50d4042dfb9f47a224bd4d21a : SUCCESS in 4m 26s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/99067389faf04edf9466139d90440210 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":8},{"id":"b986f45b60e7cd716831d6666ce753af7f20d003","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 14:40:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"40d7230ffa2a74b36a572c769e9195090361f59a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 14:42:51.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4550aff08574f4da08c4dcd774446a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d43b008dd5b4aebb48879d724ef6d2e : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"f5ca9050baf795546e3001a5001517d8c5130904","revisions":{"2ce1b387c4634c25edecf05ceff8fe354fd4743a":{"kind":"REWORK","_number":1,"created":"2025-11-04 09:12:30.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/14/966014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack-plugin-ceph","ref":"refs/changes/14/966014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/1"}}},"commit":{"parents":[{"commit":"536ebea559673d694c8f8deeafe1362e2c41c021","subject":"move ceph jobs to debian 12","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/536ebea559673d694c8f8deeafe1362e2c41c021"}]}],"author":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2025-11-03 13:09:48.000000000","tz":60},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2025-11-04 09:12:23.000000000","tz":60},"subject":"Separate replicated and erasure coded pool handling in cephadm","message":"Separate replicated and erasure coded pool handling in cephadm\n\nReplace single \"POOLS\" array with \"POOLS_REPLICATED\" and \"POOLS_EC\" arrays to\ndistinguish between replicated and erasure coded pools. Update the \"add_pools()\"\nfunction to accept a \"pool mode\" parameter and create pools with the appropriate\nreplication strategy, while maintaining backward compatibility to existing\nreplicated-only deployment scenarios. In addition, this change adds support\nto configure Glance with erasure coded pools via \"GLANCE_EC_ENABLE\" parameter.\nThe data_pool can be customized through the \"GLANCE_EC_DATA_POOL\" variable.\n\nChange-Id: I589adb435799d558b97da454e2cf754494c0feb4\nSigned-off-by: Francesco Pantano \u003cfpantano@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/2ce1b387c4634c25edecf05ceff8fe354fd4743a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/2ce1b387c4634c25edecf05ceff8fe354fd4743a"}]},"branch":"refs/heads/master"},"faf669648bcb61a05d880abd9f5d32873ce56adb":{"kind":"REWORK","_number":2,"created":"2025-11-19 20:19:06.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/14/966014/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack-plugin-ceph","ref":"refs/changes/14/966014/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/2"}}},"commit":{"parents":[{"commit":"536ebea559673d694c8f8deeafe1362e2c41c021","subject":"move ceph jobs to debian 12","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/536ebea559673d694c8f8deeafe1362e2c41c021"}]}],"author":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2025-11-03 13:09:48.000000000","tz":60},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2025-11-19 20:18:51.000000000","tz":60},"subject":"Separate replicated and erasure coded pool handling in cephadm","message":"Separate replicated and erasure coded pool handling in cephadm\n\nReplace single \"POOLS\" array with \"POOLS_REPLICATED\" and \"POOLS_EC\" arrays to\ndistinguish between replicated and erasure coded pools. Update the \"add_pools()\"\nfunction to accept a \"pool mode\" parameter and create pools with the appropriate\nreplication strategy, while maintaining backward compatibility to existing\nreplicated-only deployment scenarios. In addition, this change adds support\nto configure Glance with erasure coded pools via \"GLANCE_EC_ENABLE\" parameter.\nThe data_pool can be customized through the \"GLANCE_EC_DATA_POOL\" variable.\n\nChange-Id: I589adb435799d558b97da454e2cf754494c0feb4\nSigned-off-by: Francesco Pantano \u003cfpantano@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/faf669648bcb61a05d880abd9f5d32873ce56adb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/faf669648bcb61a05d880abd9f5d32873ce56adb"}]},"branch":"refs/heads/master"},"7ae1df02320d8c64ee42d38ffa370446003f0c54":{"kind":"REWORK","_number":3,"created":"2026-04-17 11:48:52.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/14/966014/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack-plugin-ceph","ref":"refs/changes/14/966014/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/3"}}},"commit":{"parents":[{"commit":"1324730e88d9af1230a1390c1d244579efa13a84","subject":"Add nova ceph jobs for more coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/1324730e88d9af1230a1390c1d244579efa13a84"}]}],"author":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2025-11-03 13:09:48.000000000","tz":60},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2026-04-17 11:47:54.000000000","tz":120},"subject":"Separate replicated and erasure coded pool handling in cephadm","message":"Separate replicated and erasure coded pool handling in cephadm\n\nReplace single \"POOLS\" array with \"POOLS_REPLICATED\" and \"POOLS_EC\" arrays to\ndistinguish between replicated and erasure coded pools. Update the \"add_pools()\"\nfunction to accept a \"pool mode\" parameter and create pools with the appropriate\nreplication strategy, while maintaining backward compatibility to existing\nreplicated-only deployment scenarios. In addition, this change adds support\nto configure both Glance and Manila (cephfs) with erasure coded pools via\n\"\u003cSERVICE\u003e_EC_ENABLE\" parameters.\nThe data_pool can be customized through the \"\u003cSERVICE\u003e_EC_DATA_POOL\" variable.\n\nChange-Id: I589adb435799d558b97da454e2cf754494c0feb4\nSigned-off-by: Francesco Pantano \u003cfpantano@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/7ae1df02320d8c64ee42d38ffa370446003f0c54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/7ae1df02320d8c64ee42d38ffa370446003f0c54"}]},"branch":"refs/heads/master"},"18f019dc37405452ae11c7371ea23f80251b7657":{"kind":"REWORK","_number":4,"created":"2026-04-17 11:51:10.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/14/966014/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack-plugin-ceph","ref":"refs/changes/14/966014/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/4"}}},"commit":{"parents":[{"commit":"1324730e88d9af1230a1390c1d244579efa13a84","subject":"Add nova ceph jobs for more coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/1324730e88d9af1230a1390c1d244579efa13a84"}]}],"author":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2025-11-03 13:09:48.000000000","tz":60},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2026-04-17 11:51:04.000000000","tz":120},"subject":"Separate replicated and erasure coded pool handling in cephadm","message":"Separate replicated and erasure coded pool handling in cephadm\n\nReplace single \"POOLS\" array with \"POOLS_REPLICATED\" and \"POOLS_EC\" arrays to\ndistinguish between replicated and erasure coded pools. Update the \"add_pools()\"\nfunction to accept a \"pool mode\" parameter and create pools with the appropriate\nreplication strategy, while maintaining backward compatibility to existing\nreplicated-only deployment scenarios. In addition, this change adds support\nto configure both Glance and Manila (cephfs) with erasure coded pools via\n\"\u003cSERVICE\u003e_EC_ENABLE\" parameters.\nThe data_pool can be customized through the \"\u003cSERVICE\u003e_EC_DATA_POOL\" variable.\n\nChange-Id: I589adb435799d558b97da454e2cf754494c0feb4\nSigned-off-by: Francesco Pantano \u003cfpantano@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/18f019dc37405452ae11c7371ea23f80251b7657"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/18f019dc37405452ae11c7371ea23f80251b7657"}]},"branch":"refs/heads/master"},"362b1575cf58087c23fa0d2944a178d3eacf8666":{"kind":"REWORK","_number":5,"created":"2026-04-20 07:39:15.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/14/966014/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack-plugin-ceph","ref":"refs/changes/14/966014/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/5"}}},"commit":{"parents":[{"commit":"1324730e88d9af1230a1390c1d244579efa13a84","subject":"Add nova ceph jobs for more coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/1324730e88d9af1230a1390c1d244579efa13a84"}]}],"author":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2025-11-03 13:09:48.000000000","tz":60},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2026-04-20 07:38:52.000000000","tz":120},"subject":"Separate replicated and erasure coded pool handling in cephadm","message":"Separate replicated and erasure coded pool handling in cephadm\n\nReplace single \"POOLS\" array with \"POOLS_REPLICATED\" and \"POOLS_EC\" arrays to\ndistinguish between replicated and erasure coded pools. Update the \"add_pools()\"\nfunction to accept a \"pool mode\" parameter and create pools with the appropriate\nreplication strategy, while maintaining backward compatibility to existing\nreplicated-only deployment scenarios. In addition, this change adds support\nto configure both Glance and Manila (cephfs) with erasure coded pools via\n\"\u003cSERVICE\u003e_EC_ENABLE\" parameters.\nThe data_pool can be customized through the \"\u003cSERVICE\u003e_EC_DATA_POOL\" variable.\n\nChange-Id: I589adb435799d558b97da454e2cf754494c0feb4\nSigned-off-by: Francesco Pantano \u003cfpantano@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/362b1575cf58087c23fa0d2944a178d3eacf8666"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/362b1575cf58087c23fa0d2944a178d3eacf8666"}]},"branch":"refs/heads/master"},"45bb04f4457fc85fe9ddffc3d5f1f4554f308641":{"kind":"REWORK","_number":6,"created":"2026-05-13 12:50:07.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/14/966014/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack-plugin-ceph","ref":"refs/changes/14/966014/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/6"}}},"commit":{"parents":[{"commit":"1324730e88d9af1230a1390c1d244579efa13a84","subject":"Add nova ceph jobs for more coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/1324730e88d9af1230a1390c1d244579efa13a84"}]}],"author":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2025-11-03 13:09:48.000000000","tz":60},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2026-05-13 12:49:59.000000000","tz":120},"subject":"Separate replicated and erasure coded pool handling in cephadm","message":"Separate replicated and erasure coded pool handling in cephadm\n\nReplace single \"POOLS\" array with \"POOLS_REPLICATED\" and \"POOLS_EC\" arrays to\ndistinguish between replicated and erasure coded pools. Update the \"add_pools()\"\nfunction to accept a \"pool mode\" parameter and create pools with the appropriate\nreplication strategy, while maintaining backward compatibility to existing\nreplicated-only deployment scenarios. In addition, this change adds support\nto configure both Glance and Manila (cephfs) with erasure coded pools via\n\"\u003cSERVICE\u003e_EC_ENABLE\" parameters.\nThe data_pool can be customized through the \"\u003cSERVICE\u003e_EC_DATA_POOL\" variable.\n\nChange-Id: I589adb435799d558b97da454e2cf754494c0feb4\nSigned-off-by: Francesco Pantano \u003cfpantano@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/45bb04f4457fc85fe9ddffc3d5f1f4554f308641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/45bb04f4457fc85fe9ddffc3d5f1f4554f308641"}]},"branch":"refs/heads/master"},"be3e98ff1f9a7f630b5b8acd676044be6799e6cd":{"kind":"REWORK","_number":7,"created":"2026-08-19 06:41:11.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/14/966014/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack-plugin-ceph","ref":"refs/changes/14/966014/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/7"}}},"commit":{"parents":[{"commit":"86e24060411379597edf13aaa7af8dd6131b7d00","subject":"Remove logic for old releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/86e24060411379597edf13aaa7af8dd6131b7d00"}]}],"author":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2025-11-03 13:09:48.000000000","tz":60},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-08-19 06:24:11.000000000","tz":0},"subject":"Separate replicated and erasure coded pool handling in cephadm","message":"Separate replicated and erasure coded pool handling in cephadm\n\nReplace single \"POOLS\" array with \"POOLS_REPLICATED\" and \"POOLS_EC\" arrays to\ndistinguish between replicated and erasure coded pools. Update the \"add_pools()\"\nfunction to accept a \"pool mode\" parameter and create pools with the appropriate\nreplication strategy, while maintaining backward compatibility to existing\nreplicated-only deployment scenarios. In addition, this change adds support\nto configure both Glance and Manila (cephfs) with erasure coded pools via\n\"\u003cSERVICE\u003e_EC_ENABLE\" parameters.\nThe data_pool can be customized through the \"\u003cSERVICE\u003e_EC_DATA_POOL\" variable.\n\nChange-Id: I589adb435799d558b97da454e2cf754494c0feb4\nSigned-off-by: Francesco Pantano \u003cfpantano@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/be3e98ff1f9a7f630b5b8acd676044be6799e6cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/be3e98ff1f9a7f630b5b8acd676044be6799e6cd"}]},"branch":"refs/heads/master"},"f5ca9050baf795546e3001a5001517d8c5130904":{"kind":"REWORK","_number":8,"created":"2026-08-19 08:05:05.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/14/966014/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack-plugin-ceph","ref":"refs/changes/14/966014/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/14/966014/8"}}},"commit":{"parents":[{"commit":"86e24060411379597edf13aaa7af8dd6131b7d00","subject":"Remove logic for old releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/86e24060411379597edf13aaa7af8dd6131b7d00"}]}],"author":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2025-11-03 13:09:48.000000000","tz":60},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-08-19 07:49:12.000000000","tz":0},"subject":"Separate replicated and erasure coded pool handling in cephadm","message":"Separate replicated and erasure coded pool handling in cephadm\n\nReplace single \"POOLS\" array with \"POOLS_REPLICATED\" and \"POOLS_EC\" arrays to\ndistinguish between replicated and erasure coded pools. Update the \"add_pools()\"\nfunction to accept a \"pool mode\" parameter and create pools with the appropriate\nreplication strategy, while maintaining backward compatibility to existing\nreplicated-only deployment scenarios. In addition, this change adds support\nto configure both Glance and Manila (cephfs) with erasure coded pools via\n\"\u003cSERVICE\u003e_EC_ENABLE\" parameters.\nThe data_pool can be customized through the \"\u003cSERVICE\u003e_EC_DATA_POOL\" variable.\n\nChange-Id: I589adb435799d558b97da454e2cf754494c0feb4\nSigned-off-by: Francesco Pantano \u003cfpantano@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/f5ca9050baf795546e3001a5001517d8c5130904"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/f5ca9050baf795546e3001a5001517d8c5130904"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
