)]}'
{"id":"openstack%2Fdevstack-plugin-ceph~985325","triplet_id":"openstack%2Fdevstack-plugin-ceph~master~I4b3621ad5abdb57bbfdec863d84a4e24220203df","project":"openstack/devstack-plugin-ceph","branch":"master","topic":"ec_pools","attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-04-20 07:39:17.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I4b3621ad5abdb57bbfdec863d84a4e24220203df","subject":"DNM: Test Manila EC pool support with CephFS NFS cephadm","status":"NEW","created":"2026-04-20 07:05:57.000000000","updated":"2026-08-21 12:28:23.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8bad88fd4154fef3f53b0d487610664f32b426c9","_number":985325,"virtual_id_number":985325,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-20 22:59:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-20 07:39:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-21 12:28:23.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"CC"}],"messages":[{"id":"2796888f9b5e774299adcb4cad6174358592b2e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-20 07:05:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cf7eb096147fd129a273071d5c50250b9c63420","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 07:39:17.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/966014 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"4a8037899e06486d60b3b1ec944bc379f44dc8ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-20 20:46:30.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"4a1231754a3773abb3355b62810a160925383dda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 22:59:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8f87d930c42418fb1608bf16a4f18f7\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d0bb92f82fe248b18edf75339e2ce707 : SUCCESS in 49m 32s\n- devstack-plugin-ceph-multinode-tempest-py3 https://zuul.opendev.org/t/openstack/build/babd268184344e8d8a662f61fa0b9c1f : SUCCESS in 2h 07m 30s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/aa95d331cf0a45bb8a483583f7a0ae05 : SUCCESS in 25m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/14519df2fae848bf81c36add9773849d : SUCCESS in 53m 07s\n- devstack-plugin-ceph-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/aad09fc8e507469fbffc77cd97adda15 : FAILURE in 1h 23m 28s (non-voting)\n- devstack-plugin-ceph-cephfs-native https://zuul.opendev.org/t/openstack/build/6a12c79ef8f34c6ea20786765ef2c2d5 : FAILURE in 18m 52s (non-voting)\n- devstack-plugin-ceph-cephfs-nfs-standalone https://zuul.opendev.org/t/openstack/build/38c79dab474c41c4b4b122f029d44cec : FAILURE in 1h 03m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d05d14b563c490bba02eac777f45639 : SUCCESS in 6m 12s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/8aa3578a9b334c3a9f67e758fd3e992f : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"37b1a901e0c564a6dc20105263c36f5eaf1a1080","revisions":{"081db0c4ec76ce92e343ddba214cee90834b456f":{"kind":"REWORK","_number":1,"created":"2026-04-20 07:05:57.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/25/985325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack-plugin-ceph","ref":"refs/changes/25/985325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/25/985325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/25/985325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/25/985325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/25/985325/1"}}},"commit":{"parents":[{"commit":"1324730e88d9af1230a1390c1d244579efa13a84","subject":"Add nova ceph jobs for more coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/1324730e88d9af1230a1390c1d244579efa13a84"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-20 07:05:17.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-20 07:05:47.000000000","tz":-420},"subject":"DNM: Test Manila EC pool support with CephFS NFS cephadm","message":"DNM: Test Manila EC pool support with CephFS NFS cephadm\n\nEnable MANILA_EC_ENABLE in the multinode cephfs-nfs-cephadm job\nto verify erasure coded data pool support with Manila.\n\nDepends-On: I589adb435799d558b97da454e2cf754494c0feb4\nChange-Id: I4b3621ad5abdb57bbfdec863d84a4e24220203df\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/081db0c4ec76ce92e343ddba214cee90834b456f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/081db0c4ec76ce92e343ddba214cee90834b456f"}]},"branch":"refs/heads/master"},"37b1a901e0c564a6dc20105263c36f5eaf1a1080":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2026-08-20 20:46:30.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/25/985325/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack-plugin-ceph","ref":"refs/changes/25/985325/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/25/985325/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/25/985325/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/25/985325/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/25/985325/2"}}},"commit":{"parents":[{"commit":"6c125515781a00df0e58cad5d9c0200b4d19db5e","subject":"Merge \"Fix Glance EC pool config for RBD and multi-store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/6c125515781a00df0e58cad5d9c0200b4d19db5e"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-04-20 07:05:17.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-08-20 20:46:16.000000000","tz":-420},"subject":"DNM: Test Manila EC pool support with CephFS NFS cephadm","message":"DNM: Test Manila EC pool support with CephFS NFS cephadm\n\nEnable MANILA_EC_ENABLE in the multinode cephfs-nfs-cephadm job\nto verify erasure coded data pool support with Manila.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/1001739\nChange-Id: I4b3621ad5abdb57bbfdec863d84a4e24220203df\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/37b1a901e0c564a6dc20105263c36f5eaf1a1080"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/37b1a901e0c564a6dc20105263c36f5eaf1a1080"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
