)]}'
{"id":"openstack%2Fdevstack-plugin-ceph~986473","triplet_id":"openstack%2Fdevstack-plugin-ceph~master~I3ed729bdedf3bbf3b43306790633ed0568983638","project":"openstack/devstack-plugin-ceph","branch":"master","topic":"ipv6-multinode","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-09-15 15:12:24.000000000","reason":"Change was marked work in progress"},"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2026-09-15 14:59:07.000000000","reason":"\u003cGERRIT_ACCOUNT_4690\u003e replied on the change","reason_account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}}},"hashtags":[],"change_id":"I3ed729bdedf3bbf3b43306790633ed0568983638","subject":"Use IPv6 for ceph mon when SERVICE_IP_VERSION\u003d6","status":"NEW","created":"2026-04-28 00:41:07.000000000","updated":"2026-09-15 19:48:08.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"ad306eb471e7c9dc569441f5658a6e72ed9621e8","_number":986473,"virtual_id_number":986473,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-15 17:21:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-28 03:24:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-04 14:24:12.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"}],"messages":[{"id":"26457124a495a1a62fff29a2d5a570278ab691d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-04-28 00:41:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0c16bcab065dcf5d4b5ff47e8e6727cba989e4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 03:24:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1311ced9814f40f2890ec7d440cf538b\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/06b3eecbfee74302be2942f42bbd9849 : SUCCESS in 1h 02m 33s\n- devstack-plugin-ceph-multinode-tempest-py3 https://zuul.opendev.org/t/openstack/build/39c7b5ceac2b4d6e988ae93b63dbb33f : SUCCESS in 2h 38m 55s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0f4412c27ec548a2914176b9d9d7812b : SUCCESS in 35m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4b868b4a33bb4cb08045cef925201f65 : SUCCESS in 1h 41m 21s\n- devstack-plugin-ceph-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/87144a8dbbc04bc7bfdde7bbfd0ef2d5 : FAILURE in 1h 53m 35s (non-voting)\n- devstack-plugin-ceph-cephfs-native https://zuul.opendev.org/t/openstack/build/0dc55d9597884a4e894f795d7845f9c3 : FAILURE in 59m 07s (non-voting)\n- devstack-plugin-ceph-cephfs-nfs-standalone https://zuul.opendev.org/t/openstack/build/ee28af6f76e24ac19648ead9adf59a82 : FAILURE in 1h 46m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c8469f081734588afd73cbc168671b5 : SUCCESS in 6m 00s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5e5d3fca161e4ebeb3e31fbf0c311a17 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":1},{"id":"95de78be6f97a89f9c1252451d2d3ea31840900f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-06-23 19:10:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f833117bac48809060984b052089c76096a69cb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 21:04:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4255ae435544073b08316fb955b0f79\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a7ff7f090f544c9688937f5bc072a935 : SUCCESS in 1h 43m 39s\n- devstack-plugin-ceph-multinode-tempest-py3 https://zuul.opendev.org/t/openstack/build/d992423dfc4246958a2e3f190bada47f : SUCCESS in 1h 27m 01s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/489f48a8b7034b4398e4a49b97655b54 : SUCCESS in 41m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9b73be16d1ae420d81597de7577425e9 : SUCCESS in 1h 29m 17s\n- devstack-plugin-ceph-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a750d51b1c7c486c81b93701d7a76a26 : SUCCESS in 1h 46m 48s (non-voting)\n- devstack-plugin-ceph-cephfs-native https://zuul.opendev.org/t/openstack/build/ef3666ec90914a9dae1b942af12ce752 : SUCCESS in 28m 03s (non-voting)\n- devstack-plugin-ceph-cephfs-nfs-standalone https://zuul.opendev.org/t/openstack/build/47a09a056609493486075e8178a6bfc4 : SUCCESS in 59m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55ffa54b0a554d64898700c8e85f272b : SUCCESS in 5m 28s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/d35059c99b5c48f491efa697aa203c5c : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":2},{"id":"52e28f3b9c20dfaa5f44ee670a6ee2a312a7ce83","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-04 14:24:12.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f0fc501caa06955788a52c76a8bd60118a5e1bad","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-09-15 14:59:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"203e302d35a79afb1a0adfd0537ffae8efcd05cb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-09-15 15:12:24.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"bd047907f401e107846f0c5bcf8d5ed080728c6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 17:21:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ea55233db7943898061ca07caad61f4\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ef203277a41c4a109c97860860e6e272 : SUCCESS in 1h 10m 54s\n- devstack-plugin-ceph-multinode-tempest-py3 https://zuul.opendev.org/t/openstack/build/d39736a9048d45fdb62e9405993fc59a : SUCCESS in 1h 02m 57s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/cee7b86386db4e3fb7ca850e8f9e1473 : SUCCESS in 49m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9b351a79e61b42ecb86464f4ceead283 : SUCCESS in 1h 48m 48s\n- devstack-plugin-ceph-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ab03c015dfe64f5782e038df10593e60 : FAILURE in 1h 25m 20s (non-voting)\n- devstack-plugin-ceph-cephfs-native https://zuul.opendev.org/t/openstack/build/4663f30273b84a459c7c0f4a3fbdcd32 : FAILURE in 40m 00s (non-voting)\n- devstack-plugin-ceph-cephfs-nfs-standalone https://zuul.opendev.org/t/openstack/build/9637676ba7ab490d95e93993a789a13e : FAILURE in 1h 27m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab51c5fd9c644725b6928bb26206b269 : SUCCESS in 6m 05s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/7134469291c94156b628e05b398117ac : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":3},{"id":"ad306eb471e7c9dc569441f5658a6e72ed9621e8","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-09-15 19:48:08.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c02720dd939511a1bc85baca5d31da79bc6e73d4","revisions":{"e20f624f33b3f1bda8824b89f25a9b9b6c2fc944":{"kind":"REWORK","_number":1,"created":"2026-04-28 00:41:07.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/73/986473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack-plugin-ceph","ref":"refs/changes/73/986473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/73/986473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/73/986473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/73/986473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/73/986473/1"}}},"commit":{"parents":[{"commit":"1324730e88d9af1230a1390c1d244579efa13a84","subject":"Add nova ceph jobs for more coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/1324730e88d9af1230a1390c1d244579efa13a84"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2026-04-28 00:39:59.000000000","tz":-420},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2026-04-28 00:40:50.000000000","tz":-420},"subject":"DNM Use IPv6 for ceph mon when SERVICE_IP_VERSION\u003d6","message":"DNM Use IPv6 for ceph mon when SERVICE_IP_VERSION\u003d6\n\nWhen devstack is configured with SERVICE_IP_VERSION\u003d6, use\nHOST_IPV6 instead of HOST_IP for the ceph monitor address\nand the ingress VIP. This allows ceph to work with IPv6\nservice endpoints in multinode deployments.\n\nAssisted-by: Claude claude-opus-4-6 claude-code-2.1.116\nChange-Id: I3ed729bdedf3bbf3b43306790633ed0568983638\nSigned-off-by: melanie witt \u003cmelwittt@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/e20f624f33b3f1bda8824b89f25a9b9b6c2fc944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/e20f624f33b3f1bda8824b89f25a9b9b6c2fc944"}]},"branch":"refs/heads/master"},"9a3afb1dc2e08a91bca2a2a1158f8af1df7fcfc1":{"kind":"REWORK","_number":2,"created":"2026-06-23 19:10:32.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/73/986473/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack-plugin-ceph","ref":"refs/changes/73/986473/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/73/986473/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/73/986473/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/73/986473/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/73/986473/2"}}},"commit":{"parents":[{"commit":"7763f48d0f271e13e6fb718c64b679cd34c308f9","subject":"Add support for ceph replication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/7763f48d0f271e13e6fb718c64b679cd34c308f9"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2026-04-28 00:39:59.000000000","tz":-420},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2026-06-23 19:09:52.000000000","tz":-420},"subject":"Use IPv6 for ceph mon when SERVICE_IP_VERSION\u003d6","message":"Use IPv6 for ceph mon when SERVICE_IP_VERSION\u003d6\n\nWhen devstack is configured with SERVICE_IP_VERSION\u003d6, use\nHOST_IPV6 instead of HOST_IP for the ceph monitor address\nand the ingress VIP. This allows ceph to work with IPv6\nservice endpoints in multinode deployments.\n\nAssisted-by: Claude claude-opus-4-6 claude-code-2.1.116\nChange-Id: I3ed729bdedf3bbf3b43306790633ed0568983638\nSigned-off-by: melanie witt \u003cmelwittt@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/9a3afb1dc2e08a91bca2a2a1158f8af1df7fcfc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/9a3afb1dc2e08a91bca2a2a1158f8af1df7fcfc1"}]},"branch":"refs/heads/master"},"c02720dd939511a1bc85baca5d31da79bc6e73d4":{"kind":"REWORK","_number":3,"created":"2026-09-15 15:12:24.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/73/986473/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack-plugin-ceph","ref":"refs/changes/73/986473/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/73/986473/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/73/986473/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/73/986473/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/73/986473/3"}}},"commit":{"parents":[{"commit":"7763f48d0f271e13e6fb718c64b679cd34c308f9","subject":"Add support for ceph replication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/7763f48d0f271e13e6fb718c64b679cd34c308f9"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2026-04-28 00:39:59.000000000","tz":-420},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2026-09-15 15:11:56.000000000","tz":-420},"subject":"Use IPv6 for ceph mon when SERVICE_IP_VERSION\u003d6","message":"Use IPv6 for ceph mon when SERVICE_IP_VERSION\u003d6\n\nWhen devstack is configured with SERVICE_IP_VERSION\u003d6, use\nHOST_IPV6 instead of HOST_IP for the ceph monitor address\nand the ingress VIP. This allows ceph to work with IPv6\nservice endpoints in multinode deployments.\n\nAssisted-by: Claude claude-opus-4-6 claude-code-2.1.116\nChange-Id: I3ed729bdedf3bbf3b43306790633ed0568983638\nSigned-off-by: melanie witt \u003cmelwittt@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/c02720dd939511a1bc85baca5d31da79bc6e73d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/c02720dd939511a1bc85baca5d31da79bc6e73d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
