)]}'
{"id":"openstack%2Fdevstack-plugin-ceph~994503","triplet_id":"openstack%2Fdevstack-plugin-ceph~unmaintained%2F2024.1~Ic3c17e6bc1249d608218c63207100a1403f3d741","project":"openstack/devstack-plugin-ceph","branch":"unmaintained/2024.1","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-06-24 10:59:39.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-06-24 12:43:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic3c17e6bc1249d608218c63207100a1403f3d741","subject":"Pin cinder-tempest-plugin in ceph jobs","status":"MERGED","created":"2026-06-23 17:29:05.000000000","updated":"2026-06-24 12:46:41.000000000","submitted":"2026-06-24 12:43:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994503","meta_rev_id":"271c36030d1a2652eb503ab49a83a05a3db2864f","_number":994503,"virtual_id_number":994503,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-24 12:43:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-06-24 09:40:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-06-24 10:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-06-24 10:59:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-23 18:59:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-24 09:40:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-06-24 10:59:39.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"fd32768a0b62ea06ed8fb97384cb9cd5eef1f134","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-23 17:29:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10f6bec9bbd4d8913f2d3c3e1cdfa36b4498a1b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 18:59:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a790441f1a05411ca4c4b81f477c5d8c\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/28994af5e82d44b58693a003c18f06d3 : SUCCESS in 1h 20m 01s\n- devstack-plugin-ceph-tempest-ubuntu https://zuul.opendev.org/t/openstack/build/87505bb1b30c470d9982359114313c8f : FAILURE in 16m 45s (non-voting)\n- devstack-plugin-ceph-cephfs-native https://zuul.opendev.org/t/openstack/build/3346ebfc04324cf0878135848e8018a9 : FAILURE in 52m 42s (non-voting)\n- devstack-plugin-ceph-cephfs-nfs-standalone https://zuul.opendev.org/t/openstack/build/4f3b920106ef4033a2c3e291384304e5 : FAILURE in 35m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d07cb95d4be4afb850d8dc50383e022 : FAILURE in 4m 11s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f18abf52793047baaa9c64b540dd0ee5 : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":1},{"id":"63bf389574b335b2e64ac736b9a10e70728eafe0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-23 19:34:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"897382e48a3b3512778571f7d881ee597d27bfc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-06-24 08:59:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7c05a3e195d4e6637fbac83044a885a6be4e7b8e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-24 09:40:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a63909fc592beec13888d826b2460c86ba68a4a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 10:58:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf33e03f69904d8d919a017492031e95\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/5291c58981db4e17879bf472fda76168 : SUCCESS in 1h 40m 38s\n- devstack-plugin-ceph-tempest-ubuntu https://zuul.opendev.org/t/openstack/build/774f3f827ce8487992b055a6e9940616 : FAILURE in 1h 52m 54s (non-voting)\n- devstack-plugin-ceph-cephfs-native https://zuul.opendev.org/t/openstack/build/88628ab305b743aea8838d262a42c20c : FAILURE in 34m 38s (non-voting)\n- devstack-plugin-ceph-cephfs-nfs-standalone https://zuul.opendev.org/t/openstack/build/dbdf1a4a9f214f3bb372fa3cf5efccba : FAILURE in 17m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5251eea00d2d4aefbd066c37798ae28a : SUCCESS in 4m 08s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/1a1bbebb18b0412eab5a295a5ed1b0ce : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"4d6c8a534a7982e3b92a83b0d6b1f0c74b868934","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-06-24 10:59:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0a2d3a097de935325add024dbb078e29b6b1a918","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 11:00:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d7f34dea220d328cb3647c0ed6d9b282890275cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 12:43:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e778ad7f5f04d1f952f35dc98d188e5\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/5cc409100a7d41d0b7528da6c6dd9f2a : SUCCESS in 1h 38m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a946a19911d434a8d3c5982f3526abb : SUCCESS in 4m 42s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/2930baa2ab09416f8e9a196bf661e7a2 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":2},{"id":"69c41b3cc597f06b0771bb8e3178898829c8f4ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 12:43:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"271c36030d1a2652eb503ab49a83a05a3db2864f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 12:46:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11c6c83df22b4e55b24e01643b9f8378\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64b4258f9312405ba0b5114ad160469c : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ceb12617696312b578003cda01f3a63ba71bddb9","revisions":{"52382f04456c00b50615812ef4f4bcfa56867ce0":{"kind":"REWORK","_number":1,"created":"2026-06-23 17:29:05.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/03/994503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack-plugin-ceph","ref":"refs/changes/03/994503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/03/994503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/03/994503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/03/994503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/03/994503/1"}}},"commit":{"parents":[{"commit":"1ec8a65ede4fc413a5b2933d50251659ca338170","subject":"Update .gitreview for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/1ec8a65ede4fc413a5b2933d50251659ca338170"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-06-23 17:27:51.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-06-23 17:29:04.000000000","tz":0},"subject":"Pin cinder-tempest-plugin in ceph jobs","message":"Pin cinder-tempest-plugin in ceph jobs\n\n2024.1 use old tempest 45.0.0 but the master\ncinder-tempest-plugin which is not compatible\nand break the jobs. Pinning the cinder-tempest-plugin\nalso with compatible tag.\n\nChange-Id: Ic3c17e6bc1249d608218c63207100a1403f3d741\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/52382f04456c00b50615812ef4f4bcfa56867ce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/52382f04456c00b50615812ef4f4bcfa56867ce0"}]},"branch":"refs/heads/unmaintained/2024.1"},"ceb12617696312b578003cda01f3a63ba71bddb9":{"kind":"REWORK","_number":2,"created":"2026-06-24 08:59:41.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/03/994503/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack-plugin-ceph","ref":"refs/changes/03/994503/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/03/994503/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/03/994503/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/03/994503/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/03/994503/2"}}},"commit":{"parents":[{"commit":"1ec8a65ede4fc413a5b2933d50251659ca338170","subject":"Update .gitreview for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/1ec8a65ede4fc413a5b2933d50251659ca338170"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-06-23 17:27:51.000000000","tz":0},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-06-24 08:56:21.000000000","tz":120},"subject":"Pin cinder-tempest-plugin in ceph jobs","message":"Pin cinder-tempest-plugin in ceph jobs\n\n2024.1 use old tempest 45.0.0 but the master\ncinder-tempest-plugin which is not compatible\nand break the jobs. Pinning the cinder-tempest-plugin\nalso with compatible tag.\n\nNOTE(elod.illes): tox-docs job fails [1] if setuptools\nis not capped to \u003c82.0.0, so setting it in tox.ini.\n\n[1] Could not import extension openstackdocstheme (exception: No module named \u0027pkg_resources\u0027)\n\nChange-Id: Ic3c17e6bc1249d608218c63207100a1403f3d741\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/ceb12617696312b578003cda01f3a63ba71bddb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/ceb12617696312b578003cda01f3a63ba71bddb9"}]},"branch":"refs/heads/unmaintained/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
