)]}'
{"id":"openstack%2Fdevstack-plugin-ceph~998427","triplet_id":"openstack%2Fdevstack-plugin-ceph~master~I229c0f7539362fae02bee18d0768705b417c94f1","project":"openstack/devstack-plugin-ceph","branch":"master","topic":"support-multi-ceph","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-07-30 14:22:09.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-09-04 15:03:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I229c0f7539362fae02bee18d0768705b417c94f1","subject":"Add set-require-min-compat-client to secondary ceph cluster","status":"MERGED","created":"2026-07-23 09:32:30.000000000","updated":"2026-09-04 15:04:15.000000000","submitted":"2026-09-04 15:03:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998427-support-multi-ceph","meta_rev_id":"a42bdf4cbe51451f796d833c7a0c98d2abc30a0c","_number":998427,"virtual_id_number":998427,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-04 15:03:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-09-04 14:13:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2026-07-30 14:22:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2026-09-04 14:13:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-23 11:32:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-27 19:05:01.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-09-04 14:13:56.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"998613eb7b2b4b592f4ff6e4d638c7df4ad52867","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-07-23 09:32:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b5f32471f548bbc60c6f0f543e41f2f6fb7b581","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-07-23 10:33:32.000000000","message":"Topic set to support-multi-ceph","accounts_in_message":[],"_revision_number":1},{"id":"4558e379a5ebabca30a9d13d0db476a5c19c2af5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 11:32:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3579181af59248d28e61af55f952e235\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1f750e18cff5426ebfb6aba10cab5311 : SUCCESS in 48m 34s\n- devstack-plugin-ceph-multinode-tempest-py3 https://zuul.opendev.org/t/openstack/build/6cbfdcfed86245e982e9183e2ae0e397 : SUCCESS in 1h 52m 37s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/3a76e01a0bfc4695b2a841fd6d80eb65 : SUCCESS in 50m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2e61324f3a3347cfb3c6bcf2b539fe77 : SUCCESS in 1h 35m 00s\n- devstack-plugin-ceph-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c35ba3a872eb4f6e9e1b5a8176599625 : SUCCESS in 1h 15m 30s (non-voting)\n- devstack-plugin-ceph-cephfs-native https://zuul.opendev.org/t/openstack/build/8b42a75c83cb45ce9eead95f65d0ce5d : SUCCESS in 1h 07m 24s (non-voting)\n- devstack-plugin-ceph-cephfs-nfs-standalone https://zuul.opendev.org/t/openstack/build/5465adb62bc145758e222052e10f01de : SUCCESS in 1h 29m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bf2236937e6493f884d95f25d58c055 : SUCCESS in 5m 59s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/3c174f4ae3c64143a2269138a53d4eda : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7c7494a38445c4c84a9c551a9da8ea46d54d62f1","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-07-30 14:22:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8ceb0fc9a91cdf239d8870803e72c683d6015bf6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-04 14:13:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1844c9252c966f2573c172b037318a45fdffb01c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 14:14:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4929ab455a35533b38bc850540157eb11fcb40b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 15:03:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fcba1efdc834c6e94663a8d4b1d7e63\n\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e6c4dc26a4f04265942e71f74e7b3126 : SUCCESS in 48m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66f0d5548423473d97c36bd7a45e76f6 : SUCCESS in 3m 24s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/beb9103645f448469266f1c2c8302998 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f45f3577096ad4858311c24446e7e70aaade3115","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 15:03:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a42bdf4cbe51451f796d833c7a0c98d2abc30a0c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 15:04:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c21700dc8ac04f7fbef6efab13b4ab06\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40f73027fc104cb2aab0677d812f35c8 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f2bd5709af145c382e13dcf666d006109f4a6468","revisions":{"f2bd5709af145c382e13dcf666d006109f4a6468":{"kind":"REWORK","_number":1,"created":"2026-07-23 09:32:30.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/27/998427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack-plugin-ceph","ref":"refs/changes/27/998427/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/27/998427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/27/998427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/27/998427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack-plugin-ceph refs/changes/27/998427/1"}}},"commit":{"parents":[{"commit":"daed912296e89835b5c240c10e2ebecdc1eaef2c","subject":"Merge \"Allow customizing NFS-Ganesha port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/daed912296e89835b5c240c10e2ebecdc1eaef2c"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2026-07-23 09:26:19.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2026-07-23 09:29:34.000000000","tz":0},"subject":"Add set-require-min-compat-client to secondary ceph cluster","message":"Add set-require-min-compat-client to secondary ceph cluster\n\nWhen configuring 2 ceph clusters with MULTI_CEPH\u003dTrue, we match\nthe configurations of the primary ceph clusters.\nHowever, in the change[1], we missed setting\nset-require-min-compat-client in secondary cluster which enables\nclone v2 capabilities so fixing it in this patch.\n\n[1] https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/951810\n\nChange-Id: I229c0f7539362fae02bee18d0768705b417c94f1\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/f2bd5709af145c382e13dcf666d006109f4a6468"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack-plugin-ceph/commit/f2bd5709af145c382e13dcf666d006109f4a6468"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
