)]}'
{"id":"openstack%2Fdevstack~1002953","triplet_id":"openstack%2Fdevstack~master~Ib4cbfed227c754c6849d66cc278709c9638f616a","project":"openstack/devstack","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-30 09:34:05.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib4cbfed227c754c6849d66cc278709c9638f616a","subject":"tempest: drop redundant [network] public_router_id","status":"NEW","created":"2026-08-30 08:33:43.000000000","updated":"2026-08-30 10:20:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"801cfe099b6d7892dd0720cac8cc3a1b9e5efb91","_number":1002953,"virtual_id_number":1002953,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-30 10:20:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2026-08-30 09:34:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-30 09:34:05.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-08-30 10:20:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"574c09bf8cca944698740b428986266f8dd52f79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-30 08:33:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7419cbe6ceac0563ef5099275360d6ea970805a3","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-30 09:34:05.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/4e2ffafc55524deebb580e7ef2fb0cfa\n\n- devstack-platform-centos-10-stream https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/0b84905ebd584b4f9d6db95f6a894869 : SUCCESS in 59m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"801cfe099b6d7892dd0720cac8cc3a1b9e5efb91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 10:20:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82139c6165aa4965b75970f6bb5cf19f\n\n- grenade https://zuul.opendev.org/t/openstack/build/fb00e5fae7584d788fbc8d0f1c3e16c9 : SUCCESS in 32m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/023dd79060e047babda8740c23bdebeb : SUCCESS in 1h 40m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95d7b41d7f4e43d3adee33d5651751ed : SUCCESS in 3m 51s\n- devstack https://zuul.opendev.org/t/openstack/build/0b22774c1a314e98b7f5f46527cee679 : SUCCESS in 26m 22s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/9f1062792a194abf8bdf0c21c991b793 : SUCCESS in 13m 31s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/5f1cdb103263410fa352a8fdbb5719a2 : SUCCESS in 25m 13s\n- devstack-platform-almalinux-purple-lion-ovn-source https://zuul.opendev.org/t/openstack/build/6f3ecd41c6d344aa824310fd3fda8b0f : FAILURE in 16m 53s (non-voting)\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/61533697ec344b5f8237ed1d789a68e7 : SUCCESS in 44m 44s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/bdcad855ff9c44059c9992910d558641 : SUCCESS in 1h 27m 10s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/52b8762cfdf54c248bb3e40f221d50b4 : SUCCESS in 47m 47s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/7d14d0a66abb48e4a11b345a5a38e76e : SUCCESS in 1h 21m 04s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/acda47f613a343e5903ba65f9d9cbb53 : SUCCESS in 1h 33m 32s (non-voting)\n- devstack-platform-rocky-red-quartz https://zuul.opendev.org/t/openstack/build/6cadaeeed35a4cdf9d41e67fc2d2a499 : FAILURE in 26m 27s (non-voting)\n- devstack-platform-ubuntu-noble-ovn-source https://zuul.opendev.org/t/openstack/build/07bc7a015a4849a6ae1c577e6038a229 : SUCCESS in 1h 32m 26s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/e22c95c5cf66443485800102910eaa97 : SUCCESS in 1h 19m 17s (non-voting)\n- devstack-platform-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/a4818727513b4b07ae9c25044e3b8a9a : SUCCESS in 1h 21m 26s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/05f98e49be68424e84e162f48d6008cc : SUCCESS in 35m 18s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/c3849a17b36e4b4db2e373233d90704e : SUCCESS in 2m 34s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/3cf0df225078414fa8d0260786e47011 : SUCCESS in 2m 12s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/cd821bffc5404af991ee30f762a8d6db : SUCCESS in 57m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c68e463c25e458fbda35677c9fb1512 : SUCCESS in 50m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b1b53896a924ec6bbb6bbef4f2c8f04 : SUCCESS in 1h 10m 54s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2c18ffe5d3194c0d91a553379519ba00 : SUCCESS in 1h 20m 22s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/79358347464b4ac4916bb87c0c96c9f6 : SUCCESS in 1h 08m 11s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a1605f0e8ae49cb9d079300452c6f65 : SUCCESS in 46m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d569d0353cd42ddabd1959cb5d70635 : SUCCESS in 1h 00m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b0823e5acbff46ce9768a67a535fe4a2 : SUCCESS in 50m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af98d6e02ddf1a31d79757e1b22d50c6937f9f86","revisions":{"af98d6e02ddf1a31d79757e1b22d50c6937f9f86":{"kind":"REWORK","_number":1,"created":"2026-08-30 08:33:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/1002953/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/53/1002953/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/53/1002953/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/53/1002953/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/53/1002953/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/53/1002953/1"}}},"commit":{"parents":[{"commit":"9b0a00865c2537338f72ed5b9b44a0d6bff16eb4","subject":"ovs: Fix OVS path mismatch when building from source with OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/9b0a00865c2537338f72ed5b9b44a0d6bff16eb4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-30 16:04:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-30 16:04:17.000000000","tz":540},"subject":"tempest: drop redundant [network] public_router_id","message":"tempest: drop redundant [network] public_router_id\n\nThis option is always set to an empty string, which exactly matches\nthe default value.\n\nChange-Id: Ib4cbfed227c754c6849d66cc278709c9638f616a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/af98d6e02ddf1a31d79757e1b22d50c6937f9f86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/af98d6e02ddf1a31d79757e1b22d50c6937f9f86"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
