)]}'
{"id":"openstack%2Fdevstack~1002954","triplet_id":"openstack%2Fdevstack~master~I9b46c7ba47ed2e0ebb40e861c31ffba84389497a","project":"openstack/devstack","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-30 09:29:46.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9b46c7ba47ed2e0ebb40e861c31ffba84389497a","subject":"Create neutron resources only when neutron is enabled","status":"NEW","created":"2026-08-30 08:35:42.000000000","updated":"2026-08-30 10:32:04.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e97cd1a3d28ba54b89e65b2310a4296ca19820d7","_number":1002954,"virtual_id_number":1002954,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-30 10:32:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2026-08-30 09:29:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-30 09:29:46.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-08-30 10:32:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58a81fde7ab9223876a0137b0bcf0578e7f17160","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-30 08:35:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d6849d81b9b660a32530d34b70f3047ca2e0375","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-30 09:29:46.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/5e11e7db9d544aac9c9831e1b51b883e\n\n- devstack-platform-centos-10-stream https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/6aaa15b298fb4ca790ee8b3096aced06 : SUCCESS in 52m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e97cd1a3d28ba54b89e65b2310a4296ca19820d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 10:32:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6cdeff2efaa4a7da4deef6c0fa2c302\n\n- grenade https://zuul.opendev.org/t/openstack/build/85ff9188ef214329a2425e2252366d22 : SUCCESS in 34m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b2959044cd34a38bcea12ad04129afd : SUCCESS in 1h 32m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2432bfae92f14d76b8178e289f8d7c65 : SUCCESS in 6m 05s\n- devstack https://zuul.opendev.org/t/openstack/build/8c57fefa1aea450f8c634a3e9ff0be4b : SUCCESS in 22m 29s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/45dd6692dd1e4390ab75e290fce87f59 : SUCCESS in 19m 13s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/c9c0c68a9777464395a560439642f337 : SUCCESS in 14m 22s\n- devstack-platform-almalinux-purple-lion-ovn-source https://zuul.opendev.org/t/openstack/build/bd238f1e2d514586a13debf1bd5dfc58 : FAILURE in 19m 05s (non-voting)\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/aa1c9fcc554d413897453861eb9b047e : SUCCESS in 1h 00m 40s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ecff782c3f64512b8b570ea396599e7 : SUCCESS in 1h 17m 19s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/21d1a86b2c914ec38fe4f19191af1637 : SUCCESS in 42m 55s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/0033095333d04136bb1fb244ed942f95 : SUCCESS in 55m 13s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/a9be645db3244d90ad8c39c71386ef8d : FAILURE in 1h 49m 51s (non-voting)\n- devstack-platform-rocky-red-quartz https://zuul.opendev.org/t/openstack/build/b298f22de2be4e658d8dd2c64f7b233b : FAILURE in 17m 51s (non-voting)\n- devstack-platform-ubuntu-noble-ovn-source https://zuul.opendev.org/t/openstack/build/cb6f8426c4534787867e59effd6bb401 : SUCCESS in 1h 11m 08s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/dee1e23d5a4e46c982163e45953965f4 : SUCCESS in 1h 21m 10s (non-voting)\n- devstack-platform-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/17f9b79787d24c82a629212ce854b3ff : SUCCESS in 47m 45s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/2c202c0108a143e3b7f88a16dfc816ac : SUCCESS in 34m 42s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/dc3653ed273949b1815a54018464160e : SUCCESS in 2m 48s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/a3ce023f84084ac582f45d6e2473dbfc : SUCCESS in 2m 27s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/bfb51bab33aa4e229abda8b52aefaa2d : SUCCESS in 1h 16m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/afc6b7ea470f432aa98288b14933ec01 : SUCCESS in 38m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5b7621565df94c8fbf90f52fed36f3f2 : SUCCESS in 1h 02m 37s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5b61763f72f947f48b859e6c5168854e : SUCCESS in 1h 07m 34s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/949dd9d92ad54d27bc930ec491f789c5 : SUCCESS in 35m 15s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/797f22f03dd14d4ba9fa7246c8f5aa3c : SUCCESS in 57m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4484ddd38ed04959aa46df73998c6b82 : SUCCESS in 26m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6a4fa19781ad484cae593857ad7db430 : SUCCESS in 1h 23m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9711bdfd9088c5b469da1408316c6a4c65bf44ba","revisions":{"9711bdfd9088c5b469da1408316c6a4c65bf44ba":{"kind":"REWORK","_number":1,"created":"2026-08-30 08:35:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/1002954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/54/1002954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/54/1002954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/54/1002954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/54/1002954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/54/1002954/1"}}},"commit":{"parents":[{"commit":"9b0a00865c2537338f72ed5b9b44a0d6bff16eb4","subject":"ovs: Fix OVS path mismatch when building from source with OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/9b0a00865c2537338f72ed5b9b44a0d6bff16eb4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-30 16:07:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-30 16:07:48.000000000","tz":540},"subject":"Create neutron resources only when neutron is enabled","message":"Create neutron resources only when neutron is enabled\n\n... following how the other resources such as flavors are created.\n\nChange-Id: I9b46c7ba47ed2e0ebb40e861c31ffba84389497a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/9711bdfd9088c5b469da1408316c6a4c65bf44ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/9711bdfd9088c5b469da1408316c6a4c65bf44ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
