)]}'
{"id":"openstack%2Fdevstack~1002955","triplet_id":"openstack%2Fdevstack~master~If5fbf5a15904d0c214efa0e9cab0ae8412f00f5e","project":"openstack/devstack","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-30 09:36:31.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"If5fbf5a15904d0c214efa0e9cab0ae8412f00f5e","subject":"Remove deprecated/removed identity-feature-enabled options","status":"NEW","created":"2026-08-30 08:40:07.000000000","updated":"2026-08-30 10:22:01.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3ad290c14cbff38aed9480fa6ba20562ebcec979","_number":1002955,"virtual_id_number":1002955,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-30 10:22:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2026-08-30 09:36:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-30 09:36:31.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-08-30 10:22:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b6b1a3437b4b5cc0b03bdd74ce8dc700c5a5779","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-30 08:40:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7fee7705db86ec34782b6f2600023e5add23012","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-30 09:36:31.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/2dcb268fa3234d0e925626cc1cbf0b78\n\n- devstack-platform-centos-10-stream https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/bb12edfb56cd4f84bbbbf1f171dcb5e9 : SUCCESS in 55m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ad290c14cbff38aed9480fa6ba20562ebcec979","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 10:22:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5be5e4ac96d74c31b735d0b345fbb19b\n\n- grenade https://zuul.opendev.org/t/openstack/build/ed959a73481643268c271813dde0aed6 : SUCCESS in 1h 07m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93bd885c9be6422a855e92f258267eda : SUCCESS in 1h 31m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4be31c4c51a443286d32930b2468d5b : SUCCESS in 3m 35s\n- devstack https://zuul.opendev.org/t/openstack/build/d6c15b9877ac49068840f74ac66dbeb5 : SUCCESS in 13m 26s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/e493ce3f354245feb77f6034e1b70723 : SUCCESS in 12m 43s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/51636d15d07841178f25474fb9b5e11b : SUCCESS in 26m 50s\n- devstack-platform-almalinux-purple-lion-ovn-source https://zuul.opendev.org/t/openstack/build/6f72e457385545538e648fde8c2ca321 : FAILURE in 23m 15s (non-voting)\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/23aa55d9fc8444518f95913ec10bbee3 : SUCCESS in 1h 23m 09s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e33fe82ff924fd8ac99d25ebebf44c6 : SUCCESS in 1h 36m 46s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/35847b4f8b76413db2ecfef53e8bdd5e : SUCCESS in 46m 41s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/dcdfe56d9e524fbf9d1589b88e8a493b : SUCCESS in 42m 04s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/6ddb9367ba4a4bdfa64438805ccb1761 : SUCCESS in 42m 43s (non-voting)\n- devstack-platform-rocky-red-quartz https://zuul.opendev.org/t/openstack/build/421ba574c8d147e0ae59c58bdc823a08 : FAILURE in 18m 24s (non-voting)\n- devstack-platform-ubuntu-noble-ovn-source https://zuul.opendev.org/t/openstack/build/d053492f35bc44cf9e2b89585b755895 : SUCCESS in 1h 28m 52s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/bbbb27f1fdcc4ec081a7caecb0c25976 : SUCCESS in 46m 25s (non-voting)\n- devstack-platform-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/a33cda1fcfb8420c9504f69e47441bc6 : SUCCESS in 1h 03m 20s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/164b1c5d1c3d4ba8a9b9b459f474954f : SUCCESS in 33m 54s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/fc558f0257614a4baebaaf6bb7a7c2f3 : SUCCESS in 2m 14s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/a3b2973389b34c80b56ca10203a3f873 : SUCCESS in 2m 16s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a6ae6862f6f94cd7a9c194d415776989 : SUCCESS in 57m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/76535a2ee8db43cdbc8d1c51f097a92d : SUCCESS in 1h 00m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bbcf492cabb94fb18af0bc0357eb21fb : SUCCESS in 38m 04s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d47dd8a1ca0141b79be19e960a62244c : SUCCESS in 58m 23s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0a5ff4b9661c4672a0608b08a2e81db3 : SUCCESS in 1h 09m 50s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8330312538cf4e40b317cf13a786d10c : SUCCESS in 29m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/502de1c70a5a4dedb29a9a71926863ca : SUCCESS in 38m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/01f0ce92ab9c481d9a8ed1944dc71d05 : SUCCESS in 1h 24m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"add892c6a5324f8b4cbf217aedd6ecdae1ca3ea7","revisions":{"add892c6a5324f8b4cbf217aedd6ecdae1ca3ea7":{"kind":"REWORK","_number":1,"created":"2026-08-30 08:40:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/1002955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/55/1002955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/55/1002955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/55/1002955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/55/1002955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/55/1002955/1"}}},"commit":{"parents":[{"commit":"9b0a00865c2537338f72ed5b9b44a0d6bff16eb4","subject":"ovs: Fix OVS path mismatch when building from source with OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/9b0a00865c2537338f72ed5b9b44a0d6bff16eb4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-30 16:09:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-30 16:11:07.000000000","tz":540},"subject":"Remove deprecated/removed identity-feature-enabled options","message":"Remove deprecated/removed identity-feature-enabled options\n\nDepends-on: https://review.opendev.org/c/openstack/tempest/+/906160\nChange-Id: If5fbf5a15904d0c214efa0e9cab0ae8412f00f5e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/add892c6a5324f8b4cbf217aedd6ecdae1ca3ea7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/add892c6a5324f8b4cbf217aedd6ecdae1ca3ea7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
