)]}'
{"id":"openstack%2Fdevstack~1004684","triplet_id":"openstack%2Fdevstack~master~Ie2c11de70f62bbe7f647e57f39192fb07005a861","project":"openstack/devstack","branch":"master","attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-09-08 21:10:12.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie2c11de70f62bbe7f647e57f39192fb07005a861","subject":"make devstack-platform-ubuntu-resolute voting","status":"NEW","created":"2026-09-08 19:00:12.000000000","updated":"2026-09-08 21:10:12.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ec40043621905a999af9257b497dacea26306415","_number":1004684,"virtual_id_number":1004684,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-08 21:10:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":-1,"date":"2026-09-08 19:00:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-08 19:00:24.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-09-08 21:10:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23efe93f7d38f070aca64b776e6914e0586687f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-08 19:00:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81aaabb4b3aab7c06c4890445d820b679577a1e0","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-08 19:00:24.000000000","message":"Patch Set 1: Code-Review-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/devstack on branch master.  The\nproblem was:\n\n  The nodeset \"devstack-single-node-debian-trixie\" was not found.\n\nThe problem appears in the \"devstack-platform-debian-trixie\" job stanza:\n\n  job:\n      name: devstack-platform-debian-trixie\n      parent: tempest-full-py3\n      description: Debian Trixie platform test\n      nodeset: devstack-single-node-debian-trixie\n      timeout: 9000\n      vars:\n        configure_swap_size: 4096\n\n  in \"openstack/devstack/.zuul.yaml@master\", line 809","accounts_in_message":[],"_revision_number":1},{"id":"ec40043621905a999af9257b497dacea26306415","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 21:10:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca07201491694b7b8369bbd643ae3816\n\n- grenade https://zuul.opendev.org/t/openstack/build/8d63f3fba07d48d1a769e6527c6634bc : SUCCESS in 37m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c95c2ff374ef49bdaf054761a5b355da : SUCCESS in 1h 44m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d309b0a2d9fa4080bce1ab7e9f1a3a7c : SUCCESS in 3m 53s\n- devstack https://zuul.opendev.org/t/openstack/build/2f0a03fb5b3e4aa09a47a968a5417f65 : SUCCESS in 14m 26s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/a5f48cd703064d3184508069282cf33f : SUCCESS in 23m 14s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/a3fdc15b9c6d487d98c3a2c8419c351e : SUCCESS in 28m 24s\n- devstack-platform-almalinux-purple-lion-ovn-source https://zuul.opendev.org/t/openstack/build/77773c3657114516bfe092aa3d1ee9ab : FAILURE in 28m 39s (non-voting)\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/cef23855d8e2425f93155681afe228f0 : FAILURE in 39m 31s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c0a1851e6584b3390c7e471deecc91d : FAILURE in 14m 01s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/611399a7deab4bebad68a385842708c0 : SUCCESS in 1h 38m 36s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/5e9b35acd0784a1ab7a2d275c1932713 : SUCCESS in 57m 11s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/eb706c7fda4242a183ff0fa4f053ae79 : FAILURE in 34m 38s (non-voting)\n- devstack-platform-rocky-red-quartz https://zuul.opendev.org/t/openstack/build/5f2baf4f501445c1918b38f1054c2b88 : FAILURE in 25m 22s (non-voting)\n- devstack-platform-ubuntu-noble-ovn-source https://zuul.opendev.org/t/openstack/build/e6bdeb41e60445528e67b1fb157ad41e : SUCCESS in 1h 34m 24s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/24ca2f10ec394244980335eeee47ea87 : SUCCESS in 56m 11s (non-voting)\n- devstack-platform-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/6e9c43bb0d204549ba60ae1350e884e5 : SUCCESS in 55m 00s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/ca08da69b5ef48ceb7eba8bfdfcc5693 : FAILURE in 43m 10s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/be2537be53a24ac0945792b011b9a13c : SUCCESS in 3m 19s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/8d28bb996b6545489a13e14240ec5688 : SUCCESS in 3m 58s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/d86c1486cc9844358ab3c06373e0163d : SUCCESS in 36m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aa0cdbdacb4a454dbf44dc170976b5bc : SUCCESS in 59m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9a2908ebd9be4e27b065ee58df94a2ea : SUCCESS in 57m 12s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/689356861d1e446389d756c75e746df4 : SUCCESS in 1h 10m 41s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e4285f3fc75b44deb87770066bd44b3a : SUCCESS in 1h 16m 40s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebe7dc45e3044fde99b6dcd73cce34a6 : SUCCESS in 1h 00m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50806d3ea8b7482a82ca8165a54c2e82 : SUCCESS in 1h 02m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5213fae66fa64a74a9040823df5bbd52 : SUCCESS in 1h 15m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa66a40cfc2402d78a0ccff82c22765159f78c7f","revisions":{"aa66a40cfc2402d78a0ccff82c22765159f78c7f":{"kind":"REWORK","_number":1,"created":"2026-09-08 19:00:12.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/84/1004684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/84/1004684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/84/1004684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/84/1004684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/84/1004684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/84/1004684/1"}}},"commit":{"parents":[{"commit":"9b0a00865c2537338f72ed5b9b44a0d6bff16eb4","subject":"ovs: Fix OVS path mismatch when building from source with OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/9b0a00865c2537338f72ed5b9b44a0d6bff16eb4"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-09-08 18:58:29.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-09-08 18:58:29.000000000","tz":60},"subject":"make devstack-platform-ubuntu-resolute voting","message":"make devstack-platform-ubuntu-resolute voting\n\nThe ubuntu 26.04 job has been running most most of the cycle\nwithouth issues. Lets make it voting so that it does not regress.\n\nChange-Id: Ie2c11de70f62bbe7f647e57f39192fb07005a861\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/aa66a40cfc2402d78a0ccff82c22765159f78c7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/aa66a40cfc2402d78a0ccff82c22765159f78c7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
