)]}'
{"id":"openstack%2Fdevstack~1004804","triplet_id":"openstack%2Fdevstack~master~I50cf937a399d020bf0f5b08c33675d123994f895","project":"openstack/devstack","branch":"master","attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-09-17 21:25:59.000000000","reason":"\u003cGERRIT_ACCOUNT_13425\u003e replied on the change","reason_account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I50cf937a399d020bf0f5b08c33675d123994f895","subject":"Install boto3 for cinder s3 backup","status":"NEW","created":"2026-09-09 12:47:54.000000000","updated":"2026-09-17 21:25:59.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"71078c1e88bf1f225a90cb308366e7ea103a4b20","_number":1004804,"virtual_id_number":1004804,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-09 14:47:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2026-09-17 21:25:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2026-09-09 13:42:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 13:42:13.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-09-09 14:47:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-17 21:25:59.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"}],"messages":[{"id":"c2a4ef97db83f361c3adfb63c274d3d5ffda241e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-09 12:47:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e99d1b4b8d9d502f19675432a02289adb028854d","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-09 13:42:13.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/96d8410be983416886c02095795bc88d\n\n- devstack-platform-centos-10-stream https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/9ca358a60abb4c1e94fe4b6db160d69e : SUCCESS in 53m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d5f43b5659d802d6153785cd9ca516298fee76c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 14:47:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d8243758bc045a6a5b3b73c762f5d63\n\n- grenade https://zuul.opendev.org/t/openstack/build/f0ad512ad6064502a7bad06a757034a5 : SUCCESS in 1h 06m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/81dfabd8b00f4cd4a33e35b01a314ad4 : SUCCESS in 1h 41m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24dc300f8c724fbb9bb043d863076773 : SUCCESS in 3m 35s\n- devstack https://zuul.opendev.org/t/openstack/build/cc97adddff7049fe9ee92b957ac8e0ed : SUCCESS in 13m 31s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/f601930055a5417b83074ce20ebce93f : SUCCESS in 25m 53s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/e3149e8f2e1c4c7fbc5ca5f4d6287ef3 : SUCCESS in 27m 33s\n- devstack-platform-almalinux-purple-lion-ovn-source https://zuul.opendev.org/t/openstack/build/a700d5363c81489c95bd729aea4152cb : FAILURE in 25m 56s (non-voting)\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/0f0285a169e54547803ae1e6e80f3dc9 : SUCCESS in 58m 12s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6e8457fad124f2baa08c67d6f2b6e91 : SUCCESS in 1h 02m 33s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/e4016e7b89ed4cbda33fff1ce9274965 : SUCCESS in 1h 37m 33s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/3d55c6f0edb04c3fa4d63afe4e5c47c8 : SUCCESS in 38m 23s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/70427845702745d6820171b26e2d5d19 : SUCCESS in 1h 37m 01s (non-voting)\n- devstack-platform-rocky-red-quartz https://zuul.opendev.org/t/openstack/build/93d8ca43636240fdb1cf1a46b522ce85 : FAILURE in 17m 57s (non-voting)\n- devstack-platform-ubuntu-noble-ovn-source https://zuul.opendev.org/t/openstack/build/2b526258a8df44fcaf6df705ec065de2 : SUCCESS in 1h 49m 07s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/fd64f3712b7647628120a68209dd747b : SUCCESS in 45m 08s (non-voting)\n- devstack-platform-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e70e6f2475824a85b6b6a8a47e0a4324 : SUCCESS in 40m 13s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/25ba8c93cc59447f9670d67a408dc9ea : SUCCESS in 27m 41s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/83cf7cd72f954ea0870cf4ee8912fbad : SUCCESS in 3m 05s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/bc4fe6dd61664b75821fcd596543f50f : SUCCESS in 1m 48s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/7c96978544d543eb945e9931c7932ac4 : SUCCESS in 1h 13m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9a9b4f0ce65c4aacbdf863d4c6bd15e5 : SUCCESS in 24m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d7dfa2094f3412496ee7582447e59d7 : SUCCESS in 1h 01m 19s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3061c0f9a6a34924af3f973464140055 : SUCCESS in 54m 54s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d285c91261b14ddcae61f01a3626751c : SUCCESS in 1h 16m 33s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91153b88685042908e56121bfaed0ce7 : SUCCESS in 43m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f082f8fe8eb84971a687189c2ec0af5e : SUCCESS in 59m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1fa49aab23cf41dfa14e0add638c439b : SUCCESS in 1h 35m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e6fe6af3aa258a9e14d0f3d6ba8bd32ddbc9b350","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-09 18:34:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"71078c1e88bf1f225a90cb308366e7ea103a4b20","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-09-17 21:25:59.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"807818e43a3bdf6e0d7593833fcbfb4d7321c6f6","revisions":{"807818e43a3bdf6e0d7593833fcbfb4d7321c6f6":{"kind":"REWORK","_number":1,"created":"2026-09-09 12:47:54.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/04/1004804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/04/1004804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/04/1004804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/04/1004804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/04/1004804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/04/1004804/1"}}},"commit":{"parents":[{"commit":"9b0a00865c2537338f72ed5b9b44a0d6bff16eb4","subject":"ovs: Fix OVS path mismatch when building from source with OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/9b0a00865c2537338f72ed5b9b44a0d6bff16eb4"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-09 12:45:28.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-09-09 12:46:32.000000000","tz":-240},"subject":"Install boto3 for cinder s3 backup","message":"Install boto3 for cinder s3 backup\n\nChange Ie9846b moved boto3 out of cinder\u0027s\nrequirements.txt. This means that the\ncinder-tempest-plugin-cbak-s3 job fails because\nboto3 is not installed.\n\nInstall it when configuring the s3 backup backend.\n\nChange-Id: I50cf937a399d020bf0f5b08c33675d123994f895\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/807818e43a3bdf6e0d7593833fcbfb4d7321c6f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/807818e43a3bdf6e0d7593833fcbfb4d7321c6f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
