)]}'
{"id":"openstack%2Fdevstack~1006721","triplet_id":"openstack%2Fdevstack~stable%2F2026.2~I600ab2603d9a240f9f1825f15df1384cb0ccc501","project":"openstack/devstack","branch":"stable/2026.2","topic":"qa-2026-2-release","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-09-23 14:22:18.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2026-09-23 19:24:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I600ab2603d9a240f9f1825f15df1384cb0ccc501","subject":"Update devstack stable/2026.2 branch setting","status":"MERGED","created":"2026-09-22 13:36:34.000000000","updated":"2026-09-23 19:25:17.000000000","submitted":"2026-09-23 19:24:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"1006721-qa-2026-2-release","meta_rev_id":"9533e536bdd2279bcc97bb2ca3e03311746d999a","_number":1006721,"virtual_id_number":1006721,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-23 19:24:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-09-23 01:32:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2026-09-23 02:24:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2026-09-23 07:26:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2026-09-23 07:26:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2026-09-23 02:24:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-22 15:16:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-22 16:40:46.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-09-22 21:00:11.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"CC"},{"updated":"2026-09-22 21:15:31.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2026-09-23 01:32:55.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2026-09-23 07:26:32.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"3abc28cf346dd06c3be8e480cd40027fefed593c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-22 13:36:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f17e592427c48bb6cb168d411d5ff60ba7025d84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 15:16:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b53d76fa034420d9f4da3dd5bc4869d\n\n- grenade https://zuul.opendev.org/t/openstack/build/9f406c4eeb01468b9341de915fcd3741 : SUCCESS in 1h 02m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eafd3bcea7974db7a3f70bed7128b81a : SUCCESS in 43m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac947a73fec40b5a9659b66e71462ae : SUCCESS in 6m 10s\n- devstack https://zuul.opendev.org/t/openstack/build/193433614e0c48128979826657297376 : SUCCESS in 12m 15s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/1fe38c5e80b245588349fc8a7c7c76ff : SUCCESS in 15m 24s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/e9e3ca794c804e5a8d5ceb3297c56a54 : SUCCESS in 15m 50s\n- devstack-platform-almalinux-purple-lion-ovn-source https://zuul.opendev.org/t/openstack/build/0c78839c95454b4d905e703740d86992 : FAILURE in 10m 43s (non-voting)\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/40c289a4baf44d98ae9c1db6bde14f1f : FAILURE in 7m 21s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a2f2216878242d985151a789922dee5 : FAILURE in 15m 46s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/0b7fd2c7bb0e44248544d08754bd9fc9 : SUCCESS in 1h 22m 55s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/767a83edc99a41acad0f631ee1e423ea : SUCCESS in 1h 02m 49s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/95a92cbb4cf74d04aa8ab3108086f2ea : FAILURE in 23m 06s (non-voting)\n- devstack-platform-rocky-red-quartz https://zuul.opendev.org/t/openstack/build/0e3281085e0744cd8e5e09314ed96780 : FAILURE in 12m 03s (non-voting)\n- devstack-platform-ubuntu-noble-ovn-source https://zuul.opendev.org/t/openstack/build/3418c64667474f96bb5b70144ef0b049 : SUCCESS in 46m 51s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/ca9972f62f55492ea11c2a8e44e16078 : SUCCESS in 1h 25m 05s (non-voting)\n- devstack-platform-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e15eac094beb4b949cb2ce40504f92f5 : SUCCESS in 1h 02m 43s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/c44c34a6559749cc9126310049bb7b82 : SUCCESS in 37m 53s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/bd8160154745475ea4662a6c29688a65 : SUCCESS in 2m 04s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/1b445016937f440195ad4e83e64f78a4 : SUCCESS in 2m 31s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/526600d05f6e4bff918596d32ed0714c : FAILURE in 29m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0dbeb7814c1a46ac80801384d766768b : SUCCESS in 50m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc2dde3a862444fbae4bd29a850b5e3d : SUCCESS in 1h 05m 44s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d063072199dc46af99614d72011f8d40 : SUCCESS in 1h 11m 03s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b4b0cdfb86d540efa0c630d99bb66ca0 : SUCCESS in 32m 30s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5575362f7b8c4521ac76fc2cd97cab96 : SUCCESS in 59m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6fb96d974db04307aef3b3b463fb54e1 : SUCCESS in 1h 03m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/55a94d43493b42129a62a956e06fbed9 : SUCCESS in 1h 34m 05s","accounts_in_message":[],"_revision_number":1},{"id":"549eaf0a3996b5ea752f4b391f389d840897d689","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-22 16:40:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b1721f1bb2e4617e31759f4921351c3619133c38","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-22 17:59:34.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e9015853e1704c3d92a63b2a6bb5e47abf42bfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 18:19:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc8d1ec3bddc49fb9e4283f0bf21ac77\n\n- grenade https://zuul.opendev.org/t/openstack/build/89cfcb9a2b0f4abca435af127243e94a : SUCCESS in 31m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/081b93a321684b1b96eb57ce2c323fab : SUCCESS in 45m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a0b1d88f40d48978019f9c732e77f74 : SUCCESS in 3m 50s\n- devstack https://zuul.opendev.org/t/openstack/build/98e6bf56c18f42fc8d228bb44f2ed5bd : SUCCESS in 32m 40s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/d3af7560ce0e4c19ac9f618358ba7c1a : SUCCESS in 12m 40s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/2895dca8390c4ccfb03f890a37b490ff : SUCCESS in 18m 43s\n- devstack-platform-almalinux-purple-lion-ovn-source https://zuul.opendev.org/t/openstack/build/7a9c07757f754b4bb05604fea1e35995 : FAILURE in 10m 46s (non-voting)\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/2eb15e2d53d14f57b89770b9c602c548 : FAILURE in 10m 29s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c6dc9f237f641b39fb0e63ab80322c9 : FAILURE in 7m 41s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/e4dc1262d9884672bc5f4993c0451936 : SUCCESS in 1h 18m 10s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/55cd81813b3b40f6b66ae7a3ffa1f42c : SUCCESS in 1h 14m 20s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/15205d8d35cf405688c78affc13afe66 : FAILURE in 9m 54s (non-voting)\n- devstack-platform-rocky-red-quartz https://zuul.opendev.org/t/openstack/build/47dd9a5962cb45f984d3c8b1cc25ae95 : FAILURE in 9m 54s (non-voting)\n- devstack-platform-ubuntu-noble-ovn-source https://zuul.opendev.org/t/openstack/build/8000edb458cf43c083cd64ea9ba82a4e : SUCCESS in 1h 35m 24s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/40e65bba405d454ab233680c380f1bb6 : SUCCESS in 1h 31m 08s (non-voting)\n- devstack-platform-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/a66c5efed9f64fa492120f6495e1a196 : SUCCESS in 40m 20s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/335eb168915c45d2b5be9924e76f6867 : SUCCESS in 34m 42s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/435c44c1af2644b69da5a3109057cb8e : SUCCESS in 4m 18s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e60750a02b9b421cab5dd9ced29a06ec : SUCCESS in 2m 33s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/daeb4535070b46d383a29b304fd90ff3 : FAILURE in 32m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/65947a6c697145c38519af8e868aba0d : SUCCESS in 1h 02m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cbe245a3bc6643319f77f178642b4c47 : SUCCESS in 1h 09m 37s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c4a10501c7a74da784ec979e4201de28 : SUCCESS in 32m 43s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c0217d53479f473595b293a7ccf5d31b : SUCCESS in 1h 10m 48s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63f35b8389b245eeb89648aea6772271 : SUCCESS in 59m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90f58f5ce4e446139299cc7f098570bb : SUCCESS in 1h 12m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4f73ce1491f14d3bb51da92875267ca1 : SUCCESS in 1h 29m 37s","accounts_in_message":[],"_revision_number":1},{"id":"87228dea45a89032cca9866355c8c3b2d5c17b55","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-09-22 21:00:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8861b201826609263e8b13a0ae10e6f80046b8c5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-22 21:11:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1054dc70cc86d138cfa78842bce1f36c83fb016c","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-09-22 21:15:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d710323af93c5da84261e14362ffe0481954867","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-09-22 21:17:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"fd1df1390517482f3db318b8ab3ef62bfc930284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 22:53:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e80b401515d497e9c4e1e63eb9c20eb\n\n- grenade https://zuul.opendev.org/t/openstack/build/6cd67d4e09784a4bb28e3122c2324260 : SUCCESS in 41m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3737f1e3882c4968a2309ec7eb01c5c2 : SUCCESS in 1h 27m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b311effdb284e9fba60921069c0cdad : SUCCESS in 6m 52s\n- devstack https://zuul.opendev.org/t/openstack/build/93840b55ba324d7d9a272af2c8bcaf55 : SUCCESS in 25m 15s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/3e9794476bbb49dc9b2c09e3d14b6ecc : SUCCESS in 13m 06s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/4614dc6711fe45cc81a1578325c56e3e : SUCCESS in 26m 29s\n- devstack-platform-almalinux-purple-lion-ovn-source https://zuul.opendev.org/t/openstack/build/8820c6323391468aa39d15ad2dd3e5ba : FAILURE in 9m 02s (non-voting)\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/803f31793bbc4f72852ab9989e087875 : FAILURE in 8m 05s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b0585d00c9b4f6cb1370de682b54273 : FAILURE in 10m 47s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/caaa7d3cadb8431d919e66b012083dab : SUCCESS in 1h 29m 41s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/278b1b3c48a54fa3b50aa190b6f0fcbd : SUCCESS in 1h 24m 21s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/2df3ae1cf86f4a0da0496d3375116546 : FAILURE in 7m 21s (non-voting)\n- devstack-platform-rocky-red-quartz https://zuul.opendev.org/t/openstack/build/d7fdb6390e2c40b4b60bd86aa273a1e9 : FAILURE in 12m 54s (non-voting)\n- devstack-platform-ubuntu-noble-ovn-source https://zuul.opendev.org/t/openstack/build/84362c69700b46af81d20183fb8d1d29 : SUCCESS in 1h 13m 08s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/cc373862974745cabdb78a48ca428384 : SUCCESS in 1h 27m 30s (non-voting)\n- devstack-platform-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/197b0b8390db4b21b54b6ceeca4d6fcc : SUCCESS in 52m 32s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/3c4c5a65af374ee7b8f5a1f2832cbd7e : SUCCESS in 28m 43s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/c4541b2778b74b35af31e5b8d2608698 : SUCCESS in 2m 20s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c1739cd7073e4eacb285bb293fe51ede : SUCCESS in 3m 06s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/b425f406b38c47889524ed41efcf7cac : SUCCESS in 42m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/324738e611564c6c821abe7134efd5b2 : SUCCESS in 36m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/02d3533e07d540f2a6c1bad5dd7970f2 : SUCCESS in 1h 13m 47s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ccab25fed68b4729906aa08324dc5baf : SUCCESS in 43m 10s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d883e9ca23af4aa796f1acf622be591f : SUCCESS in 1h 18m 53s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/134376ca4bd7460497a17c53989a455c : SUCCESS in 56m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48235e687e784deb82ce977100e4d541 : SUCCESS in 56m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/38c711ebadb1429298c35e10ede8d96e : SUCCESS in 1h 02m 14s","accounts_in_message":[],"_revision_number":2},{"id":"b6d0b942f0e416f1adc82f194777b73bae8581b2","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-23 01:32:55.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9e3fb947a8541f4a07253655eb8f6d226860bc02","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-23 02:24:10.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9482f3242eb8b98ce4035ca9e2e555438148eb18","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-23 02:26:32.000000000","message":"Topic target-branch removed","accounts_in_message":[],"_revision_number":2},{"id":"aac8bea89b92fb06fe013ae955ad2d140929e06b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-23 02:26:40.000000000","message":"Topic set to qa-2026-2-release","accounts_in_message":[],"_revision_number":2},{"id":"e30479b06c5f80c8bd4af30eab7e9136a5bb5fae","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-23 07:26:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5865876e108479c7de5a97d80dc4689413429bda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 07:29:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d00717606a7c45fcb7d55aec8431238c347149f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 08:45:50.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b3dc2057c07475f9102c077f4e0475f\n\n- grenade https://zuul.opendev.org/t/openstack/build/12d3aac06a5147288920d932cb19e2a6 : SUCCESS in 49m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ca273043f4d4c11886876a032a96217 : SUCCESS in 57m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f41e032314934e1fbd6edfce569b15f2 : SUCCESS in 6m 30s\n- devstack https://zuul.opendev.org/t/openstack/build/10571f6ded614e668833416a05aa735d : SUCCESS in 29m 02s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/2fe17ce4fd7e47b2bdf1c9efac150f23 : RETRY_LIMIT in 1m 28s\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/01c9bcce241a44fa88e2280dd68d2ecb : FAILURE in 41m 14s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/cc086079d97644cb855ecb5066fa3072 : FAILURE in 29m 28s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/1ca51afc4c6d401cbaff53d43da65fbb : SUCCESS in 14m 06s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/cc10eed8551141b6a45c4522cf46ed2b : SUCCESS in 34m 50s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/7e49df4b9fc94aa6870e994c6170fa47 : SUCCESS in 3m 19s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/a40aa33c9fa3496cbc1595a9ed633892 : SUCCESS in 2m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/542c7ec718b84d97918b948f7c356f9b : FAILURE in 30m 41s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/940fbf319020431f9f8011a51e109ae1 : SUCCESS in 1h 08m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5ee3752eb7ca4560ac71782166642c45 : SUCCESS in 37m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2943608d93314b928d50ead31259b303 : SUCCESS in 59m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8c3a3fba70c4a94ace39cc97e670007 : SUCCESS in 41m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/68d2867d8aff4ed8a5bdb8baece4fb1b : FAILURE in 31m 59s","accounts_in_message":[],"_revision_number":2},{"id":"05c84f9c50e509df510d8d2f6fdb5a1b2717c319","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-23 14:22:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6c2ae8d291cd0026fec01dbc01c5ceedbe510517","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 16:13:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f07dbceaa2a34ec5b13c13da4a9035ff\n\n- grenade https://zuul.opendev.org/t/openstack/build/614a5ff5d9ea4ce480785f4c8c19b4e1 : SUCCESS in 1h 08m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/856f685dcc954d8dbee3eac8bfc92a86 : SUCCESS in 1h 31m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c97ffa460f4d448f222b46dfa240bb : SUCCESS in 3m 40s\n- devstack https://zuul.opendev.org/t/openstack/build/7cb593da20dd4dda9b17288e268cc05b : SUCCESS in 15m 17s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/24261f448e1c4c21973ef3ea748a2de2 : SUCCESS in 27m 20s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/be69c3d80e8143c0a748516465b75214 : SUCCESS in 25m 49s\n- devstack-platform-almalinux-purple-lion-ovn-source https://zuul.opendev.org/t/openstack/build/12d031ebec22472893e5804e80bad905 : FAILURE in 8m 57s (non-voting)\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/53972c342a8f40ebb816b289dd3f2a04 : FAILURE in 9m 10s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/88c48ada63b642dd816d1ad34eafe8b7 : FAILURE in 17m 22s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/8d3d27d03f2947518b478efc99e67016 : SUCCESS in 1h 30m 56s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/9410f047776e4e96a666649ec69d0f54 : SUCCESS in 1h 45m 10s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/1ae1bf3968e7426cabaffd6c6a9fac59 : FAILURE in 19m 50s (non-voting)\n- devstack-platform-rocky-red-quartz https://zuul.opendev.org/t/openstack/build/7bb52e2a0fef49f493ba8fc5a93b3583 : FAILURE in 9m 51s (non-voting)\n- devstack-platform-ubuntu-noble-ovn-source https://zuul.opendev.org/t/openstack/build/3e04b3d9dea444a69b61802d2bec476f : SUCCESS in 1h 43m 44s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/82379a8c381b43f8b922315d52e2219a : SUCCESS in 1h 03m 01s (non-voting)\n- devstack-platform-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/2e54f0bc7eb549a29ef1514cb9db45c7 : SUCCESS in 56m 10s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/a4347566b6ab4950ae4d63e0a07fea67 : SUCCESS in 35m 59s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/1fe9a4b480704c479ee42d600936b321 : SUCCESS in 3m 21s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/d440142477cf4d79acdfe6e6e76d390d : SUCCESS in 2m 14s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/4caae64da5b1442e84e7cbc5689601a5 : SUCCESS in 1h 07m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/83ff3b19e4cb45538dee51064cbe63a5 : SUCCESS in 34m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/797f4d81c4ce4a17ba6c5900e433791f : SUCCESS in 1h 14m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f9d88603cc094559b05d0595c1d6844f : SUCCESS in 1h 14m 30s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8299f0d55a3d43aab2fff74ea0ed7545 : SUCCESS in 1h 10m 57s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d0c53fcf0b34a768d33b96b47a5726b : SUCCESS in 1h 00m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c582fd71829a422898f652439f99f30f : SUCCESS in 59m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8fbb873f03b24d989bc8319266bbb788 : SUCCESS in 1h 26m 06s","accounts_in_message":[],"_revision_number":2},{"id":"fc68d3343baf11b495f160dc5eb2b4ce0119acd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 16:17:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e4e09bf48065ee5034487170dd68c835de94b7db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 19:24:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34b828d853b24fc69caaf75c961bb95b\n\n- grenade https://zuul.opendev.org/t/openstack/build/5a37437846664ca3bfa123e9b18a0211 : SUCCESS in 1h 00m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a931883e215749bca84f6c1f639c62b2 : SUCCESS in 1h 32m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a28e6a41a9824c1b86423b731750fb92 : SUCCESS in 6m 41s\n- devstack https://zuul.opendev.org/t/openstack/build/cc3cc2e3640546b088c342af67e90b62 : SUCCESS in 27m 26s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/8adeea1442da40b3aa25f02f056232d5 : SUCCESS in 27m 10s\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/d2dca1b5bc29407ebf3782c47f6bff7e : SUCCESS in 1h 04m 16s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/bf3a09bb3f76406087fbc7f108890708 : SUCCESS in 1h 23m 22s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/a37a647352cc4b93a15f7b15741d971d : SUCCESS in 13m 47s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/4b647f7ad4194e35b0268312890bb056 : SUCCESS in 18m 29s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/9a590787ef1a4234b74a6d421970c8a2 : SUCCESS in 3m 44s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f36f9b44b516456ca269e196f1f5bbef : SUCCESS in 4m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/018d2ba881f14fdda1f9d304cbcbc450 : SUCCESS in 40m 45s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/fbf561d184c3406aa80954dd466cb312 : SUCCESS in 1h 10m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d20e996b6ffe4cf3831fdd02a5edfe21 : SUCCESS in 25m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e6500c6bf0d47a2a610706cd79f1e32 : SUCCESS in 1h 02m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa94e5bd27244f38950b7f2012195e14 : SUCCESS in 57m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/77a76ce5e66243be9c6f41821fc0b66d : SUCCESS in 1h 46m 29s","accounts_in_message":[],"_revision_number":2},{"id":"44fd51357aaee045d901b0ddba98824a4c2312b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 19:24:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9533e536bdd2279bcc97bb2ca3e03311746d999a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 19:25:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25b97600144d45d3bc8e263194b7f020\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d83c0561946a4402a70a3681ef5a4bc9 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bc469c043cdd68d4099f33159ded4006e92256c5","revisions":{"e9d14e9eab0ec846219324485b4a9e6ebea94486":{"kind":"REWORK","_number":1,"created":"2026-09-22 13:36:34.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/21/1006721/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/21/1006721/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/21/1006721/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/21/1006721/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/21/1006721/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/21/1006721/1"}}},"commit":{"parents":[{"commit":"8078f5870841065080a3282b1fa59b4f3c2cf2fb","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/8078f5870841065080a3282b1fa59b4f3c2cf2fb"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-09-22 13:26:43.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-09-22 13:26:43.000000000","tz":120},"subject":"Update devstack stable/2026.2 branch setting","message":"Update devstack stable/2026.2 branch setting\n\nChange-Id: I600ab2603d9a240f9f1825f15df1384cb0ccc501\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/e9d14e9eab0ec846219324485b4a9e6ebea94486"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/e9d14e9eab0ec846219324485b4a9e6ebea94486"}]},"branch":"refs/heads/stable/2026.2"},"bc469c043cdd68d4099f33159ded4006e92256c5":{"kind":"REWORK","_number":2,"created":"2026-09-22 21:17:21.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/21/1006721/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/21/1006721/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/21/1006721/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/21/1006721/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/21/1006721/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/21/1006721/2"}}},"commit":{"parents":[{"commit":"8078f5870841065080a3282b1fa59b4f3c2cf2fb","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/8078f5870841065080a3282b1fa59b4f3c2cf2fb"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-09-22 13:26:43.000000000","tz":120},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2026-09-22 21:17:20.000000000","tz":-420},"subject":"Update devstack stable/2026.2 branch setting","message":"Update devstack stable/2026.2 branch setting\n\nNOTE(JayF): Modified patchset 2 to add \"stable/\" qualifier,\nwhich matches stable/2026.1 equivalent change and should fix\nIronic CI jobs.\n\nChange-Id: I600ab2603d9a240f9f1825f15df1384cb0ccc501\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: Jay Faulkner \u003cjay@jvf.cc\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/bc469c043cdd68d4099f33159ded4006e92256c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/bc469c043cdd68d4099f33159ded4006e92256c5"}]},"branch":"refs/heads/stable/2026.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
