)]}'
{"id":"openstack%2Fdevstack~1006722","triplet_id":"openstack%2Fdevstack~stable%2F2026.2~I6f778e58e76187393096611c18b6a5cdde1d3224","project":"openstack/devstack","branch":"stable/2026.2","topic":"qa-2026-2-release","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-09-23 14:22:40.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I6f778e58e76187393096611c18b6a5cdde1d3224","subject":"Stop installing Tempest at system wide for stable branch","status":"MERGED","created":"2026-09-22 13:36:34.000000000","updated":"2026-09-23 19:24:21.000000000","submitted":"2026-09-23 19:24:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006722-qa-2026-2-release","meta_rev_id":"6707c15b8e902d4aacb56eae68c5683d373f65ec","_number":1006722,"virtual_id_number":1006722,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-23 19:24:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":2,"date":"2026-09-23 02:24:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2026-09-23 07:27:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2026-09-23 07:27:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2026-09-23 02:24:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-22 15:28:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-23 02:24:32.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-09-23 07:26:46.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2026-09-23 07:27:07.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"c58db47e3b44b36c4f0ed0cebd8036d8cc996b4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-22 13:36:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afdf121b2a4febe6f0d8c033036c8a523f766c27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 15:28:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bfef228f3674b659a324539a6453e5a\n\n- grenade https://zuul.opendev.org/t/openstack/build/c6da823c39ec475a8f8d3e3753739910 : SUCCESS in 1h 10m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69ea4ecab22f43d8a07fdf2dcac5928c : SUCCESS in 1h 24m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e807807c0b1a4db78c8ed0e011b5b77d : SUCCESS in 6m 11s\n- devstack https://zuul.opendev.org/t/openstack/build/d4f5e77ae80548509aef465d0e94b0b0 : SUCCESS in 29m 22s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/20fe02b15c284db6a2d4bfa87f6dcf8a : SUCCESS in 18m 50s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/30992c28df154f7e974db8aef6fd4743 : SUCCESS in 20m 22s\n- devstack-platform-almalinux-purple-lion-ovn-source https://zuul.opendev.org/t/openstack/build/bb54859cefbe4c2382a6a95d1ec8edf1 : FAILURE in 7m 19s (non-voting)\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/4b98c2035e264485955c52d5b4334d0f : FAILURE in 7m 26s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a4b83c7f440479d9959fb8b58e4ab74 : FAILURE in 7m 19s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/86f7b46730024bf79760fc54645d49ef : SUCCESS in 1h 10m 00s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/b16de37afe2a4a61ab5201e7ea4381dd : SUCCESS in 1h 02m 58s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/f22537c2e57c49b8b804cc9bc10e19fb : FAILURE in 20m 35s (non-voting)\n- devstack-platform-rocky-red-quartz https://zuul.opendev.org/t/openstack/build/aacd9c50e06940b49773d58e7d824978 : FAILURE in 9m 25s (non-voting)\n- devstack-platform-ubuntu-noble-ovn-source https://zuul.opendev.org/t/openstack/build/f0e9d59cc232427d8de1a6c91dc54fd2 : SUCCESS in 1h 44m 26s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/d5c5da2f2b4f4ba6b23e96f5d1693f48 : SUCCESS in 1h 39m 01s (non-voting)\n- devstack-platform-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/0afce6de89c04380992511969a348679 : FAILURE in 55m 22s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/5863b7a2167348149f9a7a67bc1e7f10 : SUCCESS in 39m 43s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/d854a2d9d5ae4d32a3207f179f77f67c : SUCCESS in 3m 24s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/d642ebef2a5f484aa3690091c6147a60 : SUCCESS in 2m 47s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/86710164e1ef4df19fef821f024ac8b9 : FAILURE in 33m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5245d0fd328a4a7ca7225e601361c4ba : SUCCESS in 36m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7d780a2313d4d409d4994dc3bfdc2de : SUCCESS in 1h 00m 21s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/eccc9fc0ee8e4aff813c616e7a99b0dd : SUCCESS in 1h 12m 42s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3d2d173b69494891b855ed5b5dc16309 : SUCCESS in 58m 46s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e6fd3a719064992907bbc9654f1b4d4 : SUCCESS in 1h 03m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad6d4494b16a401cb3df68085ec7ff24 : SUCCESS in 51m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cb8352111a4b4088a0bdea86ae64e6a6 : SUCCESS in 1h 23m 12s","accounts_in_message":[],"_revision_number":1},{"id":"ed29d9c42d4e051cca8957b25f7510cd5537a848","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"real_author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-23 02:24:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_13252\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"_revision_number":2},{"id":"6b016b8644a7ec9c209f48b18ba885e3031ac6f3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-23 02:24:32.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"589cd1d042615e893bcd4f9b28363b54d4894aac","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-23 02:27:39.000000000","message":"Topic target-branch removed","accounts_in_message":[],"_revision_number":2},{"id":"dd3082e6d2ccfb9b1a584dbf6dbe6025945bb5ea","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-23 02:27:50.000000000","message":"Topic set to qa-2026-2-release","accounts_in_message":[],"_revision_number":2},{"id":"1a6d20b3c8a373487aefc2c9a983ea0a577161c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 05:36:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb501722fe2e404390cb755f5da987fa\n\n- grenade https://zuul.opendev.org/t/openstack/build/1cb36b012ba442bf9ef0392955912646 : SUCCESS in 1h 03m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/61b8db632dc84ca797cd463e9f52619c : SUCCESS in 1h 35m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f00b5a11dff411494abe39472438979 : SUCCESS in 9m 34s\n- devstack https://zuul.opendev.org/t/openstack/build/973988ba093e49499f57af74073514b0 : SUCCESS in 26m 56s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/49f80caec876437d9b63fdd242d95287 : SUCCESS in 24m 25s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/e812bc6bb225422dacd233a88421b647 : RETRY_LIMIT in 3m 04s\n- devstack-platform-almalinux-purple-lion-ovn-source https://zuul.opendev.org/t/openstack/build/a731014bf34b410296f8b6b3699b12c1 : FAILURE in 11m 49s (non-voting)\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/6ae677c8591f4b57a6d335c6cc843128 : FAILURE in 7m 33s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/859b03d5a6084eb592c19ed7d0f71162 : FAILURE in 16m 37s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/e19a8347153f4eb0842d55e1c36984dd : SUCCESS in 1h 45m 09s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/c014614053be4a2b9ef3cc6990a214c6 : SUCCESS in 1h 27m 59s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/622aa49b004d45ab80363d1052688502 : FAILURE in 7m 53s (non-voting)\n- devstack-platform-rocky-red-quartz https://zuul.opendev.org/t/openstack/build/d84a1ddee9de41d9a948c52786725587 : FAILURE in 8m 58s (non-voting)\n- devstack-platform-ubuntu-noble-ovn-source https://zuul.opendev.org/t/openstack/build/4d013a6bf69449e89dbba45c763d6bde : SUCCESS in 1h 44m 06s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/4a9d6f07c19b417b970f578fbe89c58c : SUCCESS in 1h 25m 31s (non-voting)\n- devstack-platform-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/041fcc98c78c499ea09b0f1d6bb93a76 : SUCCESS in 1h 22m 26s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/f5741c9c9da94254915ae7a38001913f : SUCCESS in 35m 27s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/e8ac8846b9da49fd88538278508ee5f9 : SUCCESS in 3m 04s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/27e231967a6441259df2df6765902444 : SUCCESS in 3m 03s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/7479ec05bb4d4098bb04f2488f82cb0e : SUCCESS in 1h 05m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/99386f71a9714830881abddefbc42fb5 : SUCCESS in 54m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/34c6f45efcb9473b85980eac8b1f23ab : SUCCESS in 1h 07m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8ae160619777425185fc7974d049e64e : SUCCESS in 37m 25s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/749ff381b8eb4ef8b42db5b0adf0c840 : SUCCESS in 1h 03m 47s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed589245108143d49f8f290f598bbeff : SUCCESS in 1h 07m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/054e7f7ee74849c8a981d2cde071e3d1 : SUCCESS in 53m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e5dfb4ae266243df9bd772523750d9ab : SUCCESS in 1h 23m 39s","accounts_in_message":[],"_revision_number":2},{"id":"5add7835df7c768e23e7b847ab60ee44ff207663","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-23 07:26:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"76c6c80a00659d72fc7948600e8fb5caf3be8b89","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-23 07:27:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"01491287b446014606fbbfc9d5571aacbd5db8fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 09:19:25.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/04b6782e5c6e407195152ce49b064e6a\n\n- grenade https://zuul.opendev.org/t/openstack/build/91782df711de4eb8b7f1c76a712859a8 : SUCCESS in 1h 04m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/800311ff3d8b4c629d9c1cc34e68bdb3 : SUCCESS in 1h 30m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c729e95fba234f6396119ae7903c142b : SUCCESS in 7m 46s\n- devstack https://zuul.opendev.org/t/openstack/build/221d3a5a8d8b4e56a1da7618e14a7c44 : RETRY_LIMIT in 1m 47s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/169255d51a834986ba2dfc4fab7bcb6d : SUCCESS in 27m 33s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/62a8f4941cc94743b2698ab6c207e722 : SUCCESS in 26m 20s\n- devstack-platform-almalinux-purple-lion-ovn-source https://zuul.opendev.org/t/openstack/build/9e1efcab8e634919bc89f920921de984 : FAILURE in 9m 48s (non-voting)\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/57c75575a4f6470a9fe956376b24986b : FAILURE in 6m 30s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/97c7a3955704402395c98afbbd1e9b46 : FAILURE in 7m 00s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/dde36dc6769c43a29f0b4996c078bc90 : RETRY_LIMIT in 1m 42s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/44c248edcc594a9d8012ec424e23cfeb : RETRY_LIMIT in 1m 30s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/1b18c0b628b84035ae847441f62f8f92 : FAILURE in 6m 19s (non-voting)\n- devstack-platform-rocky-red-quartz https://zuul.opendev.org/t/openstack/build/73ee928ee12f4e848709be10173597be : FAILURE in 8m 38s (non-voting)\n- devstack-platform-ubuntu-noble-ovn-source https://zuul.opendev.org/t/openstack/build/b7310649f71d4fcda2c27637df5fec7c : SUCCESS in 58m 48s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/90f3c1870c1a43ffa8befa6d06a905b4 : SUCCESS in 50m 16s (non-voting)\n- devstack-platform-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/93a71554809c42c28cb260bc1fcb7146 : SUCCESS in 1h 31m 26s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/76db3801adfd488b928ae370468a38b0 : SUCCESS in 33m 28s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/a96b60dc481b450da5db9f2451d6a30f : SUCCESS in 3m 16s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/64ee8e2c2c1546d88b5dcb65aaefc074 : SUCCESS in 2m 12s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/23edf6cbd7aa45a5a51b5ff77f758a43 : SUCCESS in 37m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/74736412ad66446a9b93e2265e5aac55 : SUCCESS in 33m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/349729d04d85470c87a4bbc0fe32bce1 : SUCCESS in 1h 10m 02s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5074570bd03e4f67b5eb27fa7250b844 : SUCCESS in 39m 20s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3360950064fb4bf9b9b0eeab2d3e9dd9 : SUCCESS in 1h 00m 10s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0faab72ecf9414e976714fe9b4cfea3 : SUCCESS in 56m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5de3fe3304774e978aa5365faab9a7c9 : SUCCESS in 32m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d49c83100a5b4e51b36f00618c0657eb : SUCCESS in 58m 55s","accounts_in_message":[],"_revision_number":2},{"id":"26aa6d84a8e7a5d060cc34c0a20b64b8c7164c0e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-23 14:22:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6c1ce8823a711d6dc6086c533594aaf4f9451835","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 16:09:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8e778406a464adea483951bdfe95bc6\n\n- grenade https://zuul.opendev.org/t/openstack/build/ee43e3e07e3f496d8ec155b88ac3fb85 : SUCCESS in 34m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8b5862cac83472d8f5f4054fcc3b78d : SUCCESS in 1h 34m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/891731ea4a1d49e69de9adb6f449408a : SUCCESS in 6m 46s\n- devstack https://zuul.opendev.org/t/openstack/build/322dd1c6237f448ab2f12522054fd393 : SUCCESS in 17m 43s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/f349c7e338354d2a8e124b87dd4e0755 : SUCCESS in 25m 04s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/d18444a0801e46adb7ceb4282fb92f0a : SUCCESS in 12m 07s\n- devstack-platform-almalinux-purple-lion-ovn-source https://zuul.opendev.org/t/openstack/build/ddc722ee1dd84459bd139d575afd8f14 : FAILURE in 7m 54s (non-voting)\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/61b39146061a400d89a33b3eb7c91153 : FAILURE in 10m 18s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c5dc136ffda4c6cba5effac9aca4e8e : FAILURE in 15m 29s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/fc3f86d92df84319aaac6c5abf3b0c9b : SUCCESS in 1h 25m 23s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/703d31d326504b179815418cd5d101d1 : SUCCESS in 1h 36m 23s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/5b63058a090547cd8520b2be3867a616 : FAILURE in 6m 10s (non-voting)\n- devstack-platform-rocky-red-quartz https://zuul.opendev.org/t/openstack/build/d07ba008cb4a4e8ca9b5d932ca301fad : FAILURE in 6m 54s (non-voting)\n- devstack-platform-ubuntu-noble-ovn-source https://zuul.opendev.org/t/openstack/build/15b1327bf01b4cfdb4fe99555b671853 : SUCCESS in 1h 33m 21s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/e4beea9c527648dca00e7da71400ae6e : SUCCESS in 1h 21m 54s (non-voting)\n- devstack-platform-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/ea870c328cae4b6595a7331ce35a18fd : SUCCESS in 1h 38m 45s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/448c8dce849845e0ab7674db7b25e3d4 : SUCCESS in 37m 25s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/49a40e71790d477db3a12f4e18e9db0c : SUCCESS in 2m 04s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/4ab006acf1624162939bc29935d2ec40 : SUCCESS in 2m 56s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/0b6f3cb9075840098842828335115634 : SUCCESS in 1h 06m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bb52fb2fec8a4cabb19aed1a367dc3e9 : SUCCESS in 57m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ec20f5c6a6542dba76edf2666aae1da : SUCCESS in 1h 04m 54s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0ec764b2682f4c2891b6cf557970a1af : SUCCESS in 47m 50s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e7d1e23440a64669a11445c1d8dbd33a : SUCCESS in 1h 09m 57s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3ef03b370c1424e8e8cc90aa3b92d82 : SUCCESS in 1h 01m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/097accbed9eb4ae8a7ac3bb20940e7ea : SUCCESS in 27m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/50f09df8c710446c8dda97bcfcfc8697 : SUCCESS in 1h 33m 11s","accounts_in_message":[],"_revision_number":2},{"id":"ca8e8909045a468a2e0c8a15983aa165df285887","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 16:15:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8ca52f29ed165d066b146f98cfccc5d6447ac34a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 19:24:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4015023d0d534c34b82d85c820cd4afb\n\n- grenade https://zuul.opendev.org/t/openstack/build/303fd05fda2443699a46dd326c5772d2 : SUCCESS in 39m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e912a061366c45ef9f01916a722abd4a : SUCCESS in 55m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6452f0fc4a6a474ab769687c9e5fbe43 : SUCCESS in 8m 05s\n- devstack https://zuul.opendev.org/t/openstack/build/de319a6a1dc74d90b185f168735c4158 : SUCCESS in 23m 29s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/81c9261549eb44cebff207f1c5698813 : SUCCESS in 15m 21s\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/0dcb971a38824179a87ea4117f62acb7 : SUCCESS in 1h 20m 55s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/7a68147c52f44a79a2bcad92bac85b27 : SUCCESS in 1h 40m 42s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/c404eb9e75e24b7682186f7c25ea5d2c : SUCCESS in 18m 11s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/7ac10e58af0740b390cc5c744b606576 : SUCCESS in 35m 43s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/3521a6044f4a443ebd1fa4e404505ab8 : SUCCESS in 2m 50s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/31e1627fd93f4c829b5c948d2b15310e : SUCCESS in 3m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/048e92c24466498580573405375d6e89 : SUCCESS in 1h 08m 59s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/f1912fbb33fb4ac4888e0d4d52388ed6 : SUCCESS in 37m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6164cc9a0f2a4adbbf2fec7493e9f2f5 : SUCCESS in 37m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4ef82fc198b409dbd2eb449e71e88e3 : SUCCESS in 55m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06872a18089b4d3b90f984b1cc04cf8e : SUCCESS in 56m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e7ea0d45236548bd8fb01dcfccf5f0c8 : SUCCESS in 1h 38m 17s","accounts_in_message":[],"_revision_number":2},{"id":"6707c15b8e902d4aacb56eae68c5683d373f65ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 19:24:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d103bdfe4b38cdb9c85046aa361bb78e4a49770f","revisions":{"8ff3f5d8858fb61eca43c76d6e2f527a271cc281":{"kind":"REWORK","_number":1,"created":"2026-09-22 13:36:34.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/22/1006722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/22/1006722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/22/1006722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/22/1006722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/22/1006722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/22/1006722/1"}}},"commit":{"parents":[{"commit":"e9d14e9eab0ec846219324485b4a9e6ebea94486","subject":"Update devstack stable/2026.2 branch setting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/e9d14e9eab0ec846219324485b4a9e6ebea94486"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-09-22 13:27:49.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-09-22 13:28:09.000000000","tz":120},"subject":"Stop installing Tempest at system wide for stable branch","message":"Stop installing Tempest at system wide for stable branch\n\nAs added in notes for INSTALL_TEMPEST variable we need to set\nthis as False for stable branch so that devstack does not\ninstall Tempest at system wide.\n\nChange-Id: I6f778e58e76187393096611c18b6a5cdde1d3224\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/8ff3f5d8858fb61eca43c76d6e2f527a271cc281"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/8ff3f5d8858fb61eca43c76d6e2f527a271cc281"}]},"branch":"refs/heads/stable/2026.2"},"d103bdfe4b38cdb9c85046aa361bb78e4a49770f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-23 02:24:23.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"real_uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/22/1006722/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/22/1006722/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/22/1006722/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/22/1006722/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/22/1006722/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/22/1006722/2"}}},"commit":{"parents":[{"commit":"bc469c043cdd68d4099f33159ded4006e92256c5","subject":"Update devstack stable/2026.2 branch setting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/bc469c043cdd68d4099f33159ded4006e92256c5"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-09-22 13:27:49.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-09-23 02:24:23.000000000","tz":0},"subject":"Stop installing Tempest at system wide for stable branch","message":"Stop installing Tempest at system wide for stable branch\n\nAs added in notes for INSTALL_TEMPEST variable we need to set\nthis as False for stable branch so that devstack does not\ninstall Tempest at system wide.\n\nChange-Id: I6f778e58e76187393096611c18b6a5cdde1d3224\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/d103bdfe4b38cdb9c85046aa361bb78e4a49770f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/d103bdfe4b38cdb9c85046aa361bb78e4a49770f"}]},"branch":"refs/heads/stable/2026.2","description":"Rebase","conflicts":{"base":"e9d14e9eab0ec846219324485b4a9e6ebea94486","ours":"8ff3f5d8858fb61eca43c76d6e2f527a271cc281","theirs":"bc469c043cdd68d4099f33159ded4006e92256c5","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
