)]}'
{"id":"openstack%2Fdevstack~1006880","triplet_id":"openstack%2Fdevstack~master~I212bdf230cee425077668e8e0320c5d2b2dab377","project":"openstack/devstack","branch":"master","topic":"qa-2026-2-release","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-09-30 19:35:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I212bdf230cee425077668e8e0320c5d2b2dab377","subject":"Update master DEVSTACK_SERIES to 2027.1","status":"MERGED","created":"2026-09-23 02:42:19.000000000","updated":"2026-09-30 19:36:23.000000000","submitted":"2026-09-30 19:35:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006880-qa-2026-2-release","meta_rev_id":"9d377c8994442906292f83713f8c7ee69f224b6e","_number":1006880,"virtual_id_number":1006880,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-30 19:35:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2026-09-23 15:20:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2026-09-30 17:58:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-09-23 03:39:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2026-09-30 17:58:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2026-09-30 17:58:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-23 03:39:54.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-09-23 05:16:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-23 15:20:00.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2026-09-30 17:58:32.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"9dd4796f433ae14eb66078895da663290e84c2c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-23 02:42:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d4ebd6991faf85fc78c8ac818f87e9a350ec969","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-23 03:39:54.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/b5109f413570409496462156319fbd72\n\n- devstack-platform-centos-10-stream https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/37ac741bb2154ec2b17a128d3e870817 : SUCCESS in 56m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b041b7c5175d361d0df474cafb20675ab8219125","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 05:16:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ab3ccfc219b43e294320d840e558be8\n\n- grenade https://zuul.opendev.org/t/openstack/build/1dcbe8f1242f42f087c64ff5086557ab : SUCCESS in 1h 05m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a809d463696430ea926576be60910c4 : SUCCESS in 1h 21m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f550302ffb2e488ba09e8aeec4db5cc0 : SUCCESS in 7m 50s\n- devstack https://zuul.opendev.org/t/openstack/build/f2875df3a0cf487bbb034bff92a82f41 : SUCCESS in 29m 33s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/3a1d503d4ac349359b45c6550e32395b : SUCCESS in 27m 56s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/a1c31f7fcfc7471dbe98f483e53a1655 : SUCCESS in 29m 17s\n- devstack-platform-almalinux-purple-lion-ovn-source https://zuul.opendev.org/t/openstack/build/ba199535bd4548e1885c16187e0917d9 : FAILURE in 30m 53s (non-voting)\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/2298f62fda48454ebb6b65f57348d86a : SUCCESS in 48m 09s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/24bee5fd1da24d46a92422d7e5648885 : SUCCESS in 1h 51m 11s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/381d9baa2373476aa593280073c021bb : RETRY_LIMIT in 2m 22s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/22f16a03b3cd4f6f8189435ce5c8e8ee : SUCCESS in 1h 26m 34s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/cb15b0d07cd04b889dee11d47fa964da : FAILURE in 1h 03m 05s (non-voting)\n- devstack-platform-rocky-red-quartz https://zuul.opendev.org/t/openstack/build/31599f51665a4a439337caa34d60331b : FAILURE in 15m 33s (non-voting)\n- devstack-platform-ubuntu-noble-ovn-source https://zuul.opendev.org/t/openstack/build/7fadc34c707240a98a0f3c3e0f747f2b : RETRY_LIMIT in 2m 23s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/e1a388fe60e242e89ee7d1ce756f4028 : SUCCESS in 49m 03s (non-voting)\n- devstack-platform-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f4d5a685be134c239ae16c42a5e88813 : SUCCESS in 1h 00m 48s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/36521bde10854f7f9f41f48b66515f04 : SUCCESS in 38m 16s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/d9db9dcb5a62425aa95dd8100420f2fa : SUCCESS in 4m 23s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f836b5b70fb1481080a1ee4b39427bfe : RETRY_LIMIT in 1m 25s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/949dc70f4805484fb44af630aafa1462 : SUCCESS in 39m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/64572e6e6e934b18b9b833c216965ddc : SUCCESS in 56m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/095478fa51764a65bb2292c6b15f8daf : SUCCESS in 1h 22m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/130e44535f574ee99c7fa493d72c146e : SUCCESS in 1h 08m 51s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e6a3aa624a5a47019bac8557c353c35e : SUCCESS in 1h 25m 17s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8fd5896b8bdc4aa6a02b7c790ad90c89 : SUCCESS in 1h 01m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d9c9d21e5514d1e83bbb56f185c8511 : SUCCESS in 59m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1289b4099d53454dbe40001efa7b738c : SUCCESS in 1h 05m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d807142f722eab61b8879102ba45a1b5530d081c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-09-23 15:20:00.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"716adf7ac7d675e294028bf862c74c7b62b16cb5","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-23 16:10:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/a42a968e39ae4e108fd945c16dcc410b\n\n- devstack-platform-centos-10-stream https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/f1573bddcc68452d87ac1705ee53b127 : SUCCESS in 49m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"797fe5a1b675e0e20c546ae2f7889634e8c78265","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 17:04:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b612ad615c04b84ba81a6ebfb58aea2\n\n- grenade https://zuul.opendev.org/t/openstack/build/35cf2132fed7458aa026b11782410763 : SUCCESS in 53m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4eb767cb74bd4203a525cbabed5fc7bb : SUCCESS in 1h 28m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a66dc426679a4f0181ddb3165a9d4130 : SUCCESS in 6m 11s\n- devstack https://zuul.opendev.org/t/openstack/build/ae0ac17eaef64d3faf077884aeb306d7 : SUCCESS in 26m 44s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/a5ebbb04d0b74441b1fc559f6628e018 : SUCCESS in 19m 36s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/64cd6f87b86443008181c73b68f40a0e : SUCCESS in 16m 39s\n- devstack-platform-almalinux-purple-lion-ovn-source https://zuul.opendev.org/t/openstack/build/157a8335f6734e01be21b40fdd151d29 : FAILURE in 22m 22s (non-voting)\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/09a52107cf5245a09ce81d0cef5f8a1e : SUCCESS in 43m 52s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/d91c0aa0afe642059dba84c78d9d17aa : SUCCESS in 1h 33m 59s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/ca663455c6f244b4a021f6a3ce88077a : SUCCESS in 1h 31m 13s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/4b594c73f0274005a1607325d20edc3d : SUCCESS in 1h 29m 23s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/f1c879a20bba4e85a6fea369c0bd307d : SUCCESS in 1h 36m 07s (non-voting)\n- devstack-platform-rocky-red-quartz https://zuul.opendev.org/t/openstack/build/bc9cdd0b6e0340569774102e2ed889ee : FAILURE in 22m 22s (non-voting)\n- devstack-platform-ubuntu-noble-ovn-source https://zuul.opendev.org/t/openstack/build/dfa62003b8bd420cb320209324a5cb54 : SUCCESS in 1h 23m 40s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/d3a94e8e0d634907955b1a9badf67de6 : SUCCESS in 1h 35m 12s (non-voting)\n- devstack-platform-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/03c0c2d9cf9a40da962a555e3eadf88b : SUCCESS in 1h 36m 29s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/109a75b19b554fc488a682a3fbbb5c99 : SUCCESS in 29m 42s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/e543705a4c3842c8abd66f4067c982d0 : SUCCESS in 3m 09s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/851bad6579d64a2e8a2eb015ef6d644b : SUCCESS in 2m 51s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/6ec3e8edf40d4aa29bd0955d0fa78bef : SUCCESS in 1h 04m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3144d655fd664586bab56a808c689300 : SUCCESS in 28m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d6d0767694cf4a1e99e01bcd3ca8d610 : SUCCESS in 1h 10m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/eb8045b434d24525a9dec01733af06c7 : SUCCESS in 58m 15s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/da729f815b1947c5aa147d32465a0fc1 : SUCCESS in 1h 18m 08s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d9fee51ffa447f0bc50363340914b4b : SUCCESS in 1h 06m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b44b7ece79f4d788e90f7e5c640c789 : SUCCESS in 1h 00m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/89a78784fbf74e0aa0c2aa84c73176f8 : SUCCESS in 1h 33m 50s","accounts_in_message":[],"_revision_number":1},{"id":"091a3c6ed4052be54d2234df0b601b874dc2f564","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-30 17:58:32.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4fadf0b2da3b2eba01055a3c080cf269b22c49d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 17:59:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f92fb89279e7b21173a923e5b254d018543b9aa1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 19:35:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f77a0f9b02144c095863cc3c8b80ade\n\n- grenade https://zuul.opendev.org/t/openstack/build/6827be425ee643958536963543e94242 : SUCCESS in 40m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ab3332c54444aeabcf2b322948c6a8e : SUCCESS in 42m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74dfbe8b45534c079b2a62fdd953dd10 : SUCCESS in 5m 52s\n- devstack https://zuul.opendev.org/t/openstack/build/75b3238d4e6f4c6a812eb316124d5fd7 : SUCCESS in 15m 36s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/79cf00c16d904e28a49145a1c7227365 : SUCCESS in 11m 56s\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/8527d91fee2046baa5d8b7aa14a71cd4 : SUCCESS in 53m 16s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/e5623dba50e14ccc93c6c0521470fb62 : SUCCESS in 1h 28m 30s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/3e7c6db6ac9c4e82852b234b807fd8df : SUCCESS in 27m 57s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/829b109085ee4df7a0eecb4335a5889b : SUCCESS in 23m 46s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/a9d2130851084812826bfdd9205b22ee : SUCCESS in 2m 37s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/90a9ab927e7f47c0854467a4b93be759 : SUCCESS in 2m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/20112557aee24d9390adc8c247cbcde7 : SUCCESS in 1h 12m 10s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/f209e17c1c184a23812f24d8b4a3819c : SUCCESS in 1h 01m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7bb1584da415451cbc55ee73a652baed : SUCCESS in 33m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae76bb2d9f2f4808ad61cede58ddbd8b : SUCCESS in 28m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1ef335d4861481385d0f1783142cfe0 : SUCCESS in 1h 01m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d473fde630814aa8bd53a6eb426685e4 : SUCCESS in 1h 31m 43s","accounts_in_message":[],"_revision_number":1},{"id":"edb8f37fc0e05de4e03ba66e3504be2aab44c3af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 19:35:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9d377c8994442906292f83713f8c7ee69f224b6e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 19:36:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5aca6c63266f440b9dbfc042588e5a9e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/228fe9d8e2e84f61a8d8d714c6f3189a : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d4067f38189769056e4a00e770ac8c2d9bced891","revisions":{"d4067f38189769056e4a00e770ac8c2d9bced891":{"kind":"REWORK","_number":1,"created":"2026-09-23 02:42:19.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/80/1006880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/80/1006880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/80/1006880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/80/1006880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/80/1006880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/80/1006880/1"}}},"commit":{"parents":[{"commit":"9b0a00865c2537338f72ed5b9b44a0d6bff16eb4","subject":"ovs: Fix OVS path mismatch when building from source with OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/9b0a00865c2537338f72ed5b9b44a0d6bff16eb4"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-23 02:41:02.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-23 02:41:02.000000000","tz":0},"subject":"Update master DEVSTACK_SERIES to 2027.1","message":"Update master DEVSTACK_SERIES to 2027.1\n\nstable/2026.2 branch has been created now and\ncurrent master is 2027.1.\n\nChange-Id: I212bdf230cee425077668e8e0320c5d2b2dab377\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/d4067f38189769056e4a00e770ac8c2d9bced891"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/d4067f38189769056e4a00e770ac8c2d9bced891"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
