)]}'
{"id":"openstack%2Fdevstack~1006883","triplet_id":"openstack%2Fdevstack~stable%2F2026.2~I634b3bd63b58d7ed315592ee523e458c4eacc6f2","project":"openstack/devstack","branch":"stable/2026.2","topic":"qa-2026-2-release","attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-09-24 13:25:21.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I634b3bd63b58d7ed315592ee523e458c4eacc6f2","subject":"Cap stable/2026.2 network, swift, volume api_extensions for tempest","status":"NEW","created":"2026-09-23 02:56:16.000000000","updated":"2026-09-24 13:25:21.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f53e8a4593661568c6b469062ca97f2d3c1dfaff","_number":1006883,"virtual_id_number":1006883,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-23 05:23:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":2,"date":"2026-09-24 13:25:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-23 05:23:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-24 13:25:21.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"d978482ba21fae5a4597fc5a89c4df8e3f0c8928","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-23 02:56:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e57134d9cb50fa8ef42239faaa762a318b6f442a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 05:23:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc146ee5342a46b3943b31d2d315970e\n\n- grenade https://zuul.opendev.org/t/openstack/build/dd44d5d410ea4a8db641507ec31afab9 : SUCCESS in 1h 11m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4bd5874a4d648639dbd765276244b91 : SUCCESS in 1h 28m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a47503749a4e44a5a085a5b92228570e : SUCCESS in 7m 51s\n- devstack https://zuul.opendev.org/t/openstack/build/9258ff0c74f54281a8f6f068b57c961c : SUCCESS in 29m 19s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/ab0251bfd8324730974cf917a2894dc9 : SUCCESS in 19m 40s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/79e556b272d345949b8e52d46c4ee26f : SUCCESS in 16m 52s\n- devstack-platform-almalinux-purple-lion-ovn-source https://zuul.opendev.org/t/openstack/build/2ae6f3685e4c41e9a724b2b4074fca29 : FAILURE in 11m 38s (non-voting)\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/8a1d95478e4a41df9e88b891fa893eb6 : FAILURE in 7m 53s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9e03c516fb2486b91cdc9bd926cf0b7 : FAILURE in 22m 03s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/d9c5bbbde37341adbbbb86088442abef : SUCCESS in 1h 10m 49s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/fbc2fe495db64ab8a4e10197b33e86f1 : SUCCESS in 1h 17m 40s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/2f8dae081b81435faad534c316d3862b : FAILURE in 21m 33s (non-voting)\n- devstack-platform-rocky-red-quartz https://zuul.opendev.org/t/openstack/build/a683234ebe3546f5bc892127e8df23fd : FAILURE in 10m 19s (non-voting)\n- devstack-platform-ubuntu-noble-ovn-source https://zuul.opendev.org/t/openstack/build/bee4d74b276e4830b349e927e059ac40 : SUCCESS in 1h 39m 52s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/03495fb24da4459281234eb00efed199 : SUCCESS in 1h 30m 02s (non-voting)\n- devstack-platform-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/17681bdc419c4836b0fa7ded63014aac : SUCCESS in 1h 34m 48s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/09a130bdafce4d5cb6a5a4e9cb265b54 : SUCCESS in 36m 11s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/adc6ddd4de3348e782780a6d4dc55a4c : SUCCESS in 3m 42s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/be04d7d712a54b9ba1d23f55e06eabdd : SUCCESS in 3m 05s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/9b7bcbdc9cb846489464bfb7c83f9ddb : SUCCESS in 1h 04m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1cb3004a062f40c5b1d83b28616ee4fa : SUCCESS in 1h 03m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/41330b1cf439402c8a81cbcab6f36944 : SUCCESS in 1h 14m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3a39c757da68498580df23424de3f1e4 : SUCCESS in 1h 17m 37s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8dcd16c0d6f44ab5b8994319826d646d : SUCCESS in 1h 14m 40s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55ce12a107b54623a891a453ce041385 : SUCCESS in 1h 02m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b176280db9f149c1a24c586ccdb41da3 : SUCCESS in 52m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4670eec5a4c942169bcff41f4a957b45 : SUCCESS in 1h 27m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f53e8a4593661568c6b469062ca97f2d3c1dfaff","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-24 13:25:21.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d09f9e985c09b74df774161f721b96313180e2fc","revisions":{"d09f9e985c09b74df774161f721b96313180e2fc":{"kind":"REWORK","_number":1,"created":"2026-09-23 02:56:16.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/83/1006883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/83/1006883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/83/1006883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/83/1006883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/83/1006883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/83/1006883/1"}}},"commit":{"parents":[{"commit":"60ed1ba77a14d8683dd097b46c10233a58190e9b","subject":"Cap API max microversions for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/60ed1ba77a14d8683dd097b46c10233a58190e9b"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-23 02:55:51.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-09-23 02:55:51.000000000","tz":0},"subject":"Cap stable/2026.2 network, swift, volume api_extensions for tempest","message":"Cap stable/2026.2 network, swift, volume api_extensions for tempest\n\nThis commit cap the network, volume and swift extensions on\nTempest\u0027s config option api_extensions.\n\nand no new extensions in swift and cinder.\n\nChange-Id: I634b3bd63b58d7ed315592ee523e458c4eacc6f2\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/d09f9e985c09b74df774161f721b96313180e2fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/d09f9e985c09b74df774161f721b96313180e2fc"}]},"branch":"refs/heads/stable/2026.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
