)]}'
{"id":"openstack%2Fdevstack~561187","triplet_id":"openstack%2Fdevstack~stable%2Fqueens~Id017b169f25a7589da3eca509c28f5d867fd5962","project":"openstack/devstack","branch":"stable/queens","topic":"multinode_zuulv3","hashtags":[],"change_id":"Id017b169f25a7589da3eca509c28f5d867fd5962","subject":"Fix devstack job: the dict name is devstack_local_conf","status":"MERGED","created":"2018-04-13 10:31:17.000000000","updated":"2018-04-19 09:37:50.000000000","submitted":"2018-04-19 09:37:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"561187-1524130670328-d6052c5b","meta_rev_id":"4888b33ad2a6b0f40d53f149f202cfe20180407f","_number":561187,"virtual_id_number":561187,"owner":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-04-13 19:31:49.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2018-04-13 14:56:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2018-04-19 09:37:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2018-04-19 08:13:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2018-04-13 14:44:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":2,"date":"2018-04-18 09:22:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2018-04-19 08:13:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-04-13 14:44:18.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2018-04-13 14:56:21.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-04-13 19:31:49.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-04-18 09:22:48.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2018-04-19 08:13:26.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2018-04-19 09:37:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f9ae55aa298a73106a5b29d98f46a6b431b7ba49","author":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"date":"2018-04-13 10:31:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcdc341672eedfba0d754ecd01420283cbbcb628","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-04-13 11:33:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/devstack/87/561187/1/check/tempest-dsvm-full-xenial/3cc12f3/ : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":1},{"id":"188692242822119db9faedb1e62c76bee2277acd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-13 12:16:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/87/561187/1/check/tempest-full/f685c50/ : SUCCESS in 1h 22m 26s\n- neutron-grenade http://logs.openstack.org/87/561187/1/check/neutron-grenade/a244ba0/ : SUCCESS in 49m 22s\n- tempest-full-py3 http://logs.openstack.org/87/561187/1/check/tempest-full-py3/220c552/ : SUCCESS in 1h 21m 21s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/561187/1/check/build-openstack-sphinx-docs/fa30606/html/ : SUCCESS in 4m 04s\n- openstack-tox-bashate http://logs.openstack.org/87/561187/1/check/openstack-tox-bashate/24064fe/ : SUCCESS in 2m 59s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/87/561187/1/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/dbb3756/ : SUCCESS in 43m 07s (non-voting)\n- legacy-swift-dsvm-functional http://logs.openstack.org/87/561187/1/check/legacy-swift-dsvm-functional/b08423a/ : SUCCESS in 20m 07s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/87/561187/1/check/neutron-grenade-multinode/07d3f51/ : SUCCESS in 58m 53s\n- neutron-tempest-linuxbridge http://logs.openstack.org/87/561187/1/check/neutron-tempest-linuxbridge/27dffa7/ : SUCCESS in 1h 14m 21s\n- neutron-tempest-multinode-full http://logs.openstack.org/87/561187/1/check/neutron-tempest-multinode-full/4b2d9b8/ : SUCCESS in 1h 38m 39s (non-voting)\n- devstack http://logs.openstack.org/87/561187/1/check/devstack/6e8fe83/ : SUCCESS in 22m 12s\n- devstack-multinode http://logs.openstack.org/87/561187/1/check/devstack-multinode/ea536fb/ : FAILURE in 37m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e7c6e211af468580bbaa3f68a52ceb868313907c","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-04-13 12:35:26.000000000","message":"Patch Set 1:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/561187/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/9bc3610 : SUCCESS in 2h 03m 22s","accounts_in_message":[],"_revision_number":1},{"id":"961b199c759f1b65655d714952c3a588fb7c779b","author":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"date":"2018-04-13 13:56:11.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"290acc6cd5d8726603a44fcb5b4592f5c601efc5","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2018-04-13 14:44:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"10dd62a3a135ca4f5ece745915a3f942d9173b0d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-04-13 14:56:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/devstack/87/561187/2/check/tempest-dsvm-full-xenial/ec487c8/ : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":2},{"id":"afc7b3e885069393eb3effb9126f9ecf081a08fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-13 15:40:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/87/561187/2/check/tempest-full/f3d5737/ : SUCCESS in 1h 27m 39s\n- neutron-grenade http://logs.openstack.org/87/561187/2/check/neutron-grenade/051dc98/ : SUCCESS in 45m 47s\n- tempest-full-py3 http://logs.openstack.org/87/561187/2/check/tempest-full-py3/c791d3a/ : SUCCESS in 1h 20m 05s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/561187/2/check/build-openstack-sphinx-docs/9018f18/html/ : SUCCESS in 3m 17s\n- openstack-tox-bashate http://logs.openstack.org/87/561187/2/check/openstack-tox-bashate/1853d85/ : SUCCESS in 3m 13s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/87/561187/2/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/f45499b/ : SUCCESS in 40m 46s (non-voting)\n- legacy-swift-dsvm-functional http://logs.openstack.org/87/561187/2/check/legacy-swift-dsvm-functional/3ae89e1/ : SUCCESS in 22m 00s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/87/561187/2/check/neutron-grenade-multinode/d403970/ : SUCCESS in 1h 00m 28s\n- neutron-tempest-linuxbridge http://logs.openstack.org/87/561187/2/check/neutron-tempest-linuxbridge/5d562ac/ : SUCCESS in 1h 35m 10s\n- neutron-tempest-multinode-full http://logs.openstack.org/87/561187/2/check/neutron-tempest-multinode-full/76e038d/ : SUCCESS in 1h 33m 56s (non-voting)\n- devstack http://logs.openstack.org/87/561187/2/check/devstack/90bb41a/ : SUCCESS in 19m 18s\n- devstack-multinode http://logs.openstack.org/87/561187/2/check/devstack-multinode/634afbe/ : SUCCESS in 28m 01s","accounts_in_message":[],"_revision_number":2},{"id":"8c66b334952bfb9c572d37957937e9df7515f245","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-04-13 19:31:49.000000000","message":"Patch Set 2:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/561187/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/fb80346 : SUCCESS in 1h 15m 18s","accounts_in_message":[],"_revision_number":2},{"id":"1f0aac5f84e6e0d73d68700fc1d5712e4e2cc49d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-04-18 09:22:48.000000000","message":"Patch Set 2: Code-Review+2\n\nwas searching for its master patch - Id017b169f25a7589da3eca509c28f5d867fd5962","accounts_in_message":[],"_revision_number":2},{"id":"75e7d480f9458d02e3608861f057290e53c29584","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2018-04-19 08:13:26.000000000","message":"Patch Set 2: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"48c1b6261edb692bdb3b5a679db19e8f1e2eb300","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-19 08:14:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3ff37e3cd591672f587b25f46052f5313bf5d8b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-19 09:37:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/87/561187/2/gate/tempest-full/d92f55b/ : SUCCESS in 1h 21m 36s\n- neutron-grenade http://logs.openstack.org/87/561187/2/gate/neutron-grenade/4567c07/ : SUCCESS in 54m 05s\n- tempest-full-py3 http://logs.openstack.org/87/561187/2/gate/tempest-full-py3/497a312/ : SUCCESS in 1h 15m 48s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/561187/2/gate/build-openstack-sphinx-docs/f3617fd/html/ : SUCCESS in 5m 12s\n- openstack-tox-bashate http://logs.openstack.org/87/561187/2/gate/openstack-tox-bashate/b98f427/ : SUCCESS in 2m 43s\n- neutron-grenade-multinode http://logs.openstack.org/87/561187/2/gate/neutron-grenade-multinode/2e5f0e9/ : SUCCESS in 58m 23s\n- neutron-tempest-linuxbridge http://logs.openstack.org/87/561187/2/gate/neutron-tempest-linuxbridge/31db101/ : SUCCESS in 1h 18m 11s\n- devstack http://logs.openstack.org/87/561187/2/gate/devstack/fa51cb2/ : SUCCESS in 20m 26s","accounts_in_message":[],"_revision_number":2},{"id":"3f73363fa1dc8ce4cd276345e6a1bef317164a21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-19 09:37:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3a19ac074f290fba918d53e0a368e95789b9acfd","revisions":{"1cf68c8876ea9e499deb26fd2944573f231e4616":{"kind":"REWORK","_number":1,"created":"2018-04-13 10:31:17.000000000","uploader":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"ref":"refs/changes/87/561187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/87/561187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/87/561187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/87/561187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/87/561187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/87/561187/1"}}},"commit":{"parents":[{"commit":"1b8dd2f2cd179d88ff5b3b07f09bd770a29ad683","subject":"Enforce linear strategy for orchestrate-devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/1b8dd2f2cd179d88ff5b3b07f09bd770a29ad683"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2018-03-18 16:56:27.000000000","tz":60},"committer":{"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","date":"2018-04-13 10:30:45.000000000","tz":60},"subject":"Fix devstack job: the dict name is devstack_local_conf","message":"Fix devstack job: the dict name is devstack_local_conf\n\nIt looks like a typo.\n\nChange-Id: Id017b169f25a7589da3eca509c28f5d867fd5962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/1cf68c8876ea9e499deb26fd2944573f231e4616"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/1cf68c8876ea9e499deb26fd2944573f231e4616"}]},"branch":"refs/heads/stable/queens"},"3a19ac074f290fba918d53e0a368e95789b9acfd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2018-04-13 13:56:11.000000000","uploader":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"ref":"refs/changes/87/561187/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/87/561187/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/87/561187/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/87/561187/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/87/561187/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/87/561187/2"}}},"commit":{"parents":[{"commit":"da5c918d738dcd15a26266f4249b77f4d6eaf24b","subject":"Enforce linear strategy for orchestrate-devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/da5c918d738dcd15a26266f4249b77f4d6eaf24b"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2018-03-18 16:56:27.000000000","tz":60},"committer":{"name":"Andrea Frittoli (andreaf)","email":"andrea.frittoli@gmail.com","date":"2018-04-13 13:55:52.000000000","tz":60},"subject":"Fix devstack job: the dict name is devstack_local_conf","message":"Fix devstack job: the dict name is devstack_local_conf\n\nIt looks like a typo.\n\nChange-Id: Id017b169f25a7589da3eca509c28f5d867fd5962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/3a19ac074f290fba918d53e0a368e95789b9acfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/3a19ac074f290fba918d53e0a368e95789b9acfd"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
