)]}'
{"id":"openstack%2Fdevstack~713760","triplet_id":"openstack%2Fdevstack~stable%2Ftrain~Idac09e8259d58ed130b79b40528f733708000da7","project":"openstack/devstack","branch":"stable/train","topic":"bring-back-ld-guide-stable/train","hashtags":[],"change_id":"Idac09e8259d58ed130b79b40528f733708000da7","subject":"Bring back load balance v2 guide","status":"MERGED","created":"2020-03-18 23:48:43.000000000","updated":"2020-04-06 23:46:36.000000000","submitted":"2020-04-03 05:21:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713760-1585891301841-cde04409","meta_rev_id":"2d42c11bc81ebd80b6b063e6d6887b502997d36b","_number":713760,"virtual_id_number":713760,"owner":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 05:21:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2020-03-19 08:37:12.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-03 03:51:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-03 03:51:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 08:37:12.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-03 03:51:12.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-04-03 05:21:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"179641e9b86176d8b344c1448ab453baf31fb83d","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-03-18 23:48:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4741ec751c3addf82af25aa6d6ac02bed068b5bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 23:56:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd7bddee61f64e858687346845a9267c : SUCCESS in 6m 17s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/c3589fb3c2764c1e83691e330d6891b3 : SUCCESS in 3m 39s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/0cb6a3c5b5264e04aa10edc1320a1a2d : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":1},{"id":"d2d1dfa150a77e74d4184a6db8d5f98c3207e0dd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-19 08:37:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"479ae18335dba70ee7f7680b4f1833dac562567a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-04-03 03:51:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ea1dca269cbea6a235bd502742457e251bf14e95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 03:51:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5f7679d2e32b7248c7380ee33dcc8f6978f8ea0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 05:21:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bb7bd25225a84d7c9680a898132ad2acf4843570","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 05:21:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/440ff18986ce497ca500e552674b8ec6 : SUCCESS in 7m 24s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/b22c7f151ffc4dc78bca8ec00d6e5dfa : SUCCESS in 6m 17s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/af3021b1c43745ab94db3b5f50a7b2a7 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e6a77657fe9e537169bb2f1a0981a02da6e60ab2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 05:24:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13a6043a3d5d46ee89324df0ed94e368 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1},{"id":"2cfb48fe3c5cd8846f7d72b3474ac3abb987944c","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-04-06 23:46:36.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 4d11eef81f4fd6b9f0b0ccbbc39eda33b06fb573","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb64fe162528d35ee96fe83fc6b26aa0e0a4f445","revisions":{"eb64fe162528d35ee96fe83fc6b26aa0e0a4f445":{"kind":"REWORK","_number":1,"created":"2020-03-18 23:48:43.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/60/713760/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/60/713760/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/60/713760/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/60/713760/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/60/713760/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/60/713760/1"}}},"commit":{"parents":[{"commit":"ef44118eb8b9595d01df3e9bf50a8d7c2265604e","subject":"Removal of deprecated command and deprecated optional argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/ef44118eb8b9595d01df3e9bf50a8d7c2265604e"}]}],"author":{"name":"ushen","email":"yshxxsjt715@gmail.com","date":"2020-03-18 12:08:16.000000000","tz":480},"committer":{"name":"norman shen","email":"yshxxsjt715@gmail.com","date":"2020-03-18 23:48:43.000000000","tz":0},"subject":"Bring back load balance v2 guide","message":"Bring back load balance v2 guide\n\nPrevious commit forgets to add an entry for load\nbalancer version 2. This ps brings them back.\n\nChange-Id: Idac09e8259d58ed130b79b40528f733708000da7\n(cherry picked from commit 90f0baf3cb84118e51bc5990ba6885f07b679090)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/eb64fe162528d35ee96fe83fc6b26aa0e0a4f445"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/eb64fe162528d35ee96fe83fc6b26aa0e0a4f445"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
