)]}'
{"id":"openstack%2Fdevstack~824290","triplet_id":"openstack%2Fdevstack~stable%2Fussuri~I36751569d92fbc5084b8308d423a75318ae7d406","project":"openstack/devstack","branch":"stable/ussuri","topic":"bug/1959600-stable/ussuri","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-02-02 05:06:46.000000000","reason":"Removed by Dr. Jens Harbott (frickler) by clicking the attention icon"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-02-02 20:38:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I36751569d92fbc5084b8308d423a75318ae7d406","subject":"Drop CentOS8 job","status":"MERGED","created":"2022-01-12 02:26:47.000000000","updated":"2022-02-02 20:45:21.000000000","submitted":"2022-02-02 20:38:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824290-bug/1959600-stable/ussuri","cherry_pick_of_change":824275,"cherry_pick_of_patch_set":1,"meta_rev_id":"3174ebcdc0185564684b3c5c94b3ce55240e990b","_number":824290,"virtual_id_number":824290,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-02 20:38:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-02-02 17:50:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-02-02 17:50:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-02-02 17:50:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-12 02:27:07.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-12 02:28:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-12 05:38:13.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-01-13 14:25:04.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"1f4ee3b045dbcfb57b1509e867552db046ec02b1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-12 02:26:47.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.\n\nThe following files contain Git conflicts:\n* .zuul.yaml","accounts_in_message":[],"_revision_number":1},{"id":"c6e7d48bdb61b007e80f56b3f2455ff84052d195","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-01-12 02:27:07.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/devstack on branch stable/ussuri.  The error was:\n\n  while parsing a block collection\n  in \"openstack/devstack/.zuul.yaml@stable/ussuri\", line 1, column 1\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/devstack/.zuul.yaml@stable/ussuri\", line 570, column 1","accounts_in_message":[],"_revision_number":1},{"id":"a10e1643bd56827a5a9163219ae947519c8ae9eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 02:28:18.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/devstack on branch stable/ussuri.  The error was:\n\n  while parsing a block collection\n  in \"openstack/devstack/.zuul.yaml@stable/ussuri\", line 1, column 1\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/devstack/.zuul.yaml@stable/ussuri\", line 570, column 1","accounts_in_message":[],"_revision_number":1},{"id":"dd6308768a2672518227267d330b39134446e073","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-12 02:31:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a7b6cf8be8b223f7d5005a4478c601349feec795","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-12 02:31:44.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"f267aea5cdfa5770735a5c78e77cb45266d57b34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 04:38:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/06f3b9530d2f486ba7929fc8774228c4 : SUCCESS in 1h 17m 53s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/03d6bf1b8a3d42f5817d78f24b805a07 : SUCCESS in 1h 44m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bc0b609084c4dbe912588d8a44146d4 : SUCCESS in 11m 01s\n- devstack https://zuul.opendev.org/t/openstack/build/b08abf89ff624a8fb0e6e95c93ffa074 : SUCCESS in 28m 10s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/5f233c259a704c1eaecf7fb7c6aa77ea : SUCCESS in 27m 18s\n- devstack-platform-opensuse-15 https://zuul.opendev.org/t/openstack/build/9063cc52b99f455993f8e629fa3246dd : FAILURE in 41m 58s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/2f82b0b276724544bfba9e69da429a44 : SUCCESS in 41m 30s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/a8ae1ced4d48467b9fc2f133dbc6e3bd : SUCCESS in 5m 32s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e6931e406d1140da85cde4fe73d949e1 : SUCCESS in 6m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ea84c9e9edcb4c239aa7cf025f73b59e : SUCCESS in 1h 03m 11s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/51b584f732a741189bfac665a8b9c196 : SUCCESS in 29m 29s (non-voting)\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/43882658081d4fcf9edf25aea06b57cc : SUCCESS in 31m 59s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ff20746c17b48edb06c2a215c2f3746 : SUCCESS in 1h 16m 43s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7fbbb1c8b45a448f96acb9f3afb30484 : SUCCESS in 1h 12m 36s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b7ae0ee467b04b169fbf6763ca2eaa19 : SUCCESS in 1h 41m 28s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2ef3da9396d4fc0bf6d4ab0342bede4 : SUCCESS in 52m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/481c06932b4749818b2a27ce65c7d0ef : SUCCESS in 55m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9fd16cef7027499e9f93075688ba7b17 : FAILURE in 48m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f3f8a6515ed3379a206c748dd281370c8eed1e4f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-01-12 05:38:13.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"a4145740ebdb29d12abae66b5b8b5c19167889c3","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-01-13 14:25:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b96df506a6c7041b877dcc11cf3621d0e0e7bac7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 17:43:47.000000000","message":"Patch Set 2: Verified-2\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/nova on branch master.  The error was:\n\n  The nodeset \"devstack-single-node-centos-8\" was not found.\n\nThe error appears in the following job stanza:\n\n  job:\n      name: nova-tox-functional-centos8-py36\n      parent: openstack-tox-functional-py36\n      nodeset: devstack-single-node-centos-8\n      description: |\n        Run tox-based functional tests for the OpenStack Nova project\n        under cPython version 3.6 with Nova specific irrelevant-files list.\n        Uses tox with the ``functional-py36`` environment.\n  \n        This job also provides a parent for other projects to run the nova\n        functional tests on their own changes.\n      required-projects:\n        # including nova here makes this job reusable by other projects\n        - openstack/nova\n        - openstack/placement\n      irrelevant-files: \u0026functional-irrelevant-files\n        - ^.*\\.rst$\n        - ^api-.*$\n        - ^doc/(source|test)/.*$\n        - ^nova/locale/.*$\n        - ^releasenotes/.*$\n      vars:\n        # explicitly stating the work dir makes this job reusable by other\n        # projects\n        zuul_work_dir: src/opendev.org/openstack/nova\n        bindep_profile: test py36\n      timeout: 3600\n\n  in \"openstack/nova/.zuul.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":2},{"id":"5cb2314d43d4ca945f43facf9286927f3ad09a3e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-13 18:06:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f5bceb83dad42d1febcb01548be09e0d2323bef0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-21 21:11:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6959ed0f82ba53db935a2cb516d0f52f50249b90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 21:15:37.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/nova on branch stable/xena.  The error was:\n\n  The nodeset \"devstack-single-node-centos-8\" was not found.\n\nThe error appears in the following job stanza:\n\n  job:\n      name: nova-tox-functional-centos8-py36\n      parent: openstack-tox-functional-py36\n      nodeset: devstack-single-node-centos-8\n      description: |\n        Run tox-based functional tests for the OpenStack Nova project\n        under cPython version 3.6 with Nova specific irrelevant-files list.\n        Uses tox with the ``functional-py36`` environment.\n  \n        This job also provides a parent for other projects to run the nova\n        functional tests on their own changes.\n      required-projects:\n        # including nova here makes this job reusable by other projects\n        - openstack/nova\n        - openstack/placement\n      irrelevant-files: \u0026functional-irrelevant-files\n        - ^.*\\.rst$\n        - ^api-.*$\n        - ^doc/(source|test)/.*$\n        - ^nova/locale/.*$\n        - ^releasenotes/.*$\n      vars:\n        # explicitly stating the work dir makes this job reusable by other\n        # projects\n        zuul_work_dir: src/opendev.org/openstack/nova\n        bindep_profile: test py36\n      timeout: 3600\n\n  in \"openstack/nova/.zuul.yaml@stable/xena\", line 4, column 3","accounts_in_message":[],"_revision_number":2},{"id":"a99b7070c41cf3ee82dbd83a2681a1d5844c4f59","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-22 00:38:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"13288eda3537cdd30b4085d7ef9320562aa043e4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-30 03:44:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9cb80f119d21df51b8dbb8c96279a7d36eaa1aaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-30 03:44:33.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/python-cinderclient on branch stable/victoria.  The\nerror was:\n\n  The nodeset \"devstack-single-node-centos-8\" was not found.\n\nThe error appears in the following job stanza:\n\n  job:\n      name: python-cinderclient-functional-py36\n      parent: python-cinderclient-functional-base\n      # need to specify a platform that has python 3.6 available\n      nodeset: devstack-single-node-centos-8\n      vars:\n        python_version: 3.6\n        tox_envlist: functional-py36\n\n  in \"openstack/python-cinderclient/.zuul.yaml@stable/victoria\", line 15, column 3","accounts_in_message":[],"_revision_number":2},{"id":"a2d94b1a1e6b4b67b79e54c09acd7ee2a21a4451","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-02-01 20:21:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"466c3740c0b39aacfdbf70222cc677c756b292b0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-02-01 20:22:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ab0bdfcc7a2fda7f646052917f372b5883d20f0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 21:08:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2cb0292822a4476c979061fc85ef10b0 : FAILURE in 20m 05s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69af54b14cdf41e68c1f2f25dac0d28d : FAILURE in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a05fe8b1dd9a43868fc49fd98b9e849f : SUCCESS in 6m 02s\n- devstack https://zuul.opendev.org/t/openstack/build/5eaf0c3401d2493f915fecbdaaa1016c : FAILURE in 8m 42s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/dedcce20f1d240aab1755bcdeea7d89e : FAILURE in 7m 29s\n- devstack-platform-opensuse-15 https://zuul.opendev.org/t/openstack/build/6cab70120387433292717ab09c9acf7e : FAILURE in 15m 59s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/9baa2aa4df04451f8d2d14189276bae6 : FAILURE in 10m 30s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/7dfa83439afd47b4aa210d391f1d2fd8 : SUCCESS in 4m 05s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/504286fe6b5b4d65b8288388d99d442b : SUCCESS in 3m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e07d991feb04469d896383e53e7e7704 : FAILURE in 18m 08s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c6829c589ce742aca42a19904bc013a5 : RETRY_LIMIT in 6m 02s (non-voting)\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/563a110be4eb470ab8c6dcfacff0293f : RETRY_LIMIT in 13m 39s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/afc897c535c5458e806581ac150a434f : FAILURE in 22m 59s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/632a200f1be74851b9e9e8f39697ed16 : FAILURE in 8m 15s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6b1da08a6a044b08bfd8b6476665d33a : FAILURE in 9m 55s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5423867f511e41b78e7c420eb667e889 : POST_FAILURE in 9m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d529063f03d4aa5a57483cbabf434e9 : FAILURE in 9m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c99f7f0f6ce3400cba2c60403d87a526 : POST_FAILURE in 12m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5c7b94e4685209e91c57895c18b00daa4ee34cc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-02-02 06:46:05.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f3c5a85ecac996c7cd506f666231b73218d59a3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 08:29:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c8dffb3a04514b8cb4ccb3b8839ed82a : FAILURE in 16m 46s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/16ae6c80aa0f4c94bb2ce258f793b61b : SUCCESS in 1h 31m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87e03da64761431a9f53db8e0afe7803 : SUCCESS in 5m 08s\n- devstack https://zuul.opendev.org/t/openstack/build/70c9c2a82c1c468d8d6f7190bd8e5203 : SUCCESS in 36m 44s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/6f0094069d6045108ac7d839c0c2d059 : SUCCESS in 34m 15s\n- devstack-platform-opensuse-15 https://zuul.opendev.org/t/openstack/build/43ac477ef6d44058a67acaf939236909 : FAILURE in 39m 23s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/dd807b855b904aa981276fe8ddc7504f : SUCCESS in 46m 03s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/9ef8d6d4d13e47b5ae11ef64c50f8e9b : SUCCESS in 3m 26s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c6534c9ab776441eab301b5a1c7acf1e : SUCCESS in 2m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6cc6c8b48a6747c69e3688b768045e7a : SUCCESS in 58m 53s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cbf0bfbec1294f5f9143031b789c6bcd : SUCCESS in 34m 23s (non-voting)\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ccb7a0e537354a479fc48849ba24affe : SUCCESS in 36m 06s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/27c771b352b8439aa11f6e73fa264c2f : FAILURE in 22m 22s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f15080563c3d4c59ac2131215fcb4fc6 : SUCCESS in 1h 16m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7689ab2eef694dbdb872d20a64e9aa1e : SUCCESS in 1h 37m 07s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b932356edd441c092619a8babaae63a : SUCCESS in 49m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b84abf38bc8942d2bb7ac5c35a317ff0 : SUCCESS in 59m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d4aba87f3a93460cbe859aec46d6aedf : SUCCESS in 49m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"62ab85e8685710badc92d9e2f470e292bd4845a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-02-02 11:46:00.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent, and commit message as Patch Set 4.","accounts_in_message":[],"_revision_number":5},{"id":"24537539f2bf5bba9ed23ad0600e796a0d5854f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 13:24:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5559e547820843529c424bb76daae294 : SUCCESS in 1h 11m 29s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41abf15fdb534cc6b60bbbe16c21fc3b : SUCCESS in 1h 22m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d89f4e3c5bf145c2852b9dd83526eec9 : SUCCESS in 6m 11s\n- devstack https://zuul.opendev.org/t/openstack/build/f11f130bf8724ff7b4911201e677130d : SUCCESS in 25m 44s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/36c1846c37ec4a7cbad7fa170029e636 : SUCCESS in 25m 29s\n- devstack-platform-opensuse-15 https://zuul.opendev.org/t/openstack/build/cdfe4ee68908477da2c1acb6f6fb402a : FAILURE in 29m 35s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/08f00f91deba49f79254888229791b42 : SUCCESS in 32m 29s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/12d5df84321b45a5bc49599fce8318d5 : SUCCESS in 4m 42s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/505d9fdc5d14493f838f09c7a620467d : SUCCESS in 3m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c5e4e28c552147a38427af9c2a14d2ab : SUCCESS in 53m 50s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ab3647dd3c814adebb1daa22b9f55305 : SUCCESS in 32m 47s (non-voting)\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b7a13aa17cf24f7993a7ddc11018f69f : SUCCESS in 31m 13s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/98cbd8c6da9448a5965895f14fd9206d : SUCCESS in 1h 08m 59s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/93d895f6727248fab9cd9bf2e56f9624 : SUCCESS in 58m 15s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/aa9bba60527c43afade1f383f762a60a : SUCCESS in 1h 30m 58s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f72c4bdb4cc44adae2d7c3d67cec02f : SUCCESS in 52m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1565299ba97483584432f985ec189ba : SUCCESS in 52m 05s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f2bcc1ff979c4de8a064b27f58f160e0 : SUCCESS in 48m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4f88691ea7e80a933c80ff5295985edf64c47d8e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-02-02 17:50:20.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1 Review-Priority+2","accounts_in_message":[],"_revision_number":5},{"id":"9445fd3940e2634ecde7845ed2ae972efe3443e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 17:52:12.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a89806cd2e40e841f047bd6d2704d45822be4ef7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 20:38:22.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4a147c95014f4ae5b8e476f92b7cd876 : SUCCESS in 1h 12m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e0f649f643a494bbcc71594541726c1 : SUCCESS in 1h 19m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64cff653fd8d45e087db1d952743a618 : SUCCESS in 9m 35s\n- devstack https://zuul.opendev.org/t/openstack/build/39508cd4962849ceb53e4c912b3d76a6 : SUCCESS in 25m 06s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/8de696af4976441392916497df3df376 : SUCCESS in 33m 04s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/6753a01fe0c54a1b850b421f27c4e854 : SUCCESS in 45m 33s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/8c6845d1a4bc4a72bed2b653de8c0082 : SUCCESS in 4m 27s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/8098f942487c4a82966dce14da3cb494 : SUCCESS in 5m 53s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/48945cde38a5403296393b130ab2610b : SUCCESS in 1h 02m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fed486d6a473475ea939eb5242f662a1 : SUCCESS in 1h 01m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f2eb3e4a8f84ceab1a4b55f335d338a : SUCCESS in 1h 00m 41s","accounts_in_message":[],"_revision_number":5},{"id":"909e641966a6f64493ae5b3506b0d855fdbe6a91","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 20:38:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"3174ebcdc0185564684b3c5c94b3ce55240e990b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 20:45:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0825e605a30a44cd8f130eae35338dfd : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7d3dd00e628a48fd2e757e22ea2ec3497286027e","revisions":{"53c57717c524fe0728a93b088aeb2c66f5fe78bb":{"kind":"REWORK","_number":1,"created":"2022-01-12 02:26:47.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/90/824290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/90/824290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/90/824290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/90/824290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/90/824290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/90/824290/1"}}},"commit":{"parents":[{"commit":"d927b6017880983cf873e199bb50332fe2e3f254","subject":"nova: Ensure each compute uses a unique iSCSI initiator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/d927b6017880983cf873e199bb50332fe2e3f254"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-01-12 02:21:08.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-01-12 02:26:47.000000000","tz":0},"subject":"Drop CentOS8 job and nodeset","message":"Drop CentOS8 job and nodeset\n\nCentOS8 is going to be EOL soon and opendev\nis planning to drop the centOS8 image from CI\n- http://lists.openstack.org/pipermail/openstack-discuss/2022-January/026621.html\n\nDevstack stable/wallaby and onwards has replaced the\ncentOS testing with centOS Stream. For older stable branches\n(\u003c\u003d stable/victoria) we can either replace it with centOS Steam\nor do the removal of CentOS8 and adding CentOS Stream testing\nin two steps 1. drop the centOS8 testing as it cannot\nbe supported in CI 2. add centOS Stream job if possible.\n\nThis commit propose the step 1 as required by opendev image removal.\n\n[Stable only]\n\nChange-Id: I36751569d92fbc5084b8308d423a75318ae7d406\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/53c57717c524fe0728a93b088aeb2c66f5fe78bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/53c57717c524fe0728a93b088aeb2c66f5fe78bb"}]},"branch":"refs/heads/stable/ussuri"},"b62947beebb104752390ebc636b0ff19d13c572b":{"kind":"REWORK","_number":2,"created":"2022-01-12 02:31:17.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/90/824290/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/90/824290/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/90/824290/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/90/824290/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/90/824290/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/90/824290/2"}}},"commit":{"parents":[{"commit":"d927b6017880983cf873e199bb50332fe2e3f254","subject":"nova: Ensure each compute uses a unique iSCSI initiator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/d927b6017880983cf873e199bb50332fe2e3f254"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-01-12 02:21:08.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-01-12 02:31:01.000000000","tz":-360},"subject":"Drop CentOS8 job and nodeset","message":"Drop CentOS8 job and nodeset\n\nCentOS8 is going to be EOL soon and opendev\nis planning to drop the centOS8 image from CI\n- http://lists.openstack.org/pipermail/openstack-discuss/2022-January/026621.html\n\nDevstack stable/wallaby and onwards has replaced the\ncentOS testing with centOS Stream. For older stable branches\n(\u003c\u003d stable/victoria) we can either replace it with centOS Steam\nor do the removal of CentOS8 and adding CentOS Stream testing\nin two steps 1. drop the centOS8 testing as it cannot\nbe supported in CI 2. add centOS Stream job if possible.\n\nThis commit propose the step 1 as required by opendev image removal.\n\n[Stable only]\n\nChange-Id: I36751569d92fbc5084b8308d423a75318ae7d406\n(cherry picked from commit 3602e4349b9dcdea1865465342b192cfae3ca377)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/b62947beebb104752390ebc636b0ff19d13c572b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/b62947beebb104752390ebc636b0ff19d13c572b"}]},"branch":"refs/heads/stable/ussuri"},"2940abf991260a71a760cbf5a88c961b23a03a79":{"kind":"REWORK","_number":3,"created":"2022-02-01 20:21:16.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/90/824290/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/90/824290/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/90/824290/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/90/824290/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/90/824290/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/90/824290/3"}}},"commit":{"parents":[{"commit":"d927b6017880983cf873e199bb50332fe2e3f254","subject":"nova: Ensure each compute uses a unique iSCSI initiator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/d927b6017880983cf873e199bb50332fe2e3f254"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-01-12 02:21:08.000000000","tz":-360},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-02-01 20:20:12.000000000","tz":60},"subject":"Drop CentOS8 job","message":"Drop CentOS8 job\n\nCentOS8 is going to be EOL soon and opendev\nis planning to drop the centOS8 image from CI\n- http://lists.openstack.org/pipermail/openstack-discuss/2022-January/026621.html\n\nDevstack stable/wallaby and onwards has replaced the\ncentOS testing with centOS Stream. For older stable branches\n(\u003c\u003d stable/victoria) we can either replace it with centOS Steam\nor do the removal of CentOS8 and adding CentOS Stream testing\nin two steps 1. drop the centOS8 testing as it cannot\nbe supported in CI 2. add centOS Stream job if possible.\n\nThis commit propose the step 1 as required by opendev image removal.\n\n[Stable only]\n\nChange-Id: I36751569d92fbc5084b8308d423a75318ae7d406\n(cherry picked from commit 3602e4349b9dcdea1865465342b192cfae3ca377)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/2940abf991260a71a760cbf5a88c961b23a03a79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/2940abf991260a71a760cbf5a88c961b23a03a79"}]},"branch":"refs/heads/stable/ussuri"},"5fbee79e0a31b7d1225739560f8d7a6595b1ed9b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-02-02 06:46:05.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/90/824290/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/90/824290/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/90/824290/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/90/824290/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/90/824290/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/90/824290/4"}}},"commit":{"parents":[{"commit":"311d00da2157e6c1fec957941759658af008b7b4","subject":"Add python3.6 pip support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/311d00da2157e6c1fec957941759658af008b7b4"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-01-12 02:21:08.000000000","tz":-360},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-02-02 06:46:05.000000000","tz":0},"subject":"Drop CentOS8 job","message":"Drop CentOS8 job\n\nCentOS8 is going to be EOL soon and opendev\nis planning to drop the centOS8 image from CI\n- http://lists.openstack.org/pipermail/openstack-discuss/2022-January/026621.html\n\nDevstack stable/wallaby and onwards has replaced the\ncentOS testing with centOS Stream. For older stable branches\n(\u003c\u003d stable/victoria) we can either replace it with centOS Steam\nor do the removal of CentOS8 and adding CentOS Stream testing\nin two steps 1. drop the centOS8 testing as it cannot\nbe supported in CI 2. add centOS Stream job if possible.\n\nThis commit propose the step 1 as required by opendev image removal.\n\n[Stable only]\n\nChange-Id: I36751569d92fbc5084b8308d423a75318ae7d406\n(cherry picked from commit 3602e4349b9dcdea1865465342b192cfae3ca377)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/5fbee79e0a31b7d1225739560f8d7a6595b1ed9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/5fbee79e0a31b7d1225739560f8d7a6595b1ed9b"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"},"7d3dd00e628a48fd2e757e22ea2ec3497286027e":{"kind":"NO_CHANGE","_number":5,"created":"2022-02-02 11:46:00.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/90/824290/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/90/824290/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/90/824290/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/90/824290/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/90/824290/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/90/824290/5"}}},"commit":{"parents":[{"commit":"9c43f9fecc662444cc6b1083aa561fc00ffc8509","subject":"Add python3.6 pip support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/9c43f9fecc662444cc6b1083aa561fc00ffc8509"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-01-12 02:21:08.000000000","tz":-360},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-02-02 11:44:41.000000000","tz":60},"subject":"Drop CentOS8 job","message":"Drop CentOS8 job\n\nCentOS8 is going to be EOL soon and opendev\nis planning to drop the centOS8 image from CI\n- http://lists.openstack.org/pipermail/openstack-discuss/2022-January/026621.html\n\nDevstack stable/wallaby and onwards has replaced the\ncentOS testing with centOS Stream. For older stable branches\n(\u003c\u003d stable/victoria) we can either replace it with centOS Steam\nor do the removal of CentOS8 and adding CentOS Stream testing\nin two steps 1. drop the centOS8 testing as it cannot\nbe supported in CI 2. add centOS Stream job if possible.\n\nThis commit propose the step 1 as required by opendev image removal.\n\n[Stable only]\n\nChange-Id: I36751569d92fbc5084b8308d423a75318ae7d406\n(cherry picked from commit 3602e4349b9dcdea1865465342b192cfae3ca377)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/7d3dd00e628a48fd2e757e22ea2ec3497286027e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/7d3dd00e628a48fd2e757e22ea2ec3497286027e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
