)]}'
{"id":"openstack%2Fdevstack~838426","triplet_id":"openstack%2Fdevstack~stable%2Fwallaby~I941ef5ea90970a0901236afe81c551aaf24ac1d8","project":"openstack/devstack","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-04-19 15:47:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I941ef5ea90970a0901236afe81c551aaf24ac1d8","subject":"Write safe.directory items to system git config","status":"MERGED","created":"2022-04-19 02:31:19.000000000","updated":"2022-04-19 15:48:42.000000000","submitted":"2022-04-19 15:47:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838426","meta_rev_id":"dc87c8c890c27c9c5480227b282a3705ba8d7e5d","_number":838426,"virtual_id_number":838426,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-19 15:47:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2022-04-19 14:09:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-19 06:35:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2022-04-19 04:05:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":1,"date":"2022-04-19 14:09:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-19 02:31:19.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2022-04-19 04:05:10.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2022-04-19 04:29:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-19 06:35:23.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-04-19 14:09:09.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"9a9263807ca424ab57dfb0845ee8c5625e5166e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-04-19 02:31:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5eeebf9518c6cc18d0b6b2f0a72919984cb40221","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-04-19 04:05:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9d56a7cd265725de84bef8633815e64e398c183d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 04:29:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5da4cfdae8ad4f74b84b8da1e72c5fb4 : SUCCESS in 1h 04m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eba5930d45f94fbf83377f224ac2c471 : SUCCESS in 1h 29m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7787cad137104867af8f4293289ab763 : SUCCESS in 6m 47s\n- devstack https://zuul.opendev.org/t/openstack/build/8af18a2d1ecd4059bb54bfbc18da3bba : SUCCESS in 28m 46s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/2adcb9ba1ed14a6ab6ef99be2e66264b : SUCCESS in 32m 02s\n- devstack-platform-bionic https://zuul.opendev.org/t/openstack/build/7d880f4c137c43ad81a2bcc3dcbe9c2b : SUCCESS in 1h 28m 22s (non-voting)\n- devstack-platform-centos-8-stream https://zuul.opendev.org/t/openstack/build/770704de00b34988ae5ed0da693eb852 : SUCCESS in 1h 27m 31s (non-voting)\n- devstack-async https://zuul.opendev.org/t/openstack/build/68fb74224875460ab5294ce61d0850fc : SUCCESS in 1h 40m 05s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/28b9a02dba00472696e4655fb9ee213a : SUCCESS in 43m 16s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/e3dd362960644728a8de7ddc6925ea2c : SUCCESS in 4m 42s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/3f20a5157a384f14b82a0ea6b7d08522 : SUCCESS in 6m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3c42a4e332b5416d8edd90684f8a30cb : SUCCESS in 1h 08m 53s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2c83f4b22eca4f1e9096e7be0e865205 : SUCCESS in 56m 36s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9bd542450d464e48af3e3b5d027fbbf6 : SUCCESS in 48m 11s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/42194ede25a64f7eafc90a628091f2f6 : SUCCESS in 1h 00m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f9b154cd18dc4da6b84e10f60c281144 : SUCCESS in 1h 06m 17s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8fff03e6314b477f96879486c8ba73b6 : SUCCESS in 1h 39m 27s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/532f26b7b9904e2fbfb335ba2c02af49 : SUCCESS in 56m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/521ee945468e493f97c7da3d6dadca2c : SUCCESS in 47m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3b7e8e1df9cf4d11bea0ff3703ba9067 : SUCCESS in 1h 18m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e6876a7029f8cc0fe42921bcc170249b9a66c223","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-04-19 06:35:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"487c8999952444ab8487f20df1dc9c8a022e9e42","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-04-19 14:09:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a4e414586d23d96302d28911e1288013cd4a9bef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 14:10:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"740362cd67d5888854199ee6e895659a2d709ee9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 15:47:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/3f08c2a67af14d659a8aeda43d4dc9c5 : SUCCESS in 46m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4bc14bd1723842488c5518909141533e : SUCCESS in 1h 28m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3db42296406d4222b65bb7f957569593 : SUCCESS in 5m 33s\n- devstack https://zuul.opendev.org/t/openstack/build/2a9cd7e3dcf5415b9c8ef8708ffe4d53 : SUCCESS in 33m 06s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/dc1b53a065e8446789c665284e0009f3 : SUCCESS in 20m 37s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/402b83de32bf4334856045d2cd45aff6 : SUCCESS in 42m 04s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/e7722ecc0b394b2e81c985491643d704 : SUCCESS in 4m 24s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e32e5fbef2f94fac9db62236f1aeb73a : SUCCESS in 3m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/38e3b9bf199a44cdbc9640c180f99411 : SUCCESS in 45m 19s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9384b547c3144b3382418c2433cf9433 : SUCCESS in 1h 06m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0d6772e295747d7aca466bc5a2cb606 : SUCCESS in 58m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fab2e05181a406f963b25a110396f86 : SUCCESS in 47m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ff3a367cd9ee4516a15a45013de107f1 : SUCCESS in 1h 12m 02s","accounts_in_message":[],"_revision_number":1},{"id":"96f4f58b87fd88953ab87170984f7fbe7b130e30","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 15:47:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dc87c8c890c27c9c5480227b282a3705ba8d7e5d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 15:48:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf1979bd5f254de7b5a37971f315b2c5 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea636e0a92670353ac48274e704d30662f722691","revisions":{"ea636e0a92670353ac48274e704d30662f722691":{"kind":"REWORK","_number":1,"created":"2022-04-19 02:31:19.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/26/838426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/26/838426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/26/838426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/26/838426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/26/838426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/26/838426/1"}}},"commit":{"parents":[{"commit":"494d0b7d5fa25257ae1fa61b2cb1d6dbf9b52cd2","subject":"Mark our source trees as safe for git to use as other users","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/494d0b7d5fa25257ae1fa61b2cb1d6dbf9b52cd2"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-04-13 20:44:07.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-04-19 02:31:09.000000000","tz":-300},"subject":"Write safe.directory items to system git config","message":"Write safe.directory items to system git config\n\nThis is necessary for more consistent behavior across multiple\ndistro versions. Apparently somewhere along the way, git started\nlooking at the current user\u0027s home directory instead of $HOME.\n\nRelated-Bug: https://bugs.launchpad.net/devstack/+bug/1968798\n\nChange-Id: I941ef5ea90970a0901236afe81c551aaf24ac1d8\n(cherry picked from commit 4baeb3b51fcb6196fa311f823ad3f0ac13ccf341)\n(cherry picked from commit 9616d22938305dd1a56067810484c15912b37d8d)\n(cherry picked from commit d86f23b153acc0b8d791baddf2cd90875aa0a04d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/ea636e0a92670353ac48274e704d30662f722691"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/ea636e0a92670353ac48274e704d30662f722691"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
