)]}'
{"id":"openstack%2Fdevstack~843712","triplet_id":"openstack%2Fdevstack~master~I4808941ba4acaa15e1ec145209a88630bcac61c3","project":"openstack/devstack","branch":"master","topic":"neutron-policy","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-08-14 06:53:09.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I4808941ba4acaa15e1ec145209a88630bcac61c3","subject":"Drop legacy neutron policy file workaround","status":"ABANDONED","created":"2022-05-28 08:00:22.000000000","updated":"2024-07-08 16:51:54.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fa76a82616b5116d494d782222599fac8eac1bde","_number":843712,"virtual_id_number":843712,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-01 21:43:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-28 09:14:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-10 01:42:58.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"}],"messages":[{"id":"96e44c40bfeae6ebf76c83a55367de05066f1b82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-05-28 08:00:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92ac73191bb6f973e6a5c90973165e5ce0a8a4bc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-05-28 08:01:27.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"860efcac13986c30d9c8b674888feeccacac427a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-05-28 08:14:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4429a2117512d7178c701b2f705195bf4e1c1811","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 09:14:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/7c14d3e546834d9daadf51f2573dcfcd : SUCCESS in 54m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34e675efed3f4a56a3c16d05b06f6db6 : FAILURE in 36m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67479223bdee4648b2191d988f733e14 : SUCCESS in 4m 34s\n- devstack https://zuul.opendev.org/t/openstack/build/7a82e41309f64f12b5b882f4f85ef994 : SUCCESS in 32m 12s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/008bd64b7dd44b4196cb41496b0123cf : SUCCESS in 29m 11s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/0da3f197b491439f974a0247aa80efef : SUCCESS in 29m 59s\n- devstack-platform-fedora-latest https://zuul.opendev.org/t/openstack/build/152ac2b1b23945e5a0738743c9ebcefa : FAILURE in 54m 24s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/9305ed68e0544ffbb2a84a7dfd48755b : FAILURE in 56m 20s\n- devstack-platform-debian-bullseye https://zuul.opendev.org/t/openstack/build/d9d3d7f61e71472e873eaf6145ee4d3e : FAILURE in 52m 20s (non-voting)\n- devstack-platform-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2dd9e2e8dce4cb094bef61e2da4449f : FAILURE in 52m 45s\n- devstack-platform-ubuntu-jammy-ovn-source https://zuul.opendev.org/t/openstack/build/2de9af2223c24c01bc5693ba5d24a7d1 : FAILURE in 1h 04m 31s (non-voting)\n- devstack-platform-ubuntu-jammy-ovs https://zuul.opendev.org/t/openstack/build/b0bae9b0dfb04d3a9b0150bd44cd6891 : FAILURE in 48m 55s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/73ccaeb0ae114e92ba44634e1a427ffb : SUCCESS in 42m 16s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/e20dc26f142344ea9eb52162129491c4 : SUCCESS in 4m 12s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/428d7116481f41e0b25be4b22b6c747f : SUCCESS in 2m 50s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/58c98db65a30435baddcce81ea86692d : FAILURE in 43m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4563295e28df41d6a0c8675d28ed34e2 : SUCCESS in 33m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f5410dd0050439980d525dba9e3337a : SUCCESS in 58m 42s\n- neutron-linuxbridge-tempest https://zuul.opendev.org/t/openstack/build/813cbd8ab887474da6b80998d9c519b3 : FAILURE in 28m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f7af346217db4ab59baee6b553c67c6d : FAILURE in 31m 43s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c2db4812dd354c47a4b98364f1ddcd6b : FAILURE in 1h 08m 33s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23fd2526db8c435499064e67031a93ef : FAILURE in 37m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4985f2d22c14a4a9dd9f1a42f12f86d : FAILURE in 29m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c598a3484a9a4848af40a2f313688af0 : FAILURE in 32m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0b2edb48c77a1dc2a7b0f09ec5c3512f2292b213","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-01 19:03:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6341e7afc3222da7bed66b72a11c5bccb30a31e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-01 21:43:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7041c15abc224f37acfd738dd44a8f0e\n\n- grenade https://zuul.opendev.org/t/openstack/build/1dbaa2a4643347e3aad060c8c68f454e : SUCCESS in 40m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/227395c53e154d99bd3112a197440c0d : FAILURE in 15m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3b022064bb047c88842afc1ddf2371f : SUCCESS in 5m 23s\n- devstack https://zuul.opendev.org/t/openstack/build/85f312ecd92e4dc78e8a2fe244fb793f : FAILURE in 17m 37s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/cdc1d3d326d74e22b359a4a34c97d6ae : FAILURE in 30m 49s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/795c872fbbbc4b128726d578ec871284 : FAILURE in 14m 57s\n- devstack-platform-fedora-latest https://zuul.opendev.org/t/openstack/build/c87f179e356341ea9e697fc8515d18a1 : FAILURE in 37m 40s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/beb964ab7755456fba9769ef66a3425d : FAILURE in 18m 20s (non-voting)\n- devstack-platform-debian-bullseye https://zuul.opendev.org/t/openstack/build/46362102b02e488c85a33ad7a69cfb71 : FAILURE in 31m 15s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/e050341c048a443488bac75615ce85a2 : FAILURE in 35m 41s\n- devstack-platform-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00283377afde4cd091f96ff3a61e2970 : FAILURE in 16m 54s\n- devstack-platform-ubuntu-jammy-ovs https://zuul.opendev.org/t/openstack/build/92c9fdeb37dd4e8cb5111dee89c470d7 : FAILURE in 27m 01s (non-voting)\n- devstack-platform-openEuler-22.03-ovn-source https://zuul.opendev.org/t/openstack/build/ba280d91ad7d4bcc8bb4516aafb64b14 : TIMED_OUT in 2h 32m 21s (non-voting)\n- devstack-platform-openEuler-22.03-ovs https://zuul.opendev.org/t/openstack/build/a8cab313d4a14b7c9491e40d1eb42a79 : TIMED_OUT in 2h 33m 06s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/631e785f6f5245529b3bed61b2f37e01 : FAILURE in 29m 29s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/08f855bd9f5941e2a63434bb7f547900 : SUCCESS in 3m 39s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b4d5f04d7cc141b6a00a5ea63512ee9f : SUCCESS in 2m 08s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/63fec72030f34fac8010ca19a0ddcac9 : FAILURE in 20m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5ebec3fabc81446abdfa8dbecc273e4c : SUCCESS in 52m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/62cd35b444a248cdbb61bb58ec3da247 : SUCCESS in 38m 47s\n- neutron-linuxbridge-tempest https://zuul.opendev.org/t/openstack/build/67dee2b396ca407fb7c1fc0bc976aaee : FAILURE in 15m 54s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9bc5644873b54f07b8fc564e0d884769 : FAILURE in 14m 34s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9357699e9e1a4046bfa0c75bd2cbbf39 : FAILURE in 16m 37s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/362056172e054a75888f0d3ba2a1c81d : POST_FAILURE in 16m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/939a53db205541a8b542db4060d91a75 : FAILURE in 27m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1f2fe2074afe421b8ec13db1e6e7cf75 : FAILURE in 15m 47s","accounts_in_message":[],"_revision_number":2},{"id":"bb3d4849d66bb7f411ec7659807bbac796a9837d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-06-10 01:42:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"21aac244cbc3b7e7aaede8c0615fc6d46cfdb36c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-10 02:40:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7082687685cd40c694c65671c8f77982\n\n- grenade https://zuul.opendev.org/t/openstack/build/00ed3a9f2c5240cebed063b99b88b866 : FAILURE in 49m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d92dfb8792d04a8eab76cfdcd666f7bb : FAILURE in 32m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8539e80776ed426e9c3dd13f9b523faf : SUCCESS in 4m 48s\n- devstack https://zuul.opendev.org/t/openstack/build/31692adf03c2416c9338a7785f97ef30 : FAILURE in 31m 09s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/6fccd7c8217b4752aba27358f85038aa : FAILURE in 30m 32s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/cf900a7f73b949709790f2cd51237bac : FAILURE in 17m 51s\n- devstack-platform-fedora-latest https://zuul.opendev.org/t/openstack/build/22a1ad4cc880419cbb4d093f4d4fb4cf : FAILURE in 19m 13s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/c156d975ba4c410fb7f10afa571f88d7 : FAILURE in 27m 49s (non-voting)\n- devstack-platform-debian-bullseye https://zuul.opendev.org/t/openstack/build/c7ae1eb371274c5ca68e415397f7e476 : FAILURE in 22m 42s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/83c8e19f01b84b3f9924943f1526867d : FAILURE in 24m 57s\n- devstack-platform-ubuntu-focal https://zuul.opendev.org/t/openstack/build/079da919c7574cf0aad1015160b394ea : FAILURE in 21m 16s\n- devstack-platform-ubuntu-jammy-ovs https://zuul.opendev.org/t/openstack/build/8eb4d71573334912b69e133d2cb2dc47 : FAILURE in 16m 26s (non-voting)\n- devstack-platform-openEuler-22.03-ovn-source https://zuul.opendev.org/t/openstack/build/5e5347b2cafc42c3a8d72e43f23fc0f7 : FAILURE in 20m 23s (non-voting)\n- devstack-platform-openEuler-22.03-ovs https://zuul.opendev.org/t/openstack/build/ced8a369f7aa41b38c935bc5d28116e5 : FAILURE in 21m 26s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/518d54432f43474290697537e27d1236 : FAILURE in 18m 45s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/a39791a771374e94b68297daef6cc724 : SUCCESS in 3m 34s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/8b391ff23ed44a57be4271dfe99be0bf : SUCCESS in 2m 17s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/4983e3efecee45e08aeda78beddf7046 : FAILURE in 21m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b0a929661db641cdbc709d55f39719ce : SUCCESS in 51m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/44aaca9b7499421f8bf69fe2def00f39 : SUCCESS in 48m 06s\n- neutron-linuxbridge-tempest https://zuul.opendev.org/t/openstack/build/3b940ac5464c40418635509718ee3d35 : FAILURE in 14m 20s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e8aee6a5cd284619b716e6446c3283ce : FAILURE in 17m 04s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1ecef91438af456590fe55a3b2a972cd : FAILURE in 34m 42s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9de1edeed7a4580967f826f1eed9511 : POST_FAILURE in 28m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/edb60684451044019dafc1484e142894 : FAILURE in 18m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5570938171bb4ac18ed34df0c1dee440 : FAILURE in 21m 36s","accounts_in_message":[],"_revision_number":2},{"id":"fa76a82616b5116d494d782222599fac8eac1bde","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-08 16:51:54.000000000","message":"Abandoned\n\nduplicate of https://review.opendev.org/c/openstack/devstack/+/619554","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b864ebbee6db03172282134fb5205be9e77cfab9","revisions":{"570a8bb5e2fe0f3b05d56dce4b7aad6e3ec32078":{"kind":"REWORK","_number":1,"created":"2022-05-28 08:00:22.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/12/843712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/12/843712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/12/843712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/12/843712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/12/843712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/12/843712/1"}}},"commit":{"parents":[{"commit":"0e08e9b48c33855fe513655d5a018e891e06b49b","subject":"Merge \"Cleanup comment that should\u0027ve been removed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/0e08e9b48c33855fe513655d5a018e891e06b49b"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-05-28 07:58:59.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-05-28 07:58:59.000000000","tz":120},"subject":"Drop legacy neutron policy file workaround","message":"Drop legacy neutron policy file workaround\n\nThis is a relict from a long distant past, let\u0027s clean it up.\n\nChange-Id: I4808941ba4acaa15e1ec145209a88630bcac61c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/570a8bb5e2fe0f3b05d56dce4b7aad6e3ec32078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/570a8bb5e2fe0f3b05d56dce4b7aad6e3ec32078"}]},"branch":"refs/heads/master"},"b864ebbee6db03172282134fb5205be9e77cfab9":{"kind":"REWORK","_number":2,"created":"2023-04-01 19:03:24.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/12/843712/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/12/843712/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/12/843712/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/12/843712/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/12/843712/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/12/843712/2"}}},"commit":{"parents":[{"commit":"b10c06027273d125f2b8cd14d4b19737dfb94b94","subject":"Merge \"Add config options for cinder nfs backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/b10c06027273d125f2b8cd14d4b19737dfb94b94"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-05-28 07:58:59.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-04-01 19:01:52.000000000","tz":120},"subject":"Drop legacy neutron policy file workaround","message":"Drop legacy neutron policy file workaround\n\nThis is a relict from a long distant past, let\u0027s clean it up.\n\nChange-Id: I4808941ba4acaa15e1ec145209a88630bcac61c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/b864ebbee6db03172282134fb5205be9e77cfab9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/b864ebbee6db03172282134fb5205be9e77cfab9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
