)]}'
{"id":"openstack%2Fdevstack~867068","triplet_id":"openstack%2Fdevstack~stable%2Fxena~I9a138af94dedc0d8ce5a0d519d75779415d3c30b","project":"openstack/devstack","branch":"stable/xena","topic":"bug/1999183","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-12-09 12:24:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9a138af94dedc0d8ce5a0d519d75779415d3c30b","subject":"Pin tox\u003c4.0.0 for \u003c\u003dstable/zed branch testing","status":"MERGED","created":"2022-12-09 02:48:10.000000000","updated":"2022-12-09 12:25:25.000000000","submitted":"2022-12-09 12:24:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867068-bug/1999183","meta_rev_id":"78890eef271f35954d1e32248e0569ccc500b0ff","_number":867068,"virtual_id_number":867068,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-09 12:24:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-12-09 11:00:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-12-09 09:48:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-12-09 11:00:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-12-09 11:00:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-09 04:24:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-09 09:48:45.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-12-09 11:00:40.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"1cf9333192bebae9fc83d1f7c4f170cbf4b68839","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-09 02:48:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98cc548c6f2010359163b499f8f9a2d40dbbebc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-09 02:54:03.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"62857601c979e25253c95405575d461bf6a4906f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 04:24:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9df08fd905b94328a0af4b5b0fd017ec\n\n- grenade https://zuul.opendev.org/t/openstack/build/7a2a395251064ad886af488983369d2d : SUCCESS in 45m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dbce10cbffd841f48f08dc060a5db289 : SUCCESS in 1h 26m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c2b3c448e2444428a0bec1f1d056e02 : SUCCESS in 6m 02s\n- devstack https://zuul.opendev.org/t/openstack/build/959fff8bf3454e2298460dc74ab61805 : SUCCESS in 21m 10s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/9e42f3c064564347a1aba8f32f03f183 : SUCCESS in 27m 16s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/4aab62fb74534fe08fac0fbabc4a4ca2 : SUCCESS in 27m 27s\n- devstack-async https://zuul.opendev.org/t/openstack/build/96b50d41967d4e8d8189d52f98531bbb : SUCCESS in 1h 13m 30s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/e2831063813b4e90a93861f98f8f0e2b : SUCCESS in 31m 46s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/279ea95bf31f408bafecd647522616fe : SUCCESS in 3m 47s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e5dbf4455f994b0d9d8292525e69cd8b : SUCCESS in 2m 05s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/e0f8b41eceff448aacd7bda377f7f7de : SUCCESS in 58m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1abfc28f0361420b87b8dfe586c38923 : SUCCESS in 40m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7d0239d3d0764397b66bc55d8ac24deb : SUCCESS in 59m 55s\n- neutron-linuxbridge-tempest https://zuul.opendev.org/t/openstack/build/8af46a7084f74386906df04af248c359 : SUCCESS in 51m 38s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f879417ca1b746c7887cc6d6c93462e2 : SUCCESS in 57m 51s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5826fb04227c4b768ba7fc55109ff41c : SUCCESS in 1h 25m 36s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6b07a777a354ba98b4d674a9ffdbc7d : SUCCESS in 51m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfc432676a954a8fbb6e3012402486d1 : SUCCESS in 48m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dac3cd3ac7024f5e93a6472adff98033 : SUCCESS in 1h 11m 18s","accounts_in_message":[],"_revision_number":2},{"id":"c3afcf2083273caa4deb52f7a73d1a8629f5abae","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-12-09 09:48:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7cbd0a06cbba1bf34a512104b917bd7d7aa896e9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-12-09 11:00:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1 Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"b9c09e583f7c4c1a7023bbe058e3cd3dd9b89b71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 11:01:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8f520ba1ca5d4e6cfb1fd0c38ddda5d89874dcad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 12:24:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21850b0f96e04246affcbf1f7abf7185\n\n- grenade https://zuul.opendev.org/t/openstack/build/6ab5f82bb8744b0b901883a8bb282e8f : SUCCESS in 1h 02m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/84d57e26bef74285904ac8e303cef27c : SUCCESS in 1h 17m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9b8b7c1452b4e01b340ca0ae1a5c34a : SUCCESS in 5m 03s\n- devstack https://zuul.opendev.org/t/openstack/build/a746f0da333d4ce8a069373389f72db0 : SUCCESS in 19m 42s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/96956cbb7b9546c5b33be6f9a663acb3 : SUCCESS in 19m 34s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/320cb78a176246a387e090edda3cc656 : SUCCESS in 19m 23s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/54a4225e196748e29ea9c0a73c79a7b8 : SUCCESS in 33m 24s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/1a8a14713b8d4e38b66923cc56b75ef3 : SUCCESS in 4m 51s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e9736f6a593c44d4b8b545d9a79c0fa7 : SUCCESS in 2m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c36b69b4dbdb46cbbd60f9dbd226f6ac : SUCCESS in 59m 25s\n- neutron-linuxbridge-tempest https://zuul.opendev.org/t/openstack/build/da994a5b433544a89cf667595369b983 : SUCCESS in 53m 37s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/cebb776167aa4c2fb5d810a97e25e266 : SUCCESS in 55m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f3dcd4dbc19c44dea1bfc8bef3603226 : SUCCESS in 53m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/20919c73a0364e5c81984b752f9caa99 : SUCCESS in 56m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/787439b2e96d4dd2bf3ac36b82b6f2fb : SUCCESS in 49m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/982c71c0c698436d8f33615a79d94040 : SUCCESS in 1h 18m 22s","accounts_in_message":[],"_revision_number":2},{"id":"bc8dd08ef104d621e1e21d3a601cbe99cf9076a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 12:24:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"78890eef271f35954d1e32248e0569ccc500b0ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 12:25:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/457f6132e16345808b412717d9e66b4d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0300bdb11f0d4a9fa396aab683deccd0 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1c064ef80afa44ba096a3a6ac6130b64662d50c8","revisions":{"02f286a80a478d662e079fe066d311e4cf7264d8":{"kind":"REWORK","_number":1,"created":"2022-12-09 02:48:10.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/68/867068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/68/867068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/68/867068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/68/867068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/68/867068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/68/867068/1"}}},"commit":{"parents":[{"commit":"a298dc5fa094daa1235d223cc8453ac57e5d333e","subject":"Merge \"Stop running devstack-no-tls-proxy on stable branch\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/a298dc5fa094daa1235d223cc8453ac57e5d333e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-09 02:24:46.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-09 02:48:00.000000000","tz":-360},"subject":"Pin tox\u003c4.0.0 for \u003c\u003dstable/zed branch testing","message":"Pin tox\u003c4.0.0 for \u003c\u003dstable/zed branch testing\n\nTox 4.0.0 has some incompatible changes, epecially more\nstrict on allowlist_externals. Tempest recently changed\nallowlist_externals not to be *[1] causing the failure\non jobs where lib/tempest failing to run the tempest\nas command in virtual env.\n\n----------\nvenv: commands[0]\u003e tempest verify-config -uro /tmp/tmp.qH5KgJHTF4\nvenv: failed with tempest is not allowed, use allowlist_externals to allow it\n------\n\nWe do not need to test/fix the \u003c\u003dstable/zed branches with tox 4.0.0\nand pinning them with the compatible tox version of the time stable\nbrnaches were releaased is better way.\n\nThis commit proposes:\n1. Pinning the tox\u003c4.0.0 for \u003c\u003dstable/ze branches testing\n2. Workaround to unblock the master gate by pinning it \u003c4.0.0 but\nwe should make our testing compatible with tox 4.0.0 soon.\n\nRelated-Bug: #1999183\n\n[1] https://review.opendev.org/c/openstack/tempest/+/865314 devstack based job started failing to run tempest command on venv.\n\nChange-Id: I9a138af94dedc0d8ce5a0d519d75779415d3c30b\n(cherry picked from commit ba54baa4253f2c770760b59a4a8f28317e5e260d)\n(cherry picked from commit a3227ba0c0a6b951b301e556f78e634e6ce9e770)\n(cherry picked from commit 395f44708506b30b29b35dd655dddd434bac9a08)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/02f286a80a478d662e079fe066d311e4cf7264d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/02f286a80a478d662e079fe066d311e4cf7264d8"}]},"branch":"refs/heads/stable/xena"},"1c064ef80afa44ba096a3a6ac6130b64662d50c8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-12-09 02:54:03.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/68/867068/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/68/867068/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/68/867068/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/68/867068/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/68/867068/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/68/867068/2"}}},"commit":{"parents":[{"commit":"a298dc5fa094daa1235d223cc8453ac57e5d333e","subject":"Merge \"Stop running devstack-no-tls-proxy on stable branch\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/a298dc5fa094daa1235d223cc8453ac57e5d333e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-09 02:24:46.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-12-09 02:54:03.000000000","tz":0},"subject":"Pin tox\u003c4.0.0 for \u003c\u003dstable/zed branch testing","message":"Pin tox\u003c4.0.0 for \u003c\u003dstable/zed branch testing\n\nTox 4.0.0 has some incompatible changes, epecially more\nstrict on allowlist_externals. Tempest recently changed\nallowlist_externals not to be *[1] causing the failure\non jobs where lib/tempest failing to run the tempest\nas command in virtual env.\n\n----------\nvenv: commands[0]\u003e tempest verify-config -uro /tmp/tmp.qH5KgJHTF4\nvenv: failed with tempest is not allowed, use allowlist_externals to allow it\n------\n\nWe do not need to test/fix the \u003c\u003dstable/zed branches with tox 4.0.0\nand pinning them with the compatible tox version of the time stable\nbrnaches were releaased is better way.\n\nThis commit proposes:\n1. Pinning the tox\u003c4.0.0 for \u003c\u003dstable/ze branches testing\n2. Workaround to unblock the master gate by pinning it \u003c4.0.0 but\nwe should make our testing compatible with tox 4.0.0 soon.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/867069\n\nRelated-Bug: #1999183\n\n[1] https://review.opendev.org/c/openstack/tempest/+/865314 devstack based job started failing to run tempest command on venv.\n\nChange-Id: I9a138af94dedc0d8ce5a0d519d75779415d3c30b\n(cherry picked from commit ba54baa4253f2c770760b59a4a8f28317e5e260d)\n(cherry picked from commit a3227ba0c0a6b951b301e556f78e634e6ce9e770)\n(cherry picked from commit 395f44708506b30b29b35dd655dddd434bac9a08)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/1c064ef80afa44ba096a3a6ac6130b64662d50c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/1c064ef80afa44ba096a3a6ac6130b64662d50c8"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[]}
