)]}'
{"id":"openstack%2Fdevstack~938656","triplet_id":"openstack%2Fdevstack~master~I533b7444e5f71275ea3d5c18914e306b1dbbc5cb","project":"openstack/devstack","branch":"master","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-01-21 14:51:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I533b7444e5f71275ea3d5c18914e306b1dbbc5cb","subject":"Update glance image size limit","status":"MERGED","created":"2025-01-08 13:30:40.000000000","updated":"2025-01-21 14:52:19.000000000","submitted":"2025-01-21 14:51:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938656","meta_rev_id":"64c37ff43cd68b18c87ad546b7a6dc69e1664793","_number":938656,"virtual_id_number":938656,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-21 14:51:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"disliked":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":-1,"date":"2025-01-21 10:22:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2025-01-20 21:35:16.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2025-01-10 09:59:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":2,"date":"2025-01-13 19:42:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2025-01-20 21:35:16.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":1,"date":"2025-01-20 18:18:25.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-08 13:30:57.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-01-08 16:03:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 09:59:55.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2025-01-13 19:42:27.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2025-01-20 21:35:16.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"d547b283270168980057f2af74df28cf6dfd8281","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-01-08 13:30:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8cd55bb9fbb8c94b9c74af457de8949912016dd","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-01-08 13:30:57.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/7555735b18034b498b8df28fe81db217\n\n- noop https://softwarefactory-project.io/zuul/t/rdoproject.org/build/2eee035100c1467db24396a3ab938480 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"2aafa48d88741f1c1afd94e0768d49286d72ea66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 16:03:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01177079e62e4cc99899bb05176b789a\n\n- grenade https://zuul.opendev.org/t/openstack/build/2dddb065d8bd464aaa7affb48d8e3297 : SUCCESS in 52m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b02e0ed0b924a09bd94038ca67d296c : SUCCESS in 1h 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b211a3d2275494fb9927d62ef025413 : SUCCESS in 4m 26s\n- devstack https://zuul.opendev.org/t/openstack/build/79614eb8a95b46df9865fd8dc4e8c4b2 : SUCCESS in 22m 28s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/6d96d922a33c470fba7ae19c25a80bcd : SUCCESS in 19m 49s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/ff2c8c42b7684a8da2bed44667cff0da : SUCCESS in 30m 28s\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/406f3d95305948439401c47e741f1a54 : SUCCESS in 1h 49m 07s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/256c59b3584d4c8ab6c7f6179f287a79 : SUCCESS in 1h 37m 23s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/ea369f0611db4d1a8b931c8695527666 : SUCCESS in 1h 51m 11s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/4b83451385af4af89ffb139891ac3077 : SUCCESS in 1h 27m 08s (non-voting)\n- devstack-platform-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d090d95d54c4412694e31ccce3083b66 : SUCCESS in 1h 31m 51s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/2ccf2e42196c45f6b9730a0131a9ddc1 : SUCCESS in 28m 15s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/24ae68b1508c44999018acda7d2e7484 : SUCCESS in 3m 18s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b92d33881475473bba400afeb0c4d979 : SUCCESS in 2m 11s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/e90a80e7097d42debda88f313146a9af : SUCCESS in 1h 06m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6434ec7e0efe4dd48bd92610b6cd8281 : SUCCESS in 55m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/53c4d195f2f6404781ac272703b0c089 : SUCCESS in 1h 05m 17s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/739df063c4234fc2a4342df4587b422d : SUCCESS in 1h 21m 18s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6506d77b7a5e47228fe3e0e1a0a322c6 : SUCCESS in 1h 27m 31s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/679a39847f7a48c48064b12f20fc3bcc : SUCCESS in 1h 14m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd890e92abb84bc9abe672fd90814420 : SUCCESS in 1h 04m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f10e6eee091b4c329a5db48d1550db5b : SUCCESS in 2h 26m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ca936b0ca79f3e0895c638805dc78c250af996c6","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-01-10 09:59:55.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5f3801587b107ce666a8e43cb28a20b7265d0a0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-13 19:42:27.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da5f7a35463cbfdb6f3c0a92c032a5dae94e9d2b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-20 18:18:25.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"2b92ba4d6485adb3ce2dc3ae89bc061c870f152f","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2025-01-20 21:35:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf53721555191ae5656b0b67a3e5071dfa29fbec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-20 21:35:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"154052bffad284d01891bbb04d10be5c248e5ee8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 00:27:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce8f207e11f345a1bcc9f760af23e85f\n\n- grenade https://zuul.opendev.org/t/openstack/build/96426f36c1b8486aa20d292f56052333 : SUCCESS in 48m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/59f8f159a5be4378afafb4724c6583b8 : TIMED_OUT in 2h 37m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca027d878ac64a348fc59b14241e1aa5 : SUCCESS in 3m 22s\n- devstack https://zuul.opendev.org/t/openstack/build/cf9ecc7d41264b2880b0b58bd5c69878 : SUCCESS in 16m 55s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/177cfb851f164a57a2ad93738d4c4e8b : SUCCESS in 18m 44s\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/edec92b0a2b941b69638e1587783b405 : SUCCESS in 58m 17s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/3986c3a0387845659636448fc0d419ea : SUCCESS in 18m 59s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/337f02cb83054fa28bec1477d7382ccc : SUCCESS in 24m 46s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/c6ec127921b6497685ff1307d4300e31 : SUCCESS in 3m 22s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b3af5cef33ae44f2b9e13264aaad4ef8 : SUCCESS in 1m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/575580fda60643779b8a7426091e2319 : SUCCESS in 52m 28s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/480f690e5bde41ada080d45ba94d6115 : SUCCESS in 1h 04m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0c1b313876a8436296f33e44f337f0e5 : SUCCESS in 37m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d22b177404d74d42b4e9a004111b9ac1 : SUCCESS in 1h 01m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1218a8faaaf4bd3b0c55f1111c02017 : SUCCESS in 35m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0b6f0097ff9048f29a9333a657cf3693 : SUCCESS in 2h 11m 10s","accounts_in_message":[],"_revision_number":1},{"id":"7bdeb66952c0b537f687147a5eca0380445de14a","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2025-01-21 10:16:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd84cfca058f2004d4f4aa2256bb5d93f611c217","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-01-21 10:22:29.000000000","message":"Patch Set 1: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/72b1933fcf32479791fa322e9019388d\n\n- devstack-platform-centos-10-stream https://softwarefactory-project.io/zuul/t/rdoproject.org/build/b1644dc0c72645e6b265d8e2cc5c4eb6 : FAILURE in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"db7b666dbbc23ff5559279c43560a07daf38c89a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 12:23:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdca14f11e4848b9bfc87cb7c55d1347\n\n- grenade https://zuul.opendev.org/t/openstack/build/86179932e0e04f379abcbb3d637741f8 : SUCCESS in 58m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f212436befc41d484805f5a29287f78 : SUCCESS in 1h 05m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93aa34c909c1441ab099d1eb52323389 : SUCCESS in 3m 09s\n- devstack https://zuul.opendev.org/t/openstack/build/274983c8caa54b4783136b0a46c47612 : SUCCESS in 14m 53s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/6e5a12a4f52a4c49bdd8f40e79e013e1 : SUCCESS in 35m 57s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/c851ba55a9114e639fe8720d6dbaac43 : SUCCESS in 16m 16s\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4ea4e8d4acd463cacfc62b174cd9a85 : SUCCESS in 1h 59m 25s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/7b483e23d21a4071b8c83df15820cee8 : SUCCESS in 1h 30m 14s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/01a25c6d4f7345d6b71e879350fefc1d : SUCCESS in 1h 42m 30s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/89c730acac714d3698c29b8d85ae0638 : SUCCESS in 1h 30m 06s (non-voting)\n- devstack-platform-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d444028549f649d683049a67b68288c5 : SUCCESS in 1h 48m 48s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/cabcb2ee199a499a947d91fca20b1936 : SUCCESS in 25m 24s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/53d184a605664cb6b038474f8d921e7f : SUCCESS in 2m 57s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/347ced7b31e64bc3ba401edd8caea657 : SUCCESS in 2m 20s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/526b6236064040ee8b702b434db10e30 : SUCCESS in 54m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/75f7d8954cb747c897f9cd461bc568f0 : SUCCESS in 49m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a779579bec90419f9c3ef42d39be58ee : SUCCESS in 1h 02m 33s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2d3953e4fd5b4691aa352cd3f1d89c62 : SUCCESS in 1h 32m 14s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/497405480b0041e8ae2bab68f97bc98b : SUCCESS in 43m 25s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37ea0f7ead7d41a08640496a6c93dfcd : SUCCESS in 55m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff3ae37791874159a3d5f78573ec13a8 : SUCCESS in 1h 06m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8c6be4832dc54fe79ef67c9b15d43944 : SUCCESS in 1h 33m 27s","accounts_in_message":[],"_revision_number":1},{"id":"87b84fa6e2217e99094a97ff7b540a0ed2ab5d0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 12:24:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b401d19932b1f30bc0d01a01890f74f264fa05c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 14:51:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff9aeb392f4346149ab2f92d886b76e1\n\n- grenade https://zuul.opendev.org/t/openstack/build/f91fb6225ee24700b1257ad5a98e15d4 : SUCCESS in 58m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76626e1866b24254ab6862b8cbad0264 : SUCCESS in 1h 38m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0f5c65426d148c1bb6c71bc6690dc5f : SUCCESS in 4m 30s\n- devstack https://zuul.opendev.org/t/openstack/build/27098003c22a435a8581f0dfcd11dad5 : SUCCESS in 33m 02s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/bb72d81d37614223ae4a2415043738ca : SUCCESS in 28m 40s\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/15cdfdce61254dc595fef6ee8f12e3bc : SUCCESS in 1h 45m 24s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/11b1e7751262436b89c41c674dcb9c94 : SUCCESS in 15m 37s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/d53981c561fe42ac8755806dc1bfbdeb : SUCCESS in 22m 13s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/59f095fd2c1844f49316b247f966c0c6 : SUCCESS in 2m 57s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/46c940e8f9824e9386dc0f83d2749319 : SUCCESS in 2m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e95f8b26a774af9a061f4199a933deb : SUCCESS in 51m 11s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/3a49390cb28b412b8b9390c0a70e3d72 : SUCCESS in 1h 07m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a97aae11a60446729bde4cff322c976e : SUCCESS in 51m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/958753bfeecc45c4944487d182fc2fbc : SUCCESS in 1h 08m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4fa3b64eaf74dafa8b7d9ec200f3be1 : SUCCESS in 48m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4679d9b6f3be4b9695e6a96801e65cd5 : SUCCESS in 1h 59m 51s","accounts_in_message":[],"_revision_number":1},{"id":"2bc79dc5ebcc2b35e433b5416d5a3277df7d39e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 14:51:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"64c37ff43cd68b18c87ad546b7a6dc69e1664793","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 14:52:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccd09adc0b164f69b9a3c40c96019ac5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02ed9b11e2d34e97908d9f7a638989fa : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a1cdbc3c809f785ad01a3bbdfef8f552eafce30","revisions":{"9a1cdbc3c809f785ad01a3bbdfef8f552eafce30":{"kind":"REWORK","_number":1,"created":"2025-01-08 13:30:40.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/56/938656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/56/938656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/56/938656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/56/938656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/56/938656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/56/938656/1"}}},"commit":{"parents":[{"commit":"05f7d302cfa2da73b2887afcde92ef65b1001194","subject":"lib/cinder: Migrate cinder to WSGI module path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/05f7d302cfa2da73b2887afcde92ef65b1001194"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-01-08 13:13:40.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-01-08 13:30:12.000000000","tz":330},"subject":"Update glance image size limit","message":"Update glance image size limit\n\nThe image size limit for glance using the unified limits is\nset to 1000MB (~1GB). This is pretty low given that a volume\u0027s\nminimum size is 1GB and when uploaded to glance fills out the\nwhole limit.\n\nThe limit issue can also be seen by a recent tempest change[1]\nwhere uploading two volumes failed[2] across various jobs due\nto this limit.\n\nWe do have a config option in devstack\n``GLANCE_LIMIT_IMAGE_SIZE_TOTAL`` but that will need to be\nconfigured for various jobs and a 2GB seems to be a sensible\ndefault which this patch configures.\n\n[1] https://review.opendev.org/c/openstack/tempest/+/938592\n[2] Jan 07 23:05:33 devstack-ceph cinder-volume[909965]: ERROR\noslo_messaging.rpc.server cinder.exception.ImageLimitExceeded:\nHTTP 413 Request Entity Too Large: The request returned a 413\nRequest Entity Too Large. This generally means that rate limiting\nor a quota threshold was breached.: The response body:: Project\ndfe8648c188d46409349eac2c449c0b4 is over a limit for [Resource\nimage_size_total is over limit of 1000 due to current usage\n1024 and delta 0]\n\nChange-Id: I533b7444e5f71275ea3d5c18914e306b1dbbc5cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/9a1cdbc3c809f785ad01a3bbdfef8f552eafce30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/9a1cdbc3c809f785ad01a3bbdfef8f552eafce30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
