)]}'
{"id":"openstack%2Fdevstack~955277","triplet_id":"openstack%2Fdevstack~stable%2F2024.1~I0c48ee0b7daded1f10eda5861ae1ac0bcedd168c","project":"openstack/devstack","branch":"stable/2024.1","topic":"remove_opensuse-15","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-07-17 14:07:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0c48ee0b7daded1f10eda5861ae1ac0bcedd168c","subject":"Remove the nodeset \"devstack-single-node-opensuse-15\"","status":"MERGED","created":"2025-07-17 10:29:32.000000000","updated":"2025-07-17 14:09:21.000000000","submitted":"2025-07-17 14:07:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955277-remove_opensuse-15","meta_rev_id":"90c4016826abb4c4371f87b1ef99b05b94460366","_number":955277,"virtual_id_number":955277,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-17 14:07:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-07-17 10:32:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2025-07-17 10:29:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-07-17 10:32:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-07-17 10:32:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-17 10:29:47.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-07-17 10:32:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-07-17 12:18:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"24fdafcfe6217e099caa34b81b393e920da2a320","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-07-17 10:29:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68aa3d81f03bcebffc0cff585d20816590037b42","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-07-17 10:29:47.000000000","message":"Patch Set 1: Code-Review-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/devstack on branch stable/2024.1.  The\nproblem was:\n\n  The nodeset \"openstack-two-node-jammy\" was not found.\n\nThe problem appears in the the \"devstack-multinode\" job stanza:\n\n  job:\n      name: devstack-multinode\n      parent: devstack\n      nodeset: openstack-two-node-jammy\n      description: |\n        Simple multinode test to verify multinode functionality on devstack side.\n        This is not meant to be used as a parent job.\n  \n  # NOTE(ianw) Platform tests have traditionally been non-voting because\n  ...\n\n  in \"openstack/devstack/.zuul.yaml@stable/2024.1\", line 583","accounts_in_message":[],"_revision_number":1},{"id":"446d80c94788cb14692844066856794722749198","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-17 10:32:20.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b612ca5a3721569f1840344527ca70ebb39dd88c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 12:18:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5eb98518c5b44eb89394a2cde785e4a\n\n- grenade https://zuul.opendev.org/t/openstack/build/3a3f00431c834bd0a9e437df5378b355 : SUCCESS in 57m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e875a3eea4c24bb6aa864f78fbf12f73 : SUCCESS in 1h 24m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4996660ac67d43e7b721bcfe743ca382 : SUCCESS in 4m 16s\n- devstack https://zuul.opendev.org/t/openstack/build/0fcee69d160646618733fd83d893984b : SUCCESS in 20m 58s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/754abac9748f4929b04e6ea863693289 : SUCCESS in 20m 58s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/27ac8bc09fd6437da82d4ef2ae6f65e1 : SUCCESS in 18m 32s\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ed7bbe938194f03880787c090823f26 : SUCCESS in 1h 31m 59s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/a4e5e7690e6e48a7ac14892eb5dbeb28 : SUCCESS in 1h 27m 23s\n- devstack-platform-debian-bullseye https://zuul.opendev.org/t/openstack/build/1e3d27e3495b4f59af819bb311596ab4 : SUCCESS in 1h 34m 54s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/112db1a597ea499d8a11b66ee175875f : SUCCESS in 1h 39m 25s (non-voting)\n- devstack-platform-ubuntu-jammy-ovs https://zuul.opendev.org/t/openstack/build/1b930fb0eed546fba6cb69ef822f7bd3 : SUCCESS in 1h 24m 55s (non-voting)\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/f90404aa721d427397290494d214225a : SUCCESS in 29m 23s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/335d903e0c2c40d6978a10eaeabc57bb : SUCCESS in 2m 46s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/1fca2a696ee641df8409fb53e617298b : SUCCESS in 1m 51s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/8de3c639be4849d5b16e06ae7e7b883e : SUCCESS in 58m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/df8239d9885e45c2a50506b15049ddce : SUCCESS in 41m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/23be73e0d50242289ac1cd34b0bb79f4 : SUCCESS in 52m 03s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8b0ef545a23d4c96ab12887febfb06c6 : SUCCESS in 59m 26s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b5eb9c87c234658818207588e352d57 : SUCCESS in 53m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b317de52253e4923a7ee8cf87b2f9e41 : SUCCESS in 49m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c10e1f6d51434f96900354526e2c8e59 : SUCCESS in 1h 31m 54s","accounts_in_message":[],"_revision_number":1},{"id":"d440c5c90a85cb5eb43a26d9b8c3e41b7db88763","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 12:18:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8a6c32da7523b1fbb020a779213f74bbda0f489","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 14:07:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9db65f81def4da48250145e5bb181d6\n\n- grenade https://zuul.opendev.org/t/openstack/build/c6c9209b90ef423ca16bd67abb48f7a5 : SUCCESS in 54m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ad9fa89172a429598bb005969f8bd70 : SUCCESS in 1h 18m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cafd93f270b94f948bc71baf3fb4cad6 : SUCCESS in 5m 48s\n- devstack https://zuul.opendev.org/t/openstack/build/ea4f785f4e8f4874a59328e3b19120cc : SUCCESS in 21m 24s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/6be0d1f50dd44d0cb7709a2904923d61 : SUCCESS in 10m 17s\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/477edbc938a7477ca195fca39a254f7a : SUCCESS in 1h 36m 21s\n- devstack-platform-debian-bullseye https://zuul.opendev.org/t/openstack/build/526bc64e13af479f85da338793246f10 : SUCCESS in 1h 02m 33s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/1fdf34e4f1fd4f37a83e9903a984061d : SUCCESS in 20m 59s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/0c24e9bd9f2f4f829547ca0ca29570b6 : SUCCESS in 25m 55s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/aeb80abc30a14d02b3717155519ffa00 : SUCCESS in 3m 45s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c4fed878c11a437a8c7caac0b466a564 : SUCCESS in 3m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/71f1e7df9a4849469c2d237c6873ecc2 : SUCCESS in 51m 54s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/96790f9c9fb7458a9549337862a18f0d : SUCCESS in 56m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bc46889c459141f8a2a06cb63691cc66 : SUCCESS in 48m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4766daae637941298583f52752a6a262 : SUCCESS in 58m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a68013a3b54f4f4f853f62ffc6c10944 : SUCCESS in 25m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8f87cd94950e4e9182d5dc2fcdf8924f : SUCCESS in 1h 47m 09s","accounts_in_message":[],"_revision_number":1},{"id":"ef7bb38f1d9e458bae8d6e53ba102846f84b123d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 14:07:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"90c4016826abb4c4371f87b1ef99b05b94460366","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 14:09:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcf317ed719f4ad9829ec51face95119\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cecce250ba09482992bb4f5272e89bb1 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b04346b263a8f1c075c8c370d31a1b7cf131577b","revisions":{"b04346b263a8f1c075c8c370d31a1b7cf131577b":{"kind":"REWORK","_number":1,"created":"2025-07-17 10:29:32.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/77/955277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/77/955277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/77/955277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/77/955277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/77/955277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/77/955277/1"}}},"commit":{"parents":[{"commit":"f1c989f969e081f6262c7326ba6edaeb3500ed39","subject":"Drop OpenEuler jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/f1c989f969e081f6262c7326ba6edaeb3500ed39"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-07-17 10:28:57.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-07-17 10:28:57.000000000","tz":0},"subject":"Remove the nodeset \"devstack-single-node-opensuse-15\"","message":"Remove the nodeset \"devstack-single-node-opensuse-15\"\n\nThe nodepool \"opensuse-15\" is no longer available since [1].\n\n[1]https://review.opendev.org/c/openstack/project-config/+/955214\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I0c48ee0b7daded1f10eda5861ae1ac0bcedd168c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/b04346b263a8f1c075c8c370d31a1b7cf131577b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/b04346b263a8f1c075c8c370d31a1b7cf131577b"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
