)]}'
{"id":"openstack%2Fdevstack~956525","triplet_id":"openstack%2Fdevstack~master~I2df747d54d3cb395c245ecc2aa24dcbf395e7a46","project":"openstack/devstack","branch":"master","topic":"bug/2119496","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-08-05 12:19:47.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2025-08-05 09:20:35.000000000","reason":"removed on reply"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-06 20:07:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2df747d54d3cb395c245ecc2aa24dcbf395e7a46","subject":"Drop old libvirt group detection for Ubuntu/Debian","status":"MERGED","created":"2025-08-05 03:57:40.000000000","updated":"2025-08-06 20:08:55.000000000","submitted":"2025-08-06 20:07:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956525-bug/2119496","meta_rev_id":"f3e0aa113103275ff3b97f2394cd10e4ed268ea0","_number":956525,"virtual_id_number":956525,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-06 20:07:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-08-05 12:19:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-08-05 18:21:34.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2025-08-05 09:20:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2025-08-06 18:09:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2025-08-06 18:09:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-08-05 12:19:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-05 04:03:52.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-08-05 05:54:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-05 09:07:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-08-05 12:19:47.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-08-05 18:21:34.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-08-06 18:09:18.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"58c1d5155a597d1179da9c2539e275bbd8fe1125","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-05 03:57:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9bddbd1bf6497b53e160e0de32e9d2964f27ae9","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-08-05 04:03:52.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/ca71999949b1490eaf0526243bbc2894\n\n- devstack-platform-centos-10-stream https://softwarefactory-project.io/zuul/t/rdoproject.org/build/ce32b4dcb1044572b4916d2c8e995acf : FAILURE in 5m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06f8a8a0c48143ca5a6fc1bfc4b70886e01d72a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 05:54:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0b46df295434c1cb11c9d178b957234\n\n- grenade https://zuul.opendev.org/t/openstack/build/ada565caf5904df3b97dbedcd4a48def : SUCCESS in 1h 00m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9f6d11d45df4d04878aaf59deef80fb : SUCCESS in 50m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da2aafc669604c6893ed8914f6bb3f21 : SUCCESS in 3m 04s\n- devstack https://zuul.opendev.org/t/openstack/build/5d920354c5cd4257a4a717546486ebfa : SUCCESS in 21m 25s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/352311f0b3f640fd87c43ae08caa9bf2 : SUCCESS in 24m 06s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/125820f2de8f445fa345b5127f56ca1f : SUCCESS in 21m 50s\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e3586ba0ba84f20beeecfcc3167c42a : SUCCESS in 1h 40m 32s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/29cc3cb32d744f7f8028b53577daa1e2 : SUCCESS in 53m 25s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/8ebbbe443d68485f97ce8573d0bfae82 : SUCCESS in 1h 36m 50s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/845d392a45e945bb98eebffea210fb66 : SUCCESS in 56m 38s (non-voting)\n- devstack-platform-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7c4414741de4297b667754a558d1894 : SUCCESS in 54m 45s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/3bcb6792a97a49ea90e0c24a548ea48a : SUCCESS in 32m 47s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/5f2183785c5541ecaaed3ce3a7f02a95 : SUCCESS in 3m 37s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/bf4f18aa678b4602bc183b564ae89ba2 : SUCCESS in 2m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1290c8fbb5c8462ebdb42862a533acfc : SUCCESS in 51m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5c99ed910b854b93b06601fcc058bc21 : SUCCESS in 1h 01m 48s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6aee2107d64d4c46bd9b3ffd2b367729 : SUCCESS in 1h 11m 15s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/950006a503354057a9f95b90e2042387 : SUCCESS in 1h 13m 36s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3118dc775ade45c2858728f2b7e6f430 : SUCCESS in 31m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef9cf968ea1747c1872a58826d834e9b : SUCCESS in 45m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c6dcf35fea8045e582c7ea1c86782897 : SUCCESS in 1h 51m 53s","accounts_in_message":[],"_revision_number":1},{"id":"435af91c39c976f2646a98100ed6bd6f9bb370a3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-05 09:07:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc59dbae551becc157c55a9095f333f9c4246c70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-05 09:13:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"cd8fd37bf161009bb6e7b3fd4ddd8f4acf1c3e43","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-05 09:13:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7952e1476f7fc78add5806a64cb4d5db54878d56","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-08-05 09:20:35.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/087ac39a5206423da977a35f4e6ab7a2\n\n- devstack-platform-centos-10-stream https://softwarefactory-project.io/zuul/t/rdoproject.org/build/f1d7e915b062466f82f1a323aefbce08 : FAILURE in 5m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"709f151993e546e396d1945ec0f36b7e026ae2e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 11:01:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c94536aab17a4d7992af25ccdf7fab64\n\n- grenade https://zuul.opendev.org/t/openstack/build/460c46241fa540d3a93d873aeaaefc98 : SUCCESS in 57m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6bdef918e574903a5a2266afae7c734 : SUCCESS in 51m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bccdc3e94cd465fa6a075691848a431 : SUCCESS in 6m 13s\n- devstack https://zuul.opendev.org/t/openstack/build/7756ff1851704ef7bd56b3e19c834ad2 : SUCCESS in 9m 27s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/dc80a2a2499b4f7db568ffd73df2928b : SUCCESS in 22m 20s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/f945e480b7ff4c34b00ab78bda30e23a : SUCCESS in 32m 01s\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/81cd3d5f7c9e4a5f94fede4766821ca6 : SUCCESS in 1h 39m 10s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/9a46c8c7d52a4385b241672201a5d6f8 : SUCCESS in 1h 33m 50s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/b7f38cb94da24b4488fcb60c92546772 : FAILURE in 44m 32s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/401fa6b0620b4045b5f6660d682c8453 : SUCCESS in 1h 42m 50s (non-voting)\n- devstack-platform-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b0e55e5e9124c3fbc9a2f867ff81dc2 : SUCCESS in 1h 30m 02s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/ba03852581ae494aa263c164173decb7 : SUCCESS in 30m 34s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/6e8afe945f9e475eb0bcbd85a052725b : SUCCESS in 2m 39s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/7704384cb8ba4b788347c12c69c57746 : SUCCESS in 1m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2bbf9904c0984bfcaaa3716841bf0bca : SUCCESS in 49m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/80331360975f4985b2a597dc9ede9391 : SUCCESS in 28m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/eb5d893f90d1489c8314100e38ef55b4 : SUCCESS in 1h 12m 02s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a38ae358006d4229b08f43c663752242 : SUCCESS in 39m 57s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe0583e3d5704715bab8529b61cff073 : SUCCESS in 31m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d30865d68a624790aa408e844a9696bf : SUCCESS in 1h 25m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1961d338d28345a0a2a49f87345bfa40 : SUCCESS in 1h 43m 47s","accounts_in_message":[],"_revision_number":2},{"id":"474fbbf8c1cc1e0ad72ceeda8c850be87f0ec5dc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-05 12:19:47.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"cd71ab1f8d6845debbd06394aa7beb346393b69f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-08-05 18:21:34.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"776cd920d1564d4384d6aa74f94a80d8cdc43d86","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-06 18:09:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0b86cb2caa1040afe312cd7be11dc6057f8efe6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 18:10:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"60a68f8008309f7e5143de2ca9edd0d7fe5fc191","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 20:07:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f54b38f18bb4458badbe2134fb714280\n\n- grenade https://zuul.opendev.org/t/openstack/build/987f141af02e4753a30695d5c2b878ab : SUCCESS in 56m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d2197215dcd5480db596ab9082a2b4e9 : SUCCESS in 1h 28m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/841929ef45e048ccbbbaa8bde32b0794 : SUCCESS in 7m 31s\n- devstack https://zuul.opendev.org/t/openstack/build/bae88516e58b4bb6aa31ae5ba1b16199 : SUCCESS in 22m 26s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/5e3b54f65cf34721b8f4060adf88f4e6 : SUCCESS in 23m 35s\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/c1a214d7bdd744359e41f4e4cb00625a : SUCCESS in 1h 32m 47s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/dfc413c849194206bcff81ac5aba6657 : SUCCESS in 23m 26s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/9f07bb327f93480aa313cacb8fa4766f : SUCCESS in 32m 00s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/4a422e1781f54027b1f53033d7e8e847 : SUCCESS in 2m 45s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/1c1d19add9e949b9a04370914446956a : SUCCESS in 1m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4bfa3f56c0d944738d1fe9123392cd48 : SUCCESS in 1h 03m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/64ef8aebb90b4c568bcdb4f5eba12164 : SUCCESS in 48m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/43a1b188fd5748eeba6808488507e1cd : SUCCESS in 1h 04m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c6af6c70de845c2b3e3ad635b792972 : SUCCESS in 53m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2434a0392edf48ce968211b70b854781 : SUCCESS in 1h 53m 34s","accounts_in_message":[],"_revision_number":2},{"id":"2b816b850c296b805c13b1d1443aea26b93d88f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 20:07:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f3e0aa113103275ff3b97f2394cd10e4ed268ea0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 20:08:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1529765ac5f04c3eb70f3b7288c61eb1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aec3acf4068441ba7f8969aea34d7d0 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"90b0a6760ba95e8b2fa4a43dda1d0eeb56c00f2d","revisions":{"c7e25c8d1ccd81299108fbfd7ae41b400de6ab82":{"kind":"REWORK","_number":1,"created":"2025-08-05 03:57:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/956525/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/25/956525/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/25/956525/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/25/956525/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/25/956525/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/25/956525/1"}}},"commit":{"parents":[{"commit":"5c338f47d57fe849215d6b9f5c1f4eb53c193ab1","subject":"Fix uninialized os_VENDOR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/5c338f47d57fe849215d6b9f5c1f4eb53c193ab1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-05 03:55:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-05 03:56:01.000000000","tz":540},"subject":"Drop old libvirt group detection for Ubuntu/Debian","message":"Drop old libvirt group detection for Ubuntu/Debian\n\nThe \"libvirtd\" group was used in quite old Ubuntu such as Xenial, and\nthe \"libvirt\" group is used instead in recent versions.\n\nChange-Id: I2df747d54d3cb395c245ecc2aa24dcbf395e7a46\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/c7e25c8d1ccd81299108fbfd7ae41b400de6ab82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/c7e25c8d1ccd81299108fbfd7ae41b400de6ab82"}]},"branch":"refs/heads/master"},"90b0a6760ba95e8b2fa4a43dda1d0eeb56c00f2d":{"kind":"REWORK","_number":2,"created":"2025-08-05 09:13:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/956525/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/25/956525/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/25/956525/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/25/956525/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/25/956525/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/25/956525/2"}}},"commit":{"parents":[{"commit":"5c338f47d57fe849215d6b9f5c1f4eb53c193ab1","subject":"Fix uninialized os_VENDOR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/5c338f47d57fe849215d6b9f5c1f4eb53c193ab1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-05 03:55:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-05 09:13:02.000000000","tz":540},"subject":"Drop old libvirt group detection for Ubuntu/Debian","message":"Drop old libvirt group detection for Ubuntu/Debian\n\nThe \"libvirtd\" group was used in quite old Ubuntu such as Xenial, and\nthe \"libvirt\" group is used instead in recent versions.\n\nChange-Id: I2df747d54d3cb395c245ecc2aa24dcbf395e7a46\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/90b0a6760ba95e8b2fa4a43dda1d0eeb56c00f2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/90b0a6760ba95e8b2fa4a43dda1d0eeb56c00f2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
