)]}'
{"id":"openstack%2Fdevstack~957153","triplet_id":"openstack%2Fdevstack~master~Iec300311cab1b1a2fa124879aa3dc6a57a6a706b","project":"openstack/devstack","branch":"master","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-12-17 16:38:52.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2025-12-17 17:44:57.000000000","reason":"removed on reply"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-12-19 10:41:00.000000000","reason":"Change was submitted"},"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2025-12-19 08:25:45.000000000","reason":"\u003cGERRIT_ACCOUNT_10459\u003e replied on the change","reason_account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}}},"hashtags":[],"change_id":"Iec300311cab1b1a2fa124879aa3dc6a57a6a706b","subject":"Fix ownership of keyring file","status":"MERGED","created":"2025-08-12 16:04:33.000000000","updated":"2025-12-19 10:42:24.000000000","submitted":"2025-12-19 10:41:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957153","meta_rev_id":"c504a2334858301f168532135a5d6d06d83081ea","_number":957153,"virtual_id_number":957153,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-19 10:41:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2025-12-17 16:38:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2025-12-17 17:44:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2025-12-19 08:25:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2025-12-19 08:25:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-12 16:11:00.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-08-12 16:26:03.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2025-08-12 17:58:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-28 15:55:09.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2025-12-17 16:28:03.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"1175e9a21165f2d4175601ec365aabd8d12c0b5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-12 16:04:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dcd2c0466a1e155d0748bd765404c6d40c295f31","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-08-12 16:11:00.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/5260ff963663483d9bc996f917d5c5f0\n\n- devstack-platform-centos-10-stream https://softwarefactory-project.io/zuul/t/rdoproject.org/build/2065b9a0edbd40219f907a97d6f22c64 : FAILURE in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"335d09536267709cbaca799c29b304afe9a79877","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-08-12 16:26:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f057db1f7a7e0ae84ad411f40bf6391f0f1e278e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 17:58:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6fc19e0d3aac4677b1b2d65484fd6ce2\n\n- grenade https://zuul.opendev.org/t/openstack/build/2d754cf04bfa4ceb9ae953289dc04463 : SUCCESS in 31m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b2fb8ac3d76461b9419444c86e2a6fe : SUCCESS in 1h 44m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08772070a6ce4bfb8a1f994979ae7fba : SUCCESS in 6m 00s\n- devstack https://zuul.opendev.org/t/openstack/build/df21e01b11b94d42b9b21dfdf94dc196 : SUCCESS in 19m 16s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/23e7f8bd59ae4a999935c00717348dad : SUCCESS in 22m 33s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/8a0501c2a2ac4a049a4af75cb7b1ab55 : SUCCESS in 34m 08s\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/030ae21b087141c0aac51be1b3465650 : SUCCESS in 1h 32m 31s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/31fc48fbc5d44a9bb2f65d828b9f70a1 : SUCCESS in 1h 45m 10s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/a75304d5855f4158b37c10f9c0ec5edd : SUCCESS in 1h 46m 30s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/8a7dd6d099914a1dab0daf0b1aaa1e46 : SUCCESS in 1h 37m 00s (non-voting)\n- devstack-platform-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7248a5950f2c4d1b9f2d4e1e4b3be800 : FAILURE in 1h 16m 06s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/b8e9f71e2f6842de82d47cbf2f1424d9 : SUCCESS in 33m 37s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/18654c86b73f4b8a8bbcdea11c4bbcce : SUCCESS in 2m 50s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/96b4523bcff34cd1ba7b354987cd3646 : SUCCESS in 2m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2a4e527bf25c4070a38ce3aa54080275 : SUCCESS in 51m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fea405689cd44a65b10b287263960cde : SUCCESS in 1h 02m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fa1163d0299a403aa4a5258790a26316 : SUCCESS in 1h 13m 50s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c840bce6c6d747f8a34b1cc28ca31c87 : SUCCESS in 1h 15m 24s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0facc515a524431ab01401aedded36c9 : SUCCESS in 1h 03m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17834e53d4604dea91e5fabfa1c8b7ba : SUCCESS in 1h 01m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4cf9924ce94146a6987393f979abb6da : SUCCESS in 1h 47m 04s","accounts_in_message":[],"_revision_number":1},{"id":"657b19616778898a37cefdb1832c328888fa01ac","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 19:08:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/956802a801a5482ead1d575ff74641ef\n\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7f9f24491f9f4828ab3917c81525313a : SUCCESS in 2h 14m 22s\n- nova-next https://zuul.opendev.org/t/openstack/build/a62673e2292847d09e96281631037bc9 : SUCCESS in 2h 22m 57s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/93f36c5a5b0649d39c5684d98dcf59c3 : SUCCESS in 1h 47m 49s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/fdf89f97f92b4ee9bdca065498a5201d : SUCCESS in 1h 07m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e7fb2d9b347c4a3ab6dee27505c2daf4 : SUCCESS in 1h 44m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/233db4a6a2f640e982b885aa75e53c3f : SUCCESS in 1h 47m 34s\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/e447a5f30be74831a60077cafb84478d : SUCCESS in 50m 18s\n- devstack-no-tls-proxy https://zuul.opendev.org/t/openstack/build/24cdc00225744250b9f7a12f95f6c296 : SUCCESS in 2h 00m 15s","accounts_in_message":[],"_revision_number":1},{"id":"e5999c2cad8d87eabf6f74047faa2855496b0607","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-13 07:38:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"26413e400e3e419cf62f6ecbe62ee1d6479f2d4c","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-08-13 07:40:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/79a2abac0e26449b83b92273d48a1d0e\n\n- devstack-platform-centos-10-stream https://softwarefactory-project.io/zuul/t/rdoproject.org/build/9286121e35e64960bd3567714d1915f6 : NODE_FAILURE Node request 200-0007977905 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"69be8dfddcfc6df101b35f9e117a846f48df2954","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 09:36:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e31dcd525fe5419d91e9fd0e8b9f1c1d\n\n- grenade https://zuul.opendev.org/t/openstack/build/1c84bf28c36d4d8a9fa71aac15b2869b : SUCCESS in 55m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9042d14c5bce4a55aad54cc3bacf40e8 : SUCCESS in 1h 46m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4358ec6023654cd6acef9deb2c54847a : SUCCESS in 6m 51s\n- devstack https://zuul.opendev.org/t/openstack/build/c68d7a376cc54e4283f45209f5f51b53 : SUCCESS in 16m 43s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/725f515aa60d43bb92758b3c07c6fce3 : SUCCESS in 23m 08s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/eda8ba27a94f463d94980d228b3bff8b : SUCCESS in 24m 30s\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/b7b0c293b2d44817b55996b1fb5819cc : FAILURE in 1h 13m 09s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/0922079c3e58486db7c685ecf798677e : SUCCESS in 1h 51m 32s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/99384c1eb1544b608d9f8fc42753e49e : SUCCESS in 1h 27m 30s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/b9a87f8c53054c1eb6bf560f56fb8371 : FAILURE in 1h 31m 19s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/f2522278219b45b5a8502780cedf951c : SUCCESS in 1h 28m 43s (non-voting)\n- devstack-platform-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/432047045f974b349a4a2011481d1850 : SUCCESS in 1h 38m 05s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/c8e425f16adf4e8aae56dc7f30f63ee4 : SUCCESS in 34m 00s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/c020edd213324c85b6c78d8be182d534 : SUCCESS in 1m 42s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/93f1779d7d7f4079853249282aa71c07 : SUCCESS in 3m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2f59f18afec44c50ad4a7177c97ce76a : SUCCESS in 36m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/23ee9a3d68e6479b9154b046a0c21bd7 : SUCCESS in 1h 04m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/94a34673db4d45308f6df4a1cd696187 : SUCCESS in 1h 33m 08s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2f87919e4274481caa5e1bfdaffe3b75 : SUCCESS in 1h 10m 03s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca34e714656b45e5b51762570f47f306 : SUCCESS in 1h 03m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3882283e53f04e8bb61cc2dcced88a2a : SUCCESS in 57m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a02dc4abf68c4a5e946870ef55e36e03 : SUCCESS in 1h 53m 19s","accounts_in_message":[],"_revision_number":1},{"id":"fef8d10103a07c249b042bb83dcfabb571002aa4","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-10-28 15:55:09.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f4bc1479cac7460ab281180d3c55d7474f508c78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-29 14:13:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6783208067f4af954a8cd6ff0d18ec5f1bc81ed9","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-10-29 14:45:48.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f2cd3060c46632b0f3fda63a58316464fdf93ca0","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-10-29 15:31:04.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/73a7859ae6c94f4db0e44557ef5475d8\n\n- devstack-platform-centos-10-stream https://softwarefactory-project.io/zuul/t/rdoproject.org/build/aac2033a72564919a46e6b3f94f65bfd : SUCCESS in 1h 16m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"da361d206577fc8d4f48cb5e2d58aa199f597bca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 16:52:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/131a476850f24be095c909f23bf6fcad\n\n- grenade https://zuul.opendev.org/t/openstack/build/1b8c83ae3b714ab697d4b5c1a75ca9c1 : SUCCESS in 35m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ead3a9655ee44d69e9813281318023f : SUCCESS in 1h 02m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9c4ecc225574aeea5e7f07a26f2b270 : SUCCESS in 8m 02s\n- devstack https://zuul.opendev.org/t/openstack/build/260b9f0cce0c4c57a45e68c73bc47de8 : SUCCESS in 26m 01s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/9ee9cb7d91f24e7d911a03d68d12a4c7 : SUCCESS in 24m 43s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/e55d732197a24a74b11a030ce7eb82a8 : SUCCESS in 12m 39s\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/e27c4780ed3240dc97d8c2e9dfe89615 : SUCCESS in 1h 26m 17s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef2dacabd7614d6c959f554e2a3bb7d1 : SUCCESS in 1h 49m 19s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/9668026edf6e42c683c4698834a26701 : SUCCESS in 1h 52m 00s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/27ea7bd6e9bc4c11aa0a0723722d8dea : SUCCESS in 53m 57s (non-voting)\n- devstack-platform-rocky-red-quartz https://zuul.opendev.org/t/openstack/build/10d4bd3d59a34651a4c75a580fa38c58 : SUCCESS in 1h 05m 14s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/4cf997f3e1ff4a9cac850286990823be : SUCCESS in 53m 18s (non-voting)\n- devstack-platform-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b43794a8029b40909e33897f850258d0 : SUCCESS in 1h 56m 22s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/7e283412b45c40d5b7862b2adf18e06d : SUCCESS in 32m 49s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/b32f6f28da714c1eb535693c16952b81 : SUCCESS in 3m 01s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5629b16d6eb047f592dff9b70f09a2db : SUCCESS in 2m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eb80c8e3cf694dcfafea975673ce10b3 : SUCCESS in 50m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dfeaf4d7939b4b04b87574e1b6918da0 : SUCCESS in 1h 12m 03s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3556e940f08f4454b07162f2c70e2356 : SUCCESS in 1h 20m 22s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ce59d30aa43b4bc48449021540a02033 : SUCCESS in 1h 12m 34s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ca5ead2d8114824a20c6f03e240fe32 : SUCCESS in 41m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2153aebe81584fa2a62d68632bb915cb : SUCCESS in 50m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e9281a28f9ed4868b863a3b6ac93a549 : SUCCESS in 2h 34m 00s","accounts_in_message":[],"_revision_number":2},{"id":"b591c2ecd07609bb46aa5f05c4dc8a6b65738552","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-12-17 16:28:03.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a507178ebce72a2170b85797174dcbef2e6260ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-17 16:36:14.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b273d217b712ac15ad98f78ad3ffaa4479df6b0c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-17 16:36:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b4c6e2c4ed25f2830514ba38d795e77e9a7d1dff","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-12-17 16:38:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6ba222d36cd79b008c3f46f34f4bb42ad3ce2548","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-12-17 16:41:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5102a608f0fb03152366bd442c268e63e57ca408","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-12-17 17:44:57.000000000","message":"Patch Set 3: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/06e45017d94f450f9b200aea8d9cc784\n\n- devstack-platform-centos-10-stream https://softwarefactory-project.io/zuul/t/rdoproject.org/build/24692b2eb1cb4d7a860d9d8556578589 : SUCCESS in 1h 07m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4a216acaa0099b0a9fcb3bf2bed696a52358f832","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 18:26:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8450007fe8a84c4ea763f2cdb5b8b240\n\n- grenade https://zuul.opendev.org/t/openstack/build/963449a93fd843bcaa27170f031800a6 : SUCCESS in 58m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/634598f467844f2d88f501df082ff8e2 : SUCCESS in 1h 36m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2fdb8b5f80348fc8bb4d9d06b3d962d : SUCCESS in 6m 16s\n- devstack https://zuul.opendev.org/t/openstack/build/4ea754b67dc94bacaada0e327ced2ac0 : SUCCESS in 11m 33s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/de733e6dc19842009d7bf623bde48713 : SUCCESS in 19m 24s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/b1b22321ddb040c78cdad130fb0741a0 : SUCCESS in 20m 34s\n- devstack-platform-almalinux-purple-lion https://zuul.opendev.org/t/openstack/build/a952760fdc5a4c7ab21d88b005a379ea : SUCCESS in 56m 14s (non-voting)\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/3d198454d5cb440b8fc18b863319a9cf : SUCCESS in 1h 34m 01s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c9bf248cbc24f13919dd76c023577df : SUCCESS in 56m 47s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/a7bd33ef00c84e308c4f9289c97fbff5 : SUCCESS in 1h 30m 36s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/ec3ab78683ae4b628ec73705a69bf686 : FAILURE in 45m 06s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/fc0b6873e8764db688d64efad35360e1 : FAILURE in 1h 12m 28s (non-voting)\n- devstack-platform-rocky-red-quartz https://zuul.opendev.org/t/openstack/build/a39fa4cd754340f5b7264a91cc5da61d : SUCCESS in 1h 01m 48s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/94b0978fe62a4814b4383dc7e8f08a82 : SUCCESS in 1h 34m 21s (non-voting)\n- devstack-platform-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b75001ababd543298ec0182a42fb4e4b : SUCCESS in 1h 44m 47s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/39751624e38447bd9221c73499e1fad7 : SUCCESS in 19m 31s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/dbdb8aacb8ad4492a59a886a48af72cf : SUCCESS in 3m 55s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ccc23d847ec14ca4afbd41617809c0da : SUCCESS in 2m 18s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/e48180c63d004371b1ef1b1b8b506d94 : SUCCESS in 1h 21m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f8a50b752cc742ce92076a023a23f9c2 : SUCCESS in 22m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e45432fa8f344908c93f5e042a545e6 : SUCCESS in 1h 07m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/40c1d12886a94e20a2018e9f5ef8964a : SUCCESS in 55m 58s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d38010b11f1a44fabdb9a6f540f0e3b8 : SUCCESS in 1h 09m 50s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/326ef56076a043e6b80e0bbf27e8b3bd : SUCCESS in 1h 03m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9af1536afe704216b8cce2c0399a4e63 : SUCCESS in 1h 04m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a3a9cc3f30ab48f8aa520ac38a72bf50 : SUCCESS in 1h 08m 29s","accounts_in_message":[],"_revision_number":3},{"id":"cc8b7a41a932506ccb49835d9a145b309afce1e1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-18 06:30:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8163536c81f5eb0cd6b16c40a2152660a6f94d00","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-12-18 07:40:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/75d3f544e1064a22a0e11a4aea6475aa\n\n- devstack-platform-centos-10-stream https://softwarefactory-project.io/zuul/t/rdoproject.org/build/219b518b9657483aa85ca0c380df8597 : SUCCESS in 1h 08m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d7190ad67e0b749ed9a78cc2778b4a203c1318bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-18 08:34:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6e49dd09e74462ba9b07f0afa90bab0\n\n- grenade https://zuul.opendev.org/t/openstack/build/a7ce4a73897d43e4968f7fc4d3129063 : SUCCESS in 30m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0f53bd96b384d0185d462fbd6880611 : SUCCESS in 58m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df35882721f24deb8aad8dd2570c0073 : SUCCESS in 3m 12s\n- devstack https://zuul.opendev.org/t/openstack/build/2a0a9ab1c60b42869440bc48e41f5cf0 : SUCCESS in 24m 28s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/7d7671067ffb42d895d045ad285d37a6 : SUCCESS in 25m 59s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/cd8903951df74c51a83136c0471b8883 : SUCCESS in 23m 22s\n- devstack-platform-almalinux-purple-lion-ovn-source https://zuul.opendev.org/t/openstack/build/73b0a77eca7a433382b856879e3b2600 : SUCCESS in 1h 00m 10s (non-voting)\n- devstack-platform-centos-10-stream https://zuul.opendev.org/t/openstack/build/efaf503adaaf40b6aa9fa75f4b6bb0e7 : SUCCESS in 1h 37m 20s (non-voting)\n- devstack-platform-centos-9-stream https://zuul.opendev.org/t/openstack/build/64e2f26fac364030904f7eeb0a4b3398 : SUCCESS in 1h 44m 19s (non-voting)\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/20ff210299564d56999e909fc499c7ff : SUCCESS in 1h 40m 08s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/7342c6b8a7d547c9a9dec7c4c550a757 : SUCCESS in 1h 34m 36s\n- devstack-platform-rocky-blue-onyx https://zuul.opendev.org/t/openstack/build/06206ce8a1a64a239e3eeb21cf19b743 : FAILURE in 1h 20m 56s (non-voting)\n- devstack-platform-rocky-red-quartz https://zuul.opendev.org/t/openstack/build/571dd56c678b4c1d981106e063ed90cf : FAILURE in 1h 22m 37s (non-voting)\n- devstack-platform-ubuntu-noble-ovs https://zuul.opendev.org/t/openstack/build/fe4246afb6054423b9985e020bbad1e5 : SUCCESS in 1h 51m 31s (non-voting)\n- devstack-platform-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14c32d674e4a46df8429c76700ead37e : SUCCESS in 1h 47m 54s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/1e951cb3a11e45df962cf3f047a18770 : SUCCESS in 17m 21s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/6874fc62f72841cea0b96a14c0ae36eb : SUCCESS in 2m 06s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b978195964f548f1b6e39e3678c4f905 : SUCCESS in 2m 10s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/dfd74b7c3fba4c53802146909a827fcc : SUCCESS in 55m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a489725da9cc4674bc21ccd304eb6e0f : SUCCESS in 24m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d732ad048d9c4b2c97aa9c4602139430 : SUCCESS in 56m 34s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5082cf5585a44577bcd4059dcdddee56 : SUCCESS in 1h 27m 47s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/aebe35218fb54ec8afd86977f42bb9c6 : SUCCESS in 1h 12m 02s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/032d35e9ba1a424fa09732b1b7d9a277 : SUCCESS in 1h 06m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44923babeab244b48d3c5d81b74cc2b9 : SUCCESS in 58m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5a61867299d94ecca44a1cc25031c867 : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":3},{"id":"339d74f34441fbb40f4084618100784ab90ef0e9","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-12-19 08:25:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a018d325da2a590d9706e050b086d032af71fecf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 08:26:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f16b9fed9d91fadced780f47147e244e85d98bb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 10:41:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/419223f2d00e489995122a6631a92282\n\n- grenade https://zuul.opendev.org/t/openstack/build/3bda5f7acd914c5384ede6aa1c0a4128 : SUCCESS in 50m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9fea4d7dbdb5440b8f2567ba8c381609 : SUCCESS in 55m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61edf4b8165243c4889bc9368088b15d : SUCCESS in 3m 27s\n- devstack https://zuul.opendev.org/t/openstack/build/a7a12fb5fd6d464a9a5c92b151a7b3d8 : SUCCESS in 26m 08s\n- devstack-ipv6 https://zuul.opendev.org/t/openstack/build/8c83704e333546dc9896bf8492095100 : SUCCESS in 15m 06s\n- devstack-platform-debian-bookworm https://zuul.opendev.org/t/openstack/build/e7d3fd5b6ee843a8bf8472e861e8c5cb : SUCCESS in 56m 20s\n- devstack-platform-debian-trixie https://zuul.opendev.org/t/openstack/build/09181c8ce3024a65bc76ba5e5afff40b : SUCCESS in 1h 26m 43s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/2822b2d1789849c58f5f04af9f49c46e : SUCCESS in 10m 35s\n- devstack-multinode https://zuul.opendev.org/t/openstack/build/d36f6ad7e82b42469de95175193e67de : SUCCESS in 16m 36s\n- devstack-unit-tests https://zuul.opendev.org/t/openstack/build/2cd9edff4619466b9c7b930e34e11fd6 : SUCCESS in 1m 43s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ccc3a746ef4d40539ff044430589a0e6 : SUCCESS in 1m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ba94fb0c22754cd49c9664e4b54c0cb4 : SUCCESS in 41m 26s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/854909ea404f4424ab9193df43c411cc : SUCCESS in 56m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1dc02398942f45ff9cd5b628ef328c6e : SUCCESS in 48m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e0a9eec3d9645298e3d8a6e8af52b39 : SUCCESS in 1h 10m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b24581fc11e4db0b38a4593a2e1e19f : SUCCESS in 40m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c44175f3877640b687a5024bc122b7dd : SUCCESS in 2h 10m 07s","accounts_in_message":[],"_revision_number":3},{"id":"9b7d8744b64e8bb9862ad66a2cd80971b97acf9b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 10:41:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c504a2334858301f168532135a5d6d06d83081ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 10:42:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc0659356fb6433eb59662d524f60a22\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0e5238f604f4bde9d72a8dd8449c03c : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c11b16b44de613a15833e610ebf77d539e488856","revisions":{"76645f899cbf5c6265563b39dbb1a5b08c415485":{"kind":"REWORK","_number":1,"created":"2025-08-12 16:04:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/957153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/53/957153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/53/957153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/53/957153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/53/957153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/53/957153/1"}}},"commit":{"parents":[{"commit":"29cb510bd995f911a740935d097b0ddf98a8ad3d","subject":"Merge \"Fix uninialized os_VENDOR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/29cb510bd995f911a740935d097b0ddf98a8ad3d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-12 16:02:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-12 16:04:04.000000000","tz":540},"subject":"Fix ownership of keyring file","message":"Fix ownership of keyring file\n\nThe file should be owned by the user(STACK_USER) to run the process.\nNote that STACK_USER may not match the current user in case stack.sh is\nrun by root.\n\nAlso we should not assume that the group name always matches the user\nname.\n\nChange-Id: Iec300311cab1b1a2fa124879aa3dc6a57a6a706b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/76645f899cbf5c6265563b39dbb1a5b08c415485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/76645f899cbf5c6265563b39dbb1a5b08c415485"}]},"branch":"refs/heads/master"},"391e54c7223bf1cacfc6235ce15544d7a857777c":{"kind":"REWORK","_number":2,"created":"2025-10-29 14:13:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/957153/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/53/957153/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/53/957153/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/53/957153/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/53/957153/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/53/957153/2"}}},"commit":{"parents":[{"commit":"2801938292d76002c1cb2f6b93a3b7e68b7fab0c","subject":"Merge \"Remove dead checks for Fedora 36\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/2801938292d76002c1cb2f6b93a3b7e68b7fab0c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-12 16:02:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-29 14:13:18.000000000","tz":540},"subject":"Fix ownership of keyring file","message":"Fix ownership of keyring file\n\nThe file should be owned by the user(STACK_USER) to run the process.\nNote that STACK_USER may not match the current user in case stack.sh is\nrun by root.\n\nAlso we should not assume that the group name always matches the user\nname.\n\nChange-Id: Iec300311cab1b1a2fa124879aa3dc6a57a6a706b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/391e54c7223bf1cacfc6235ce15544d7a857777c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/391e54c7223bf1cacfc6235ce15544d7a857777c"}]},"branch":"refs/heads/master"},"c11b16b44de613a15833e610ebf77d539e488856":{"kind":"REWORK","_number":3,"created":"2025-12-17 16:36:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/957153/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/devstack","ref":"refs/changes/53/957153/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/devstack refs/changes/53/957153/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/devstack refs/changes/53/957153/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/devstack refs/changes/53/957153/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/devstack refs/changes/53/957153/3"}}},"commit":{"parents":[{"commit":"2801938292d76002c1cb2f6b93a3b7e68b7fab0c","subject":"Merge \"Remove dead checks for Fedora 36\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/2801938292d76002c1cb2f6b93a3b7e68b7fab0c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-12 16:02:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-17 16:36:14.000000000","tz":0},"subject":"Fix ownership of keyring file","message":"Fix ownership of keyring file\n\nThe file should be owned by the user(STACK_USER) to run the process.\nNote that STACK_USER may not match the current user in case stack.sh is\nrun by root.\n\nAlso we should not assume that the group name always matches the user\nname.\n\nChange-Id: Iec300311cab1b1a2fa124879aa3dc6a57a6a706b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/c11b16b44de613a15833e610ebf77d539e488856"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/devstack/commit/c11b16b44de613a15833e610ebf77d539e488856"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
