)]}'
{"id":"openstack%2Fdiskimage-builder~745665","triplet_id":"openstack%2Fdiskimage-builder~master~I3c96ae7ebb814973b77b4a1b966c39658acc5683","project":"openstack/diskimage-builder","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"I3c96ae7ebb814973b77b4a1b966c39658acc5683","subject":"Do not install python2 packages in ubuntu focal","status":"MERGED","created":"2020-08-11 12:52:05.000000000","updated":"2020-08-18 11:29:55.000000000","submitted":"2020-08-18 11:28:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745665-1597750086383-145709ed","meta_rev_id":"e9961b13b7e0cb9c05c05d21c635bf94b740ecd6","_number":745665,"virtual_id_number":745665,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-18 08:22:56.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"date":"2020-08-18 08:16:41.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-18 11:28:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-18 08:39:09.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-13 10:25:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-17 10:36:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-17 10:36:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 10:25:57.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-08-17 10:36:40.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-08-18 08:22:56.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-18 08:39:09.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-18 11:28:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a637e9874198f75b0da4f9944f9b021efbb0e1b8","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-11 12:52:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"298e4a5edce71ff6f14dc316148a68930db8519b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-11 13:01:19.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- dib-functests-bionic-python3-p9 dib-functests-bionic-python3-p9 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"6c057b8e872a14663eea89009cae41ee540d9759","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 13:15:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- dib-functests-arm64-bionic https://zuul.opendev.org/t/openstack/build/b30eb21d2afd4589a53ac5e9ccfb8bfc : SUCCESS in 22m 01s","accounts_in_message":[],"_revision_number":1},{"id":"45a267903eda59322d5d4770130a6dcd173032d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 14:27:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/938347d82bd7405b97d15fc42ce16cf7 : SUCCESS in 6m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dea8ed80e0124ef9a2855eba0f701921 : SUCCESS in 4m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b51f8c5c12f48f3adeec043797b09d3 : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1277f6a001c24469bd84f0da1c5b28d5 : SUCCESS in 4m 30s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/68d5965c6df843bc8055a82a81f834c0 : SUCCESS in 48m 29s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/48ab30e5e9184c81a5fe4e1f35e4f6d3 : SUCCESS in 33m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fde8e98f45cc4d3cb699b805088f4520 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bea8dc8b073439a91669ba93b3aeaf4 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/783b8574c73a41d78957d70f6232ee9d : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f7871b7a61f4dd98936596ffadc90b0 : SUCCESS in 4m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a37ab422f1764e25b66e3e74f4edc375 : SUCCESS in 3m 07s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/229403fdeb714d829cd80bb2b96fa469 : SUCCESS in 1h 29m 25s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/67e0bbc96e044eb7898970532653c1e2 : SUCCESS in 16m 32s\n- dib-nodepool-functional-openstack-centos-7-src https://zuul.opendev.org/t/openstack/build/c4958ae866e54f3085cea48ecaf8740d : SUCCESS in 31m 19s\n- dib-nodepool-functional-openstack-centos-8-src https://zuul.opendev.org/t/openstack/build/96c57d806cc346b7badd79987e052b2c : SUCCESS in 33m 41s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/26ee36bb73a441e6a17ce24bf170f6c7 : SUCCESS in 33m 28s\n- dib-nodepool-functional-openstack-fedora-31-src https://zuul.opendev.org/t/openstack/build/05c544a10a7041738f1f3b2ab980b246 : SUCCESS in 35m 47s\n- dib-nodepool-functional-openstack-ubuntu-xenial-src https://zuul.opendev.org/t/openstack/build/48502f9b5ea946eba13cf0f87d43853d : SUCCESS in 28m 51s\n- dib-nodepool-functional-openstack-ubuntu-bionic-src https://zuul.opendev.org/t/openstack/build/ad1715465cf34ec08ae8d893ee55ccca : SUCCESS in 32m 00s\n- dib-nodepool-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/c1301a6234e2482990daea17d7bec002 : SUCCESS in 31m 34s\n- dib-nodepool-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/ffe2b8ed5bf9449c99c53b86162f1575 : SUCCESS in 50m 13s (non-voting)\n- dib-nodepool-functional-openstack-opensuse-tumbleweed-src https://zuul.opendev.org/t/openstack/build/5b91442a13114dd59dfea734a82b98c4 : SUCCESS in 40m 37s (non-voting)\n- dib-functests-bionic-python3 https://zuul.opendev.org/t/openstack/build/3591d297c08749ea8800817242e13080 : SUCCESS in 32m 11s\n- dib-functests-bionic-python3-image https://zuul.opendev.org/t/openstack/build/da40ca01681f44e3b3738739a9b3527f : SUCCESS in 33m 56s (non-voting)\n- dib-functests-bionic-python3-extras https://zuul.opendev.org/t/openstack/build/24e088c32e934ad3bbbccb412f2ff9ba : SUCCESS in 1h 04m 27s (non-voting)\n- dib-functests-bionic-python3-containers https://zuul.opendev.org/t/openstack/build/7e1326d1fea246f187c82b95b264aabe : SUCCESS in 9m 31s (non-voting)\n- ironic-python-agent-check-image-dib-centos8 https://zuul.opendev.org/t/openstack/build/e88c5aab703644c6b60e2686f18c7815 : SUCCESS in 9m 52s\n- ipa-tempest-wholedisk-bios-ipmi-direct-dib-src https://zuul.opendev.org/t/openstack/build/879035ca54d743f788df7f922effcf47 : SUCCESS in 1h 23m 53s (non-voting)\n- ipa-tempest-partition-bios-ipmi-iscsi-dib-src https://zuul.opendev.org/t/openstack/build/faa68280f8d04e608b897ba53ef0acf2 : SUCCESS in 1h 27m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c90ffa17cb38066920dffba7790d47a7fe8e118e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-11 15:52:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b85cafca72d54673ba981c2140391bae : FAILURE in 2h 01m 21s","accounts_in_message":[],"_revision_number":1},{"id":"98b34ab04501807485290ebf126b365dc412f939","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-12 10:16:46.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7174cff4ddddab7eb4123b02b890366a7b7f2fe2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-12 10:24:14.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- dib-functests-bionic-python3-p9 dib-functests-bionic-python3-p9 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"8c193edd3f1b5b85f9db7bcd5145a61d77cf0b82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 10:39:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- dib-functests-arm64-bionic https://zuul.opendev.org/t/openstack/build/c9ca360f18664a24963643d17cddd63c : SUCCESS in 21m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e622688399592c63f6e8822cd05c991b612d9c23","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-12 12:21:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/3a0f18ee719942018089111f1b3fa21f : SUCCESS in 2h 04m 12s","accounts_in_message":[],"_revision_number":2},{"id":"4f7f048e551006b57544e23d95142ecd92677db3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 13:46:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74052dfbd8874c21baeece98e8b28cc9 : SUCCESS in 6m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a266bb9d165465fbfd838756eafe2de : SUCCESS in 4m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d7b06d2d5704bdab10c60f54a3c3033 : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3941532548f146109e291dfbca7de44c : SUCCESS in 4m 02s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/5a53009e72b648f08aa8f402afb203b9 : SUCCESS in 53m 58s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/9fcfede8e4464cddb9f3f5b974eef510 : SUCCESS in 27m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/64155df5a24b446a9778f1c9374ca60b : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4584bc2768cb4ec39998b661a53776c1 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b680d8f8d4142cc91a4e1bb897eb5f6 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/449997b927a64cbf882b00bcc2e41c0c : SUCCESS in 4m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/951ce44c3c4d440f84012397d16b50c0 : SUCCESS in 3m 52s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/6068659c70e74aa0b3611bbed0fbad4b : SUCCESS in 3h 27m 51s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/eec3c8c47963413d899218b78a41889a : SUCCESS in 13m 25s\n- dib-nodepool-functional-openstack-centos-7-src https://zuul.opendev.org/t/openstack/build/816a9695f3ad4b628f2c12b101399d09 : RETRY_LIMIT in 15m 46s\n- dib-nodepool-functional-openstack-centos-8-src https://zuul.opendev.org/t/openstack/build/a88f6b6b250a445facca6c39953d6438 : RETRY_LIMIT in 33m 52s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/4cec4870978f445c9a04e3bfab199685 : RETRY_LIMIT in 30m 58s\n- dib-nodepool-functional-openstack-fedora-31-src https://zuul.opendev.org/t/openstack/build/977aa57f05484bd9b1cfb0c03f6bab0c : RETRY_LIMIT in 36m 25s\n- dib-nodepool-functional-openstack-ubuntu-xenial-src https://zuul.opendev.org/t/openstack/build/85a67e7756294e36a028695156deb3a3 : RETRY_LIMIT in 19m 41s\n- dib-nodepool-functional-openstack-ubuntu-bionic-src https://zuul.opendev.org/t/openstack/build/0eed193c3f7142898aec4b4287841034 : RETRY_LIMIT in 33m 59s\n- dib-nodepool-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/a7d4bd8e94d04a4ba1b8e27be8a83b47 : SUCCESS in 1h 20m 54s\n- dib-nodepool-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/646629ec6cfb4f189889b39e85329584 : RETRY_LIMIT in 1h 00m 35s (non-voting)\n- dib-nodepool-functional-openstack-opensuse-tumbleweed-src https://zuul.opendev.org/t/openstack/build/3ce9e1116aa149079a212f7d65743c57 : RETRY_LIMIT in 26m 35s (non-voting)\n- dib-functests-bionic-python3 https://zuul.opendev.org/t/openstack/build/6638874df4ef4728adc94b9643e3f4b5 : SUCCESS in 28m 46s\n- dib-functests-bionic-python3-image https://zuul.opendev.org/t/openstack/build/878d0bd419bc43b39b574f3013785f63 : SUCCESS in 29m 04s (non-voting)\n- dib-functests-bionic-python3-extras https://zuul.opendev.org/t/openstack/build/1dfd915797a9484fadf3abae902331c7 : SUCCESS in 41m 31s (non-voting)\n- dib-functests-bionic-python3-containers https://zuul.opendev.org/t/openstack/build/934244e525cd4fbeaf6eb2d3886d14ba : SUCCESS in 10m 20s (non-voting)\n- ironic-python-agent-check-image-dib-centos8 https://zuul.opendev.org/t/openstack/build/b9cc53fedf6643d1b86cdc5213fa41f9 : SUCCESS in 11m 05s\n- ipa-tempest-wholedisk-bios-ipmi-direct-dib-src https://zuul.opendev.org/t/openstack/build/e29ee6f630294766aa8ee7e0d78d3957 : SUCCESS in 1h 43m 04s (non-voting)\n- ipa-tempest-partition-bios-ipmi-iscsi-dib-src https://zuul.opendev.org/t/openstack/build/b4074c6fd22249dc87b9c91a5f36e012 : SUCCESS in 1h 23m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7ef5015931fe691785e8455670f2cb08d4eb385a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-12 14:35:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5892cef1c3d7321661e40aa1dc35d55cc8b81e7c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-12 14:40:49.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- dib-functests-bionic-python3-p9 dib-functests-bionic-python3-p9 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"e3e5423e3ff6d0c6facfeba40973ef0594ce2f00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 15:02:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- dib-functests-arm64-bionic https://zuul.opendev.org/t/openstack/build/53769ef0fa7e4cbf86e351314699c095 : SUCCESS in 22m 46s","accounts_in_message":[],"_revision_number":2},{"id":"22c1b762c9586e6a6b64ab22566e62a5c25af72c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 16:20:35.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba36affa65bb4663818425542d437523 : SUCCESS in 12m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/269031141a4e439e935b4ddf5a069f81 : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05b5a0250507406094b573cbb9579808 : FAILURE in 10m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c88cae623fb4434a1b565b13eeb8684 : SUCCESS in 4m 16s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/3eb20dc8e05d422d9918f52d6ff353ce : TIMED_OUT in 1h 11m 54s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/88f16fafa668473f9a3b94df1f336e3e : SUCCESS in 29m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5ee326f9354840b09351d34a9389ed2b : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11dfbbd868b946098e233e5696cf5140 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db3db60935f5457a8a2d01ade9c2efa4 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f176b9a6cfdb4157afab83a4b415bca6 : SUCCESS in 3m 50s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/edc065f6acdd49e599edc422bc9f369a : SUCCESS in 3m 20s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/720a48c9c1a940a6bc632b63ab41c6e1 : SUCCESS in 1h 27m 34s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/92920888e8f1457ea1bd75fc53b3b11f : SUCCESS in 17m 23s\n- dib-nodepool-functional-openstack-centos-7-src https://zuul.opendev.org/t/openstack/build/6027f771eeb14fe49f5970f5463841c2 : SUCCESS in 37m 48s\n- dib-nodepool-functional-openstack-centos-8-src https://zuul.opendev.org/t/openstack/build/8786c2b4118f486eac66daf1df2c98a7 : SUCCESS in 35m 26s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/8efb228413cd4fa2a7965fd2472c7084 : SUCCESS in 35m 43s\n- dib-nodepool-functional-openstack-fedora-31-src https://zuul.opendev.org/t/openstack/build/51ae493cfba043f7b634b315a6bd0b4b : SUCCESS in 34m 32s\n- dib-nodepool-functional-openstack-ubuntu-xenial-src https://zuul.opendev.org/t/openstack/build/366da61379fe459fab2b87ff8416a7d5 : SUCCESS in 29m 13s\n- dib-nodepool-functional-openstack-ubuntu-bionic-src https://zuul.opendev.org/t/openstack/build/6b45e376ad0748b19fc09f7dfabb4580 : SUCCESS in 35m 42s\n- dib-nodepool-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/426aff8d40ce44e3ba0bea9f149a2faf : SUCCESS in 32m 48s\n- dib-nodepool-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/0e9e7d1ed1734aef98e541f095a38196 : SUCCESS in 52m 40s (non-voting)\n- dib-nodepool-functional-openstack-opensuse-tumbleweed-src https://zuul.opendev.org/t/openstack/build/1b23bae235d44c5c890b705a3ac3f379 : SUCCESS in 43m 38s (non-voting)\n- dib-functests-bionic-python3 https://zuul.opendev.org/t/openstack/build/1cf2bd085b354fcc9d33b0afb725db9c : SUCCESS in 28m 00s\n- dib-functests-bionic-python3-image https://zuul.opendev.org/t/openstack/build/56595188c85a4d0db9abe2b0bc05651f : SUCCESS in 32m 59s (non-voting)\n- dib-functests-bionic-python3-extras https://zuul.opendev.org/t/openstack/build/a63ebdcbccc24427b9a0c927ea76afad : SUCCESS in 38m 18s (non-voting)\n- dib-functests-bionic-python3-containers https://zuul.opendev.org/t/openstack/build/5d8af554b9d9412484b82f4e9757d34f : SUCCESS in 9m 36s (non-voting)\n- ironic-python-agent-check-image-dib-centos8 https://zuul.opendev.org/t/openstack/build/96eb3807ce9743728c5db73579735d85 : SUCCESS in 10m 03s\n- ipa-tempest-wholedisk-bios-ipmi-direct-dib-src https://zuul.opendev.org/t/openstack/build/b826d5ff7ee349e5be14181d2b15a591 : SUCCESS in 1h 31m 07s (non-voting)\n- ipa-tempest-partition-bios-ipmi-iscsi-dib-src https://zuul.opendev.org/t/openstack/build/25899c26c2734ae9b261554466309819 : SUCCESS in 1h 30m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fe643365f68b956b9455cd8025658b2a237b93b2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-12 16:45:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2806c34f57504c53a27f50f0f6ad231c : SUCCESS in 2h 08m 48s","accounts_in_message":[],"_revision_number":2},{"id":"58b7be392a872e8ff69dff5cdec4ab550cf310be","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-13 08:34:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"29b5c113b41e9a3aad61cdd2d04200a78cd84976","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 08:39:31.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- dib-functests-arm64-bionic https://zuul.opendev.org/t/openstack/build/cf86a15a67ac43ca82945fbbeaa52f0c : FAILURE in 4m 23s","accounts_in_message":[],"_revision_number":2},{"id":"3f9326726409391f2fc58d7f8dd09ea66c78a0b8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-13 08:42:18.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- dib-functests-bionic-python3-p9 dib-functests-bionic-python3-p9 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"b20888f5ca3ef2e43f9ab2b40eafabddfb81cc85","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-13 09:26:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0ca71ebdd38263c4e21ec995c8e62791ea7fb744","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-13 09:33:11.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- dib-functests-bionic-python3-p9 dib-functests-bionic-python3-p9 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"9d8875c9d301a0e07c90e8e87b8bbdf45eaa0daa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 09:44:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- dib-functests-arm64-bionic https://zuul.opendev.org/t/openstack/build/07bdfe22b4aa4e19a444d2c518b1d6b9 : SUCCESS in 17m 52s","accounts_in_message":[],"_revision_number":2},{"id":"a64d0e69793f0b078a56425a35057ef8e87830d5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-13 10:25:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f72cc90b579a245342366e810991bf2b70bd34e2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-13 10:35:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ecd3ead309f0444fb36ddbb3603f2f6d : SUCCESS in 2h 00m 45s","accounts_in_message":[],"_revision_number":2},{"id":"cb6809e8da4ea1e1b4de9cd734c787f862881937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 10:49:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1b532c7fd134ffa9a4c28d4db37b8f2 : SUCCESS in 5m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6392c43b2aff41e6b5a98ac71d948825 : SUCCESS in 3m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c1edebfb8c840e1ba5cf92c2ac191c3 : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5026e5340c4e4645ac96420504640fe3 : SUCCESS in 3m 31s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/aaa101481277496b98b2595f6cc8cd65 : SUCCESS in 51m 44s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/f34fe74c31294510bf2bfbc78e78f0ad : SUCCESS in 18m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c586202164a5497a8081cf025e953f2d : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c73cd7b7a6f84610a368a15a0bcc7ddf : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dff9c8dc8144f179fd878e7562a8651 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/845a7140f8cf4fe885100dc6da7fee26 : SUCCESS in 6m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/52015ed1cbc14409b1d04c7609c06487 : SUCCESS in 3m 56s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/80400ee642be4e58a9866be8a37a30f9 : SUCCESS in 1h 36m 37s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/1e77a7b8d0304509bdb70dc23e323984 : SUCCESS in 18m 40s\n- dib-nodepool-functional-openstack-centos-7-src https://zuul.opendev.org/t/openstack/build/30ee575f524f4d789b2ae2ead9d3b23d : SUCCESS in 35m 25s\n- dib-nodepool-functional-openstack-centos-8-src https://zuul.opendev.org/t/openstack/build/bf65e3c9f69d4dd28eb71ab911220f00 : SUCCESS in 37m 12s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/ea5382be97f847d0ad74592603667f63 : SUCCESS in 37m 17s\n- dib-nodepool-functional-openstack-fedora-31-src https://zuul.opendev.org/t/openstack/build/ce5b878116d9425d91a0fcdcf6f52904 : SUCCESS in 43m 52s\n- dib-nodepool-functional-openstack-ubuntu-xenial-src https://zuul.opendev.org/t/openstack/build/1c70094b8e28499dac919903cc36bee8 : SUCCESS in 33m 47s\n- dib-nodepool-functional-openstack-ubuntu-bionic-src https://zuul.opendev.org/t/openstack/build/83ec97bc51cd49059a22f19e1365aa5c : SUCCESS in 37m 22s\n- dib-nodepool-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/d0c10a028b0a48e989b47da37a1fabbe : SUCCESS in 37m 45s\n- dib-nodepool-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/673dfb561f1e49c5853368cbfeb25937 : SUCCESS in 1h 07m 46s (non-voting)\n- dib-nodepool-functional-openstack-opensuse-tumbleweed-src https://zuul.opendev.org/t/openstack/build/c01fbfb192bc4c84a4c183271d6e65d8 : SUCCESS in 43m 23s (non-voting)\n- dib-functests-bionic-python3 https://zuul.opendev.org/t/openstack/build/95d909283f784900a18b8c021773b058 : SUCCESS in 28m 08s\n- dib-functests-bionic-python3-image https://zuul.opendev.org/t/openstack/build/012789e6a6ec4053b017917804d88fce : SUCCESS in 36m 46s (non-voting)\n- dib-functests-bionic-python3-extras https://zuul.opendev.org/t/openstack/build/3c1a163f4c5045e1a7282e406bfb3ccd : SUCCESS in 47m 29s (non-voting)\n- dib-functests-bionic-python3-containers https://zuul.opendev.org/t/openstack/build/df3b19eaa2974126a7bc54680ccf505c : SUCCESS in 9m 59s (non-voting)\n- ironic-python-agent-check-image-dib-centos8 https://zuul.opendev.org/t/openstack/build/91d7201e0300431a81ff5ca9a9e3f98b : SUCCESS in 10m 26s\n- ipa-tempest-wholedisk-bios-ipmi-direct-dib-src https://zuul.opendev.org/t/openstack/build/3962226fb3a64cd7b893a89cb79fb3d3 : SUCCESS in 1h 35m 36s (non-voting)\n- ipa-tempest-partition-bios-ipmi-iscsi-dib-src https://zuul.opendev.org/t/openstack/build/2ffb2a64b75c42bbaac46bdc69358191 : SUCCESS in 1h 44m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7e3948705d05c9fc02c00050faa54411426aaa58","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-08-17 10:36:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dd57aa13bafaa4eec36a48665361a03ed6c2c9e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 10:36:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a5212665084f5b6528cdd804e0c75378d1317ef0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 11:42:09.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea1daec8211544a8aee366f68d3ba92b : SUCCESS in 5m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba5732ce56734a02b3ff3b4d4dee1f16 : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46693a57ebb448fda229a7610107a6f9 : SUCCESS in 3m 51s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/2f23b81969624cb6bfb564954be07d9a : SUCCESS in 44m 30s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/5ee5f0d4b68845ac82fa00ff0dc81935 : SUCCESS in 23m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c64206f3682843f4b3e82831f4d560b9 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cd31258063242218c63071c3b4d883b : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2285f36ed604f03b0c976c48fb2398c : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6309bb4278c344e6a32bf9f23e7fba48 : SUCCESS in 3m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d8860824a17f4fd399b85ba88295dec2 : SUCCESS in 4m 03s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/9c531211a16545b99dfd9916e4e8889a : SUCCESS in 1h 02m 57s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/a16af2e6786545b991f3b95e79d28a1d : SUCCESS in 18m 48s\n- dib-nodepool-functional-openstack-centos-7-src https://zuul.opendev.org/t/openstack/build/c136132e589148b29134505f97c57cd5 : SUCCESS in 34m 05s\n- dib-nodepool-functional-openstack-centos-8-src https://zuul.opendev.org/t/openstack/build/2da7277e6bfa4535b3099d526f8801da : SUCCESS in 38m 46s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/9640cfb9ec5243f598a59452dcf66e5f : SUCCESS in 36m 14s\n- dib-nodepool-functional-openstack-fedora-31-src https://zuul.opendev.org/t/openstack/build/264e52ceefe9420eaefb0f5c71e7fd46 : SUCCESS in 37m 39s\n- dib-nodepool-functional-openstack-ubuntu-xenial-src https://zuul.opendev.org/t/openstack/build/1f906a4f218045a4baa08a799a970399 : SUCCESS in 29m 55s\n- dib-nodepool-functional-openstack-ubuntu-bionic-src https://zuul.opendev.org/t/openstack/build/5b770c7d67164ba2b05198932b00e8c2 : SUCCESS in 31m 04s\n- dib-nodepool-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/c8225850bfc941dc88a67e00357949c5 : SUCCESS in 32m 31s\n- dib-functests-bionic-python3 https://zuul.opendev.org/t/openstack/build/f39e5fb12c824fc9b80839762ffce465 : FAILURE in 13m 23s\n- ironic-python-agent-check-image-dib-centos8 https://zuul.opendev.org/t/openstack/build/c52409a14cac4f198838d4c3ec0b734f : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":2},{"id":"818371469ec2d639cf8508e05aefa93cd29084ea","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-17 12:29:43.000000000","message":"Patch Set 2:\n\nretest","accounts_in_message":[],"_revision_number":2},{"id":"6f9e4765a48908336ecda51fd32a7a45dc0bdbe4","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-17 12:29:50.000000000","message":"Patch Set 2:\n\nrecheck maybe","accounts_in_message":[],"_revision_number":2},{"id":"6fef9a7514d0d8dc917adac441cfc4c0fd279949","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-17 12:37:04.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- dib-functests-bionic-python3-p9 dib-functests-bionic-python3-p9 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"04534639194cc630a184eab284c8ec8d1f9fefe6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 12:52:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- dib-functests-arm64-bionic https://zuul.opendev.org/t/openstack/build/f2dc564d29904152b2a00b4047492b5e : SUCCESS in 22m 30s","accounts_in_message":[],"_revision_number":2},{"id":"370277f320c86f6204a10b4e2acb8ee0343c9353","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 14:24:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a38370b3777d443b88258899471bee42 : SUCCESS in 4m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7961d5dba42b4d7c85dcd903073bbdf5 : SUCCESS in 4m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efffc4e5d4534effb312b5eb9b72053b : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/853523355b7d4a9eaeff2db871b66ace : SUCCESS in 4m 43s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/c43fc9b3782e47889100f6ca4740c791 : SUCCESS in 58m 02s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/b3382c8384bc4dfc9b55d08f6ecdda9c : SUCCESS in 23m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6ce6cb3ba5c54bae95e85eedb82f1578 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8910a7429fcb4a8aa68c07e98be2b507 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/684fa8f7f874479c8a59ece9ddef24c8 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6f89139ce4843db944d260946b45df3 : SUCCESS in 3m 33s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bdb2d76ea30c42a29338cb565afcbe78 : SUCCESS in 3m 24s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ce40eba2172d46a3bd8c260c591b9bc4 : SUCCESS in 1h 49m 31s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/88baa68b05ee4fe9b2bfbbb6f3b2d107 : SUCCESS in 17m 52s\n- dib-nodepool-functional-openstack-centos-7-src https://zuul.opendev.org/t/openstack/build/ea6ab0a3438148a9a542b73703db3bff : SUCCESS in 32m 38s\n- dib-nodepool-functional-openstack-centos-8-src https://zuul.opendev.org/t/openstack/build/6f0778b7500c42d3946857f292dfd6a6 : SUCCESS in 35m 47s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/c1a82d44059645d99577da0fd7b97dec : SUCCESS in 36m 15s\n- dib-nodepool-functional-openstack-fedora-31-src https://zuul.opendev.org/t/openstack/build/be2106ed539d48f48a7b61462bb18f8e : SUCCESS in 39m 50s\n- dib-nodepool-functional-openstack-ubuntu-xenial-src https://zuul.opendev.org/t/openstack/build/cca9b44c880940028e9aac05bbba95e8 : SUCCESS in 30m 50s\n- dib-nodepool-functional-openstack-ubuntu-bionic-src https://zuul.opendev.org/t/openstack/build/d365701edbfc4ce8942f451f79f35375 : SUCCESS in 33m 23s\n- dib-nodepool-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/cfcd3cb185a34faea8651218307befcf : SUCCESS in 34m 43s\n- dib-nodepool-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/e0008d9ada234e4282dbf74cd335fa74 : SUCCESS in 1h 15m 08s (non-voting)\n- dib-nodepool-functional-openstack-opensuse-tumbleweed-src https://zuul.opendev.org/t/openstack/build/12f6e4d5d57e459ba98424e55a926305 : SUCCESS in 45m 27s (non-voting)\n- dib-functests-bionic-python3 https://zuul.opendev.org/t/openstack/build/87bb22561e5c4ab7b61ab73131c174c4 : FAILURE in 11m 15s\n- dib-functests-bionic-python3-image https://zuul.opendev.org/t/openstack/build/ff030fdd1fff4a75aa5e79312e6cd4e1 : FAILURE in 15m 37s (non-voting)\n- dib-functests-bionic-python3-extras https://zuul.opendev.org/t/openstack/build/4283c83fa1584863b49b49f7c6f434b7 : SUCCESS in 57m 42s (non-voting)\n- dib-functests-bionic-python3-containers https://zuul.opendev.org/t/openstack/build/c1b7ce40ad2e43fdaf6b53ad3419b4cb : SUCCESS in 8m 23s (non-voting)\n- ironic-python-agent-check-image-dib-centos8 https://zuul.opendev.org/t/openstack/build/4c271d141f67475b9a5a5851d33e1b4e : SUCCESS in 10m 18s\n- ipa-tempest-wholedisk-bios-ipmi-direct-dib-src https://zuul.opendev.org/t/openstack/build/bca847815efd4a3eae9d826d38ea32c2 : SUCCESS in 1h 29m 49s (non-voting)\n- ipa-tempest-partition-bios-ipmi-iscsi-dib-src https://zuul.opendev.org/t/openstack/build/655a30bc7578407da7a77aa5373a74e1 : SUCCESS in 1h 29m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c56948fcb4df4224b283a01c192cfafd8db23469","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-17 14:34:33.000000000","message":"Patch Set 2:\n\nrecheck issue with Fedora mirrors","accounts_in_message":[],"_revision_number":2},{"id":"be56765d8c3b72ea7ebacacdea4e46318b6600db","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-17 14:36:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/aa5fa4af22c44d4c847b36093f431045 : SUCCESS in 2h 05m 30s","accounts_in_message":[],"_revision_number":2},{"id":"bf3b03db964839a29e12ad0bf34e6a018e31f20d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-17 14:40:04.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- dib-functests-bionic-python3-p9 dib-functests-bionic-python3-p9 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"9cb40c6271026219124e5095dfb71d613c1e413f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 15:00:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- dib-functests-arm64-bionic https://zuul.opendev.org/t/openstack/build/b4a01d0586084f6abe0cef3ddf65b2e0 : SUCCESS in 21m 44s","accounts_in_message":[],"_revision_number":2},{"id":"e9b80a338376e23a05b5908d322b3bf0f1eba9ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 16:07:18.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cfb1ab786374cc4a4b7948d8c72ed67 : SUCCESS in 6m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58773ccfbf8c493cba99270b55b8f107 : SUCCESS in 4m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4f08837fd064d69a1a9034b3b74c236 : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ae8e87ce4a04659a7b2dd7311f18595 : SUCCESS in 4m 25s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/7d4c2f1b4d1247869d8b7e28d6f8862d : SUCCESS in 53m 10s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/48366e9de2704342ab4643ee3d1c8e09 : SUCCESS in 26m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/db6219cba03240c28f48ba6e71d0bdc9 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2598bcf4c09e4b4a82c51bd9d380e7eb : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa6a9553c0aa4ecf8db62e642114518a : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce0d8b8543214558a5a044cc58ad6062 : SUCCESS in 6m 01s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b2e8d138eb934b79840c4b4e00e052cf : SUCCESS in 4m 42s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/406307abe92e47e58cee520a9ff8572d : SUCCESS in 1h 18m 26s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/58808268dc1c4716acbcb8a6c950d9b4 : SUCCESS in 21m 34s\n- dib-nodepool-functional-openstack-centos-7-src https://zuul.opendev.org/t/openstack/build/b87c91fd0bfe43fa83bd0327e4b417d1 : SUCCESS in 32m 05s\n- dib-nodepool-functional-openstack-centos-8-src https://zuul.opendev.org/t/openstack/build/414b18a032b4498bba968505fb1425c3 : SUCCESS in 33m 21s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/531c536cf2a3443f99350efca060c734 : SUCCESS in 38m 30s\n- dib-nodepool-functional-openstack-fedora-31-src https://zuul.opendev.org/t/openstack/build/e57abfbd6a0e41f7b6212b0c968f86f2 : SUCCESS in 40m 31s\n- dib-nodepool-functional-openstack-ubuntu-xenial-src https://zuul.opendev.org/t/openstack/build/bb25589ab08c43c1ae86a04bb0a60b1a : SUCCESS in 35m 11s\n- dib-nodepool-functional-openstack-ubuntu-bionic-src https://zuul.opendev.org/t/openstack/build/c5b36ccd8337471cb0af57551c4f78a4 : SUCCESS in 32m 04s\n- dib-nodepool-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/6debcd4dbd744d5b95cfb5035adde898 : SUCCESS in 36m 44s\n- dib-nodepool-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/644509ac4ea64ee2b87521f47c3b2be3 : SUCCESS in 45m 35s (non-voting)\n- dib-nodepool-functional-openstack-opensuse-tumbleweed-src https://zuul.opendev.org/t/openstack/build/17e63f3459be4eeda0e99765c1f28bf0 : SUCCESS in 42m 21s (non-voting)\n- dib-functests-bionic-python3 https://zuul.opendev.org/t/openstack/build/e6795afd8c6643539b91d79d39818d5f : FAILURE in 11m 12s\n- dib-functests-bionic-python3-image https://zuul.opendev.org/t/openstack/build/d27383949d264ee8a2f6b4f0c8503679 : FAILURE in 13m 44s (non-voting)\n- dib-functests-bionic-python3-extras https://zuul.opendev.org/t/openstack/build/ff865c8ac9f74706826e9a761ba8a4da : SUCCESS in 42m 53s (non-voting)\n- dib-functests-bionic-python3-containers https://zuul.opendev.org/t/openstack/build/9359ff1d0f724fbfba37f9716b3f7007 : SUCCESS in 8m 02s (non-voting)\n- ironic-python-agent-check-image-dib-centos8 https://zuul.opendev.org/t/openstack/build/fedda95057e54c4b9a4785a70e92fff3 : SUCCESS in 12m 35s\n- ipa-tempest-wholedisk-bios-ipmi-direct-dib-src https://zuul.opendev.org/t/openstack/build/79146aaaa33b436d857164688c8e6b9d : SUCCESS in 1h 23m 48s (non-voting)\n- ipa-tempest-partition-bios-ipmi-iscsi-dib-src https://zuul.opendev.org/t/openstack/build/37eea4e97c884b6a96a229c49a37268a : SUCCESS in 1h 19m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ff268d943b609bf13b282c1417103d734d63229b","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-08-18 04:27:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0c09204b8daf83ce77244260bfae9485a279faa3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-18 04:33:55.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- dib-functests-bionic-python3-p9 dib-functests-bionic-python3-p9 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"b04c396d14826fdab407daa7c9661fb9fcd0d577","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 04:46:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- dib-functests-arm64-bionic https://zuul.opendev.org/t/openstack/build/74befd6d6d6a4cf68fc6817b3dd800c3 : SUCCESS in 18m 24s","accounts_in_message":[],"_revision_number":2},{"id":"edf9e8cf10a426a78e4f56dea3b5dc6405e7f023","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 06:09:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dd1440039be4ad5bf4dd347d386f7b2 : SUCCESS in 4m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0499ce55fee49a2b7c7aa9056e4820c : SUCCESS in 3m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27b283b1c12a499bb9c4041e855b525a : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fc0740da4e04afc943805d1ae52c00a : SUCCESS in 3m 18s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/32aefc107d6f46c7abaaead78bdbef6e : SUCCESS in 50m 23s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/9398f8186c5c4de3b581c6431e13f2bc : SUCCESS in 26m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e009e1ee3ac044a3a198995c748b7c6a : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fc261a27b504fe5b5c78088ea6cef1d : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f7f8e49cdc44a8db381dfa80c0635c2 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf37b68391584f2980f2358723d68640 : SUCCESS in 3m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0005e97af2e649d096f371c63a958816 : SUCCESS in 3m 03s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/9528d3cb56c74180aa3ad8dd53fb8812 : SUCCESS in 1h 34m 00s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/0733c44b32384528ba6ff7b6ee637105 : SUCCESS in 16m 51s\n- dib-nodepool-functional-openstack-centos-7-src https://zuul.opendev.org/t/openstack/build/a014115857924467ace156c439f55a36 : SUCCESS in 32m 34s\n- dib-nodepool-functional-openstack-centos-8-src https://zuul.opendev.org/t/openstack/build/4720f18827604527a46b195f74075b8f : SUCCESS in 33m 10s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/85c6b5892fae41d1aca8c6b165e2a748 : SUCCESS in 31m 11s\n- dib-nodepool-functional-openstack-fedora-31-src https://zuul.opendev.org/t/openstack/build/81e90504c7294c5f870f3b874f435357 : SUCCESS in 32m 33s\n- dib-nodepool-functional-openstack-ubuntu-xenial-src https://zuul.opendev.org/t/openstack/build/0b34739fe2624b13abb82e4336dbd72a : SUCCESS in 31m 08s\n- dib-nodepool-functional-openstack-ubuntu-bionic-src https://zuul.opendev.org/t/openstack/build/aaf2a83f6c854cd585864bfe365e0c18 : SUCCESS in 44m 17s\n- dib-nodepool-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/d0031f0ea29149ed8a24b2915f0f8fa1 : SUCCESS in 30m 36s\n- dib-nodepool-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/8934ad016a5a418eb5dbc8254d648619 : SUCCESS in 1h 03m 31s (non-voting)\n- dib-nodepool-functional-openstack-opensuse-tumbleweed-src https://zuul.opendev.org/t/openstack/build/dd8da8ec05ca444089012bc3c339715a : SUCCESS in 35m 16s (non-voting)\n- dib-functests-bionic-python3 https://zuul.opendev.org/t/openstack/build/8cebdbd788fd418288afffd91d845443 : SUCCESS in 26m 28s\n- dib-functests-bionic-python3-image https://zuul.opendev.org/t/openstack/build/61dbe4a00a084f2d9ddfb204c7087486 : SUCCESS in 39m 37s (non-voting)\n- dib-functests-bionic-python3-extras https://zuul.opendev.org/t/openstack/build/37bcadbe161f4524bd6e324a278c3044 : SUCCESS in 32m 52s (non-voting)\n- dib-functests-bionic-python3-containers https://zuul.opendev.org/t/openstack/build/95e843b01552464885d17f871784f1fc : SUCCESS in 6m 41s (non-voting)\n- ironic-python-agent-check-image-dib-centos8 https://zuul.opendev.org/t/openstack/build/8f03d612a507485ea82077c4facf27a9 : SUCCESS in 7m 19s\n- ipa-tempest-wholedisk-bios-ipmi-direct-dib-src https://zuul.opendev.org/t/openstack/build/6b96a6bcb7554620a7af5e798fbe092b : SUCCESS in 56m 16s (non-voting)\n- ipa-tempest-partition-bios-ipmi-iscsi-dib-src https://zuul.opendev.org/t/openstack/build/4e3572305a734bc796e0e79e508bc48d : SUCCESS in 55m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6720808b274acd0325b1dd16824d00ef458128b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 06:09:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f93ca7f9648e219cb39b90caab7ccfc25f7d7076","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 07:26:28.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/092dd5f061ce4eef93982f1659d3df86 : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b48c2cc2412437195beb8fc2e23043a : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d776dd906d27400684fc861e17addb4f : SUCCESS in 4m 01s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/b134d7df895d49078dae9f8933352e5c : SUCCESS in 48m 52s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/c54eb4d526b944d28ec23cee08fa74f8 : SUCCESS in 25m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a5e89ce8a1d6469aad23e09c6b3142b1 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23f440f2e4004ef7b466a9762f2acc98 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e43e0836598341eead71d53ffd3db959 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4c56a8e72f744439ab60c8a087f18d8 : SUCCESS in 4m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/baaef25aa9514497a41c1eb66c13abfe : SUCCESS in 5m 28s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b5765b875c4f4cdeb57e89b1d393395f : SUCCESS in 1h 13m 43s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/80915adea055442b8e2a6f0a0b1a0d5d : SUCCESS in 17m 37s\n- dib-nodepool-functional-openstack-centos-7-src https://zuul.opendev.org/t/openstack/build/ea5adfbe570e4e3eb4566a2934aeeeb9 : SUCCESS in 45m 40s\n- dib-nodepool-functional-openstack-centos-8-src https://zuul.opendev.org/t/openstack/build/d89b6c73527d4758a1e2a536652ab692 : SUCCESS in 45m 14s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/22596a79885246f7904edfe4b9a9ed58 : SUCCESS in 46m 57s\n- dib-nodepool-functional-openstack-fedora-31-src https://zuul.opendev.org/t/openstack/build/6dbc44a7052d4b7785ae21da13f0fffd : SUCCESS in 43m 33s\n- dib-nodepool-functional-openstack-ubuntu-xenial-src https://zuul.opendev.org/t/openstack/build/411c3d2d2a1d4c6d896dd0d091653be8 : SUCCESS in 32m 44s\n- dib-nodepool-functional-openstack-ubuntu-bionic-src https://zuul.opendev.org/t/openstack/build/c1efb08678714275b19ffd6906db7775 : SUCCESS in 46m 09s\n- dib-nodepool-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/d7f8bfff8607450db1052b73bb31e1f5 : SUCCESS in 42m 43s\n- dib-functests-bionic-python3 https://zuul.opendev.org/t/openstack/build/894783e1316f4759a3ffdd2aa5865288 : FAILURE in 13m 16s\n- ironic-python-agent-check-image-dib-centos8 https://zuul.opendev.org/t/openstack/build/517d4eb49e504595aa0cdb2c88549efe : SUCCESS in 9m 43s","accounts_in_message":[],"_revision_number":2},{"id":"8bc54b72c273a7c4f568e51c71d50bd5ee98f339","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-18 08:16:41.000000000","message":"Patch Set 2:\n\nrecheck seriously zuul ?","accounts_in_message":[],"_revision_number":2},{"id":"ac3f26ec1c6407fbc41772f855b826c8ad74b2d6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-18 08:22:56.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- dib-functests-bionic-python3-p9 dib-functests-bionic-python3-p9 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"af52d865933ad3dc271e18d76391e4c632cb572f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 08:36:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- dib-functests-arm64-bionic https://zuul.opendev.org/t/openstack/build/2d10c8ecd04e40438a9dcf07ec0d556d : SUCCESS in 18m 46s","accounts_in_message":[],"_revision_number":2},{"id":"db01a1c7ab878427397a4789451eed39f524cefc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-18 08:39:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/7fe31c32dc574fbaa36c3d69dd22bd75 : TIMED_OUT in 4h 10m 42s","accounts_in_message":[],"_revision_number":2},{"id":"a17e51a2fe7c4da1ccb8355c75acc26e0fb9295c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 10:04:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df3280e3eaa243d593baef1b902521e5 : SUCCESS in 6m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7051bc457c7341c48298f9546034edf4 : SUCCESS in 6m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58db17c51069429eb63e3eb06fdcb6b6 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94568c58800a4163af3d47db7342d694 : SUCCESS in 4m 17s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/e8ade1cb858b4fb5af2c64e49ed46f98 : SUCCESS in 49m 59s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/c43ef13712a84155b7c6c870b56ac2fc : SUCCESS in 25m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9d525de13ffb49c4ba519bfe7c3d1834 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/892e0c898c224caabc04ed35a7614f6d : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d77c56d20ff4f87ba03ceb5447514a3 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/983bbe991c644e4982eaafa15350d273 : SUCCESS in 3m 11s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cb4c43447a3e45b68d4ffb45651fd070 : SUCCESS in 3m 28s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b65251f1d3ff431b999077883ba2c4ea : SUCCESS in 1h 46m 45s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/5aa7acec77af401c86159e23314bd82b : SUCCESS in 20m 40s\n- dib-nodepool-functional-openstack-centos-7-src https://zuul.opendev.org/t/openstack/build/e7fc2f53cb2c40cfada755272b12439d : SUCCESS in 32m 23s\n- dib-nodepool-functional-openstack-centos-8-src https://zuul.opendev.org/t/openstack/build/eb3a7b9aa31645d2a540c42c30b0ce10 : SUCCESS in 34m 30s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/0ed63ec437a940db95206e1f74d948f4 : SUCCESS in 37m 01s\n- dib-nodepool-functional-openstack-fedora-31-src https://zuul.opendev.org/t/openstack/build/9b1c17be4caf49748c6970c0c46a0b88 : SUCCESS in 35m 18s\n- dib-nodepool-functional-openstack-fedora-32-src https://zuul.opendev.org/t/openstack/build/5ce2df9d74f74cec92d6d4cc1627ad1d : SUCCESS in 37m 39s\n- dib-nodepool-functional-openstack-ubuntu-xenial-src https://zuul.opendev.org/t/openstack/build/e80811786297445a821ea652818cddb8 : SUCCESS in 32m 28s\n- dib-nodepool-functional-openstack-ubuntu-bionic-src https://zuul.opendev.org/t/openstack/build/cdd78a8f06d74ee08c8d58393bd1f5cd : SUCCESS in 29m 18s\n- dib-nodepool-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/932c52bda5d8441cb7baf7e177432df1 : SUCCESS in 46m 18s\n- dib-nodepool-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/496871ed8b314d5593d038d16e0c1dad : SUCCESS in 52m 49s (non-voting)\n- dib-nodepool-functional-openstack-opensuse-tumbleweed-src https://zuul.opendev.org/t/openstack/build/4c6096492c8840a2a3fcdfd04197bc2c : SUCCESS in 42m 21s (non-voting)\n- dib-functests-bionic-python3 https://zuul.opendev.org/t/openstack/build/a82449906afd494689383efc6b238951 : SUCCESS in 30m 43s\n- dib-functests-bionic-python3-image https://zuul.opendev.org/t/openstack/build/52c67265f0c043bda0dc9be54e4e2209 : SUCCESS in 35m 14s (non-voting)\n- dib-functests-bionic-python3-extras https://zuul.opendev.org/t/openstack/build/74211ff64aa0452ba62d426642a2a78e : SUCCESS in 48m 51s (non-voting)\n- dib-functests-bionic-python3-containers https://zuul.opendev.org/t/openstack/build/5df46e6b3a2343e6bfebd7f8022849cc : SUCCESS in 9m 49s (non-voting)\n- ironic-python-agent-check-image-dib-centos8 https://zuul.opendev.org/t/openstack/build/5dd4348c188842c9ae754f8a4d1536fa : SUCCESS in 10m 31s\n- ipa-tempest-wholedisk-bios-ipmi-direct-dib-src https://zuul.opendev.org/t/openstack/build/5a0250a52d43457483a16a1920bb022c : SUCCESS in 1h 33m 04s (non-voting)\n- ipa-tempest-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/f3002d9b53d94c7d9377df66aa120df4 : SUCCESS in 1h 33m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"995c8ff90808b49671b2707b10810087c3fa2494","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 10:04:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"58c1a36f490a59c272cdb85563d2d629e2255080","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 11:28:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9c4d0e0bbdeaf727577f3786bf24781c22f54713","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 11:28:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f56c16c9342542bea0c1052f4edcd738 : SUCCESS in 6m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18d82e6c1c614992b2d4497edea65f8d : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8e87dc938934467a1626e5f2122f458 : SUCCESS in 3m 24s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/9941ca26a1204adfbe4a53e54c46ec5d : SUCCESS in 57m 11s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/9e7e807c4e094357890106c376259c20 : SUCCESS in 25m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1ddfae5e74094e728b3c13fa845e4e80 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eb64a4a5bc24f53a98bfea58e6a310e : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17a3d1148d714e29b2c629831e805b63 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05a1cb904ffb47bfab7d7d286ef2a347 : SUCCESS in 3m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e19bc4a402dc4444a73e3a4071f627af : SUCCESS in 3m 04s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/86912488d3a24590a9534e85aea8b98b : SUCCESS in 1h 22m 05s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/a7fbbbdbb02e4e65b856bf976f2fa6b5 : SUCCESS in 18m 46s\n- dib-nodepool-functional-openstack-centos-7-src https://zuul.opendev.org/t/openstack/build/c7cea83410f64ded8920931560423412 : SUCCESS in 41m 21s\n- dib-nodepool-functional-openstack-centos-8-src https://zuul.opendev.org/t/openstack/build/638e847018df42bba2a38e729a5c3d0b : SUCCESS in 43m 05s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/e86436a2dc8947dfa478bfe453ec2382 : SUCCESS in 57m 06s\n- dib-nodepool-functional-openstack-fedora-31-src https://zuul.opendev.org/t/openstack/build/9a2093f2080a49199b1dad7b048fbac0 : SUCCESS in 54m 33s\n- dib-nodepool-functional-openstack-fedora-32-src https://zuul.opendev.org/t/openstack/build/a5721ba44dd04f6f8c670f9098a9c5c0 : SUCCESS in 52m 48s\n- dib-nodepool-functional-openstack-ubuntu-xenial-src https://zuul.opendev.org/t/openstack/build/fce87938970a459898f1ef67afb7972c : SUCCESS in 46m 57s\n- dib-nodepool-functional-openstack-ubuntu-bionic-src https://zuul.opendev.org/t/openstack/build/34fc3be9b2084302a5e78a32ae974d28 : SUCCESS in 51m 49s\n- dib-nodepool-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/c2ae97e3876a461c988f61b69c1b6e5a : SUCCESS in 51m 49s\n- dib-functests-bionic-python3 https://zuul.opendev.org/t/openstack/build/7220bb60abb34363abfc6c0ef9aadfda : SUCCESS in 42m 18s\n- ironic-python-agent-check-image-dib-centos8 https://zuul.opendev.org/t/openstack/build/3299c943b38e46d6bf10c3eeef6717cd : SUCCESS in 9m 15s","accounts_in_message":[],"_revision_number":2},{"id":"75c2b16093d5812ae71c7a4a543f6b9d414288ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 11:29:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31c3fc630cd64e9bbce03b86321c7c2d : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92d3b09ed16345a7a0847978f049e9bd : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d6aa43731d89b4d2201fe9252a4e680fcadbba50","revisions":{"722da6f1f0a951869d9fa673af2091d79e8fec96":{"kind":"REWORK","_number":1,"created":"2020-08-11 12:52:05.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/65/745665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/diskimage-builder","ref":"refs/changes/65/745665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/65/745665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/65/745665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/65/745665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/diskimage-builder refs/changes/65/745665/1"}}},"commit":{"parents":[{"commit":"1291dcba516a95fce9d3a51da7e5c73d9da5f584","subject":"Merge \"Deprecate dib-python; remove from in-tree elements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/1291dcba516a95fce9d3a51da7e5c73d9da5f584"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-11 12:49:47.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-11 12:49:47.000000000","tz":120},"subject":"Do not install python2 packages in ubuntu focal","message":"Do not install python2 packages in ubuntu focal\n\nThe packages yum-utils and python-lzma have python 2 dependencies\nand they\u0027re not present for Ubuntu Focal.\nThe package zypper should not be installed as it\u0027s not used.\n\nChange-Id: I3c96ae7ebb814973b77b4a1b966c39658acc5683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/722da6f1f0a951869d9fa673af2091d79e8fec96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/722da6f1f0a951869d9fa673af2091d79e8fec96"}]},"branch":"refs/heads/master"},"d6aa43731d89b4d2201fe9252a4e680fcadbba50":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-12 10:16:46.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/65/745665/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/diskimage-builder","ref":"refs/changes/65/745665/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/65/745665/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/65/745665/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/65/745665/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/diskimage-builder refs/changes/65/745665/2"}}},"commit":{"parents":[{"commit":"1291dcba516a95fce9d3a51da7e5c73d9da5f584","subject":"Merge \"Deprecate dib-python; remove from in-tree elements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/1291dcba516a95fce9d3a51da7e5c73d9da5f584"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-11 12:49:47.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-12 10:16:44.000000000","tz":0},"subject":"Do not install python2 packages in ubuntu focal","message":"Do not install python2 packages in ubuntu focal\n\nThe packages yum-utils and python-lzma have python 2 dependencies\nand they\u0027re not present for Ubuntu Focal.\nThe package zypper should not be installed as it\u0027s not used.\n\nCloses-Bug: 1891299\n\nChange-Id: I3c96ae7ebb814973b77b4a1b966c39658acc5683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/d6aa43731d89b4d2201fe9252a4e680fcadbba50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/d6aa43731d89b4d2201fe9252a4e680fcadbba50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
