)]}'
{"id":"openstack%2Fdiskimage-builder~805960","triplet_id":"openstack%2Fdiskimage-builder~master~I1694924963183e404a79f06b8fe943d9d2377a06","project":"openstack/diskimage-builder","branch":"master","topic":"ironic-centos8-stream","hashtags":[],"change_id":"I1694924963183e404a79f06b8fe943d9d2377a06","subject":"Use Centos 8 Stream jobs for ironic jobs","status":"ABANDONED","created":"2021-08-25 07:54:47.000000000","updated":"2021-09-22 08:42:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d5463263ada287981384c571e4903aea3e598a20","_number":805960,"virtual_id_number":805960,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-25 14:27:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":-1,"date":"2021-08-26 08:29:20.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-25 08:00:36.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-25 08:36:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-25 09:28:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c516d3af4bccac3b89090dfeb750a775ad4d8c0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-08-25 07:54:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1291d404a781da5f6b7c57ea9de269b20a8e7fd9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-25 08:00:36.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- dib-functests-bionic-python3 finger://zuul-v3-executor.novalocal/2b0aff722486498796bfc9fe427ae64c : RETRY_LIMIT in 1m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f98ba732cf91bf457628edbbbd1dc19ea0985628","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 08:36:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- dib-functests-arm64-bionic https://zuul.opendev.org/t/openstack/build/f0f3c84937df4f21a4734730502889f5 : SUCCESS in 39m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c22e0d1c389e75c21516f1ba25e66bc0c8486500","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 09:28:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e20c6fdb7aae463ab4a5feb1b51b0aa8 : SUCCESS in 6m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c4b412c65ef4075b2ef9e2e097c482b : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6116554c54784c11bdbe7c4dfc88875b : SUCCESS in 5m 13s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/5dbc229048d44658b69740dd843d06db : SUCCESS in 54m 29s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/2704daa3e7df49729da2877b5d34c45e : SUCCESS in 30m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cc3ff8b9146c44d1a722e1f425384096 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9af4b1d960614d8d8d53b925b06a7821 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3822b8680c2f4d32b6f5d9e95acb16e7 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f801fbcfe5ea4786b85f82bc0a3d5199 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e39883878d54967964dedadc86aaa1f : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aff498792a874da49d326a5564c62bd1 : SUCCESS in 4m 43s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/aafc14e800e94490a897de585e094e7d : SUCCESS in 1h 08m 37s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/4a1068159be349afb6f33e1d7f225315 : SUCCESS in 18m 23s\n- dib-nodepool-functional-openstack-centos-7-src https://zuul.opendev.org/t/openstack/build/c5021c3b5bfd4ada81667498515fe96e : SUCCESS in 44m 03s\n- dib-nodepool-functional-openstack-centos-8-src https://zuul.opendev.org/t/openstack/build/792975446b0243d595bda5dfc8ac0871 : SUCCESS in 43m 37s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/78a74c8778ba49fb8de440c977e69dc0 : SUCCESS in 35m 31s\n- dib-nodepool-functional-openstack-fedora-34-containerfile-src https://zuul.opendev.org/t/openstack/build/cacea13057414f4486e4a6176c707e1f : SUCCESS in 40m 01s\n- dib-nodepool-functional-openstack-ubuntu-xenial-src https://zuul.opendev.org/t/openstack/build/eeb9e51c85974ef1ba0aed796a24f432 : SUCCESS in 37m 03s\n- dib-nodepool-functional-openstack-ubuntu-bionic-src https://zuul.opendev.org/t/openstack/build/d233f7e613994560942aba5cd48d1367 : SUCCESS in 34m 00s\n- dib-nodepool-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/335854ef80f14db7b1549e4d5ece7fc3 : SUCCESS in 40m 46s\n- dib-nodepool-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/cc24c67ccc4746e3805115ac21fd81f4 : SUCCESS in 42m 14s (non-voting)\n- dib-nodepool-functional-openstack-opensuse-tumbleweed-src https://zuul.opendev.org/t/openstack/build/66777f936b1c4297aec04817875ce63c : FAILURE in 42m 42s (non-voting)\n- dib-functests-bionic-python3 https://zuul.opendev.org/t/openstack/build/09e8cd366f8d47bfaadae26e738cb22b : FAILURE in 35m 35s\n- dib-functests-bionic-python3-image https://zuul.opendev.org/t/openstack/build/5f18a1549c864ed8b9059c0adf91b478 : FAILURE in 11m 57s (non-voting)\n- dib-functests-bionic-python3-extras https://zuul.opendev.org/t/openstack/build/1ea2d9c84fd549759333c36a9b7b981c : TIMED_OUT in 1h 31m 20s (non-voting)\n- dib-functests-bionic-python3-containers https://zuul.opendev.org/t/openstack/build/198bd99b6847492e9309d606b344c67d : SUCCESS in 11m 48s (non-voting)\n- ironic-python-agent-check-image-dib-centos8-stream https://zuul.opendev.org/t/openstack/build/d8ccdda1037c4f4ea95cec82bccde40c : SUCCESS in 13m 18s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/a74881a0f50c4317aa7c4b15603ce67b : SUCCESS in 1h 14m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9bdb583b0c28cda6c7a89d653856a420956fc4f2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-08-25 12:52:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ada59607328d1dbbf159eb3f1c53c18d6e19e9c2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-25 12:58:53.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- dib-functests-bionic-python3 finger://zuul-v3-executor.novalocal/94ad8157bd2447fb93cdaf4eab8a13f1 : RETRY_LIMIT in 1m 00s","accounts_in_message":[],"_revision_number":1},{"id":"972dd652bb3e52e7c9cf06010bc7cd21a2ea21fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 13:33:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- dib-functests-arm64-bionic https://zuul.opendev.org/t/openstack/build/0ffa1d1de81348f7a75863adcb2a5aef : SUCCESS in 38m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ba29204152b3672973fbbd55df4e4b7334b585d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 14:27:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63db52fe5a4a422790675e587d022748 : SUCCESS in 5m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e0af21bd72d4dc7a56ea88a3f127437 : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68e7e708125148af8ecac73b4f688d36 : SUCCESS in 5m 35s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/4a2e4d879f704f6da374c29202043171 : SUCCESS in 1h 02m 46s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/44651e5a29544b3fbc90f6c2e9949d5a : SUCCESS in 37m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4791e25f57f5458eb03894171ab0d224 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04263b1a117444e2a7c9b6a95293fae1 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcf12f4433ff4a818952fce1c9e3d150 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd9cfaa02594433cb5571f7dede05205 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dafe914951354a0181e509dfa5fb1691 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7b674e2819964eeb92a915229a932127 : SUCCESS in 5m 08s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/4851c1f7e5be487490377e01ce0b4c0b : SUCCESS in 1h 26m 47s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/7a2ec11d508e4b878dcba132481daf9a : SUCCESS in 18m 55s\n- dib-nodepool-functional-openstack-centos-7-src https://zuul.opendev.org/t/openstack/build/8e6723f2834c41369de58f20975fe24a : SUCCESS in 39m 44s\n- dib-nodepool-functional-openstack-centos-8-src https://zuul.opendev.org/t/openstack/build/0580f98612c8485e8d3af8c86534b99f : SUCCESS in 35m 27s\n- dib-nodepool-functional-openstack-centos-8-stream-src https://zuul.opendev.org/t/openstack/build/4c257f41174245e5af759412f96b6f48 : SUCCESS in 36m 36s\n- dib-nodepool-functional-openstack-fedora-34-containerfile-src https://zuul.opendev.org/t/openstack/build/f2f7fd2a82364cb69e5fbbf7b5d727dc : SUCCESS in 37m 01s\n- dib-nodepool-functional-openstack-ubuntu-xenial-src https://zuul.opendev.org/t/openstack/build/45fb8f39defb48d0b1af774293f584f9 : SUCCESS in 41m 20s\n- dib-nodepool-functional-openstack-ubuntu-bionic-src https://zuul.opendev.org/t/openstack/build/996e27b26ac645e0b865e3022fd7c2f0 : SUCCESS in 38m 11s\n- dib-nodepool-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/b33debb55bc648f6a45d8cfb915849f6 : SUCCESS in 43m 08s\n- dib-nodepool-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/ec56987934c5459c87c3e806e66e4840 : FAILURE in 28m 46s (non-voting)\n- dib-nodepool-functional-openstack-opensuse-tumbleweed-src https://zuul.opendev.org/t/openstack/build/c8b8cf779a194a4890e527d7943458f5 : FAILURE in 45m 06s (non-voting)\n- dib-functests-bionic-python3 https://zuul.opendev.org/t/openstack/build/860550eddb1548d9ae6049bdb9cc0c43 : SUCCESS in 1h 04m 55s\n- dib-functests-bionic-python3-image https://zuul.opendev.org/t/openstack/build/80659f83697c4a1b9da9cae083af5e66 : FAILURE in 13m 54s (non-voting)\n- dib-functests-bionic-python3-extras https://zuul.opendev.org/t/openstack/build/17559eb3b0ef4e56a3677a0accf6b2f0 : TIMED_OUT in 1h 31m 32s (non-voting)\n- dib-functests-bionic-python3-containers https://zuul.opendev.org/t/openstack/build/be010d075ff6423698ef39e012072fed : SUCCESS in 11m 54s (non-voting)\n- ironic-python-agent-check-image-dib-centos8-stream https://zuul.opendev.org/t/openstack/build/bfe4b12178094126898e41511dfeb154 : SUCCESS in 13m 00s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/d629a5f2a96843f5bce7ab9c863cb41f : SUCCESS in 1h 14m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92718e2261abb07167ca2884b15ca44d56c8a71d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-08-26 08:29:20.000000000","message":"Patch Set 1: Workflow-1\n\nwe won\u0027t rename the jobs, but this is still valid as a test\nand it looks good!","accounts_in_message":[],"_revision_number":1},{"id":"d5463263ada287981384c571e4903aea3e598a20","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-09-22 08:42:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f77dfc748433640f1b0a9fb45f2d39b3159da1de","revisions":{"f77dfc748433640f1b0a9fb45f2d39b3159da1de":{"kind":"REWORK","_number":1,"created":"2021-08-25 07:54:47.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/60/805960/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/diskimage-builder","ref":"refs/changes/60/805960/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/60/805960/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/60/805960/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/60/805960/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/diskimage-builder refs/changes/60/805960/1"}}},"commit":{"parents":[{"commit":"a6ee4d0c217335cf83ca9a42de78263d085ab839","subject":"Introduce openEuler distro","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/a6ee4d0c217335cf83ca9a42de78263d085ab839"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2021-08-25 07:54:07.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2021-08-25 07:54:07.000000000","tz":120},"subject":"Use Centos 8 Stream jobs for ironic jobs","message":"Use Centos 8 Stream jobs for ironic jobs\n\nDepends-On: https://review.opendev.org/c/openstack/ironic-python-agent-builder/+/805957\nChange-Id: I1694924963183e404a79f06b8fe943d9d2377a06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/f77dfc748433640f1b0a9fb45f2d39b3159da1de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/f77dfc748433640f1b0a9fb45f2d39b3159da1de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
