)]}'
{"id":"openstack%2Fdiskimage-builder~89531","triplet_id":"openstack%2Fdiskimage-builder~master~I3a351d0f9854f6f320ba145a71ffc66fd02c231c","project":"openstack/diskimage-builder","branch":"master","topic":"map-packages/openjdk","hashtags":[],"change_id":"I3a351d0f9854f6f320ba145a71ffc66fd02c231c","subject":"Map openjdk-7-jre-headless to RHEL+SUSE","status":"MERGED","created":"2014-04-22 10:42:12.000000000","updated":"2014-04-29 09:34:49.000000000","submitted":"2014-04-29 08:10:37.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f1396a7cc0b9b43529f47a3d03618c976f6eb322","_number":89531,"virtual_id_number":89531,"owner":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},{"value":2,"date":"2014-04-28 16:27:41.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9369,"name":"Steve Kowalik","email":"steven@wedontsleep.org","username":"stevenk"},{"value":0,"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},{"value":0,"_account_id":1726,"name":"Ghe Rivero","email":"ghe.rivero@gmail.com","username":"ghe.rivero"},{"value":0,"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":0,"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},{"value":2,"date":"2014-04-28 16:31:24.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2014-04-24 07:49:19.000000000","_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2014-04-24 08:07:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9369,"name":"Steve Kowalik","email":"steven@wedontsleep.org","username":"stevenk"},{"value":1,"date":"2014-04-24 07:49:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},{"value":2,"date":"2014-04-24 07:49:19.000000000","_account_id":1726,"name":"Ghe Rivero","email":"ghe.rivero@gmail.com","username":"ghe.rivero"},{"value":2,"date":"2014-04-29 07:29:58.000000000","_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},"all":[{"value":0,"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2014-04-29 09:34:49.000000000","post_submit":true,"_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9369,"name":"Steve Kowalik","email":"steven@wedontsleep.org","username":"stevenk"},{"value":0,"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},{"value":0,"_account_id":1726,"name":"Ghe Rivero","email":"ghe.rivero@gmail.com","username":"ghe.rivero"},{"value":1,"date":"2014-04-29 07:29:58.000000000","_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1726,"name":"Ghe Rivero","email":"ghe.rivero@gmail.com","username":"ghe.rivero"},{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},{"_account_id":9369,"name":"Steve Kowalik","email":"steven@wedontsleep.org","username":"stevenk"},{"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},{"_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-04-24 07:49:19.000000000","updated_by":{"_account_id":1726,"name":"Ghe Rivero","email":"ghe.rivero@gmail.com","username":"ghe.rivero"},"reviewer":{"_account_id":1726,"name":"Ghe Rivero","email":"ghe.rivero@gmail.com","username":"ghe.rivero"},"state":"REVIEWER"},{"updated":"2014-04-24 07:49:19.000000000","updated_by":{"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},"reviewer":{"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},"state":"REVIEWER"},{"updated":"2014-04-24 07:49:19.000000000","updated_by":{"_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},"reviewer":{"_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},"state":"REVIEWER"},{"updated":"2014-04-24 08:07:44.000000000","updated_by":{"_account_id":9369,"name":"Steve Kowalik","email":"steven@wedontsleep.org","username":"stevenk"},"reviewer":{"_account_id":9369,"name":"Steve Kowalik","email":"steven@wedontsleep.org","username":"stevenk"},"state":"REVIEWER"},{"updated":"2014-04-28 16:27:41.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2014-04-28 16:31:24.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2014-04-29 07:29:58.000000000","updated_by":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"reviewer":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"state":"REVIEWER"}],"messages":[{"id":"2afcb27926cf5516f944cc41b644a3871fe3145c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-22 10:42:12.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"7810bacfb826324911b0e928b12ad019fc018345","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-22 10:45:54.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- gate-diskimage-builder-pep8 http://logs.openstack.org/31/89531/1/check/gate-diskimage-builder-pep8/e79e644 : SUCCESS in 31s\n- gate-diskimage-builder-python26 http://logs.openstack.org/31/89531/1/check/gate-diskimage-builder-python26/d79dcdc : SUCCESS in 20s\n- gate-diskimage-builder-python27 http://logs.openstack.org/31/89531/1/check/gate-diskimage-builder-python27/27bfe5e : SUCCESS in 45s\n- check-oslotest-dsvm-diskimage-builder http://logs.openstack.org/31/89531/1/check/check-oslotest-dsvm-diskimage-builder/bba39c1 : SUCCESS in 3m 40s\n","accounts_in_message":[],"_revision_number":1},{"id":"ceb4a5feba78c12471c829987b1b28b4b2d823d1","author":{"_account_id":1726,"name":"Ghe Rivero","email":"ghe.rivero@gmail.com","username":"ghe.rivero"},"date":"2014-04-22 10:55:46.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\nSmall nit. Other than that, lgtm","accounts_in_message":[],"_revision_number":1},{"id":"f545e534e54f6ecfb38934e7b83f4922f9c3e045","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-23 05:30:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-tripleo-seed-precise http://logs.openstack.org/31/89531/1/check-tripleo/check-tripleo-seed-precise/14d3ec1 : SUCCESS in 39m 34s\n- check-tripleo-ironic-seed-precise http://logs.openstack.org/31/89531/1/check-tripleo/check-tripleo-ironic-seed-precise/bf9d4ec : SUCCESS in 37m 26s\n- check-tripleo-undercloud-precise http://logs.openstack.org/31/89531/1/check-tripleo/check-tripleo-undercloud-precise/2eb9818 : SUCCESS in 1h 07m 05s\n- check-tripleo-ironic-undercloud-precise http://logs.openstack.org/31/89531/1/check-tripleo/check-tripleo-ironic-undercloud-precise/f8bd619 : FAILURE in 55m 47s (non-voting)\n- check-tripleo-overcloud-precise http://logs.openstack.org/31/89531/1/check-tripleo/check-tripleo-overcloud-precise/b91920d : SUCCESS in 1h 34m 03s\n- check-tripleo-overcloud-f20 http://logs.openstack.org/31/89531/1/check-tripleo/check-tripleo-overcloud-f20/92ec839 : FAILURE in 49m 50s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"bd9a9b16838f66959a5f9d24af54770726ab735f","author":{"_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},"date":"2014-04-23 07:50:45.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\nAlso agree with Ghe on breaking alpha ordering into another patch.","accounts_in_message":[],"_revision_number":1},{"id":"b1f8a4c0b48f35a7a1d774ce23cb1f2a6d89f8c1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-23 07:54:25.000000000","message":"Patch Set 2:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"a6e3e4fe24add9417ca70751d1aa699c1f5edc8b","author":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"date":"2014-04-23 07:54:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f00382ae48b10f2954e42aa7afeec5414926af2a","author":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"date":"2014-04-23 08:13:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7bd03fb079485381203e5f1d758550fc6b85b4ea","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-23 08:13:22.000000000","message":"Patch Set 3:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"1e393a4fdeb5607a3eedbdbfcb5097093b9ad89d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-23 08:30:46.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- gate-diskimage-builder-pep8 http://logs.openstack.org/31/89531/3/check/gate-diskimage-builder-pep8/82e1b81 : SUCCESS in 53s\n- gate-diskimage-builder-python26 http://logs.openstack.org/31/89531/3/check/gate-diskimage-builder-python26/c5d1e73 : SUCCESS in 42s\n- gate-diskimage-builder-python27 http://logs.openstack.org/31/89531/3/check/gate-diskimage-builder-python27/40bb3b2 : SUCCESS in 1m 01s\n- check-oslotest-dsvm-diskimage-builder http://logs.openstack.org/31/89531/3/check/check-oslotest-dsvm-diskimage-builder/6bc3304 : SUCCESS in 3m 55s\n","accounts_in_message":[],"_revision_number":3},{"id":"2e94f67849cc38e1ee8ec98c23caba80120b5e70","author":{"_account_id":1726,"name":"Ghe Rivero","email":"ghe.rivero@gmail.com","username":"ghe.rivero"},"date":"2014-04-23 08:40:13.000000000","message":"Patch Set 3: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":3},{"id":"e3939fc406fccbb83f4bbc859d824a7ecff16d8a","author":{"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},"date":"2014-04-23 08:56:59.000000000","message":"Patch Set 3: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":3},{"id":"6c313fed54c7c225d4f214db9b2220efd6950b45","author":{"_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},"date":"2014-04-23 20:23:42.000000000","message":"Patch Set 3: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":3},{"id":"447460d16b091583fb433130fce77c86c875cb2a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-23 20:23:43.000000000","message":"Patch Set 3: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"9f188515fb116a43044ee3c9861f78958fb2bfb3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-23 23:18:51.000000000","message":"Patch Set 3: Fails\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"814d8229670d911c515afa3585ffc9844de2879f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-24 04:28:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- check-tripleo-seed-precise http://logs.openstack.org/31/89531/3/check-tripleo/check-tripleo-seed-precise/102007e : SUCCESS in 38m 10s\n- check-tripleo-ironic-seed-precise http://logs.openstack.org/31/89531/3/check-tripleo/check-tripleo-ironic-seed-precise/8b93c13 : SUCCESS in 32m 01s\n- check-tripleo-undercloud-precise http://logs.openstack.org/31/89531/3/check-tripleo/check-tripleo-undercloud-precise/17e9080 : SUCCESS in 45m 34s\n- check-tripleo-ironic-undercloud-precise http://logs.openstack.org/31/89531/3/check-tripleo/check-tripleo-ironic-undercloud-precise/f5f6d0d : FAILURE in 40m 21s (non-voting)\n- check-tripleo-overcloud-precise http://logs.openstack.org/31/89531/3/check-tripleo/check-tripleo-overcloud-precise/86fc7a6 : SUCCESS in 1h 07m 23s\n- check-tripleo-overcloud-f20 http://logs.openstack.org/31/89531/3/check-tripleo/check-tripleo-overcloud-f20/a850124 : FAILURE in 32m 50s (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"1223b6e7a45cfa323b38bc952652befa2c1c8d30","author":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"date":"2014-04-24 07:49:12.000000000","message":"Patch Set 3: Rebased","accounts_in_message":[],"_revision_number":3},{"id":"835a3c0a733d08e13aec27c855a75a2cc5f9d05a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-24 07:49:14.000000000","message":"Patch Set 4:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"0ebabe835d2b2aa3fab99f33fe8aa169a44750d4","author":{"_account_id":1726,"name":"Ghe Rivero","email":"ghe.rivero@gmail.com","username":"ghe.rivero"},"date":"2014-04-24 07:49:19.000000000","message":"Patch Set 4: Looks good to me (core reviewer)\n\nAutomatically re-added by Gerrit trivial rebase detection script.","accounts_in_message":[],"_revision_number":4},{"id":"f605eaf785ad3bf618adeff9f7cb1733b91b726f","author":{"_account_id":9420,"name":"Matthew Gilliard","email":"matthew.gilliard@hp.com","username":"gilliard"},"date":"2014-04-24 07:49:19.000000000","message":"Patch Set 4: Looks good to me, but someone else must approve\n\nAutomatically re-added by Gerrit trivial rebase detection script.","accounts_in_message":[],"_revision_number":4},{"id":"565a4ed8f861289fa4147319d019a375b6162900","author":{"_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},"date":"2014-04-24 07:49:19.000000000","message":"Patch Set 4: Looks good to me (core reviewer)\n\nAutomatically re-added by Gerrit trivial rebase detection script.","accounts_in_message":[],"_revision_number":4},{"id":"a0e4e7f734bb64e73bed7c575ce5065d3453363d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-24 07:54:03.000000000","message":"Patch Set 4: Works for me\n\nBuild succeeded.\n\n- gate-diskimage-builder-pep8 http://logs.openstack.org/31/89531/4/check/gate-diskimage-builder-pep8/e0fecd5 : SUCCESS in 40s\n- gate-diskimage-builder-python26 http://logs.openstack.org/31/89531/4/check/gate-diskimage-builder-python26/e52d5a4 : SUCCESS in 24s\n- gate-diskimage-builder-python27 http://logs.openstack.org/31/89531/4/check/gate-diskimage-builder-python27/1b5d99a : SUCCESS in 21s\n- check-oslotest-dsvm-diskimage-builder http://logs.openstack.org/31/89531/4/check/check-oslotest-dsvm-diskimage-builder/5cc9f76 : SUCCESS in 4m 43s\n","accounts_in_message":[],"_revision_number":4},{"id":"8dedc4099cf943701710db9026d3e23e55ba5501","author":{"_account_id":9369,"name":"Steve Kowalik","email":"steven@wedontsleep.org","username":"stevenk"},"date":"2014-04-24 08:07:44.000000000","message":"Patch Set 4: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":4},{"id":"bc17de3c73b0c4d18052de22321ca3ae07621050","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-24 09:48:27.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n\n- check-tripleo-seed-precise http://logs.openstack.org/31/89531/4/check-tripleo/check-tripleo-seed-precise/3381269 : FAILURE in 9m 28s\n- check-tripleo-ironic-seed-precise http://logs.openstack.org/31/89531/4/check-tripleo/check-tripleo-ironic-seed-precise/f2fe76f : SUCCESS in 35m 12s\n- check-tripleo-undercloud-precise http://logs.openstack.org/31/89531/4/check-tripleo/check-tripleo-undercloud-precise/8ddb76d : SUCCESS in 46m 33s\n- check-tripleo-ironic-undercloud-precise http://logs.openstack.org/31/89531/4/check-tripleo/check-tripleo-ironic-undercloud-precise/836f8ed : FAILURE in 56m 44s (non-voting)\n- check-tripleo-overcloud-precise http://logs.openstack.org/31/89531/4/check-tripleo/check-tripleo-overcloud-precise/5697593 : SUCCESS in 1h 07m 16s\n- check-tripleo-overcloud-f20 http://logs.openstack.org/31/89531/4/check-tripleo/check-tripleo-overcloud-f20/feeea23 : SUCCESS in 1h 52m 00s (non-voting)\n","accounts_in_message":[],"_revision_number":4},{"id":"613f6558201145266b332325a8af8fac82aaef44","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2014-04-28 16:27:40.000000000","message":"Patch Set 4:\n\nrecheck bug 1292141","accounts_in_message":[],"_revision_number":4},{"id":"535991f565516100082ce894508cc38290b61299","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-28 16:27:41.000000000","message":"Patch Set 4: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"9d0b6d8f5de5b567e69d6a58e4bdc1205745d717","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2014-04-28 16:31:24.000000000","message":"Patch Set 4: Looks good to me (core reviewer); Approved\n\nOh, this passed CI on Apr. 23 and the rebase just hit a pypi bug.  And since this is just adding package mappings it should be safe.","accounts_in_message":[],"_revision_number":4},{"id":"de47135057522cf163ec4610a68ace32621e4ca2","author":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"date":"2014-04-29 07:29:58.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"09cd492bae7ad552da22267edbd2e58a14249b58","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-29 07:30:56.000000000","message":"Patch Set 4:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"efa479cc3fd9323d78d999faa7ebe724485101df","author":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"date":"2014-04-29 07:30:56.000000000","message":"Patch Set 4:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":4},{"id":"5a57be148c72b9d61944db076119915e72d365fb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-29 07:48:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- gate-diskimage-builder-pep8 http://logs.openstack.org/31/89531/4/check/gate-diskimage-builder-pep8/c330d03 : SUCCESS in 45s\n- gate-diskimage-builder-python26 http://logs.openstack.org/31/89531/4/check/gate-diskimage-builder-python26/47a07bc : SUCCESS in 1m 07s\n- gate-diskimage-builder-python27 http://logs.openstack.org/31/89531/4/check/gate-diskimage-builder-python27/9aec3f7 : SUCCESS in 50s\n- check-oslotest-dsvm-diskimage-builder http://logs.openstack.org/31/89531/4/check/check-oslotest-dsvm-diskimage-builder/aa1533e : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":4},{"id":"750687107141f8ed820ca5c985a17bd3469cbecf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-29 07:48:26.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"dd60001c7ee2b1a4d5a1815c0a9b6c3e4ea5e540","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-29 08:10:37.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded.\n\n- gate-diskimage-builder-pep8 http://logs.openstack.org/31/89531/4/gate/gate-diskimage-builder-pep8/ca5c47f : SUCCESS in 44s\n- gate-diskimage-builder-python26 http://logs.openstack.org/31/89531/4/gate/gate-diskimage-builder-python26/71a127d : SUCCESS in 39s\n- gate-diskimage-builder-python27 http://logs.openstack.org/31/89531/4/gate/gate-diskimage-builder-python27/db10ec1 : SUCCESS in 1m 12s\n- gate-oslotest-dsvm-diskimage-builder http://logs.openstack.org/31/89531/4/gate/gate-oslotest-dsvm-diskimage-builder/033ace0 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":4},{"id":"30fec15c006306fefe59d5a6f161131db0068293","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-29 08:10:37.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":4},{"id":"edf4abfb6aa1a894e3fd645541b14c37100846dd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-29 09:34:49.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- check-tripleo-seed-precise http://logs.openstack.org/31/89531/4/check-tripleo/check-tripleo-seed-precise/021c9e4 : SUCCESS in 36m 17s\n- check-tripleo-ironic-seed-precise http://logs.openstack.org/31/89531/4/check-tripleo/check-tripleo-ironic-seed-precise/5f4e31d : SUCCESS in 41m 07s\n- check-tripleo-undercloud-precise http://logs.openstack.org/31/89531/4/check-tripleo/check-tripleo-undercloud-precise/1720f24 : SUCCESS in 1h 17m 21s\n- check-tripleo-ironic-undercloud-precise http://logs.openstack.org/31/89531/4/check-tripleo/check-tripleo-ironic-undercloud-precise/ec0f8ea : FAILURE in 1h 03m 55s (non-voting)\n- check-tripleo-overcloud-precise http://logs.openstack.org/31/89531/4/check-tripleo/check-tripleo-overcloud-precise/3d71d17 : SUCCESS in 1h 07m 25s\n- check-tripleo-overcloud-f20 http://logs.openstack.org/31/89531/4/check-tripleo/check-tripleo-overcloud-f20/54d2ff9 : SUCCESS in 2h 03m 51s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"21f01e937cfbdbaadefee2631c5f8c31f8f76c3b","revisions":{"9da37da35a64629f3d05a9421c894490b19bad85":{"kind":"REWORK","_number":1,"created":"2014-04-22 10:42:12.000000000","uploader":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"ref":"refs/changes/31/89531/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/diskimage-builder","ref":"refs/changes/31/89531/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/31/89531/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/31/89531/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/31/89531/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/diskimage-builder refs/changes/31/89531/1"}}},"commit":{"parents":[{"commit":"b588e4af05f52926b6ed49500c377460c86d0872","subject":"Merge \"Use provides to note an element provides deploy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/b588e4af05f52926b6ed49500c377460c86d0872"}]}],"author":{"name":"Alexis Lee","email":"alexisl@hp.com","date":"2014-04-22 10:38:55.000000000","tz":60},"committer":{"name":"Alexis Lee","email":"alexisl@hp.com","date":"2014-04-22 10:42:08.000000000","tz":60},"subject":"Map openjdk-7-jre to RHEL+SUSE","message":"Map openjdk-7-jre to RHEL+SUSE\n\nProvide package name mappings for openjdk-7-jre:\n        RHEL/Fedora: java-1.7.0-openjdk\n        SUSE: java-1_7_0-openjdk\n\nSee https://review.openstack.org/#/c/86314/3\n\nChange-Id: I3a351d0f9854f6f320ba145a71ffc66fd02c231c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/9da37da35a64629f3d05a9421c894490b19bad85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/9da37da35a64629f3d05a9421c894490b19bad85"}]},"branch":"refs/heads/master"},"9c52599c911074aa9a9d86b77ad6c421f4736045":{"kind":"REWORK","_number":2,"created":"2014-04-23 07:54:25.000000000","uploader":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"ref":"refs/changes/31/89531/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/diskimage-builder","ref":"refs/changes/31/89531/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/31/89531/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/31/89531/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/31/89531/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/diskimage-builder refs/changes/31/89531/2"}}},"commit":{"parents":[{"commit":"6b2a78f3abdcb7133ded96324f30907739f8f855","subject":"Sort rhel/bin/map-packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/6b2a78f3abdcb7133ded96324f30907739f8f855"}]}],"author":{"name":"Alexis Lee","email":"alexisl@hp.com","date":"2014-04-22 10:38:55.000000000","tz":60},"committer":{"name":"Alexis Lee","email":"alexisl@hp.com","date":"2014-04-23 07:54:00.000000000","tz":60},"subject":"Map openjdk-7-jre to RHEL+SUSE","message":"Map openjdk-7-jre to RHEL+SUSE\n\nProvide package name mappings for openjdk-7-jre:\n        RHEL/Fedora: java-1.7.0-openjdk\n        SUSE: java-1_7_0-openjdk\n\nSee https://review.openstack.org/#/c/86314/3\n\nChange-Id: I3a351d0f9854f6f320ba145a71ffc66fd02c231c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/9c52599c911074aa9a9d86b77ad6c421f4736045"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/9c52599c911074aa9a9d86b77ad6c421f4736045"}]},"branch":"refs/heads/master"},"2b09776674a1c156adfa42611a0996cd56188e64":{"kind":"REWORK","_number":3,"created":"2014-04-23 08:13:21.000000000","uploader":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"ref":"refs/changes/31/89531/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/diskimage-builder","ref":"refs/changes/31/89531/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/31/89531/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/31/89531/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/31/89531/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/diskimage-builder refs/changes/31/89531/3"}}},"commit":{"parents":[{"commit":"0fc955b77e839704f9c769f95a49b25eb5229d18","subject":"Sort rhel/bin/map-packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/0fc955b77e839704f9c769f95a49b25eb5229d18"}]}],"author":{"name":"Alexis Lee","email":"alexisl@hp.com","date":"2014-04-22 10:38:55.000000000","tz":60},"committer":{"name":"Alexis Lee","email":"alexisl@hp.com","date":"2014-04-23 08:13:05.000000000","tz":60},"subject":"Map openjdk-7-jre-headless to RHEL+SUSE","message":"Map openjdk-7-jre-headless to RHEL+SUSE\n\nProvide package name mappings for openjdk-7-jre-headless:\n        RHEL/Fedora: java-1.7.0-openjdk-headless\n        SUSE: java-1_7_0-openjdk-headless\n\nSee https://review.openstack.org/#/c/86314\n\nChange-Id: I3a351d0f9854f6f320ba145a71ffc66fd02c231c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/2b09776674a1c156adfa42611a0996cd56188e64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/2b09776674a1c156adfa42611a0996cd56188e64"}]},"branch":"refs/heads/master"},"21f01e937cfbdbaadefee2631c5f8c31f8f76c3b":{"kind":"NO_CHANGE","_number":4,"created":"2014-04-23 08:13:21.000000000","uploader":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"ref":"refs/changes/31/89531/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/diskimage-builder","ref":"refs/changes/31/89531/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/31/89531/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/31/89531/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/31/89531/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/diskimage-builder refs/changes/31/89531/4"}}},"commit":{"parents":[{"commit":"6b2a78f3abdcb7133ded96324f30907739f8f855","subject":"Sort rhel/bin/map-packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/6b2a78f3abdcb7133ded96324f30907739f8f855"}]}],"author":{"name":"Alexis Lee","email":"alexisl@hp.com","date":"2014-04-22 10:38:55.000000000","tz":60},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2014-04-24 07:49:12.000000000","tz":0},"subject":"Map openjdk-7-jre-headless to RHEL+SUSE","message":"Map openjdk-7-jre-headless to RHEL+SUSE\n\nProvide package name mappings for openjdk-7-jre-headless:\n        RHEL/Fedora: java-1.7.0-openjdk-headless\n        SUSE: java-1_7_0-openjdk-headless\n\nSee https://review.openstack.org/#/c/86314\n\nChange-Id: I3a351d0f9854f6f320ba145a71ffc66fd02c231c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/21f01e937cfbdbaadefee2631c5f8c31f8f76c3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/21f01e937cfbdbaadefee2631c5f8c31f8f76c3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
