)]}'
{"id":"openstack%2Fdiskimage-builder~983813","triplet_id":"openstack%2Fdiskimage-builder~master~I8183931b1b962b7916265e3b5c3ea1175bd57c4d","project":"openstack/diskimage-builder","branch":"master","topic":"tarball","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-09-29 00:19:30.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-08-31 22:34:26.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"4571":{"account":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"last_update":"2026-09-29 00:19:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8183931b1b962b7916265e3b5c3ea1175bd57c4d","subject":"Refactor 02-set-machine-id into an element","status":"MERGED","created":"2026-04-09 05:02:17.000000000","updated":"2026-09-29 00:20:43.000000000","submitted":"2026-09-29 00:19:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983813-tarball","meta_rev_id":"e9dbd7599d0ece8e008219d70c078affee48f14f","_number":983813,"virtual_id_number":983813,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-29 00:19:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2026-09-28 22:51:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2026-09-28 21:58:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2026-09-28 21:58:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2026-09-28 22:51:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-09 06:34:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-09 06:39:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-10 01:41:27.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"CC"},{"updated":"2026-08-31 21:12:32.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2026-08-31 21:17:59.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-09-18 14:33:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-09-18 14:33:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-09-28 22:51:34.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"bd6ba478433f0777d4a24178fcd89931ca55080b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-04-09 05:02:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7dacb6e26b52554417b56b5551bb7730c5bae5b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 06:34:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df9e41181f15438b90e38dded92e6d47\n\n- dib-functests-arm64-deb https://zuul.opendev.org/t/openstack/build/af69000f62414485b19c987b47ed5690 : TIMED_OUT in 1h 30m 56s\n- dib-functests-arm64-rpm https://zuul.opendev.org/t/openstack/build/ff388e5bf64b41b8b078ff4f2979438c : SUCCESS in 29m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6c41555209119dee9daf4166fba034c7db84476c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 06:39:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f925d887aa942809e7f4c60b87608f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0098514f155449f9abfad21a29eab040 : SUCCESS in 4m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df0b3587a226418aa55ddbedfae9c5eb : SUCCESS in 3m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b0978c7d8c534656b17a9a23ce7a41ac : SUCCESS in 3m 28s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/8d066d2bcc0a44ac83617c7ae671d9e6 : SUCCESS in 3m 47s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/2495a7df90ed4264bbd4af0b54b51d80 : SUCCESS in 3m 08s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/9132095176d94fe197dbfda2b06e47a9 : SUCCESS in 2m 49s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/685a741bb2084c80865b9097d7c708c6 : SUCCESS in 3m 57s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/9f2b089548d04c7799afd73ecf9025cd : SUCCESS in 2m 51s\n- tox-py313 https://zuul.opendev.org/t/openstack/build/25b380bcc31f4865b892904562c36368 : SUCCESS in 2m 24s\n- dib-functests https://zuul.opendev.org/t/openstack/build/cb77d6c982aa4bd7a02b7a1e045a6707 : SUCCESS in 48m 42s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/7fdcf28176da4bb498d7b2076fcb2774 : SUCCESS in 3m 24s\n- dib-devstack-functional-openstack-almalinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/4762871026524c16b2b0e99299a027e2 : FAILURE in 18m 25s (non-voting)\n- dib-devstack-functional-openstack-almalinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/d2fc62c347ab4663a797470c3f93631d : FAILURE in 24m 28s (non-voting)\n- dib-devstack-functional-openstack-almalinux-10-v2-containerfile-src https://zuul.opendev.org/t/openstack/build/3794099d3b034f3e8243712fd0fa1c90 : FAILURE in 26m 53s (non-voting)\n- dib-devstack-functional-openstack-centos-9-stream-src https://zuul.opendev.org/t/openstack/build/ada6bba6ae014287961b6ec77a3524b2 : FAILURE in 33m 31s\n- dib-devstack-functional-openstack-centos-10-stream-src https://zuul.opendev.org/t/openstack/build/56acbb6a34bc49c3b8a0f340c7b2f4be : FAILURE in 22m 56s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/e2e268ed74ed429385c572b000a6b022 : FAILURE in 19m 19s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/be7c272377514d21ad8ea145f6681f00 : FAILURE in 19m 12s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-10-containerfile-src https://zuul.opendev.org/t/openstack/build/3355f7b0f980440e930da32244405b2b : FAILURE in 24m 18s (non-voting)\n- dib-devstack-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/91467c912f90456e82acc306d613df8e : FAILURE in 35m 45s\n- dib-devstack-functional-openstack-ubuntu-jammy-src https://zuul.opendev.org/t/openstack/build/d024f4731c61452184e2513f5369d4ed : FAILURE in 32m 30s\n- dib-devstack-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/0fd8a6ef9af146fdbd87feeb1279b1b2 : FAILURE in 15m 24s\n- dib-devstack-functional-openstack-ubuntu-resolute-src https://zuul.opendev.org/t/openstack/build/e0119b829a804cf89bc60b8f68285e3e : FAILURE in 53m 55s\n- dib-devstack-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/2a53e26b025848f8bd1a397c2369f283 : FAILURE in 30m 25s (non-voting)\n- dib-devstack-functional-openstack-gentoo-src https://zuul.opendev.org/t/openstack/build/ce46c4c04e5f4df589b7a569f542d5e8 : FAILURE in 1h 35m 06s (non-voting)\n- dib-devstack-functional-openstack-debian-bullseye-src https://zuul.opendev.org/t/openstack/build/e3a967fbcce7425598784f74416de4de : FAILURE in 28m 14s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/e4d66ee064d34e41b0d53af98b8e56f8 : SUCCESS in 7m 44s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/093c1717833a4a589f54529a93ec8e33 : SUCCESS in 1h 08m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d22cfd4b534da4b8fe75d8d1d77be77c4c5e8f1","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2026-04-10 01:41:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec414ad810c07d4723346335f672e4f43111e627","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2026-04-10 01:41:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"56547fde650c1b9770e5320a6a735a6178dbe7a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 03:12:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc0bca5f8e4349b59fa722e6d3f8ffb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8287272f1144f7db248b5751bdbb1ba : SUCCESS in 4m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35701b3f51594a7ebcd9c61c8e440494 : SUCCESS in 3m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9941dab6c5014518a0522631d42abb47 : SUCCESS in 3m 35s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/51ab1a6da2754a5cb1931569879c54d7 : SUCCESS in 3m 39s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/6024f0bb8aeb46b3984996f50113f362 : SUCCESS in 3m 29s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/680b5e3772fb4dc69ac18da9e9709911 : SUCCESS in 6m 33s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/960ef6ca210b4965bc7e2eb4019bcafc : SUCCESS in 4m 25s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/e5c7b68a878a41959ab8ea86db447fa4 : SUCCESS in 3m 30s\n- tox-py313 https://zuul.opendev.org/t/openstack/build/0feee39d1879475e85ca73749db6d947 : SUCCESS in 4m 16s\n- dib-functests https://zuul.opendev.org/t/openstack/build/7de0e534128b4207be9530039bbea65f : SUCCESS in 1h 03m 16s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/9df643763ebc48828bc7e4bcabc70142 : SUCCESS in 3m 15s\n- dib-devstack-functional-openstack-almalinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/c7fecba912fb4d949a1def37b3adad93 : FAILURE in 21m 03s (non-voting)\n- dib-devstack-functional-openstack-almalinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/02923369633d4f578e39b2fc82e04982 : FAILURE in 41m 16s (non-voting)\n- dib-devstack-functional-openstack-almalinux-10-v2-containerfile-src https://zuul.opendev.org/t/openstack/build/28ead06a6e96475da742933224e01602 : FAILURE in 34m 34s (non-voting)\n- dib-devstack-functional-openstack-centos-9-stream-src https://zuul.opendev.org/t/openstack/build/1bcd6e24f0464381abfe7754214ee5b9 : FAILURE in 36m 09s\n- dib-devstack-functional-openstack-centos-10-stream-src https://zuul.opendev.org/t/openstack/build/59111624c99944aeae69e4fc1dfff20a : FAILURE in 25m 06s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/787c9f40a67a49039e0a1ac3a1d6a583 : FAILURE in 27m 38s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/ff3f25f565424e04a1c9aa01958a5e2f : FAILURE in 35m 12s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-10-containerfile-src https://zuul.opendev.org/t/openstack/build/66909be744e94b8c894fefe3b2fff1f5 : FAILURE in 25m 28s (non-voting)\n- dib-devstack-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/8788558b5e614eec9f620ca7f9b164fc : FAILURE in 1h 01m 05s\n- dib-devstack-functional-openstack-ubuntu-jammy-src https://zuul.opendev.org/t/openstack/build/fe8489bd40144e80821281f18f738db8 : FAILURE in 16m 04s\n- dib-devstack-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/04046b518e9444d4903ed807105490fd : FAILURE in 16m 09s\n- dib-devstack-functional-openstack-ubuntu-resolute-src https://zuul.opendev.org/t/openstack/build/c707ca883a734af8a504921708294603 : FAILURE in 51m 47s\n- dib-devstack-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/cd6b7d8aa0a0460f978de92a6a1c5ac5 : FAILURE in 29m 23s (non-voting)\n- dib-devstack-functional-openstack-gentoo-src https://zuul.opendev.org/t/openstack/build/84f7765b9ccc4fdeab3e09dae07b9997 : FAILURE in 1h 26m 20s (non-voting)\n- dib-devstack-functional-openstack-debian-bullseye-src https://zuul.opendev.org/t/openstack/build/9ee0c845445b4e62855fddb23a062daf : FAILURE in 14m 49s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/682a889f85e74dbd88d51ad2ffc481b1 : SUCCESS in 5m 43s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/954349dad3be49488243f7e9d2b9c177 : SUCCESS in 30m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6ff6a61c36042b66ec1bb8c4982acd5fbf1c572","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 03:14:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20f6b78b27154ae8af738f0074ae2969\n\n- dib-functests-arm64-deb https://zuul.opendev.org/t/openstack/build/826ca55ab8534530b7f89c351e2e8ca5 : TIMED_OUT in 1h 30m 46s\n- dib-functests-arm64-rpm https://zuul.opendev.org/t/openstack/build/e0bb29e3d0754e4187865ded2b257469 : SUCCESS in 34m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b1ddab3eb6325c7376fe56995336ae7035293000","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-04-14 23:07:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"73dca95e7f4bd6bfafeb364c30a0f80d5e99e36b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 23:53:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2b0aebf1eda49fd89992adca3aecd9c\n\n- dib-functests-arm64-deb https://zuul.opendev.org/t/openstack/build/cf481a3e41b64868a8b6a5281d2dd006 : SUCCESS in 34m 06s\n- dib-functests-arm64-rpm https://zuul.opendev.org/t/openstack/build/65607982f766446fa475760db41d2117 : SUCCESS in 36m 37s","accounts_in_message":[],"_revision_number":1},{"id":"9e3821393d4130be0049717d02448054d19b6a1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 01:04:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18eb56a509ed47bea8b9d33b9eadebab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/412286e300574717b5c5ece10d7f688b : SUCCESS in 4m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd212b15734e424b8d5262cccee1609a : SUCCESS in 2m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7f0f033d89a448e3a047cde140595a7b : SUCCESS in 5m 23s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/f8137877a4ba4c78b4af05e116d88b39 : SUCCESS in 3m 39s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/ebee81cd8bbd41f4a8454126180ffd61 : SUCCESS in 8m 47s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/50e86f4b00184d1ea165e0b1df056fed : SUCCESS in 3m 50s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/b9315aa317b245b5a5dd25309f052556 : SUCCESS in 3m 09s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/7ef8ec5838e848629837e7e4a05b6626 : SUCCESS in 4m 12s\n- tox-py313 https://zuul.opendev.org/t/openstack/build/e043c16248254696bc52c8da933d3c42 : SUCCESS in 6m 57s\n- dib-functests https://zuul.opendev.org/t/openstack/build/4d0a68fc3e8b45d1b5509c761f95b4d7 : SUCCESS in 49m 06s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/aa9c842ad16b45c79607837ebd4337db : SUCCESS in 3m 17s\n- dib-devstack-functional-openstack-almalinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/f0a6880b1f20462b83f9ed1a3431ba04 : FAILURE in 22m 10s (non-voting)\n- dib-devstack-functional-openstack-almalinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/8b57b29464eb42c9bc2157481704a300 : FAILURE in 45m 23s (non-voting)\n- dib-devstack-functional-openstack-almalinux-10-v2-containerfile-src https://zuul.opendev.org/t/openstack/build/b569f572d88e4e6988b6df8311183d15 : FAILURE in 34m 55s (non-voting)\n- dib-devstack-functional-openstack-centos-9-stream-src https://zuul.opendev.org/t/openstack/build/28cc5420d1ce444c88bba42dea9354f8 : FAILURE in 20m 20s\n- dib-devstack-functional-openstack-centos-10-stream-src https://zuul.opendev.org/t/openstack/build/71559b1ab33944e489dc8c622d47fb4d : FAILURE in 23m 21s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/5cdf20e2e7064d54886d9f214e9fafa4 : FAILURE in 22m 45s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/e782d282d8ea4c9591462cd0f5f23f15 : FAILURE in 21m 17s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-10-containerfile-src https://zuul.opendev.org/t/openstack/build/0bcecbfcf67c4f1fbe489dd6db4e9714 : FAILURE in 23m 21s (non-voting)\n- dib-devstack-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/ea27d6e73627402b95c4b175cecfbf10 : FAILURE in 15m 31s\n- dib-devstack-functional-openstack-ubuntu-jammy-src https://zuul.opendev.org/t/openstack/build/598a3c6420ee4a7d8a77792c1a72613b : FAILURE in 38m 15s\n- dib-devstack-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/a7608474cf424300b9cc89b0bb287396 : FAILURE in 21m 34s\n- dib-devstack-functional-openstack-ubuntu-resolute-src https://zuul.opendev.org/t/openstack/build/ce2f97aec8e14062a2e6a36021547af5 : FAILURE in 48m 26s\n- dib-devstack-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/6ba6081d672840169c0b7088ebf29182 : FAILURE in 13m 34s (non-voting)\n- dib-devstack-functional-openstack-gentoo-src https://zuul.opendev.org/t/openstack/build/eb26ef951688424b87270ffc5a9d5d67 : FAILURE in 1h 50m 35s (non-voting)\n- dib-devstack-functional-openstack-debian-bullseye-src https://zuul.opendev.org/t/openstack/build/4eff67a6162c4e5bb882f74216fab583 : FAILURE in 34m 13s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/8b061d274cdd4806be6f82307d65130b : SUCCESS in 10m 18s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/d2c8014b40454be2ab0bf51baba10295 : SUCCESS in 1h 14m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"74b966764e995b983fb46e31b303296e594e6dca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-08-24 02:34:30.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"9aff32c9163a23e613b4a9a80f03b8e482a02757","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 04:04:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ec2b13e477145c6a440053b7ea332b7\n\n- dib-functests-arm64-deb https://zuul.opendev.org/t/openstack/build/7abea88f86ad49d78cae26f6213b2196 : SUCCESS in 52m 33s\n- dib-functests-arm64-rpm https://zuul.opendev.org/t/openstack/build/86d3fbd9fec242c5a61d585c4afbc55a : SUCCESS in 55m 46s","accounts_in_message":[],"_revision_number":2},{"id":"bc362ff9af745282fdec0294065fa7b2bc394712","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 04:15:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8046f5d4924a4d0c92bd4ee90acc7bcf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8769282069548fbbc7f617062eeb574 : SUCCESS in 6m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11aff6dc508f4b5684b27faa66be0a84 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99d0ff8ea3224cf3b3552743bebf36d0 : SUCCESS in 5m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f3ae9754393844dab7e34acec7a90bbb : SUCCESS in 4m 30s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/521d6a215fe044b8bfcd79f5dfd68399 : SUCCESS in 5m 24s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/7f7fe522b6b14d3bb6483861acbcb98e : SUCCESS in 3m 48s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/95c2fa52841f4645bc8bf36d02b6e837 : SUCCESS in 5m 47s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/4695a27b6c9d44388a29bcc0b81de55a : SUCCESS in 4m 59s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/7f0a283ff21a42ed856e9f553df63155 : SUCCESS in 5m 44s\n- tox-py313 https://zuul.opendev.org/t/openstack/build/972a293bcc314c67835d75bac1e84263 : SUCCESS in 7m 18s\n- dib-functests https://zuul.opendev.org/t/openstack/build/466cfaf868f7498488aed096175e8c07 : SUCCESS in 44m 24s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/5dac0e50936b4091aa2981925eb76d4f : SUCCESS in 3m 15s\n- dib-devstack-functional-openstack-almalinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/31fbf0a6a0a0479e89f9143155b0f817 : SUCCESS in 52m 11s (non-voting)\n- dib-devstack-functional-openstack-almalinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/58936fcaa13a4ee7ba5f66a2abf3e1ca : SUCCESS in 39m 51s (non-voting)\n- dib-devstack-functional-openstack-almalinux-10-v2-containerfile-src https://zuul.opendev.org/t/openstack/build/eec51739baba4b6e978c56bc129c4786 : SUCCESS in 32m 12s (non-voting)\n- dib-devstack-functional-openstack-centos-9-stream-src https://zuul.opendev.org/t/openstack/build/b3191041e9ba457aa7bed7f4896bb8de : SUCCESS in 31m 20s\n- dib-devstack-functional-openstack-centos-10-stream-src https://zuul.opendev.org/t/openstack/build/f373faa3fcbe438fb2b12f83b51b49dd : SUCCESS in 25m 43s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/59d94c8614344bf2ae72db5f4256a23e : SUCCESS in 1h 07m 38s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/3758cb163d364d5ea14e585d0f89cb95 : SUCCESS in 51m 01s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-10-containerfile-src https://zuul.opendev.org/t/openstack/build/d76f93cbd58c43988e588fce511398a1 : SUCCESS in 27m 08s (non-voting)\n- dib-devstack-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/13b52f1464934a6284adf32383e7a37a : SUCCESS in 40m 12s\n- dib-devstack-functional-openstack-ubuntu-jammy-src https://zuul.opendev.org/t/openstack/build/2828e3e1718f49da98e567f0094e679b : SUCCESS in 18m 16s\n- dib-devstack-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/fc2b756345994195b38c204ecfcba430 : SUCCESS in 32m 54s\n- dib-devstack-functional-openstack-ubuntu-resolute-src https://zuul.opendev.org/t/openstack/build/56fd38a021134a70809c2f7a8a5f7eac : SUCCESS in 45m 55s\n- dib-devstack-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/2ffe995d3f8e4e38adc48450dd93caa0 : FAILURE in 13m 23s (non-voting)\n- dib-devstack-functional-openstack-gentoo-src https://zuul.opendev.org/t/openstack/build/69a85de5d6804cb4a9c0dc1da7ba9a20 : FAILURE in 58m 33s (non-voting)\n- dib-devstack-functional-openstack-debian-bullseye-src https://zuul.opendev.org/t/openstack/build/8c8de458eea9432cb0f19ce5f1482d4c : SUCCESS in 40m 57s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/ddaa2ef0264d4df7903f634f00c6dfb7 : SUCCESS in 9m 58s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/2102a13b855148aeaa278ff84e7008a1 : SUCCESS in 29m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6aad81daaa61402d9d1559b4579488dd0e999ce3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-08-24 21:47:57.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"5f5987ef1de47fd08459e3b901d829830a96665e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 22:52:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/038091cf148e4d97b5f5b2407ecaba0a\n\n- dib-functests-arm64-deb https://zuul.opendev.org/t/openstack/build/90c26bfb32984e41851b76ea1d6f3568 : SUCCESS in 39m 07s\n- dib-functests-arm64-rpm https://zuul.opendev.org/t/openstack/build/17c848a738d94d0792d558ca49c9f2e8 : SUCCESS in 40m 36s","accounts_in_message":[],"_revision_number":3},{"id":"84cbfc2cbd3a3dfe5a87e42b2f011a527ea6aa1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 22:53:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ed0a4c6b9ad4dbfabf20ca3ca375b99\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be8632620e4341e6a80240d17cf0425f : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ec1c83a202448479b2ea02b7dc77d89 : SUCCESS in 2m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7f579f12aa64fd8b67fce4faa24d7cf : SUCCESS in 4m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/87bf83e1e2d849ed9d1dbe485d1f543b : SUCCESS in 2m 56s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/a25b681bd92b49a79c021b9862f5cdaf : SUCCESS in 2m 06s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/c212e3ad17d24832a8b0dbd8ac388981 : SUCCESS in 3m 43s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/c352a08d649540c5b85190129fc84670 : SUCCESS in 4m 12s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/18670935fd9d47688f1cb1a31af4f44f : SUCCESS in 4m 19s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/9ae068e28da14ca59ab9eb47959e2dda : SUCCESS in 3m 29s\n- tox-py313 https://zuul.opendev.org/t/openstack/build/096282963ab5496bb424d5ae4cc2799f : SUCCESS in 2m 31s\n- dib-functests https://zuul.opendev.org/t/openstack/build/8b447761491747228bc1a6ba72075781 : SUCCESS in 41m 07s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/e6d3a3846f5845f081ad0bae763238aa : SUCCESS in 2m 52s\n- dib-devstack-functional-openstack-almalinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/d66738392d014da784053e1284754f4d : SUCCESS in 21m 10s (non-voting)\n- dib-devstack-functional-openstack-almalinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/72f76ed490254337967d8140110513f0 : FAILURE in 16m 17s (non-voting)\n- dib-devstack-functional-openstack-almalinux-10-v2-containerfile-src https://zuul.opendev.org/t/openstack/build/8cdf4c6cf0fe412ca70f7f9061caa73d : SUCCESS in 30m 39s (non-voting)\n- dib-devstack-functional-openstack-centos-9-stream-src https://zuul.opendev.org/t/openstack/build/5339965a470c4269994edd30fd438c0f : FAILURE in 33m 40s\n- dib-devstack-functional-openstack-centos-10-stream-src https://zuul.opendev.org/t/openstack/build/9bcdec7f25e14f578f8c31dfd58f9432 : SUCCESS in 24m 58s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/166ca723285e4525800d6b0af95b4b3c : SUCCESS in 58m 57s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/97dc55a5e6b44c00b211e81078441393 : FAILURE in 33m 57s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-10-containerfile-src https://zuul.opendev.org/t/openstack/build/dc4c01996e434926a4332f4fc19afa56 : SUCCESS in 30m 42s (non-voting)\n- dib-devstack-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/1508ae5471254d948d3c91fdc5f973e4 : FAILURE in 21m 12s\n- dib-devstack-functional-openstack-ubuntu-jammy-src https://zuul.opendev.org/t/openstack/build/cba4903e7fe74cccb1999d5ba867f107 : FAILURE in 32m 23s\n- dib-devstack-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/e7204136a00641e196fdaf1e5f1735ab : SUCCESS in 34m 57s\n- dib-devstack-functional-openstack-ubuntu-resolute-src https://zuul.opendev.org/t/openstack/build/cc3f5374f90c4cb3aa2b67ecec588b07 : SUCCESS in 26m 26s\n- dib-devstack-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/9a694a34318a4e4b89a767a6091cf0f2 : FAILURE in 20m 02s (non-voting)\n- dib-devstack-functional-openstack-gentoo-src https://zuul.opendev.org/t/openstack/build/9cd4c92dd95e44e1bcd1068bc6dcf002 : FAILURE in 46m 08s (non-voting)\n- dib-devstack-functional-openstack-debian-bullseye-src https://zuul.opendev.org/t/openstack/build/a41f4dd631674e89be9fd2420abc8afe : FAILURE in 20m 03s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/c91842f8b54b4c548e0ae28e96151262 : SUCCESS in 9m 58s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/5460335027724d558b58ed83e5a08b54 : SUCCESS in 44m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"72bad5ff775ce2f8b2f219087bf15941331c13cf","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-08-31 21:12:32.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"007b29e25ae6e7d076b92a1d979efd16020b4555","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-08-31 21:32:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d878c03eb4ffc2b0fa2c956ab37bb2fd92737740","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 22:26:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/293f8bb6ad1c4c1a8ba45231b41050be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0a349639c6043e9a758b2cdeb91ebe5 : SUCCESS in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5a7d942f8ce4f09ab3820dfb3ef1dee : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2c18775a099482b83b2d00323ce3053 : SUCCESS in 8m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/329d36929fc14ff58eaec779f269c9ba : SUCCESS in 5m 41s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/452b4752302348d0a5af6aa3381360c0 : SUCCESS in 2m 27s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/eeacdfd3eacd43cba70272a8d3fac199 : SUCCESS in 4m 42s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/15c1e94e07b34f1fb72fce03e8c5078f : SUCCESS in 3m 43s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/1d100b5986c24157b1b20712f8ae63aa : SUCCESS in 2m 30s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/e610c2868ed645778de5be57b6ef525d : SUCCESS in 7m 44s\n- tox-py313 https://zuul.opendev.org/t/openstack/build/031470490b714f25919ba39ffd03fa42 : SUCCESS in 2m 33s\n- dib-functests https://zuul.opendev.org/t/openstack/build/a90a4f129a7245869a515f71aad76a23 : SUCCESS in 45m 13s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/31290f3a7ddf444ebfb8d92cfc68b63d : SUCCESS in 2m 54s\n- dib-devstack-functional-openstack-almalinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/9e8a7cd23df14a03a32e7b79ce1faa44 : SUCCESS in 23m 54s (non-voting)\n- dib-devstack-functional-openstack-almalinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/ef54e330ab0c433dac444f0f2474bf33 : SUCCESS in 34m 49s (non-voting)\n- dib-devstack-functional-openstack-almalinux-10-v2-containerfile-src https://zuul.opendev.org/t/openstack/build/70b2ded4b5e647b6adf7fccb5722b289 : SUCCESS in 19m 55s (non-voting)\n- dib-devstack-functional-openstack-centos-9-stream-src https://zuul.opendev.org/t/openstack/build/347f9879dd454ab7af02c364d6299d40 : SUCCESS in 37m 19s\n- dib-devstack-functional-openstack-centos-10-stream-src https://zuul.opendev.org/t/openstack/build/99bdb3b8e8cb4dccb15f17dd21240525 : SUCCESS in 26m 07s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/2c8a817d6df74b07868b0d4a78df96e9 : SUCCESS in 23m 53s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/b7f44ce770fa4fc894ebffb9d13c370d : SUCCESS in 48m 41s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-10-containerfile-src https://zuul.opendev.org/t/openstack/build/ae1ce97527224cbf9132e73237008901 : SUCCESS in 27m 10s (non-voting)\n- dib-devstack-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/eeeb5b18546e41b1a29a49a19798613c : SUCCESS in 44m 47s\n- dib-devstack-functional-openstack-ubuntu-jammy-src https://zuul.opendev.org/t/openstack/build/3e055e7e54b949e8b1043eda9a49919b : SUCCESS in 43m 25s\n- dib-devstack-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/afe642ec06354ac1900a5b8a9934730d : SUCCESS in 17m 52s\n- dib-devstack-functional-openstack-ubuntu-resolute-src https://zuul.opendev.org/t/openstack/build/0a7daa1c17ca4249acf43aa6163e3154 : SUCCESS in 33m 36s\n- dib-devstack-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/ae2133709cb545368968342b1f3a79ee : FAILURE in 26m 09s (non-voting)\n- dib-devstack-functional-openstack-gentoo-src https://zuul.opendev.org/t/openstack/build/c2c8d95485d547549e99d091d0ddc995 : FAILURE in 43m 36s (non-voting)\n- dib-devstack-functional-openstack-debian-bullseye-src https://zuul.opendev.org/t/openstack/build/223d9387d5db41e599a0764725247c13 : SUCCESS in 16m 02s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/42a8dcb1a79d4bf395cf9e05eb78f1c7 : SUCCESS in 14m 57s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/8de1386adcd44f5e85885d4b07dcd195 : SUCCESS in 29m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b2857a6d6d7aa192e7f3887521a713dd826b71ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 22:32:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfc58bdccddb4e0cbccbcf10f9e53bbc\n\n- dib-functests-arm64-deb https://zuul.opendev.org/t/openstack/build/22622a43e9074310abf45e76d8bcf582 : SUCCESS in 49m 40s\n- dib-functests-arm64-rpm https://zuul.opendev.org/t/openstack/build/ec234b3bd8bb4aed920c90d50a7f2f0d : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":3},{"id":"4d003a75e11f835da9641a01524a569f20bd6f72","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-31 22:34:26.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"66d7afbdb74dc1acaf8a335a483dcdfdfe1d727e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-09-01 03:26:15.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"65d7134e428e5137fb05d33dd27ac73f39f01b3b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 04:47:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88df863c7d044a22bf6a202d9d8b7e38\n\n- dib-functests-arm64-deb https://zuul.opendev.org/t/openstack/build/4e53a855481242bb8e3ab3540fd8951f : SUCCESS in 50m 11s\n- dib-functests-arm64-rpm https://zuul.opendev.org/t/openstack/build/939bb3c669a64df08850989b97aaa92a : SUCCESS in 39m 00s","accounts_in_message":[],"_revision_number":4},{"id":"5595b90cf61263849087df4b63aa468a0dfce6e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 04:49:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/264d379b8a8d417cbd4fdc63557201a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c045c43a491c48018ffa3b50ae8cbe75 : SUCCESS in 5m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e72be07ac55440afa0828686d6789bce : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8a5797c64ca4f468270eb39dc718e6e : SUCCESS in 4m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e61f3cf3082b4ba3952e6d5df94015f4 : SUCCESS in 5m 25s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/2dbef1e8a5d94b56bcb3db982453fd06 : SUCCESS in 4m 42s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/3ef8e7b4dbdb467d80e923ca694b08f4 : SUCCESS in 2m 21s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/9e17524dc12748fbaa02dece31857c14 : SUCCESS in 4m 10s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/be11784555ee4bc5945905128ba22173 : SUCCESS in 2m 55s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/a2aca7ea223141e58e69f0586c63606f : SUCCESS in 4m 55s\n- tox-py313 https://zuul.opendev.org/t/openstack/build/8e7213a00196439cb1d26a14944cab15 : SUCCESS in 2m 50s\n- dib-functests https://zuul.opendev.org/t/openstack/build/8ba8ee483c9048728a0dac0b908571c2 : SUCCESS in 47m 15s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/1a63454a78414771b92f9429db27f2d9 : SUCCESS in 3m 39s\n- dib-devstack-functional-openstack-almalinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/c50f2d93026a47f8a056617dff40331a : SUCCESS in 1h 02m 27s (non-voting)\n- dib-devstack-functional-openstack-almalinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/3fce14811a694d70ae8d118da13c7827 : SUCCESS in 19m 33s (non-voting)\n- dib-devstack-functional-openstack-almalinux-10-v2-containerfile-src https://zuul.opendev.org/t/openstack/build/1480268a81264dfb8783a7fcc27d3550 : SUCCESS in 51m 39s (non-voting)\n- dib-devstack-functional-openstack-centos-9-stream-src https://zuul.opendev.org/t/openstack/build/c5d4d2b9843741a9a04fb48c3ed1d220 : SUCCESS in 24m 33s\n- dib-devstack-functional-openstack-centos-10-stream-src https://zuul.opendev.org/t/openstack/build/9a7f6b57e21e4c358527b9be7293f7e6 : SUCCESS in 27m 54s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/c99a45d3716b42139db179df4eb64179 : SUCCESS in 22m 38s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/6e7a69c36f6b4cf998276cf6eaaa90ba : SUCCESS in 45m 18s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-10-containerfile-src https://zuul.opendev.org/t/openstack/build/2eb06c94a8c6426895e91c76615468d4 : SUCCESS in 29m 04s (non-voting)\n- dib-devstack-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/38c9afdfcb68403ea7cabca30684c4d7 : SUCCESS in 35m 17s\n- dib-devstack-functional-openstack-ubuntu-jammy-src https://zuul.opendev.org/t/openstack/build/90d0010ba51c41fa82417fb86fdc2bdc : SUCCESS in 39m 45s\n- dib-devstack-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/f7d00809eb124e738cc05d95f84ee929 : SUCCESS in 18m 18s\n- dib-devstack-functional-openstack-ubuntu-resolute-src https://zuul.opendev.org/t/openstack/build/2b528d97fc1540439ed231c1c407ad40 : SUCCESS in 51m 26s\n- dib-devstack-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/5da3f26251a34dbfbfadd1022e6e162f : FAILURE in 17m 50s (non-voting)\n- dib-devstack-functional-openstack-gentoo-src https://zuul.opendev.org/t/openstack/build/eea1286939c24167a4014c7d1a360757 : FAILURE in 29m 10s (non-voting)\n- dib-devstack-functional-openstack-debian-bullseye-src https://zuul.opendev.org/t/openstack/build/5fe6fb5361234c6b941df9e7ac467bd5 : SUCCESS in 20m 53s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/ea19da17436d4c28a1a24a030a7cadfa : SUCCESS in 14m 17s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/3b7ad056303a4066a246c2cc4c786b7c : SUCCESS in 1h 15m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"43eec59405f38e7dbedfcf49efa1fecb11d980ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-09-28 21:58:50.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"b3abe9cd47b848df9810243506d2983e5fb3849a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-09-28 22:51:34.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5b05a47c56d92d89ee586cfcae13122b9da24cde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 22:54:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2b77aba13ce4f7bad1062740a9b4a32\n\n- dib-functests-arm64-deb https://zuul.opendev.org/t/openstack/build/ddee6dc060894cf5b63ba82ffa923cf4 : SUCCESS in 52m 23s\n- dib-functests-arm64-rpm https://zuul.opendev.org/t/openstack/build/cda14d37dadf4be79b5dd589ede54c67 : SUCCESS in 45m 48s","accounts_in_message":[],"_revision_number":5},{"id":"b1a1fef0ad0f50459603fa0bc5ce480a473bba99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 23:23:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31c498bab34d4dc982a4e39e780b6222\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/773fd54065ed41648f682dd3af473f5b : SUCCESS in 5m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ef88b35f64744faab808df78d74d76c : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11bbc147ee1549f9b0fefabd15920cc6 : SUCCESS in 5m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ee88b2e421594e68900fd054eac403c8 : SUCCESS in 3m 44s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/84d0bb0cc3214d9995e24294dbe77e9a : SUCCESS in 3m 00s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/74a3aaac52fe4aada27689189eab03e1 : SUCCESS in 3m 51s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/88e49a03bcb149baade6188ba9ae4186 : SUCCESS in 3m 12s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/a784493ec13e404dabedc473ccbcb008 : SUCCESS in 4m 32s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/1d9b7e6266f2459dbbe48b27c9a6e4ca : SUCCESS in 5m 50s\n- tox-py313 https://zuul.opendev.org/t/openstack/build/a15bf79d1cab44898c5a3b121a32ffd4 : SUCCESS in 4m 02s\n- dib-functests https://zuul.opendev.org/t/openstack/build/dee7fe8e9b4f4d7fbe8ffeadd1259fa7 : SUCCESS in 50m 55s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b1de469e9a3840e4b0e66604fa11fe95 : SUCCESS in 4m 33s\n- dib-devstack-functional-openstack-almalinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/d7019191bcad4fbaa13d95cbfb4a9e8e : SUCCESS in 26m 48s (non-voting)\n- dib-devstack-functional-openstack-almalinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/87cf0fea39e34960a80a5cde0991ce69 : SUCCESS in 30m 52s (non-voting)\n- dib-devstack-functional-openstack-almalinux-10-v2-containerfile-src https://zuul.opendev.org/t/openstack/build/d875888616a3486daea0ba8255f82742 : SUCCESS in 55m 36s (non-voting)\n- dib-devstack-functional-openstack-centos-9-stream-src https://zuul.opendev.org/t/openstack/build/f586b4cb4666456eaeacdb24895802d8 : SUCCESS in 43m 37s\n- dib-devstack-functional-openstack-centos-10-stream-src https://zuul.opendev.org/t/openstack/build/5de108717c2f49748df6a655156d6f4c : SUCCESS in 26m 12s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/b6fc4d7184a1457a9c1707d3ee9839b0 : SUCCESS in 52m 42s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/bee6b61c73464e77aa3e17acbe10cf2b : SUCCESS in 27m 05s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-10-containerfile-src https://zuul.opendev.org/t/openstack/build/5d40ec00792a41d8a7a78c3f5be69c7b : SUCCESS in 28m 27s (non-voting)\n- dib-devstack-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/13f65a9f6ee8445f953b2447db9554a7 : SUCCESS in 39m 23s\n- dib-devstack-functional-openstack-ubuntu-jammy-src https://zuul.opendev.org/t/openstack/build/f616898a75f0428dad5b116cd8437c6c : SUCCESS in 34m 36s\n- dib-devstack-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/f439a5290ecb4701bd93c889cae9cfa8 : SUCCESS in 22m 53s\n- dib-devstack-functional-openstack-ubuntu-resolute-src https://zuul.opendev.org/t/openstack/build/c1e530da68264fc6b7dfdb5d45a7d499 : SUCCESS in 47m 39s\n- dib-devstack-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/d61cd3714718452ca67ead7d35b3133d : FAILURE in 31m 25s (non-voting)\n- dib-devstack-functional-openstack-gentoo-src https://zuul.opendev.org/t/openstack/build/11e09429d97145beaef9be5cc75b5ada : FAILURE in 48m 36s (non-voting)\n- dib-devstack-functional-openstack-debian-bullseye-src https://zuul.opendev.org/t/openstack/build/1f13bc62687f4d1a90e8925929502f2f : FAILURE in 25m 32s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/f59cc24a2072407682f74e19fe6274ab : SUCCESS in 10m 12s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/68d599f5b736474c9fab30df3e2759ff : SUCCESS in 1h 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1229d5ed897b9f3ba0bedfc41704c99277ea2c59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 23:26:06.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5fabd292bde69fef1cf8788eee93582ac64bd42b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 00:19:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e4f4cb920c345d8a3d884fb56c6a0bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19a86f7e4e654b818e1d42f6daeccc89 : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/716071e0ed304c03821536dffcd8fbe7 : SUCCESS in 5m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d4a008a97ecf4a5581013d999727689e : SUCCESS in 3m 01s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/996695ec4cac48899ad990cbafb2fab1 : SUCCESS in 3m 41s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/9d8980c25b0542fe91a924abee6eff28 : SUCCESS in 3m 23s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/16a724e267e34c1f98555b49f42421e0 : SUCCESS in 2m 29s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/f8b5b93edcdc46a69d2983af3fe2e152 : SUCCESS in 3m 40s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/20455e44e7fb4cceaf44160a00b083dc : SUCCESS in 3m 49s\n- tox-py313 https://zuul.opendev.org/t/openstack/build/ef03cfe205294622b6ee4fac54d6a324 : SUCCESS in 3m 45s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/dd60902f310e466699985906d49dfc52 : SUCCESS in 27m 45s\n- nodepool-build-image-siblings https://zuul.opendev.org/t/openstack/build/6af99ed52fd741039c269f2190b8a641 : SUCCESS in 16m 32s\n- dib-devstack-functional-openstack-centos-9-stream-src https://zuul.opendev.org/t/openstack/build/998b066d7f0b43398c7d16b329f1bdc4 : SUCCESS in 17m 36s\n- dib-devstack-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/37e958e6c57f41eebf6300e6a6c4b1c1 : SUCCESS in 39m 08s\n- dib-devstack-functional-openstack-ubuntu-jammy-src https://zuul.opendev.org/t/openstack/build/5cd98e51172f437dae8272621b522b03 : SUCCESS in 42m 49s\n- dib-devstack-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/60e9a27291d64b5396a9995d7eb7f5a5 : SUCCESS in 33m 14s\n- dib-functests https://zuul.opendev.org/t/openstack/build/11ad5a5d58114f80ac57b28076f84ea5 : SUCCESS in 52m 20s\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/073342b2e53d45a790dc8c6bce8fe74f : SUCCESS in 9m 42s","accounts_in_message":[],"_revision_number":5},{"id":"309e51e40c39e144583755a68f25c68a291a6b34","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 00:19:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"e9dbd7599d0ece8e008219d70c078affee48f14f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 00:20:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99aa4356da98480aae7076b494b0aea0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c445f224c0c4920b80c086a6977ae70 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1af83478cef40368b4d188f73f296b415b537251","revisions":{"342165b8f8edd887216ef84eac175640eeb3729b":{"kind":"REWORK","_number":1,"created":"2026-04-09 05:02:17.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/13/983813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/diskimage-builder","ref":"refs/changes/13/983813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/1"}}},"commit":{"parents":[{"commit":"68f8b8f31779d3f747a611f1809724bf40b92ba6","subject":"Merge \"Add support for building Ubuntu Resolute (26.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/68f8b8f31779d3f747a611f1809724bf40b92ba6"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-04-09 04:10:03.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-04-09 05:01:44.000000000","tz":720},"subject":"Refactor 02-set-machine-id into an element","message":"Refactor 02-set-machine-id into an element\n\nThis script is needlessly duplicated across 3 elements. This change\nconsolidates those.\n\nAssisted-By: Opencode - Claude Sonnet 4.6\nSigned-off-by: Steve Baker \u003csbaker@redhat.com\u003e\nChange-Id: I8183931b1b962b7916265e3b5c3ea1175bd57c4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/342165b8f8edd887216ef84eac175640eeb3729b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/342165b8f8edd887216ef84eac175640eeb3729b"}]},"branch":"refs/heads/master"},"2dc202f7ef145eef6a4262732abf6920315720e3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-08-24 02:34:30.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/13/983813/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/diskimage-builder","ref":"refs/changes/13/983813/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/2"}}},"commit":{"parents":[{"commit":"012ca670bfc186c8d7b591fb87e0a0bf8798cad3","subject":"Increase devstack Glance total image size limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/012ca670bfc186c8d7b591fb87e0a0bf8798cad3"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-04-09 04:10:03.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-08-24 02:30:02.000000000","tz":720},"subject":"Refactor 02-set-machine-id into an element","message":"Refactor 02-set-machine-id into an element\n\nThis script is needlessly duplicated across 3 elements. This change\nconsolidates those.\n\nAssisted-By: Opencode - Claude Sonnet 4.6\nSigned-off-by: Steve Baker \u003csbaker@redhat.com\u003e\nChange-Id: I8183931b1b962b7916265e3b5c3ea1175bd57c4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/2dc202f7ef145eef6a4262732abf6920315720e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/2dc202f7ef145eef6a4262732abf6920315720e3"}]},"branch":"refs/heads/master"},"97f112de139623110dec023201fc125a9a40d3a0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-08-24 21:47:57.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/13/983813/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/diskimage-builder","ref":"refs/changes/13/983813/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/3"}}},"commit":{"parents":[{"commit":"ece6a6e98c588104c70fadbf4bf3d97770c239e6","subject":"CI: Bump dib flavor disk to 6GB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/ece6a6e98c588104c70fadbf4bf3d97770c239e6"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-04-09 04:10:03.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-08-24 21:47:44.000000000","tz":720},"subject":"Refactor 02-set-machine-id into an element","message":"Refactor 02-set-machine-id into an element\n\nThis script is needlessly duplicated across 3 elements. This change\nconsolidates those.\n\nAssisted-By: Opencode - Claude Sonnet 4.6\nSigned-off-by: Steve Baker \u003csbaker@redhat.com\u003e\nChange-Id: I8183931b1b962b7916265e3b5c3ea1175bd57c4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/97f112de139623110dec023201fc125a9a40d3a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/97f112de139623110dec023201fc125a9a40d3a0"}]},"branch":"refs/heads/master"},"b5accd5e683f68981ba799d055627c98c2f28c0a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-09-01 03:26:15.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/13/983813/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/diskimage-builder","ref":"refs/changes/13/983813/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/4"}}},"commit":{"parents":[{"commit":"908418886e1a96faf08abb62566d4f32ae32127b","subject":"Merge \"Bump upper version of flake8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/908418886e1a96faf08abb62566d4f32ae32127b"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-04-09 04:10:03.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-09-01 03:13:35.000000000","tz":720},"subject":"Refactor 02-set-machine-id into an element","message":"Refactor 02-set-machine-id into an element\n\nThis script is needlessly duplicated across 3 elements. This change\nconsolidates those.\n\nAssisted-By: Opencode - Claude Sonnet 4.6\nSigned-off-by: Steve Baker \u003csbaker@redhat.com\u003e\nChange-Id: I8183931b1b962b7916265e3b5c3ea1175bd57c4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/b5accd5e683f68981ba799d055627c98c2f28c0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/b5accd5e683f68981ba799d055627c98c2f28c0a"}]},"branch":"refs/heads/master"},"1af83478cef40368b4d188f73f296b415b537251":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-09-28 21:58:50.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/13/983813/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/diskimage-builder","ref":"refs/changes/13/983813/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/diskimage-builder refs/changes/13/983813/5"}}},"commit":{"parents":[{"commit":"73764519c23e338f88b9daf642d80376c416ae75","subject":"Fix centos element BASE_IMAGE_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/73764519c23e338f88b9daf642d80376c416ae75"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-04-09 04:10:03.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-09-28 21:58:03.000000000","tz":780},"subject":"Refactor 02-set-machine-id into an element","message":"Refactor 02-set-machine-id into an element\n\nThis script is needlessly duplicated across 3 elements. This change\nconsolidates those.\n\nAssisted-By: Opencode - Claude Sonnet 4.6\nSigned-off-by: Steve Baker \u003csbaker@redhat.com\u003e\nChange-Id: I8183931b1b962b7916265e3b5c3ea1175bd57c4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/1af83478cef40368b4d188f73f296b415b537251"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/1af83478cef40368b4d188f73f296b415b537251"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
