)]}'
{"id":"openstack%2Fdiskimage-builder~983814","triplet_id":"openstack%2Fdiskimage-builder~master~If8cbb38bd739169afcbe8e99325249f86f453bc5","project":"openstack/diskimage-builder","branch":"master","topic":"tarball","attention_set":{"4571":{"account":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"last_update":"2026-08-24 23:10:02.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-08-31 23:52:01.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"If8cbb38bd739169afcbe8e99325249f86f453bc5","subject":"Refactor 03-reset-bls-entries into an element","status":"NEW","created":"2026-04-09 05:02:17.000000000","updated":"2026-09-01 05:46:29.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b19208806568a45a9a34724a234a55a7cf688bb7","_number":983814,"virtual_id_number":983814,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-01 04:56:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":2,"date":"2026-09-01 03:26:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2026-09-01 03:26:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-09 06:34:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-09 06:44:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-31 21:13:16.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2026-08-31 21:18:37.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"56b3370decd295adcfdf205120a778996a368836","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-04-09 05:02:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d531445cbdc2adbb9ee9c12c143d4dfaad74e78","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 06:34:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68e541ec832345e4a6a8d45f1273fb45\n\n- dib-functests-arm64-deb https://zuul.opendev.org/t/openstack/build/a99cfc40b7dc4468995e812e37e3e58f : TIMED_OUT in 1h 30m 57s\n- dib-functests-arm64-rpm https://zuul.opendev.org/t/openstack/build/f5bfaa1e7dba44ffa83c85f4511d02f6 : SUCCESS in 30m 14s","accounts_in_message":[],"_revision_number":1},{"id":"bc396c185ad4c1945d9ea7a190882c16c70445cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 06:44:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b604a11a12ff427fa35f0eb1c091b453\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25b3f96ed251455fb792b0e519654cca : SUCCESS in 5m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c268569ad50e4050b336f62ea31c5a59 : SUCCESS in 4m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/10e2b4acbc3144f985e610a78274e71b : SUCCESS in 5m 02s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/562cfbed412144c496e8a836f66eb0af : SUCCESS in 3m 48s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/2affea0c119347478941ad93860661a4 : SUCCESS in 3m 48s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/dca6dbbfd92e4a6dbbb2ce100aae0dcd : SUCCESS in 3m 31s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/c79481c307414d6d9593ed776ad99707 : SUCCESS in 3m 45s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/96aed6a4fd4a4303b0c44bdef739167d : SUCCESS in 3m 53s\n- tox-py313 https://zuul.opendev.org/t/openstack/build/f525cc728878460ca94f621dd947181a : SUCCESS in 5m 07s\n- dib-functests https://zuul.opendev.org/t/openstack/build/96952f29de28421aae64f2f9911a082c : SUCCESS in 49m 44s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/e5ae0118ec5548b5962e86a7704a8d71 : SUCCESS in 3m 41s\n- dib-devstack-functional-openstack-almalinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/1dacab9154fa4cd5be717d37a1094773 : FAILURE in 31m 57s (non-voting)\n- dib-devstack-functional-openstack-almalinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/af138546ace04a13a11ffb106fa039e9 : FAILURE in 41m 29s (non-voting)\n- dib-devstack-functional-openstack-almalinux-10-v2-containerfile-src https://zuul.opendev.org/t/openstack/build/1a01983383d948408e13dfd7bdbf481b : FAILURE in 23m 24s (non-voting)\n- dib-devstack-functional-openstack-centos-9-stream-src https://zuul.opendev.org/t/openstack/build/affcb900c210445384678e84d881615f : FAILURE in 40m 39s\n- dib-devstack-functional-openstack-centos-10-stream-src https://zuul.opendev.org/t/openstack/build/3049d2f5f31043458e0c6cd42822b0aa : FAILURE in 23m 44s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/57ee749233964c8d9de351eb05549f6b : FAILURE in 44m 59s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/2cc7352299004c61b394ada6d493ee61 : FAILURE in 42m 26s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-10-containerfile-src https://zuul.opendev.org/t/openstack/build/caa81ad51f5a479dbc3057c549c28d99 : FAILURE in 25m 16s (non-voting)\n- dib-devstack-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/77a6968e09a34d258f34e1a08e93aae1 : FAILURE in 39m 01s\n- dib-devstack-functional-openstack-ubuntu-jammy-src https://zuul.opendev.org/t/openstack/build/34e5a0ce7d1947f998b3f44e84a75168 : FAILURE in 36m 56s\n- dib-devstack-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/05e1918efb994c148dbd3736c4eaf766 : FAILURE in 34m 28s\n- dib-devstack-functional-openstack-ubuntu-resolute-src https://zuul.opendev.org/t/openstack/build/4d5a05c32da0421c81a98bb56b9b68cc : FAILURE in 52m 55s\n- dib-devstack-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/4edb3ee347ad4e38975497530f25f9c3 : FAILURE in 29m 16s (non-voting)\n- dib-devstack-functional-openstack-gentoo-src https://zuul.opendev.org/t/openstack/build/290114471bb04a61b3864a20633cdba6 : FAILURE in 1h 36m 24s (non-voting)\n- dib-devstack-functional-openstack-debian-bullseye-src https://zuul.opendev.org/t/openstack/build/cf34c17aecb4429d9c5bebccd0898aee : FAILURE in 31m 33s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/fa4974f7b9d5449f960f2f2de9b1520d : SUCCESS in 10m 20s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/4ea0ebcc4eb941fc8be2016ab38ae1e0 : SUCCESS in 1h 26m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"55e5a553b3e0bdd7269a54f021a6e59f0a756956","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-04-14 23:07:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a96689aac250c72ae03344742670d13c41fa80a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 00:35:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3f45e1d794e4f6080096a1b840a6bb1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eab7f98a405542b0a6bd3ed5f0a1d387 : SUCCESS in 3m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/634352b013184a6cb1a2cd110148cdda : SUCCESS in 4m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/717cde9ea3424ad0a472511fd6d53d00 : SUCCESS in 5m 00s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/5d64a69207614169a72e27d7da2ae4b9 : SUCCESS in 3m 26s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/bd1feeca1f8a413e8816b39fca6f1f68 : SUCCESS in 5m 54s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/b2b47c710daa4383b5511c28aa3f9bfb : SUCCESS in 2m 21s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/604e6cc7776849dea284e4f394d3f49d : SUCCESS in 2m 48s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/4ef459dde333438494e718f8d725fbaa : SUCCESS in 2m 23s\n- tox-py313 https://zuul.opendev.org/t/openstack/build/2c63d6c220c84e7f8554a54c2794e765 : SUCCESS in 3m 32s\n- dib-functests https://zuul.opendev.org/t/openstack/build/09dfa56541ff45a9a9c99e9ecf1797df : SUCCESS in 51m 55s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/a1ba816f32734491b0a6b82b7042fa72 : SUCCESS in 3m 33s\n- dib-devstack-functional-openstack-almalinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/91ac41a212d248d1a042077a7829c1da : FAILURE in 48m 16s (non-voting)\n- dib-devstack-functional-openstack-almalinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/4e983734008f4159a70d535674a3f818 : FAILURE in 46m 50s (non-voting)\n- dib-devstack-functional-openstack-almalinux-10-v2-containerfile-src https://zuul.opendev.org/t/openstack/build/9c97dab7c3eb4d6cadfe7dee519b4793 : FAILURE in 19m 50s (non-voting)\n- dib-devstack-functional-openstack-centos-9-stream-src https://zuul.opendev.org/t/openstack/build/64628190dc224428b8b39c341f767996 : FAILURE in 37m 54s\n- dib-devstack-functional-openstack-centos-10-stream-src https://zuul.opendev.org/t/openstack/build/c95b48688c9a4dc0b4539bd4274fe8d0 : FAILURE in 23m 53s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/1b35ded2f0be480089ba3495a914eede : FAILURE in 46m 25s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/51c668c8d5164c85856261c8e75a6a9e : FAILURE in 45m 43s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-10-containerfile-src https://zuul.opendev.org/t/openstack/build/baf1cbcc9a484cf58652d60f893532c8 : FAILURE in 24m 40s (non-voting)\n- dib-devstack-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/ad183c867ebf447b8b512db039cfbfd7 : FAILURE in 35m 19s\n- dib-devstack-functional-openstack-ubuntu-jammy-src https://zuul.opendev.org/t/openstack/build/f1e7a19f238345f685a6eced144e72c1 : FAILURE in 37m 01s\n- dib-devstack-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/8033971f598c4de693fcce1d912da7c5 : FAILURE in 47m 21s\n- dib-devstack-functional-openstack-ubuntu-resolute-src https://zuul.opendev.org/t/openstack/build/d504cc66335e452f81dcd3633d2f7984 : FAILURE in 37m 08s\n- dib-devstack-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/bddfbdf3aeb04e27a9b5640d03445c43 : FAILURE in 27m 18s (non-voting)\n- dib-devstack-functional-openstack-gentoo-src https://zuul.opendev.org/t/openstack/build/7aa81fc71e8941fa88812ce2483dda11 : FAILURE in 50m 40s (non-voting)\n- dib-devstack-functional-openstack-debian-bullseye-src https://zuul.opendev.org/t/openstack/build/fc05b22abf66451c9ab31411a05742dc : FAILURE in 34m 03s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/e7c93cd08bc949d9887db7c1b97cbfd4 : SUCCESS in 9m 35s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/bcaabc7074d742bba393cab74680c4b8 : SUCCESS in 1h 21m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee05f16a3926a4af1f19782798d69031da63a5d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 00:47:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acbffa0ab75242679b06a054763375f3\n\n- dib-functests-arm64-deb https://zuul.opendev.org/t/openstack/build/973e78e8bc024607b0dffa8da029225d : SUCCESS in 36m 44s\n- dib-functests-arm64-rpm https://zuul.opendev.org/t/openstack/build/9f87dad328bf4b64927c94b3e70926f8 : SUCCESS in 36m 42s","accounts_in_message":[],"_revision_number":1},{"id":"14390e3360182dd9052a3ced7058dfd898b251fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-08-24 02:34:30.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"90fe7cc035d01c91edfe3ce804dd95a926b47d9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 04:14:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06bd4b082f0049f1a2771985ee4311b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07a54e7e6de641098b2b83629bd7a5ab : SUCCESS in 4m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09c62da71e9443128009d4860f559619 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4caecc2b23d845e396cfaa5cb475b255 : SUCCESS in 3m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9775d24b29894eb2894270ad232e9c02 : SUCCESS in 7m 03s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/4f7c84df0de34d6d990bc6e6d8076f94 : SUCCESS in 4m 24s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/45ae3629594d46ce89858e5034423a4b : SUCCESS in 5m 56s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/e8554933792546e9a54a3c00648a262d : SUCCESS in 4m 26s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/9527d535a7f445b392048fb4ba6a8fc0 : SUCCESS in 4m 20s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/1a649ef237b046dfa59ece8125fd8e23 : SUCCESS in 3m 22s\n- tox-py313 https://zuul.opendev.org/t/openstack/build/b86d7b555eb44afd84d2283d39034117 : SUCCESS in 4m 29s\n- dib-functests https://zuul.opendev.org/t/openstack/build/564abc7b95414b49bf1b574b2147e4b9 : SUCCESS in 43m 33s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b1606d15f9f545638ab780002b36c8e5 : SUCCESS in 4m 39s\n- dib-devstack-functional-openstack-almalinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/34af39064bd9418197d0070eb16febc0 : SUCCESS in 53m 41s (non-voting)\n- dib-devstack-functional-openstack-almalinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/9ab0b7c5777f410990aaa5c2cfd67bfa : SUCCESS in 40m 59s (non-voting)\n- dib-devstack-functional-openstack-almalinux-10-v2-containerfile-src https://zuul.opendev.org/t/openstack/build/5c3b4a2af4d74e418bd0ad2a80c1abae : SUCCESS in 19m 35s (non-voting)\n- dib-devstack-functional-openstack-centos-9-stream-src https://zuul.opendev.org/t/openstack/build/a5ebc0918fe34f5181d47d7af72a3e89 : SUCCESS in 27m 02s\n- dib-devstack-functional-openstack-centos-10-stream-src https://zuul.opendev.org/t/openstack/build/d599d75292ce4a468b25ca1098d98b26 : SUCCESS in 25m 19s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/4546b75cd4de42e28de140aa00a9f36d : SUCCESS in 1h 06m 47s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/1bd2c61bf39a4def99a9e52ad7ad3c33 : SUCCESS in 1h 00m 08s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-10-containerfile-src https://zuul.opendev.org/t/openstack/build/704fbc2ef31d442795ed3081082a3c0f : SUCCESS in 27m 35s (non-voting)\n- dib-devstack-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/2117b5c777cc4fb29ac68428dce17fc4 : SUCCESS in 39m 49s\n- dib-devstack-functional-openstack-ubuntu-jammy-src https://zuul.opendev.org/t/openstack/build/558e117e8fa74e87a9b7eec0107b84e4 : SUCCESS in 39m 59s\n- dib-devstack-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/76a8a4efb9be4b56851e77eced8f2018 : SUCCESS in 19m 30s\n- dib-devstack-functional-openstack-ubuntu-resolute-src https://zuul.opendev.org/t/openstack/build/2fb1a234a89946618819a08f7908d888 : SUCCESS in 37m 01s\n- dib-devstack-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/7d61b1464d304ee993571f94a5952242 : FAILURE in 20m 39s (non-voting)\n- dib-devstack-functional-openstack-gentoo-src https://zuul.opendev.org/t/openstack/build/6d86098617184f728c36f80a88163e01 : FAILURE in 52m 47s (non-voting)\n- dib-devstack-functional-openstack-debian-bullseye-src https://zuul.opendev.org/t/openstack/build/272dc0a4d6574d01b5bb1ada723b1e44 : SUCCESS in 36m 36s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/7f7a07060f2f4e58995adf7153fb3303 : SUCCESS in 9m 24s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/f01315a0a0a1420d9b8c80caf508cc0d : SUCCESS in 46m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f7e8190c16958a7c5bbeaff790b8b0af55d0cb28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 05:20:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c08cd547f19e4d08ad2ad1fa65bbd422\n\n- dib-functests-arm64-deb https://zuul.opendev.org/t/openstack/build/6c043b3e5e1e4be0a04fddfe885ce59f : SUCCESS in 55m 13s\n- dib-functests-arm64-rpm https://zuul.opendev.org/t/openstack/build/ad581e6f32a541a28cb03b36489e7450 : SUCCESS in 36m 20s","accounts_in_message":[],"_revision_number":2},{"id":"22a0275ffbd0d7619efb77ceb519ee9d1818a208","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-08-24 19:38:21.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"006f066113c75673978dd3c553440cf157a40e87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-08-24 21:47:57.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ec8a271c8f68d4ef64cdfca75472463e32989acb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 22:59:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27aa1878ec6d4a54938f1bd40ce6cf7e\n\n- dib-functests-arm64-deb https://zuul.opendev.org/t/openstack/build/4ce420d4cec1496782e5bf0160966862 : SUCCESS in 35m 51s\n- dib-functests-arm64-rpm https://zuul.opendev.org/t/openstack/build/fd3a1cd0ae5c4969987f66757fed47ed : SUCCESS in 38m 59s","accounts_in_message":[],"_revision_number":3},{"id":"ca516de5f61eca76907eec8e3462a0ea22f4eec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 23:10:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66371d5c003a421d9f3432e1b6018aa3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b2442eac67e4229a70831001a9f23c6 : SUCCESS in 4m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aad48426aac4440aa445831411ad0864 : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b44ecf53ffe641c3ac88e279a1dcae04 : SUCCESS in 7m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1d9aa8e09f454c2689b5545092eb99d1 : SUCCESS in 3m 09s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/e7d16646c52f450388576a038227e7f6 : SUCCESS in 4m 14s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/97a593d8f58e417f8edf28e6359763e9 : SUCCESS in 3m 50s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/f3588fcbdc464423a3fc4ced54211dcd : SUCCESS in 4m 34s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/1c3385e523974be5ad1edbf3382f092e : SUCCESS in 3m 27s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/076a6372cfba427ba10704ed46fbda08 : SUCCESS in 4m 01s\n- tox-py313 https://zuul.opendev.org/t/openstack/build/31f483f0f49347f7bd7efd251b362352 : SUCCESS in 4m 05s\n- dib-functests https://zuul.opendev.org/t/openstack/build/432b2635d6374a588cbc504c3afe4682 : SUCCESS in 43m 57s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/f924590274524bf6a902ca4826b72fd9 : SUCCESS in 3m 39s\n- dib-devstack-functional-openstack-almalinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/3395107ab2824e019d3c4127ee3e950b : SUCCESS in 49m 48s (non-voting)\n- dib-devstack-functional-openstack-almalinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/fafb79f59b35448888d14c1af13472c4 : FAILURE in 34m 38s (non-voting)\n- dib-devstack-functional-openstack-almalinux-10-v2-containerfile-src https://zuul.opendev.org/t/openstack/build/afb216d0367f4889acb909da581d1866 : SUCCESS in 47m 09s (non-voting)\n- dib-devstack-functional-openstack-centos-9-stream-src https://zuul.opendev.org/t/openstack/build/cc0daabb413040469dbecb8d75e7cdae : FAILURE in 29m 12s\n- dib-devstack-functional-openstack-centos-10-stream-src https://zuul.opendev.org/t/openstack/build/bdbef52eff1e4933a33ff874da8326a5 : SUCCESS in 25m 57s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/479b0fd5584d401c85a307eccd887562 : SUCCESS in 57m 18s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/cd961f28bcfb42f7bc3c4a622914f39f : FAILURE in 34m 08s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-10-containerfile-src https://zuul.opendev.org/t/openstack/build/c026cc4109304d48b43c836e18b3b202 : SUCCESS in 27m 46s (non-voting)\n- dib-devstack-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/03cdf2f0a26c434aa3c6a719edc97347 : FAILURE in 26m 24s\n- dib-devstack-functional-openstack-ubuntu-jammy-src https://zuul.opendev.org/t/openstack/build/5e82ed6445b740d1980eaa35ca7bf191 : FAILURE in 36m 52s\n- dib-devstack-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/cfe4c1f070854091af751e0c9e07bc3c : SUCCESS in 29m 58s\n- dib-devstack-functional-openstack-ubuntu-resolute-src https://zuul.opendev.org/t/openstack/build/409f4b67761e49e191183097a6682245 : SUCCESS in 31m 42s\n- dib-devstack-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/3207eb00e07749f6a5c3deb355a23154 : FAILURE in 18m 20s (non-voting)\n- dib-devstack-functional-openstack-gentoo-src https://zuul.opendev.org/t/openstack/build/14b285a583f841c89dddac8b186eace6 : FAILURE in 1h 16m 15s (non-voting)\n- dib-devstack-functional-openstack-debian-bullseye-src https://zuul.opendev.org/t/openstack/build/a8ed1f84496f47c39890b57a101ddce5 : FAILURE in 26m 54s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/42cab62964e04b0392aac7711cf1ae35 : SUCCESS in 18m 32s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/668154216dfe4bc59bc5089d651d1b07 : SUCCESS in 49m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d6931b6ce9c57f99fc4905cfc28b36e84b399784","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-08-31 21:13:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"91f00d65621b41c1567689ed4da45162cf247526","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-08-31 21:34:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6b3e58e49e2f747c59b31ecada7cf922078d44f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 22:50:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd347b9134964d3eae4d4ea7dacc9d48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d97b4f0e0d44262baff737eb2fdd696 : SUCCESS in 5m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/128ae612fa2048aabeff488ae6eacc81 : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6979d85080934dcc83e6a4558473f17b : SUCCESS in 6m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1d4608f020cf43de95ff6ac5dfa4e3a4 : SUCCESS in 3m 12s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/613f32fb85f648cb957e85b0d316f129 : SUCCESS in 2m 26s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/faf13f9704734c4b8424a8fa461686ef : SUCCESS in 3m 48s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/ba8129817da7429fb6c5379ea5a4dcb7 : SUCCESS in 4m 17s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/72ceee45fae3499b800e74fa00e8579c : SUCCESS in 3m 56s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/4f46eb26ce9e4bc9865e03f4980dede7 : SUCCESS in 6m 11s\n- tox-py313 https://zuul.opendev.org/t/openstack/build/8e7039e7e8d644beb5201ded4cb1cc99 : SUCCESS in 3m 47s\n- dib-functests https://zuul.opendev.org/t/openstack/build/b62478b6e8b3476bb35873eee9376e43 : SUCCESS in 44m 27s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/e628e30c9a1f45428ec03a51f310a82c : SUCCESS in 3m 23s\n- dib-devstack-functional-openstack-almalinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/2a3ff50d374b4668b57181bf807b5b76 : SUCCESS in 57m 50s (non-voting)\n- dib-devstack-functional-openstack-almalinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/cd4ab819bb6a40038365a9746cc8c2d1 : SUCCESS in 41m 57s (non-voting)\n- dib-devstack-functional-openstack-almalinux-10-v2-containerfile-src https://zuul.opendev.org/t/openstack/build/58017e89f14544e2967fef6dc1b13241 : SUCCESS in 29m 48s (non-voting)\n- dib-devstack-functional-openstack-centos-9-stream-src https://zuul.opendev.org/t/openstack/build/c238ceddd31a45eab77e9df5afaa9c1c : SUCCESS in 35m 58s\n- dib-devstack-functional-openstack-centos-10-stream-src https://zuul.opendev.org/t/openstack/build/164524a87a75494187bd1a7a494a8a1a : SUCCESS in 25m 46s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/645aca13392a4a7da776371063a9cdd4 : SUCCESS in 52m 05s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/70c710365259489f97db2c3ebf0c7d3e : SUCCESS in 43m 18s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-10-containerfile-src https://zuul.opendev.org/t/openstack/build/c126faa9e63a4be48a7ea16e34fb1fc9 : SUCCESS in 27m 40s (non-voting)\n- dib-devstack-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/5fd6f37f34d14c9e820a8cfc8e6627cc : SUCCESS in 40m 13s\n- dib-devstack-functional-openstack-ubuntu-jammy-src https://zuul.opendev.org/t/openstack/build/f121c551e0c14ad19ea885a4c4df4ba3 : SUCCESS in 43m 02s\n- dib-devstack-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/0eb5cbd9c87546138927ce7c45025a87 : SUCCESS in 45m 06s\n- dib-devstack-functional-openstack-ubuntu-resolute-src https://zuul.opendev.org/t/openstack/build/ef7c52d648694de9a213578cdec0665c : SUCCESS in 42m 55s\n- dib-devstack-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/15fd981bc3ea47b7940f25c4537727fb : FAILURE in 17m 50s (non-voting)\n- dib-devstack-functional-openstack-gentoo-src https://zuul.opendev.org/t/openstack/build/02e94ea74a7a46bfb6a5933001d6249c : FAILURE in 58m 08s (non-voting)\n- dib-devstack-functional-openstack-debian-bullseye-src https://zuul.opendev.org/t/openstack/build/25481b08bdec43879eac4fa1673aedd2 : SUCCESS in 38m 54s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/361d29b3f3e446889f80c6cc29c98c86 : SUCCESS in 12m 34s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/ab7727b3bea64737995251820b7db725 : SUCCESS in 1h 08m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"25521b581aed2e25ab3c3f80e79da66093aeefc3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 23:07:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec183c3a3a134e16b084a96082fa9c09\n\n- dib-functests-arm64-deb https://zuul.opendev.org/t/openstack/build/eb2b3484b7064508bca17eae804aa05c : SUCCESS in 41m 49s\n- dib-functests-arm64-rpm https://zuul.opendev.org/t/openstack/build/83caf5cda5c14b8f91d8f412b4739a0c : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":3},{"id":"1229dec5716574035979f4cf8a1928fc8da5e2e9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-31 23:52:01.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b9dc1f7cce333f2c5a489ae67b6fd31bb3653a77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-09-01 03:26:15.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"2aceac00e80f79b6cb97217f4ba718f5e81f3cdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 04:56:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb3ab4a9ebab4d34b2f1a17eaba01839\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f5b9cb8e9a04325868efbe14d27b4cc : SUCCESS in 5m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65e9a14ef15940449a16ea63c84dfc69 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8aa7d0471f554115a1d8f342617499ab : SUCCESS in 5m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ba87da5a641b46de83a8fd4f239345b3 : SUCCESS in 5m 11s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/3e6dafc1c18049279ec9656fe98f7ecd : SUCCESS in 4m 00s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/e50200b3df034059b76da54a934acff2 : SUCCESS in 5m 03s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/04be2f22fbec4abb9eb88ae7189690d4 : SUCCESS in 4m 17s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/a70e398fff5649b5b4ec1291b12d5c74 : SUCCESS in 6m 39s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/476c2fcce1cf4f7cbaf4cc0dcb8ea93b : SUCCESS in 5m 35s\n- tox-py313 https://zuul.opendev.org/t/openstack/build/f7f3439edfb847269c52e04639952ed9 : SUCCESS in 5m 24s\n- dib-functests https://zuul.opendev.org/t/openstack/build/2b61e0c011624a73809b3b7a2dec157d : SUCCESS in 46m 22s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/5a8e866114c94985b6a64b871a42606a : SUCCESS in 3m 29s\n- dib-devstack-functional-openstack-almalinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/8bb9b207c3a843c1b655be8c19f230aa : SUCCESS in 58m 40s (non-voting)\n- dib-devstack-functional-openstack-almalinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/1a01c1e6e71648bea5ea58fe8b54f2c0 : SUCCESS in 41m 43s (non-voting)\n- dib-devstack-functional-openstack-almalinux-10-v2-containerfile-src https://zuul.opendev.org/t/openstack/build/5adfcb29cbd94e5db45a7e84bd15a60a : SUCCESS in 18m 52s (non-voting)\n- dib-devstack-functional-openstack-centos-9-stream-src https://zuul.opendev.org/t/openstack/build/1e568c0dad38428cb0c9ed37e1f47ccf : SUCCESS in 33m 24s\n- dib-devstack-functional-openstack-centos-10-stream-src https://zuul.opendev.org/t/openstack/build/708203184c02464d8400b7b98dc28bd7 : SUCCESS in 27m 31s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/667689530dcc4731b37683f0f32ecede : SUCCESS in 47m 47s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/cb1d0ba08cb74697b92d108288c55d69 : SUCCESS in 20m 03s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-10-containerfile-src https://zuul.opendev.org/t/openstack/build/a9abe48542d04eb5bfc502e698dbe5da : SUCCESS in 28m 15s (non-voting)\n- dib-devstack-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/2af1fd5da0fe4b43bf71289d247d4597 : SUCCESS in 41m 24s\n- dib-devstack-functional-openstack-ubuntu-jammy-src https://zuul.opendev.org/t/openstack/build/4755735be457497cb637a3e4b98456c6 : SUCCESS in 44m 35s\n- dib-devstack-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/59471b23a55d445c8aea51233fd1a3ab : SUCCESS in 49m 10s\n- dib-devstack-functional-openstack-ubuntu-resolute-src https://zuul.opendev.org/t/openstack/build/6a5b4607b0a44e3e999be68a462a6050 : SUCCESS in 45m 36s\n- dib-devstack-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/e18ddf4202c04574b4d2c1b45b77bd7d : FAILURE in 14m 10s (non-voting)\n- dib-devstack-functional-openstack-gentoo-src https://zuul.opendev.org/t/openstack/build/320501fba5d34cb589c50d5b9dcebab8 : FAILURE in 31m 37s (non-voting)\n- dib-devstack-functional-openstack-debian-bullseye-src https://zuul.opendev.org/t/openstack/build/9d165eb334b740279e2744cd4eec1495 : SUCCESS in 40m 36s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/7a4da4aa33a24b518ab7f0bf058e6cad : SUCCESS in 8m 01s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/dca119154a774a5c8efd5465edb7f954 : SUCCESS in 1h 19m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b19208806568a45a9a34724a234a55a7cf688bb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 05:46:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9afd6110cbd44080a9aa2311316ac886\n\n- dib-functests-arm64-deb https://zuul.opendev.org/t/openstack/build/d00bbe31293642ed8df7fac5caf46ff8 : SUCCESS in 44m 21s\n- dib-functests-arm64-rpm https://zuul.opendev.org/t/openstack/build/17adb4cdf94d461b9c6802b424fb5ad7 : SUCCESS in 51m 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"30700d8220173ce31ba7ac541061bafc59be4852","revisions":{"8292c1ca91c9742217745da9898e82724cbfd1d2":{"kind":"REWORK","_number":1,"created":"2026-04-09 05:02:17.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/14/983814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/diskimage-builder","ref":"refs/changes/14/983814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/14/983814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/14/983814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/14/983814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/diskimage-builder refs/changes/14/983814/1"}}},"commit":{"parents":[{"commit":"342165b8f8edd887216ef84eac175640eeb3729b","subject":"Refactor 02-set-machine-id into an element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/342165b8f8edd887216ef84eac175640eeb3729b"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-04-09 04:48:01.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-04-09 05:01:57.000000000","tz":720},"subject":"Refactor 03-reset-bls-entries into an element","message":"Refactor 03-reset-bls-entries into an element\n\nThis script is common across centos and rhel, so should really be in its\nown element.\n\nChange-Id: If8cbb38bd739169afcbe8e99325249f86f453bc5\nSigned-off-by: Steve Baker \u003csbaker@redhat.com\u003e\nAssisted-By: Opencode - Claude Sonnet 4.6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/8292c1ca91c9742217745da9898e82724cbfd1d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/8292c1ca91c9742217745da9898e82724cbfd1d2"}]},"branch":"refs/heads/master"},"fc93cf6ff1981dfad82ed97717faa5593c204ebd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-08-24 02:34:30.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/14/983814/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/diskimage-builder","ref":"refs/changes/14/983814/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/14/983814/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/14/983814/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/14/983814/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/diskimage-builder refs/changes/14/983814/2"}}},"commit":{"parents":[{"commit":"2dc202f7ef145eef6a4262732abf6920315720e3","subject":"Refactor 02-set-machine-id into an element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/2dc202f7ef145eef6a4262732abf6920315720e3"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-04-09 04:48:01.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-08-24 02:30:05.000000000","tz":720},"subject":"Refactor 03-reset-bls-entries into an element","message":"Refactor 03-reset-bls-entries into an element\n\nThis script is common across centos and rhel, so should really be in its\nown element.\n\nChange-Id: If8cbb38bd739169afcbe8e99325249f86f453bc5\nSigned-off-by: Steve Baker \u003csbaker@redhat.com\u003e\nAssisted-By: Opencode - Claude Sonnet 4.6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/fc93cf6ff1981dfad82ed97717faa5593c204ebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/fc93cf6ff1981dfad82ed97717faa5593c204ebd"}]},"branch":"refs/heads/master"},"9f690f382817fbde99ecf20dffa2293bcee0a285":{"kind":"REWORK","_number":3,"created":"2026-08-24 21:47:57.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/14/983814/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/diskimage-builder","ref":"refs/changes/14/983814/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/14/983814/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/14/983814/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/14/983814/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/diskimage-builder refs/changes/14/983814/3"}}},"commit":{"parents":[{"commit":"97f112de139623110dec023201fc125a9a40d3a0","subject":"Refactor 02-set-machine-id into an element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/97f112de139623110dec023201fc125a9a40d3a0"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-04-09 04:48:01.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-08-24 21:47:44.000000000","tz":720},"subject":"Refactor 03-reset-bls-entries into an element","message":"Refactor 03-reset-bls-entries into an element\n\nThis script is common across centos and rhel, so should really be in its\nown element.\n\nChange-Id: If8cbb38bd739169afcbe8e99325249f86f453bc5\nSigned-off-by: Steve Baker \u003csbaker@redhat.com\u003e\nAssisted-By: Opencode - Claude Sonnet 4.6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/9f690f382817fbde99ecf20dffa2293bcee0a285"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/9f690f382817fbde99ecf20dffa2293bcee0a285"}]},"branch":"refs/heads/master"},"30700d8220173ce31ba7ac541061bafc59be4852":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-09-01 03:26:15.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/14/983814/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/diskimage-builder","ref":"refs/changes/14/983814/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/14/983814/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/14/983814/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/14/983814/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/diskimage-builder refs/changes/14/983814/4"}}},"commit":{"parents":[{"commit":"b5accd5e683f68981ba799d055627c98c2f28c0a","subject":"Refactor 02-set-machine-id into an element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/b5accd5e683f68981ba799d055627c98c2f28c0a"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-04-09 04:48:01.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-09-01 03:13:35.000000000","tz":720},"subject":"Refactor 03-reset-bls-entries into an element","message":"Refactor 03-reset-bls-entries into an element\n\nThis script is common across centos and rhel, so should really be in its\nown element.\n\nChange-Id: If8cbb38bd739169afcbe8e99325249f86f453bc5\nSigned-off-by: Steve Baker \u003csbaker@redhat.com\u003e\nAssisted-By: Opencode - Claude Sonnet 4.6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/30700d8220173ce31ba7ac541061bafc59be4852"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/30700d8220173ce31ba7ac541061bafc59be4852"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
