)]}'
{"id":"openstack%2Fdiskimage-builder~989169","triplet_id":"openstack%2Fdiskimage-builder~master~I5cc0965917977fcb3897b7c3a5a486f8a7a21b39","project":"openstack/diskimage-builder","branch":"master","attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-07-30 10:51:55.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5cc0965917977fcb3897b7c3a5a486f8a7a21b39","subject":"Adds DIB_IMAGE_VERSION to centos element","status":"NEW","created":"2026-05-19 15:51:21.000000000","updated":"2026-07-30 11:01:50.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"95f5c747fc9e1e7f448f65d091a79d8d2a8b1894","_number":989169,"virtual_id_number":989169,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-05-19 17:27:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-19 17:27:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"119834058623e8bcb733e37bffcb976fe513b7f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-19 15:51:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a04126aa9ad228d430aea676ba473c778b52771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 17:27:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13b8ddc22e2c449382ff8d119ff754cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c03df43dd6b14f728db1a7ff9428170b : SUCCESS in 4m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b6bf6098f2d49aeae8015221a32be12 : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2b2482cd6864c10b3e95a72341a043d : SUCCESS in 2m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d874f1403ddd4ce2b226a1c5ea1ba87d : SUCCESS in 3m 27s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/246a6934f49d46be98dac880eb4575b7 : SUCCESS in 2m 34s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/1b8286daf0cf42788963b9894e4db616 : SUCCESS in 2m 40s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/29a85d0700f94221839968e94443022b : SUCCESS in 4m 12s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/504ece5f4a614f909bd8918610272fb9 : SUCCESS in 5m 00s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/4a4c2c61eb024a18aaf7dca2430b2c6d : SUCCESS in 3m 32s\n- tox-py313 https://zuul.opendev.org/t/openstack/build/0adc04f3c35540fe9cc30325599c26fb : SUCCESS in 4m 25s\n- dib-functests https://zuul.opendev.org/t/openstack/build/c6f75a616edc491cab148f439372a81a : FAILURE in 14m 13s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ac49065cd5ce4400bb6e26a335e7b6b5 : SUCCESS in 5m 24s\n- dib-devstack-functional-openstack-almalinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/c2e457172dba4e3dad7ed83852694ef2 : FAILURE in 36m 52s (non-voting)\n- dib-devstack-functional-openstack-almalinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/8289cedf3bec4b01ab550b7813550cd6 : SUCCESS in 55m 13s (non-voting)\n- dib-devstack-functional-openstack-almalinux-10-v2-containerfile-src https://zuul.opendev.org/t/openstack/build/afcff1715ba24e2d9e7cdcc878fba531 : SUCCESS in 47m 47s (non-voting)\n- dib-devstack-functional-openstack-centos-9-stream-src https://zuul.opendev.org/t/openstack/build/b608cbdf2abb4fdfba076b67a5d70e4a : SUCCESS in 18m 14s\n- dib-devstack-functional-openstack-centos-10-stream-src https://zuul.opendev.org/t/openstack/build/9f4eec801b1840fdb1888786506f7a65 : SUCCESS in 28m 15s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/64ad9fa39bb64afea69468bf03f4a2aa : FAILURE in 52m 31s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/57f768b5c71d434a9d44a57d3e30eaec : SUCCESS in 36m 03s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-10-containerfile-src https://zuul.opendev.org/t/openstack/build/6d94fc7e75804e178f4f331ef9587a0e : SUCCESS in 30m 43s (non-voting)\n- dib-devstack-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/fb0a89dee8134d42b71d7754129d79ba : SUCCESS in 23m 24s\n- dib-devstack-functional-openstack-ubuntu-jammy-src https://zuul.opendev.org/t/openstack/build/308d076bc87a4bf6b7318c5ea8fce80f : SUCCESS in 45m 59s\n- dib-devstack-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/cea2490b1d7744b9ac4e55b2421cc5a3 : SUCCESS in 45m 31s\n- dib-devstack-functional-openstack-ubuntu-resolute-src https://zuul.opendev.org/t/openstack/build/0870dd53cfd54bf28e9650989a5bb09c : SUCCESS in 23m 05s\n- dib-devstack-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/cb8a0e86d22242379edc51bee2e19254 : FAILURE in 33m 33s (non-voting)\n- dib-devstack-functional-openstack-gentoo-src https://zuul.opendev.org/t/openstack/build/75e28a238c004ebdb269fcdaa80f45c9 : FAILURE in 1h 21m 47s (non-voting)\n- dib-devstack-functional-openstack-debian-bullseye-src https://zuul.opendev.org/t/openstack/build/68e77b5978d449cd976f3c7820003adf : SUCCESS in 33m 35s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/e4a702a751c74c01822e9357c2fce971 : SUCCESS in 5m 28s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/864c6db53c574ea5983d86356a72c942 : SUCCESS in 36m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bfa8c87c463c28b0ff2f7ca4e587ea30d1ec4709","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 17:29:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb5b1dc45e324974a547bcecea9a6179\n\n- dib-functests-arm64-deb https://zuul.opendev.org/t/openstack/build/9ee9b47278e1498fa259f5ba49ed7295 : TIMED_OUT in 1h 31m 16s\n- dib-functests-arm64-rpm https://zuul.opendev.org/t/openstack/build/c6fb97cb27fb45bf80b95116d18090d7 : SUCCESS in 1h 20m 54s","accounts_in_message":[],"_revision_number":1},{"id":"09643ef9d9a70fdf227d4b8191d875332496efa2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-30 09:19:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"791b9550c778b22a253c35be16ae98a756141419","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 10:51:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/107201f0f844496ca976c1ed85d14d17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53b92803c9dd4a36b94fb17b23a834d1 : SUCCESS in 7m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a6f528f0a5c4e598ad587459bc298f2 : SUCCESS in 2m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3c2afb4e29b4485ae07e14c1e862873 : SUCCESS in 4m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/77ba8756e69d4a78949465b40a9f7030 : SUCCESS in 5m 19s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/756427947f6e42c1bfcdc32235b91cb6 : SUCCESS in 3m 28s\n- tox-py39 https://zuul.opendev.org/t/openstack/build/25a0bce03ad24e7c92614dc855222c2e : SUCCESS in 3m 53s\n- tox-py310 https://zuul.opendev.org/t/openstack/build/c6aab28ca8254c7186828d4278e6441f : SUCCESS in 4m 06s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/4cb0113b2efe4586bc49d794a9573f06 : SUCCESS in 4m 09s\n- tox-py312 https://zuul.opendev.org/t/openstack/build/823dd8f50ae049999bef407e2a829e98 : SUCCESS in 3m 49s\n- tox-py313 https://zuul.opendev.org/t/openstack/build/c2caa3c605b14781a4dbdbba0f8c9800 : SUCCESS in 2m 41s\n- dib-functests https://zuul.opendev.org/t/openstack/build/edb8d96133024d6faa69402f39d55570 : SUCCESS in 1h 26m 27s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/54e392beb6904f15a8e0a6b767581fbd : SUCCESS in 3m 21s\n- dib-devstack-functional-openstack-almalinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/963aca5938074459a77230f8e42d9d16 : FAILURE in 42m 06s (non-voting)\n- dib-devstack-functional-openstack-almalinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/04ce4c71c4784b639cc7f208cb6ba12e : FAILURE in 38m 46s (non-voting)\n- dib-devstack-functional-openstack-almalinux-10-v2-containerfile-src https://zuul.opendev.org/t/openstack/build/527788e6b5b14ba39f37cab71534d18c : FAILURE in 43m 33s (non-voting)\n- dib-devstack-functional-openstack-centos-9-stream-src https://zuul.opendev.org/t/openstack/build/3fa7727752184df3a131dd7edba625b6 : SUCCESS in 43m 38s\n- dib-devstack-functional-openstack-centos-10-stream-src https://zuul.opendev.org/t/openstack/build/f78b763d280a4c439c9ee29c4faa996c : SUCCESS in 28m 01s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-8-containerfile-src https://zuul.opendev.org/t/openstack/build/4cf0faceab28413895093963d5b90442 : FAILURE in 45m 56s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-9-containerfile-src https://zuul.opendev.org/t/openstack/build/b8ed963250194effa0a72ca30621e5f1 : FAILURE in 17m 26s (non-voting)\n- dib-devstack-functional-openstack-rockylinux-10-containerfile-src https://zuul.opendev.org/t/openstack/build/0df95015816448e1b4823821a9a13231 : FAILURE in 26m 01s (non-voting)\n- dib-devstack-functional-openstack-ubuntu-focal-src https://zuul.opendev.org/t/openstack/build/600e5256c42944fb8baf7f74a4b05f86 : FAILURE in 19m 16s\n- dib-devstack-functional-openstack-ubuntu-jammy-src https://zuul.opendev.org/t/openstack/build/9f116a6a348f4840b82e996ab0667824 : FAILURE in 45m 39s\n- dib-devstack-functional-openstack-ubuntu-noble-src https://zuul.opendev.org/t/openstack/build/35bbe440b68b4f74a1cee8d7e2cdf85b : FAILURE in 35m 27s\n- dib-devstack-functional-openstack-ubuntu-resolute-src https://zuul.opendev.org/t/openstack/build/7576d5b491404b7dbe73b56fdd3312df : FAILURE in 33m 12s\n- dib-devstack-functional-openstack-opensuse-15-src https://zuul.opendev.org/t/openstack/build/253fa25d0067488e8d9709c5176c1f13 : FAILURE in 14m 10s (non-voting)\n- dib-devstack-functional-openstack-gentoo-src https://zuul.opendev.org/t/openstack/build/4c62eb1a3ca04ab380b58efef2cb9e46 : FAILURE in 1h 05m 40s (non-voting)\n- dib-devstack-functional-openstack-debian-bullseye-src https://zuul.opendev.org/t/openstack/build/9b58403ce8f74decb11b977ba8bb570f : SUCCESS in 21m 32s (non-voting)\n- ironic-python-agent-check-image-dib-centos9 https://zuul.opendev.org/t/openstack/build/cefb856cd91b4138a932c633b729f02b : SUCCESS in 9m 58s\n- ipa-tempest-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/a8a3b83b505349aeadb82f986e18c248 : SUCCESS in 28m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"95f5c747fc9e1e7f448f65d091a79d8d2a8b1894","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 11:01:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb310c8ca6944f42a85516f6a0f92a8c\n\n- dib-functests-arm64-deb https://zuul.opendev.org/t/openstack/build/7b07d365cead4b09af3f00f0d389e8e9 : TIMED_OUT in 1h 31m 16s\n- dib-functests-arm64-rpm https://zuul.opendev.org/t/openstack/build/256278fe36c94619a9f462cabab708aa : TIMED_OUT in 1h 31m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d5277b8719b06be112cd31d13a6b89bf81027469","revisions":{"d5277b8719b06be112cd31d13a6b89bf81027469":{"kind":"REWORK","_number":1,"created":"2026-05-19 15:51:21.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/69/989169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/diskimage-builder","ref":"refs/changes/69/989169/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/69/989169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/69/989169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/diskimage-builder refs/changes/69/989169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/diskimage-builder refs/changes/69/989169/1"}}},"commit":{"parents":[{"commit":"00b49ecdfeffe7afcd3663ddfaca7ec0d99a505d","subject":"fix: add support for `cloud-init` in `Ubuntu Resolute`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/00b49ecdfeffe7afcd3663ddfaca7ec0d99a505d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-19 15:42:58.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-19 15:50:20.000000000","tz":60},"subject":"Adds DIB_IMAGE_VERSION to centos element","message":"Adds DIB_IMAGE_VERSION to centos element\n\nThis allows you to select one of the other timestamped images rather\nthan just using latest.\n\nChange-Id: I5cc0965917977fcb3897b7c3a5a486f8a7a21b39\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/d5277b8719b06be112cd31d13a6b89bf81027469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/diskimage-builder/commit/d5277b8719b06be112cd31d13a6b89bf81027469"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
