)]}'
{"id":"openstack%2Fdragonflow~414970","triplet_id":"openstack%2Fdragonflow~master~Id0dd492280911e79e07d92b5bbc78b25b318b733","project":"openstack/dragonflow","branch":"master","topic":"bp/neutron-lib","hashtags":[],"change_id":"Id0dd492280911e79e07d92b5bbc78b25b318b733","subject":"neutron-lib: use L3 constant from neutron-lib","status":"MERGED","created":"2016-12-26 13:13:09.000000000","updated":"2016-12-27 07:48:23.000000000","submitted":"2016-12-27 07:48:23.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e2a5864ed6db79d6a942820b247ba77e3f535dd5","_number":414970,"virtual_id_number":414970,"owner":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2016-12-27 07:48:23.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":11159,"name":"Hong Hui Xiao","email":"honghui_xiao@yeah.net","username":"HongHuiXiao"},{"value":0,"_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-12-27 07:43:05.000000000","_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},{"value":1,"date":"2016-12-27 06:47:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":1,"date":"2016-12-27 00:57:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11159,"name":"Hong Hui Xiao","email":"honghui_xiao@yeah.net","username":"HongHuiXiao"},{"value":2,"date":"2016-12-27 01:55:15.000000000","_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-12-27 07:43:05.000000000","_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":11159,"name":"Hong Hui Xiao","email":"honghui_xiao@yeah.net","username":"HongHuiXiao"},{"value":0,"_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"},{"_account_id":11159,"name":"Hong Hui Xiao","email":"honghui_xiao@yeah.net","username":"HongHuiXiao"},{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-12-27 00:57:38.000000000","updated_by":{"_account_id":11159,"name":"Hong Hui Xiao","email":"honghui_xiao@yeah.net","username":"HongHuiXiao"},"reviewer":{"_account_id":11159,"name":"Hong Hui Xiao","email":"honghui_xiao@yeah.net","username":"HongHuiXiao"},"state":"REVIEWER"},{"updated":"2016-12-27 01:55:15.000000000","updated_by":{"_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"},"reviewer":{"_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"},"state":"REVIEWER"},{"updated":"2016-12-27 06:47:13.000000000","updated_by":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"reviewer":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"state":"REVIEWER"},{"updated":"2016-12-27 07:43:05.000000000","updated_by":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"reviewer":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"state":"REVIEWER"},{"updated":"2016-12-27 07:48:23.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"f702a324a73f276bac12e86d98aa678e4dc02bce","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2016-12-26 13:13:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdc83923d031b8ec79c174ea64cbc19a13bfe58c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-26 14:31:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-dragonflow-docs-ubuntu-xenial http://docs-draft.openstack.org/70/414970/1/check/gate-dragonflow-docs-ubuntu-xenial/427c7ab//doc/build/html/ : SUCCESS in 2m 34s\n- gate-dragonflow-pep8-ubuntu-xenial http://logs.openstack.org/70/414970/1/check/gate-dragonflow-pep8-ubuntu-xenial/2ac0718/ : FAILURE in 4m 39s\n- gate-dragonflow-python27-ubuntu-xenial http://logs.openstack.org/70/414970/1/check/gate-dragonflow-python27-ubuntu-xenial/c46bca4/ : SUCCESS in 4m 44s\n- gate-dragonflow-python35 http://logs.openstack.org/70/414970/1/check/gate-dragonflow-python35/db86748/ : SUCCESS in 5m 46s\n- gate-tempest-dsvm-dragonflow-ubuntu-xenial-nv http://logs.openstack.org/70/414970/1/check/gate-tempest-dsvm-dragonflow-ubuntu-xenial-nv/ead77b7/ : FAILURE in 18m 41s (non-voting)\n- gate-dragonflow-dsvm-rally-ubuntu-xenial-nv http://logs.openstack.org/70/414970/1/check/gate-dragonflow-dsvm-rally-ubuntu-xenial-nv/0c63403/ : SUCCESS in 57m 53s (non-voting)\n- gate-dragonflow-dsvm-fullstack-ubuntu-xenial-nv http://logs.openstack.org/70/414970/1/check/gate-dragonflow-dsvm-fullstack-ubuntu-xenial-nv/9c9cd60/ : SUCCESS in 1h 16m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"940553a747f32e0b2fbb46c5fef2db2874a6a05e","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2016-12-26 19:47:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fe08b633aa978590c96e14bf38dda70e6312724c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-26 21:04:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-dragonflow-docs-ubuntu-xenial http://docs-draft.openstack.org/70/414970/2/check/gate-dragonflow-docs-ubuntu-xenial/7412381//doc/build/html/ : SUCCESS in 4m 58s\n- gate-dragonflow-pep8-ubuntu-xenial http://logs.openstack.org/70/414970/2/check/gate-dragonflow-pep8-ubuntu-xenial/2476528/ : SUCCESS in 5m 25s\n- gate-dragonflow-python27-ubuntu-xenial http://logs.openstack.org/70/414970/2/check/gate-dragonflow-python27-ubuntu-xenial/99df1ce/ : SUCCESS in 2m 42s\n- gate-dragonflow-python35 http://logs.openstack.org/70/414970/2/check/gate-dragonflow-python35/da676bc/ : SUCCESS in 5m 57s\n- gate-tempest-dsvm-dragonflow-ubuntu-xenial-nv http://logs.openstack.org/70/414970/2/check/gate-tempest-dsvm-dragonflow-ubuntu-xenial-nv/f5a0b7e/ : FAILURE in 33m 33s (non-voting)\n- gate-dragonflow-dsvm-rally-ubuntu-xenial-nv http://logs.openstack.org/70/414970/2/check/gate-dragonflow-dsvm-rally-ubuntu-xenial-nv/6ffb9a4/ : SUCCESS in 1h 13m 54s (non-voting)\n- gate-dragonflow-dsvm-fullstack-ubuntu-xenial-nv http://logs.openstack.org/70/414970/2/check/gate-dragonflow-dsvm-fullstack-ubuntu-xenial-nv/aba7ea1/ : SUCCESS in 1h 16m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b7ad3ad8830b3aa6e4bd2729ea2511c190e1d46a","author":{"_account_id":11159,"name":"Hong Hui Xiao","email":"honghui_xiao@yeah.net","username":"HongHuiXiao"},"date":"2016-12-27 00:57:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"086b02f1395f3eeacc49034d5840ad4c3409e5a0","author":{"_account_id":7805,"name":"Li Ma","email":"skywalker.nick@gmail.com","username":"Nick"},"date":"2016-12-27 01:55:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0c0cb29926d12a8993b2bbabcbd74c035884955a","author":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"date":"2016-12-27 06:47:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b9a368830d6c43790371730e688c49ceaeb99f0d","author":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"date":"2016-12-27 07:43:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1d0fb5bd5eafa108c14dbaee62022616aaaec5b0","author":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"date":"2016-12-27 07:43:12.000000000","message":"Patch Set 2:\n\nThanks","accounts_in_message":[],"_revision_number":2},{"id":"7bd06f5834054b2803635401c6657ca6938eb777","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-27 07:43:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"057143e79bf2c60e8b175840faad90afd0e71dc0","date":"2016-12-27 07:48:23.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":2},{"id":"b19b64035a336e822e4328e5443d1d14bc5cb2e8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-27 07:48:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-dragonflow-docs-ubuntu-xenial http://docs-draft.openstack.org/70/414970/2/gate/gate-dragonflow-docs-ubuntu-xenial/932105a//doc/build/html/ : SUCCESS in 2m 26s\n- gate-dragonflow-pep8-ubuntu-xenial http://logs.openstack.org/70/414970/2/gate/gate-dragonflow-pep8-ubuntu-xenial/033fd6b/ : SUCCESS in 3m 23s\n- gate-dragonflow-python27-ubuntu-xenial http://logs.openstack.org/70/414970/2/gate/gate-dragonflow-python27-ubuntu-xenial/3f5041e/ : SUCCESS in 4m 45s\n- gate-dragonflow-python35 http://logs.openstack.org/70/414970/2/gate/gate-dragonflow-python35/b6c25f3/ : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f7f2fd4ebb92b6a2ccb5681d23f63baac837a50b","revisions":{"3e8bb84fded4c33ae2c11ebbb7c9d62b3e257721":{"kind":"REWORK","_number":1,"created":"2016-12-26 13:13:09.000000000","uploader":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"ref":"refs/changes/70/414970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/dragonflow","ref":"refs/changes/70/414970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/70/414970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/70/414970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/70/414970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/dragonflow refs/changes/70/414970/1"}}},"commit":{"parents":[{"commit":"906fa354e7df3af88070025b21c71d00e8625462","subject":"Merge \"Dynamic register nb db models\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/906fa354e7df3af88070025b21c71d00e8625462"}]}],"author":{"name":"Gary Kotton","email":"gkotton@vmware.com","date":"2016-12-26 13:12:22.000000000","tz":120},"committer":{"name":"Gary Kotton","email":"gkotton@vmware.com","date":"2016-12-26 13:13:04.000000000","tz":120},"subject":"neutron-lib: use L3 constant from neutron-lib","message":"neutron-lib: use L3 constant from neutron-lib\n\nEnsure that the plugin uses the constant defined in neutron-lib.\nL3_ROUTER_NAT will be dropped in favor of this. This change is in\nhttps://review.openstack.org/#/c/394164/.\n\nNeutronLibImpact\nPartially-implements: blueprint neutron-lib\n\nChange-Id: Id0dd492280911e79e07d92b5bbc78b25b318b733\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/3e8bb84fded4c33ae2c11ebbb7c9d62b3e257721"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/3e8bb84fded4c33ae2c11ebbb7c9d62b3e257721"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"906fa354e7df3af88070025b21c71d00e8625462","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"f7f2fd4ebb92b6a2ccb5681d23f63baac837a50b":{"kind":"REWORK","_number":2,"created":"2016-12-26 19:47:57.000000000","uploader":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"ref":"refs/changes/70/414970/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/dragonflow","ref":"refs/changes/70/414970/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/70/414970/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/70/414970/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/70/414970/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/dragonflow refs/changes/70/414970/2"}}},"commit":{"parents":[{"commit":"906fa354e7df3af88070025b21c71d00e8625462","subject":"Merge \"Dynamic register nb db models\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/906fa354e7df3af88070025b21c71d00e8625462"}]}],"author":{"name":"Gary Kotton","email":"gkotton@vmware.com","date":"2016-12-26 13:12:22.000000000","tz":120},"committer":{"name":"Gary Kotton","email":"gkotton@vmware.com","date":"2016-12-26 19:47:43.000000000","tz":120},"subject":"neutron-lib: use L3 constant from neutron-lib","message":"neutron-lib: use L3 constant from neutron-lib\n\nEnsure that the plugin uses the constant defined in neutron-lib.\nL3_ROUTER_NAT will be dropped in favor of this. This change is in\nhttps://review.openstack.org/#/c/394164/.\n\nNeutronLibImpact\nPartially-implements: blueprint neutron-lib\n\nChange-Id: Id0dd492280911e79e07d92b5bbc78b25b318b733\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/f7f2fd4ebb92b6a2ccb5681d23f63baac837a50b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/f7f2fd4ebb92b6a2ccb5681d23f63baac837a50b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"906fa354e7df3af88070025b21c71d00e8625462","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
