)]}'
{"id":"openstack%2Fdragonflow~654999","triplet_id":"openstack%2Fdragonflow~master~Id34a36ee10fa0f30a20800b10656a85f6f3a5577","project":"openstack/dragonflow","branch":"master","topic":"654999","hashtags":[],"change_id":"Id34a36ee10fa0f30a20800b10656a85f6f3a5577","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 11:37:46.000000000","updated":"2019-06-17 11:03:42.000000000","submitted":"2019-06-17 11:03:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654999-1560769422893-91065db7","meta_rev_id":"92264c454ea9747bd362082fe781b676fd1270cc","_number":654999,"virtual_id_number":654999,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},{"value":0,"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},{"value":2,"date":"2019-06-17 11:03:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"all":[{"value":2,"date":"2019-06-17 06:12:13.000000000","_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},{"value":2,"date":"2019-06-17 06:16:18.000000000","_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"all":[{"value":0,"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},{"value":1,"date":"2019-06-17 06:16:14.000000000","_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-17 06:12:13.000000000","updated_by":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"reviewer":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"state":"REVIEWER"},{"updated":"2019-06-17 06:16:14.000000000","updated_by":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"reviewer":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"state":"REVIEWER"},{"updated":"2019-06-17 11:03:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21d01c2197599366d0266be2036056f72284040d","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 11:37:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"692060448e4cc8605bf553498e777268a46bfa8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:57:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/99/654999/1/check/test-release-openstack/3f18f4b/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 http://logs.openstack.org/99/654999/1/check/openstack-tox-pep8/c427551/ : SUCCESS in 6m 43s\n- openstack-tox-py27 http://logs.openstack.org/99/654999/1/check/openstack-tox-py27/17e7a02/ : SUCCESS in 8m 01s\n- openstack-tox-py35 http://logs.openstack.org/99/654999/1/check/openstack-tox-py35/eff1a98/ : SUCCESS in 8m 42s\n- openstack-tox-py36 http://logs.openstack.org/99/654999/1/check/openstack-tox-py36/a32022c/ : SUCCESS in 6m 58s\n- openstack-tox-docs http://logs.openstack.org/99/654999/1/check/openstack-tox-docs/afb3de9/html/ : SUCCESS in 6m 51s\n- dragonflow-tox-lower-constraints http://logs.openstack.org/99/654999/1/check/dragonflow-tox-lower-constraints/188a208/ : SUCCESS in 10m 38s\n- dragonflow-tempest http://logs.openstack.org/99/654999/1/check/dragonflow-tempest/c6aa61d/ : FAILURE in 13m 11s (non-voting)\n- dragonflow-dsvm-rally http://logs.openstack.org/99/654999/1/check/dragonflow-dsvm-rally/f9acfbd/ : RETRY_LIMIT in 17m 40s (non-voting)\n- dragonflow-dsvm-fullstack-redis finger://ze11.openstack.org/0758deafbd4f468fb890270e08aa5525 : RETRY_LIMIT in 20m 36s\n- dragonflow-dsvm-fullstack-etcd-zmq finger://ze02.openstack.org/e1872e9cf9804e3cb1b0f1123dccf00d : RETRY_LIMIT in 11m 02s\n- dragonflow-openstack-ansible-cross-repo http://logs.openstack.org/99/654999/1/check/dragonflow-openstack-ansible-cross-repo/d8ed6c6/ : FAILURE in 5m 03s (non-voting)\n- kuryr-kubernetes-tempest-dragonflow finger://ze11.openstack.org/823dddc65cc14a94b83835855d6ada56 : RETRY_LIMIT in 17m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7725260a1f8bde828708612e8bfc837ba9d361f","author":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"date":"2019-06-17 06:11:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"37890990128a775b8ad6a185b4ed07d0294b3cad","author":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"date":"2019-06-17 06:12:13.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"60afd7abff516d349762cd81cb9698634a133192","author":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"date":"2019-06-17 06:16:14.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3da325e633a4cb217abdf89f168d780458af2098","author":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"date":"2019-06-17 06:16:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"39f0ecf95302ce8226ab92fbfdf60aed6419e3b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 08:50:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/99/654999/2/check/test-release-openstack/3356bae/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/99/654999/2/check/openstack-tox-pep8/64b19b2/ : SUCCESS in 6m 05s\n- openstack-tox-py27 http://logs.openstack.org/99/654999/2/check/openstack-tox-py27/9cfa7a1/ : SUCCESS in 6m 20s\n- openstack-tox-py35 http://logs.openstack.org/99/654999/2/check/openstack-tox-py35/a17c77c/ : SUCCESS in 7m 11s\n- openstack-tox-py36 http://logs.openstack.org/99/654999/2/check/openstack-tox-py36/24f78ca/ : SUCCESS in 6m 49s\n- openstack-tox-docs http://logs.openstack.org/99/654999/2/check/openstack-tox-docs/bdd8fb6/html/ : SUCCESS in 4m 48s\n- dragonflow-tox-lower-constraints http://logs.openstack.org/99/654999/2/check/dragonflow-tox-lower-constraints/e3e6e8a/ : SUCCESS in 9m 28s\n- dragonflow-tempest http://logs.openstack.org/99/654999/2/check/dragonflow-tempest/2e7a5a3/ : FAILURE in 1h 27m 05s (non-voting)\n- dragonflow-dsvm-rally http://logs.openstack.org/99/654999/2/check/dragonflow-dsvm-rally/9eeceff/ : FAILURE in 2h 32m 11s (non-voting)\n- dragonflow-dsvm-fullstack-redis http://logs.openstack.org/99/654999/2/check/dragonflow-dsvm-fullstack-redis/3830302/ : SUCCESS in 1h 52m 57s\n- dragonflow-dsvm-fullstack-etcd-zmq http://logs.openstack.org/99/654999/2/check/dragonflow-dsvm-fullstack-etcd-zmq/2a3ed73/ : SUCCESS in 1h 43m 56s\n- dragonflow-openstack-ansible-cross-repo http://logs.openstack.org/99/654999/2/check/dragonflow-openstack-ansible-cross-repo/ae19e4b/ : FAILURE in 4m 31s (non-voting)\n- kuryr-kubernetes-tempest-dragonflow http://logs.openstack.org/99/654999/2/check/kuryr-kubernetes-tempest-dragonflow/0fc33cc/ : FAILURE in 32m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"735cf0dd25efd8d2d87552b4852848286d38c09e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 08:51:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4c56e608d2d1d359ea9923de991f920e782c4251","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 11:03:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/99/654999/2/gate/test-release-openstack/dff7072/ : SUCCESS in 3m 33s\n- openstack-tox-pep8 http://logs.openstack.org/99/654999/2/gate/openstack-tox-pep8/b94e17d/ : SUCCESS in 6m 05s\n- openstack-tox-py27 http://logs.openstack.org/99/654999/2/gate/openstack-tox-py27/679e813/ : SUCCESS in 6m 33s\n- openstack-tox-py35 http://logs.openstack.org/99/654999/2/gate/openstack-tox-py35/529d570/ : SUCCESS in 11m 30s\n- openstack-tox-py36 http://logs.openstack.org/99/654999/2/gate/openstack-tox-py36/d3af549/ : SUCCESS in 6m 43s\n- openstack-tox-docs http://logs.openstack.org/99/654999/2/gate/openstack-tox-docs/a81dc52/html/ : SUCCESS in 5m 52s\n- dragonflow-tox-lower-constraints http://logs.openstack.org/99/654999/2/gate/dragonflow-tox-lower-constraints/998a06c/ : SUCCESS in 9m 23s\n- dragonflow-dsvm-fullstack-redis http://logs.openstack.org/99/654999/2/gate/dragonflow-dsvm-fullstack-redis/cbc09cd/ : SUCCESS in 1h 59m 34s","accounts_in_message":[],"_revision_number":2},{"id":"07f74f4e716085702ae5d9763b6dca916d2ea353","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 11:03:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9573e43aa4027a3e3915c215e858fda504cafac6","revisions":{"74709639cf4df0cb50466f10585c9fbf2ac57ad2":{"kind":"REWORK","_number":1,"created":"2019-04-23 11:37:46.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/99/654999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/dragonflow","ref":"refs/changes/99/654999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/99/654999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/99/654999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/99/654999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/dragonflow refs/changes/99/654999/1"}}},"commit":{"parents":[{"commit":"c8adf28b099cda32629db0578b8af87159fcee9d","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/c8adf28b099cda32629db0578b8af87159fcee9d"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:37:28.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:37:43.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Id34a36ee10fa0f30a20800b10656a85f6f3a5577\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/74709639cf4df0cb50466f10585c9fbf2ac57ad2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/74709639cf4df0cb50466f10585c9fbf2ac57ad2"}]},"branch":"refs/heads/master"},"9573e43aa4027a3e3915c215e858fda504cafac6":{"kind":"REWORK","_number":2,"created":"2019-06-17 06:11:54.000000000","uploader":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"ref":"refs/changes/99/654999/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/dragonflow","ref":"refs/changes/99/654999/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/99/654999/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/99/654999/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/99/654999/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/dragonflow refs/changes/99/654999/2"}}},"commit":{"parents":[{"commit":"7db3e3b005470df345ffc624ec9cea6527abb849","subject":"Update the home-page for README.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/7db3e3b005470df345ffc624ec9cea6527abb849"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:37:28.000000000","tz":480},"committer":{"name":"Omer Anson","email":"omer.anson@toganetworks.com","date":"2019-06-17 06:11:41.000000000","tz":180},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Id34a36ee10fa0f30a20800b10656a85f6f3a5577\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/9573e43aa4027a3e3915c215e858fda504cafac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/9573e43aa4027a3e3915c215e858fda504cafac6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
