)]}'
{"id":"openstack%2Fdragonflow~675128","triplet_id":"openstack%2Fdragonflow~master~I71d502ebb3993bd7435378e2a9e7de24b6e8b48e","project":"openstack/dragonflow","branch":"master","topic":"update_redis","hashtags":[],"change_id":"I71d502ebb3993bd7435378e2a9e7de24b6e8b48e","subject":"Upgrade redis driver","status":"ABANDONED","created":"2019-08-07 14:41:29.000000000","updated":"2019-08-12 20:56:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0e304f7f49d0ae131d1d7cf6e4cb4eb8835ffb05","_number":675128,"virtual_id_number":675128,"owner":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},{"_account_id":23766,"name":"Dima Kuznetsov","email":"dima.kuznetsov@toganetworks.com","username":"dimak"},{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":-1,"date":"2019-08-11 11:24:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23766,"name":"Dima Kuznetsov","email":"dima.kuznetsov@toganetworks.com","username":"dimak"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},{"_account_id":23766,"name":"Dima Kuznetsov","email":"dima.kuznetsov@toganetworks.com","username":"dimak"},{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23766,"name":"Dima Kuznetsov","email":"dima.kuznetsov@toganetworks.com","username":"dimak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 14:42:06.000000000","updated_by":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"reviewer":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"state":"REVIEWER"},{"updated":"2019-08-07 14:42:08.000000000","updated_by":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"reviewer":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"state":"REVIEWER"},{"updated":"2019-08-07 14:42:11.000000000","updated_by":{"_account_id":23766,"name":"Dima Kuznetsov","email":"dima.kuznetsov@toganetworks.com","username":"dimak"},"reviewer":{"_account_id":23766,"name":"Dima Kuznetsov","email":"dima.kuznetsov@toganetworks.com","username":"dimak"},"state":"REVIEWER"},{"updated":"2019-08-11 11:24:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1f941970827887e7f7229ceb75747f9b0ce00f9","author":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"date":"2019-08-07 14:41:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24fbc1da4d6416023749e86d67b19e39435437e0","author":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"date":"2019-08-07 14:53:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"58f0f0817964dbac1af6ed094b1a75121662bbae","author":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"date":"2019-08-07 15:10:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f35b84ef4b8bf3e884c97123d1fc4dda77884448","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 16:51:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/28/675128/3/check/requirements-check/113a48b/ : SUCCESS in 2m 07s\n- openstack-tox-pep8 https://logs.opendev.org/28/675128/3/check/openstack-tox-pep8/1aaba8d/ : SUCCESS in 4m 43s\n- openstack-tox-py27 https://logs.opendev.org/28/675128/3/check/openstack-tox-py27/19dd6b8/ : FAILURE in 5m 10s\n- openstack-tox-py35 https://logs.opendev.org/28/675128/3/check/openstack-tox-py35/52ca1fe/ : SUCCESS in 7m 25s\n- openstack-tox-py36 https://logs.opendev.org/28/675128/3/check/openstack-tox-py36/65fa42e/ : FAILURE in 6m 57s\n- openstack-tox-docs https://logs.opendev.org/28/675128/3/check/openstack-tox-docs/5dcdc29/html/ : SUCCESS in 4m 37s\n- dragonflow-tox-lower-constraints https://logs.opendev.org/28/675128/3/check/dragonflow-tox-lower-constraints/1245968/ : FAILURE in 8m 47s\n- dragonflow-dsvm-fullstack-redis https://logs.opendev.org/28/675128/3/check/dragonflow-dsvm-fullstack-redis/5686273/ : FAILURE in 30m 15s\n- dragonflow-dsvm-fullstack-etcd-zmq https://logs.opendev.org/28/675128/3/check/dragonflow-dsvm-fullstack-etcd-zmq/83495a2/ : FAILURE in 1h 34m 01s","accounts_in_message":[],"_revision_number":3},{"id":"2e12df5ed27d5fac9818869aa61a103ea107ecbe","author":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"date":"2019-08-08 10:51:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3e32bc64432de9d6bdf330514818efb9c1600e49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 12:37:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/28/675128/4/check/requirements-check/02a31e3/ : SUCCESS in 2m 16s\n- openstack-tox-pep8 https://logs.opendev.org/28/675128/4/check/openstack-tox-pep8/ee03ff0/ : SUCCESS in 7m 16s\n- openstack-tox-py27 https://logs.opendev.org/28/675128/4/check/openstack-tox-py27/85b15db/ : FAILURE in 5m 34s\n- openstack-tox-py35 https://logs.opendev.org/28/675128/4/check/openstack-tox-py35/78c66e9/ : SUCCESS in 12m 34s\n- openstack-tox-py36 https://logs.opendev.org/28/675128/4/check/openstack-tox-py36/97fa710/ : FAILURE in 6m 08s\n- openstack-tox-docs https://logs.opendev.org/28/675128/4/check/openstack-tox-docs/649045b/html/ : SUCCESS in 5m 11s\n- dragonflow-tox-lower-constraints https://logs.opendev.org/28/675128/4/check/dragonflow-tox-lower-constraints/0c7772f/ : FAILURE in 9m 50s\n- dragonflow-dsvm-fullstack-redis https://logs.opendev.org/28/675128/4/check/dragonflow-dsvm-fullstack-redis/62c1770/ : FAILURE in 1h 35m 28s\n- dragonflow-dsvm-fullstack-etcd-zmq https://logs.opendev.org/28/675128/4/check/dragonflow-dsvm-fullstack-etcd-zmq/15adbf2/ : FAILURE in 1h 39m 06s","accounts_in_message":[],"_revision_number":4},{"id":"147c24a53dc5dd68c29932dea02da55d91c81401","author":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"date":"2019-08-08 14:47:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"14acad85fb4d4e3d0b656435e45b341c269c6905","author":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"date":"2019-08-08 14:58:20.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f25911a60ee4a9aa87b2c8217e57b699d2fa9bec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 16:46:50.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/28/675128/6/check/requirements-check/b494e59/ : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://logs.opendev.org/28/675128/6/check/openstack-tox-pep8/356c425/ : SUCCESS in 5m 27s\n- openstack-tox-py27 https://logs.opendev.org/28/675128/6/check/openstack-tox-py27/eb55347/ : FAILURE in 9m 29s\n- openstack-tox-py35 https://logs.opendev.org/28/675128/6/check/openstack-tox-py35/2a9b6b5/ : SUCCESS in 6m 47s\n- openstack-tox-py36 https://logs.opendev.org/28/675128/6/check/openstack-tox-py36/9c5a001/ : FAILURE in 8m 31s\n- openstack-tox-docs https://logs.opendev.org/28/675128/6/check/openstack-tox-docs/51ad73c/html/ : SUCCESS in 4m 09s\n- dragonflow-tox-lower-constraints https://logs.opendev.org/28/675128/6/check/dragonflow-tox-lower-constraints/fa59d1b/ : FAILURE in 8m 24s\n- dragonflow-dsvm-fullstack-redis https://logs.opendev.org/28/675128/6/check/dragonflow-dsvm-fullstack-redis/119a08c/ : FAILURE in 1h 28m 24s\n- dragonflow-dsvm-fullstack-etcd-zmq https://logs.opendev.org/28/675128/6/check/dragonflow-dsvm-fullstack-etcd-zmq/4eae1f6/ : FAILURE in 1h 39m 41s","accounts_in_message":[],"_revision_number":6},{"id":"e475ca50ca84a11c8f46ee48be442497e892a38e","author":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"date":"2019-08-11 07:03:18.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4ed7cf444a9cc9187320cb78a1291425192ecea5","author":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"date":"2019-08-11 07:04:43.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4b94740099338381549bca941bc62dd60f464509","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-11 08:46:03.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/28/675128/8/check/requirements-check/9e47e8f/ : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://logs.opendev.org/28/675128/8/check/openstack-tox-pep8/837ac7c/ : SUCCESS in 5m 07s\n- openstack-tox-py27 https://logs.opendev.org/28/675128/8/check/openstack-tox-py27/19c2f12/ : FAILURE in 5m 49s\n- openstack-tox-py35 https://logs.opendev.org/28/675128/8/check/openstack-tox-py35/1041952/ : SUCCESS in 6m 19s\n- openstack-tox-py36 https://logs.opendev.org/28/675128/8/check/openstack-tox-py36/5c513a7/ : FAILURE in 5m 36s\n- openstack-tox-docs https://logs.opendev.org/28/675128/8/check/openstack-tox-docs/4770f93/html/ : SUCCESS in 3m 39s\n- dragonflow-tox-lower-constraints https://logs.opendev.org/28/675128/8/check/dragonflow-tox-lower-constraints/27a291b/ : FAILURE in 8m 27s\n- dragonflow-dsvm-fullstack-redis https://logs.opendev.org/28/675128/8/check/dragonflow-dsvm-fullstack-redis/ef0c65b/ : FAILURE in 1h 37m 21s\n- dragonflow-dsvm-fullstack-etcd-zmq https://logs.opendev.org/28/675128/8/check/dragonflow-dsvm-fullstack-etcd-zmq/cfac695/ : FAILURE in 1h 33m 32s","accounts_in_message":[],"_revision_number":8},{"id":"3f646a52283b094cb9cec5249f812067cbee4743","author":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"date":"2019-08-11 09:40:58.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"14f8ad9a592ee0bd0e0d61c3c7ff5a990a6ccd84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-11 11:24:54.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/28/675128/9/check/requirements-check/5bcd011/ : SUCCESS in 1m 50s\n- openstack-tox-pep8 https://logs.opendev.org/28/675128/9/check/openstack-tox-pep8/ef231f9/ : SUCCESS in 4m 32s\n- openstack-tox-py27 https://logs.opendev.org/28/675128/9/check/openstack-tox-py27/fd6a844/ : FAILURE in 4m 35s\n- openstack-tox-py35 https://logs.opendev.org/28/675128/9/check/openstack-tox-py35/c3ff0f4/ : SUCCESS in 4m 58s\n- openstack-tox-py36 https://logs.opendev.org/28/675128/9/check/openstack-tox-py36/634a900/ : FAILURE in 4m 35s\n- openstack-tox-docs https://logs.opendev.org/28/675128/9/check/openstack-tox-docs/c0f91e3/html/ : SUCCESS in 2m 43s\n- dragonflow-tox-lower-constraints https://logs.opendev.org/28/675128/9/check/dragonflow-tox-lower-constraints/95fd4a5/ : FAILURE in 7m 23s\n- dragonflow-dsvm-fullstack-redis https://logs.opendev.org/28/675128/9/check/dragonflow-dsvm-fullstack-redis/990a42e/ : FAILURE in 1h 38m 19s\n- dragonflow-dsvm-fullstack-etcd-zmq https://logs.opendev.org/28/675128/9/check/dragonflow-dsvm-fullstack-etcd-zmq/ee22ecc/ : FAILURE in 1h 33m 54s","accounts_in_message":[],"_revision_number":9},{"id":"26ae956742f7178af6a56b33c4cd1548f538ef2b","author":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"date":"2019-08-12 20:56:24.000000000","message":"Abandoned\n\nSquashed into https://review.opendev.org/638509/","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"67bec7f5a2a8b20b46a67a72466f142cdb2c0894","revisions":{"e2cfb2f2f09064451787ea5bf3630ebf97229a45":{"kind":"REWORK","_number":1,"created":"2019-08-07 14:41:29.000000000","uploader":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"ref":"refs/changes/28/675128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/dragonflow","ref":"refs/changes/28/675128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/1"}}},"commit":{"parents":[{"commit":"5a39ecae6d522a2db93c6130760805c3dda5781c","subject":"Merge \"Move failing check jobs to experimental\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/5a39ecae6d522a2db93c6130760805c3dda5781c"}]}],"author":{"name":"Shachar Snapiri","email":"shachar.snapiri@huawei.com","date":"2019-08-07 14:41:05.000000000","tz":180},"committer":{"name":"Shachar Snapiri","email":"shachar.snapiri@huawei.com","date":"2019-08-07 14:41:05.000000000","tz":180},"subject":"Upgrade redis driver","message":"Upgrade redis driver\n\nRedis python package API was changed over time, while we did not\nchange our handling of the code.\nIn the gate the driver was updated, and the gate tests were broken.\nUpdated the code in a way that supports both python2 (soon to be\nobsoleted) and python3.\nUpdated the requirements to install new versions of the packages.\n\nChange-Id: I71d502ebb3993bd7435378e2a9e7de24b6e8b48e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/e2cfb2f2f09064451787ea5bf3630ebf97229a45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/e2cfb2f2f09064451787ea5bf3630ebf97229a45"}]},"branch":"refs/heads/master"},"c33edb50523a8b9961893cb8ca3024f1f9b23231":{"kind":"REWORK","_number":2,"created":"2019-08-07 14:53:06.000000000","uploader":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"ref":"refs/changes/28/675128/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/dragonflow","ref":"refs/changes/28/675128/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/2"}}},"commit":{"parents":[{"commit":"5a39ecae6d522a2db93c6130760805c3dda5781c","subject":"Merge \"Move failing check jobs to experimental\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/5a39ecae6d522a2db93c6130760805c3dda5781c"}]}],"author":{"name":"Shachar Snapiri","email":"shachar.snapiri@huawei.com","date":"2019-08-07 14:41:05.000000000","tz":180},"committer":{"name":"Shachar Snapiri","email":"shachar.snapiri@huawei.com","date":"2019-08-07 14:52:55.000000000","tz":180},"subject":"Upgrade redis driver","message":"Upgrade redis driver\n\nRedis python package API was changed over time, while we did not\nchange our handling of the code.\nIn the gate the driver was updated, and the gate tests were broken.\nUpdated the code in a way that supports both python2 (soon to be\nobsoleted) and python3.\nUpdated the requirements to install new versions of the packages.\n\nChange-Id: I71d502ebb3993bd7435378e2a9e7de24b6e8b48e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/c33edb50523a8b9961893cb8ca3024f1f9b23231"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/c33edb50523a8b9961893cb8ca3024f1f9b23231"}]},"branch":"refs/heads/master"},"7f5ca5db077cbc77fc5be7d9e46e7ab150879be0":{"kind":"REWORK","_number":3,"created":"2019-08-07 15:10:41.000000000","uploader":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"ref":"refs/changes/28/675128/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/dragonflow","ref":"refs/changes/28/675128/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/3"}}},"commit":{"parents":[{"commit":"5a39ecae6d522a2db93c6130760805c3dda5781c","subject":"Merge \"Move failing check jobs to experimental\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/5a39ecae6d522a2db93c6130760805c3dda5781c"}]}],"author":{"name":"Shachar Snapiri","email":"shachar.snapiri@huawei.com","date":"2019-08-07 14:41:05.000000000","tz":180},"committer":{"name":"Shachar Snapiri","email":"shachar.snapiri@huawei.com","date":"2019-08-07 15:10:32.000000000","tz":180},"subject":"Upgrade redis driver","message":"Upgrade redis driver\n\nRedis python package API was changed over time, while we did not\nchange our handling of the code.\nIn the gate the driver was updated, and the gate tests were broken.\nUpdated the code in a way that supports both python2 (soon to be\nobsoleted) and python3.\nUpdated the requirements to install new versions of the packages.\n\nChange-Id: I71d502ebb3993bd7435378e2a9e7de24b6e8b48e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/7f5ca5db077cbc77fc5be7d9e46e7ab150879be0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/7f5ca5db077cbc77fc5be7d9e46e7ab150879be0"}]},"branch":"refs/heads/master"},"6015f599f41f1509a8b893dd4aeb8f01b96f2a16":{"kind":"REWORK","_number":4,"created":"2019-08-08 10:51:18.000000000","uploader":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"ref":"refs/changes/28/675128/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/dragonflow","ref":"refs/changes/28/675128/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/4"}}},"commit":{"parents":[{"commit":"5a39ecae6d522a2db93c6130760805c3dda5781c","subject":"Merge \"Move failing check jobs to experimental\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/5a39ecae6d522a2db93c6130760805c3dda5781c"}]}],"author":{"name":"Shachar Snapiri","email":"shachar.snapiri@huawei.com","date":"2019-08-07 14:41:05.000000000","tz":180},"committer":{"name":"Shachar Snapiri","email":"shachar.snapiri@huawei.com","date":"2019-08-08 10:51:06.000000000","tz":180},"subject":"Upgrade redis driver","message":"Upgrade redis driver\n\nRedis python package API was changed over time, while we did not\nchange our handling of the code.\nIn the gate the driver was updated, and the gate tests were broken.\nUpdated the code in a way that supports both python2 (soon to be\nobsoleted) and python3.\nUpdated the requirements to install new versions of the packages.\n\nChange-Id: I71d502ebb3993bd7435378e2a9e7de24b6e8b48e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/6015f599f41f1509a8b893dd4aeb8f01b96f2a16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/6015f599f41f1509a8b893dd4aeb8f01b96f2a16"}]},"branch":"refs/heads/master"},"247e0168c0ff1d0ae7dc5bfcaf0b7e1b86372887":{"kind":"REWORK","_number":5,"created":"2019-08-08 14:47:11.000000000","uploader":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"ref":"refs/changes/28/675128/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/dragonflow","ref":"refs/changes/28/675128/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/5"}}},"commit":{"parents":[{"commit":"5a39ecae6d522a2db93c6130760805c3dda5781c","subject":"Merge \"Move failing check jobs to experimental\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/5a39ecae6d522a2db93c6130760805c3dda5781c"}]}],"author":{"name":"Shachar Snapiri","email":"shachar.snapiri@huawei.com","date":"2019-08-07 14:41:05.000000000","tz":180},"committer":{"name":"Shachar Snapiri","email":"shachar.snapiri@huawei.com","date":"2019-08-08 14:46:59.000000000","tz":180},"subject":"Upgrade redis driver","message":"Upgrade redis driver\n\nRedis python package API was changed over time, while we did not\nchange our handling of the code.\nIn the gate the driver was updated, and the gate tests were broken.\nUpdated the code in a way that supports both python2 (soon to be\nobsoleted) and python3.\nUpdated the requirements to install new versions of the packages.\n\nChange-Id: I71d502ebb3993bd7435378e2a9e7de24b6e8b48e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/247e0168c0ff1d0ae7dc5bfcaf0b7e1b86372887"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/247e0168c0ff1d0ae7dc5bfcaf0b7e1b86372887"}]},"branch":"refs/heads/master"},"1462aebd49b429d5280083a9fcc46a0df0de3e0b":{"kind":"REWORK","_number":6,"created":"2019-08-08 14:58:20.000000000","uploader":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"ref":"refs/changes/28/675128/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/dragonflow","ref":"refs/changes/28/675128/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/6"}}},"commit":{"parents":[{"commit":"5a39ecae6d522a2db93c6130760805c3dda5781c","subject":"Merge \"Move failing check jobs to experimental\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/5a39ecae6d522a2db93c6130760805c3dda5781c"}]}],"author":{"name":"Shachar Snapiri","email":"shachar.snapiri@huawei.com","date":"2019-08-07 14:41:05.000000000","tz":180},"committer":{"name":"Shachar Snapiri","email":"shachar.snapiri@huawei.com","date":"2019-08-08 14:58:10.000000000","tz":180},"subject":"Upgrade redis driver","message":"Upgrade redis driver\n\nRedis python package API was changed over time, while we did not\nchange our handling of the code.\nIn the gate the driver was updated, and the gate tests were broken.\nUpdated the code in a way that supports both python2 (soon to be\nobsoleted) and python3.\nUpdated the requirements to install new versions of the packages.\n\nChange-Id: I71d502ebb3993bd7435378e2a9e7de24b6e8b48e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/1462aebd49b429d5280083a9fcc46a0df0de3e0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/1462aebd49b429d5280083a9fcc46a0df0de3e0b"}]},"branch":"refs/heads/master"},"08a677b1c5dfcfb804a8ccf86fa7f48362d1e11b":{"kind":"REWORK","_number":7,"created":"2019-08-11 07:03:18.000000000","uploader":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"ref":"refs/changes/28/675128/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/dragonflow","ref":"refs/changes/28/675128/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/7"}}},"commit":{"parents":[{"commit":"5a39ecae6d522a2db93c6130760805c3dda5781c","subject":"Merge \"Move failing check jobs to experimental\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/5a39ecae6d522a2db93c6130760805c3dda5781c"}]}],"author":{"name":"Shachar Snapiri","email":"shachar.snapiri@huawei.com","date":"2019-08-07 14:41:05.000000000","tz":180},"committer":{"name":"Omer Anson","email":"omer.anson@toganetworks.com","date":"2019-08-11 07:03:08.000000000","tz":180},"subject":"Upgrade redis driver","message":"Upgrade redis driver\n\nRedis python package API was changed over time, while we did not\nchange our handling of the code.\nIn the gate the driver was updated, and the gate tests were broken.\nUpdated the code in a way that supports both python2 (soon to be\nobsoleted) and python3.\nUpdated the requirements to install new versions of the packages.\n\nChange-Id: I71d502ebb3993bd7435378e2a9e7de24b6e8b48e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/08a677b1c5dfcfb804a8ccf86fa7f48362d1e11b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/08a677b1c5dfcfb804a8ccf86fa7f48362d1e11b"}]},"branch":"refs/heads/master"},"6e3a7e50c4974bd1196f4df07f76c2b478f1dda6":{"kind":"REWORK","_number":8,"created":"2019-08-11 07:04:43.000000000","uploader":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"ref":"refs/changes/28/675128/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/dragonflow","ref":"refs/changes/28/675128/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/8"}}},"commit":{"parents":[{"commit":"5a39ecae6d522a2db93c6130760805c3dda5781c","subject":"Merge \"Move failing check jobs to experimental\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/5a39ecae6d522a2db93c6130760805c3dda5781c"}]}],"author":{"name":"Shachar Snapiri","email":"shachar.snapiri@huawei.com","date":"2019-08-07 14:41:05.000000000","tz":180},"committer":{"name":"Omer Anson","email":"omer.anson@toganetworks.com","date":"2019-08-11 07:04:33.000000000","tz":180},"subject":"Upgrade redis driver","message":"Upgrade redis driver\n\nRedis python package API was changed over time, while we did not\nchange our handling of the code.\nIn the gate the driver was updated, and the gate tests were broken.\nUpdated the code in a way that supports both python2 (soon to be\nobsoleted) and python3.\nUpdated the requirements to install new versions of the packages.\n\nChange-Id: I71d502ebb3993bd7435378e2a9e7de24b6e8b48e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/6e3a7e50c4974bd1196f4df07f76c2b478f1dda6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/6e3a7e50c4974bd1196f4df07f76c2b478f1dda6"}]},"branch":"refs/heads/master"},"67bec7f5a2a8b20b46a67a72466f142cdb2c0894":{"kind":"REWORK","_number":9,"created":"2019-08-11 09:40:58.000000000","uploader":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"ref":"refs/changes/28/675128/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/dragonflow","ref":"refs/changes/28/675128/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/dragonflow refs/changes/28/675128/9"}}},"commit":{"parents":[{"commit":"5a39ecae6d522a2db93c6130760805c3dda5781c","subject":"Merge \"Move failing check jobs to experimental\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/5a39ecae6d522a2db93c6130760805c3dda5781c"}]}],"author":{"name":"Shachar Snapiri","email":"shachar.snapiri@huawei.com","date":"2019-08-07 14:41:05.000000000","tz":180},"committer":{"name":"Shachar Snapiri","email":"shachar.snapiri@huawei.com","date":"2019-08-11 09:40:48.000000000","tz":180},"subject":"Upgrade redis driver","message":"Upgrade redis driver\n\nRedis python package API was changed over time, while we did not\nchange our handling of the code.\nIn the gate the driver was updated, and the gate tests were broken.\nUpdated the code in a way that supports both python2 (soon to be\nobsoleted) and python3.\nUpdated the requirements to install new versions of the packages.\n\nChange-Id: I71d502ebb3993bd7435378e2a9e7de24b6e8b48e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/67bec7f5a2a8b20b46a67a72466f142cdb2c0894"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/67bec7f5a2a8b20b46a67a72466f142cdb2c0894"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
